)]}'
{"id":"openstack%2Fkayobe~756541","triplet_id":"openstack%2Fkayobe~master~I80f9d95f02e11fda5916f9a9dd257b688a9db7e2","project":"openstack/kayobe","branch":"master","topic":"story/2008153","hashtags":[],"change_id":"I80f9d95f02e11fda5916f9a9dd257b688a9db7e2","subject":"Configure bifrost to use firewalld trusted zone","status":"MERGED","created":"2020-10-07 15:42:02.000000000","updated":"2020-10-12 12:56:19.000000000","submitted":"2020-10-12 12:54:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"756541-1602507275390-f84a88f2","meta_rev_id":"ca30cc25d3df49fe1796bcebf7df69412311846f","_number":756541,"virtual_id_number":756541,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-12 12:54:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"all":[{"value":1,"date":"2020-10-09 15:27:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":2,"date":"2020-10-07 18:15:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2020-10-09 15:51:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":1,"date":"2020-10-09 16:06:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-09 15:59:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":1,"date":"2020-10-07 18:15:56.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-09 15:27:36.000000000","updated_by":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"reviewer":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"state":"REVIEWER"},{"updated":"2020-10-09 15:51:00.000000000","updated_by":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"reviewer":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"state":"REVIEWER"},{"updated":"2020-10-09 16:06:59.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-10-12 12:54:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c09d10ae9d6a82e8514e0d77318a93092bd79a56","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-10-07 15:42:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dd816e420b3d10fdda409956408a67d5cc95d9d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 17:48:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/87b1980f890242fc92973896cf84599b : SUCCESS in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6474881a83054f818e87ccb71f4bd185 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/99e8b8d129ab439c936a77a8411ed0c4 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e32fba919c84cf081939ee55ea240b6 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/274ad9b607da4b3a87ccfdb4164caa0f : SUCCESS in 6m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3835a7bc644c48018e9a3c3b0364394f : SUCCESS in 4m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/50b9cf8be0fc4134a3286eeff80a7181 : SUCCESS in 5m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b96584eb872a472e943c2662d98347e3 : SUCCESS in 9m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0d902997692f40b29b892599d84e34f9 : SUCCESS in 19m 25s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/6c129b2ff9a348389b5f25db4f29ec6a : SUCCESS in 1h 08m 07s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/fcf9f41c77b74ae5876bcb2870f40ee6 : SUCCESS in 1h 11m 44s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/1030136cb05c4ea1b0bf24c4e8356f36 : SUCCESS in 15m 46s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/35e071dd1b434e60810a2581a298b4a5 : SUCCESS in 2h 00m 33s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/5062011fb707418f9cd64f4016d292f4 : SUCCESS in 15m 48s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/251b509537234fdbb08c07a485185536 : SUCCESS in 26m 02s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/26ce192497bd49a991d70ffb07b46e37 : SUCCESS in 10m 47s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"8f5c6bf1eec64fd2b90dc0e7e7ebb6a1a6785364","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-10-07 18:15:56.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"e51cc0177b6835f70b88af96332184d01dcd570c","author":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"date":"2020-10-09 15:27:36.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ea72286958692bd3f47960ce2f390df2f0076966","author":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"date":"2020-10-09 15:51:00.000000000","message":"Patch Set 1: Code-Review+1\n\nThanks, this is exactly what I was looking for.","accounts_in_message":[],"_revision_number":1},{"id":"06e1f00c1b1422b718e43ba587f5ac7ad8464103","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-10-09 15:59:33.000000000","message":"Patch Set 1: Backport-Candidate+2\n\nNeeds a backport to ussuri","accounts_in_message":[],"_revision_number":1},{"id":"5dcfab199b7bd09a54230d80fbd15ba656af4a4b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-10-09 16:06:59.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"803cec625a275d27ff4b44918a45eb08d80e8f89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 16:07:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"19ac03202593b1b5d67a20c1d97d26413a6abdc0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 18:25:34.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ae9a806665f48cd976c3fafabfbfcba : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a4c71128a57445ce82ae3b0dd163d0a7 : RETRY_LIMIT in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c74e520101b848989aa7b3a2244de82b : RETRY_LIMIT in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d845da869304ea79d44bd3a74a6ff41 : SUCCESS in 6m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/61e8843bb48544ff87334ca149deb448 : SUCCESS in 5m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cd9eea24415047fcb67fddc9f1cb3f0c : SUCCESS in 5m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/85d0143d859744c2946ba8a6fda862a8 : SUCCESS in 8m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2b3741455da0482ca62cc5ee0ddf8b53 : SUCCESS in 22m 17s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/e8de4d35caf84649b289f93ac5492776 : RETRY_LIMIT in 4m 55s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/354e7af16f97486884ee6e6e56999a0c : SUCCESS in 1h 05m 53s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/02ecc4e5655742c98e45c91cadc3eee2 : SUCCESS in 14m 22s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f03144bafa4d42c8bf48ef09196f36ab : SUCCESS in 1h 57m 56s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/6e4bce8eb4824a3499b287493b8aea38 : SUCCESS in 16m 17s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/048766905e8b4768a0f1fd5d6cbe5750 : FAILURE in 19m 27s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/6fc90fac439543e9aecf1bf2a43777f1 : SUCCESS in 12m 48s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"7220a285b708211af3d8c5b9702645896a6cb114","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-10-09 19:29:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"40c0d4d36dc4ab30ba9eb05df161b6a3fcd0d9ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 21:36:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/05b0992f3d6146e7a261395fdb1db6b1 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90bcd42ac66e497aa1a80809c9a75c82 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f54c2799485447e968272c6e84d5d02 : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e325ab56fada4257a24a04d1e9db39e6 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79807d4cc7a241f7958c927a88a44ff9 : SUCCESS in 7m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9818a7e68dfe4818b08c343cdabc1e82 : SUCCESS in 3m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/25320383f0ef45dc966f62306163d570 : SUCCESS in 5m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/30e087b2f85f4afbb8b6010142b84c0a : RETRY_LIMIT in 2m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/20bd9a4141524c56a6c5afb7ca60a87a : RETRY_LIMIT in 2m 54s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/0ed5e541a5a246fba4b071c710857bed : SUCCESS in 1h 02m 42s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/57acb83122c04624a96cf3b55a676b0d : SUCCESS in 1h 06m 59s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/99557a4d2166431ab99b88f86ea19e70 : SUCCESS in 16m 18s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a059ab015c9945b5bdf5ecb0725ebfdf : SUCCESS in 1h 54m 07s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/43b6688847ca487aad7dd1b4c0e6053a : SUCCESS in 16m 53s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a1fd53848bad4200b160ce905ce43157 : SUCCESS in 26m 58s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/9a506db77de34ef58f5c641e6c72ee97 : SUCCESS in 10m 53s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"dae5764454d6b30d6d100d4086ce1b9c82520a63","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-10-12 08:57:32.000000000","message":"Patch Set 1:\n\nrecheck\n\nmirror issues","accounts_in_message":[],"_revision_number":1},{"id":"15244d89c2f8fe2cb5a45046d1940680362fa13d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 10:59:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7ad03971499432f957d72812bea14db : SUCCESS in 6m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90e10178ea8e4bbd8ae4b90bdf182da9 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/98c4ddba7acf44e1802055b3d8fa976a : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c27eebf446c47e9872756b6f86880af : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c3d7ad95af4419ea3b151fedd5bde94 : SUCCESS in 6m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71f9f32036f743f9947bd409e3385f1b : SUCCESS in 4m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9776ae047575469a9ad6f58f4f5a7c68 : SUCCESS in 4m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a6fb35e0f5fd4be9b462e1149020263d : SUCCESS in 9m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4589c8b400f94a25af5bd134cfa75fd3 : SUCCESS in 21m 12s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/8c7e16d6305147a1b9258e4bea12b184 : SUCCESS in 1h 02m 27s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/3437312bf2ad4270895d381cbbec67e4 : SUCCESS in 1h 00m 57s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/473d1d45a8b843b9a74f80b4cf145ef4 : SUCCESS in 15m 03s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/95c26794dadb485f9cc79e85dcc1c11c : SUCCESS in 1h 55m 35s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/d611d5f7f4e74c16bac003d563df1ec7 : SUCCESS in 14m 13s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/34d552986127444c8ecbaf8ddcfcbfa6 : SUCCESS in 25m 21s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/bee468e6835c4038a10559cbc7f418e3 : SUCCESS in 12m 01s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"751f1f97f9b914a4106d89c2a1158da17027cc5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 10:59:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f4e1e0043912fc986ab66d15a211be4cb35e944b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 12:54:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0254c607967d4c84a0b59e9582668d8a : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b4819de4ed014fef991fd27e5d12937c : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7d924eb1d15d46f5ac631637c2bfec26 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1104d870e24942aeb0a8c90c7d258712 : SUCCESS in 7m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7200d2e28a114799a36f5efb570fa3be : SUCCESS in 3m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ddb02e42debf49ab8cf4e4fb6f0c2b3e : SUCCESS in 4m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e1ebbf0a59fb4c328503b623eb7811bf : SUCCESS in 8m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/85dfaa1686654fa6860b4a2cc47c33ff : SUCCESS in 20m 28s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/7517ea355bf743119230391c367ffd3f : SUCCESS in 1h 02m 38s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/3df4bf4864d74939a4d1bd10bac610a5 : SUCCESS in 1h 08m 25s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/3cea4267927a42568f98fefe41759b4d : SUCCESS in 14m 53s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/36fd9bac73834348a5215a8cbe74dc59 : SUCCESS in 1h 52m 08s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/ac3b8a9ce9f94a4cb4c6cfca8dc41eda : SUCCESS in 15m 24s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/7890f175eed044ad8591791045371503 : SUCCESS in 26m 24s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/bc7ba86ae7ed4c25be673f384ddac582 : SUCCESS in 10m 39s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"c98bef6bfbf01d7df54c406caedb2289bc57dd6b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 12:54:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"82f25096b2736fd2d76d8a25833f4fa487f94732","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 12:56:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1177e6bf93941808bc46a715011d886 : SUCCESS in 1m 17s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aceff1d8f9044d92b1929ea1e94fc5a8 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9df0f00bc5e38c37706eaf845f93754af7605528","revisions":{"9df0f00bc5e38c37706eaf845f93754af7605528":{"kind":"REWORK","_number":1,"created":"2020-10-07 15:42:02.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/41/756541/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/41/756541/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/756541/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/756541/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/756541/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/41/756541/1"}}},"commit":{"parents":[{"commit":"02c03839147a03ecc5da390fb3a882b09fadb282","subject":"Merge \"dev: fix test scripts when ironic is disabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/02c03839147a03ecc5da390fb3a882b09fadb282"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-10-07 15:38:43.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-10-07 15:38:43.000000000","tz":120},"subject":"Configure bifrost to use firewalld trusted zone","message":"Configure bifrost to use firewalld trusted zone\n\nWithout this setting, bifrost creates a bifrost firewalld zone only\nallowing network traffic for Ironic services and assigns the\nprovisioning network interface to it, potentially causing loss of\nconnectivity.\n\nUsing the public zone is suggested as a workaround [1] but is not\nsufficient: it allows SSH traffic, but blocks other services deployed on\nthe seed, such as Docker registry traffic.\n\n[1] https://review.opendev.org/#/c/754406/\n\nChange-Id: I80f9d95f02e11fda5916f9a9dd257b688a9db7e2\nStory: 2008153\nTask: 40899\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9df0f00bc5e38c37706eaf845f93754af7605528"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9df0f00bc5e38c37706eaf845f93754af7605528"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
