)]}'
{"id":"openstack%2Fkayobe~756586","triplet_id":"openstack%2Fkayobe~master~Ia73655812feaf5ce0598cdc71b673126423843a0","project":"openstack/kayobe","branch":"master","topic":"sync-extra-vars","hashtags":[],"change_id":"Ia73655812feaf5ce0598cdc71b673126423843a0","subject":"Synchronise global extra variables files with defaults","status":"MERGED","created":"2020-10-07 20:44:15.000000000","updated":"2020-10-09 15:04:19.000000000","submitted":"2020-10-09 15:01:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"756586-1602255708217-2e8b7072","meta_rev_id":"183666fcb3fd3886986d956cb8da7048487eb7a8","_number":756586,"virtual_id_number":756586,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-09 15:01:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2020-10-09 12:56:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-10-08 08:19:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2020-10-09 12:56:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-08 08:19:56.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-10-09 12:56:21.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-10-09 15:01:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"427d1f4bd9359d0dd4caf9b35a994a22b54cee21","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-10-07 20:44:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"914969786e4c6b032778dfa6b2e5c74f89a99904","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-10-07 20:53:59.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"27a9de25751f9a5f16f610ccd25248a431e3cee5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 22:51:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84efe0c3e819421c845827f7fc71abb8 : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3207ae4e6baa4bcf9af7d371dd8954d4 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c77d856041ea43618b9c3fcb20066269 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8767f20e943443f85f6cdc1a86c7018 : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59fb3ccae1144dfeb236411578840ef5 : SUCCESS in 6m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9fac0b0a31e342a8a7defccf3875d681 : SUCCESS in 4m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b745e07d41fb4ef5812ac4dc65d721a0 : SUCCESS in 8m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7ead6171c0c24764bb15164afe0a4624 : SUCCESS in 19m 47s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/cf4052a41dbe4350bf19cf438cd744af : SUCCESS in 59m 14s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/f5b0f1f51c88403b8ee79b76cea46fe3 : SUCCESS in 1h 11m 40s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/d6e0a868498644b989aba8d5a8aaf9af : SUCCESS in 13m 54s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/806117611b2044c2b4ab47e69a38482a : SUCCESS in 1h 51m 22s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/9e5915d0baaa43d9ae44dc0f3dfdcd3d : SUCCESS in 15m 22s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/1f386cb44c704a0e94dca52a65d675f7 : SUCCESS in 27m 50s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/72d757d018f64abc96f134b462fcae4d : SUCCESS in 10m 58s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"8da4a190ef90c55fca7ec4dc26fdf57dd60e1213","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-10-08 08:19:56.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2d2def5f2fa7a97ff80ab9a802ceceaa5bfbe43a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-10-09 12:56:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1d7b9f83e7f700413246f8009143b5633fea6932","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 12:56:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ba1ded3dd08a0004f2fa33c53ff2bc0f8819007d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 15:01:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f349bee4b38441e5845212997054f48d : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a036204de4140eba80107bd126bbd65 : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ec35f681e9e4a6dab86e813c72b1306 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c6a4b05d4784824acdb8a550e9c0560 : SUCCESS in 6m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/27a2c671564f4a3ab059c9dcc32afc65 : SUCCESS in 4m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fbbb9e03d9974990aeba7902ec0562ee : SUCCESS in 8m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5eb24306e4304dad8c0db49f408c3a0d : SUCCESS in 20m 11s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/182dae6db4ca43e388e95a9aa0edd53a : SUCCESS in 1h 05m 38s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/50cee44ca50947c58cf1ab48f46e18eb : SUCCESS in 1h 05m 27s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/1b6bb1c4d5824bd9899a16d57779c798 : SUCCESS in 15m 32s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/fbbef03b08fa4f068db5d4a05905fcbe : SUCCESS in 2h 00m 27s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/49e6d3e454b949829b3d12bf17b42c7c : SUCCESS in 19m 17s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/8cfebebd618d44a7918c82d99e25c6f5 : SUCCESS in 29m 04s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/eda7a015bfc14411a4236deadc7543ff : SUCCESS in 13m 24s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":2},{"id":"d049f020bd248900355bb513628e5f55cfe045e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 15:01:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"ac3634cf24f5d87eb5e7ae2b5e98b5e36f80d570","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 15:04:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f06c49e0afe43b8b594f9e3d886dfed : SUCCESS in 1m 33s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1b48a7130caeedc8a33deaa922419c6aea8fee46","revisions":{"700812e8ead44f69cc2235fb8c07a145c1165eb5":{"kind":"REWORK","_number":1,"created":"2020-10-07 20:44:15.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/86/756586/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/756586/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/756586/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/756586/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/756586/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/756586/1"}}},"commit":{"parents":[{"commit":"02c03839147a03ecc5da390fb3a882b09fadb282","subject":"Merge \"dev: fix test scripts when ironic is disabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/02c03839147a03ecc5da390fb3a882b09fadb282"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-10-07 20:39:47.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-10-07 20:39:47.000000000","tz":120},"subject":"Synchronise global extra variables file with defaults","message":"Synchronise global extra variables file with defaults\n\nIn particular, lvm_group_data_enabled variables were missing.\n\nChange-Id: Ia73655812feaf5ce0598cdc71b673126423843a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/700812e8ead44f69cc2235fb8c07a145c1165eb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/700812e8ead44f69cc2235fb8c07a145c1165eb5"}]},"branch":"refs/heads/master"},"1b48a7130caeedc8a33deaa922419c6aea8fee46":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-10-07 20:53:59.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/86/756586/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/756586/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/756586/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/756586/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/756586/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/756586/2"}}},"commit":{"parents":[{"commit":"02c03839147a03ecc5da390fb3a882b09fadb282","subject":"Merge \"dev: fix test scripts when ironic is disabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/02c03839147a03ecc5da390fb3a882b09fadb282"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-10-07 20:53:46.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-10-07 20:53:46.000000000","tz":120},"subject":"Synchronise global extra variables files with defaults","message":"Synchronise global extra variables files with defaults\n\nIn particular, lvm_group_data_enabled variables were missing.\n\nChange-Id: Ia73655812feaf5ce0598cdc71b673126423843a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1b48a7130caeedc8a33deaa922419c6aea8fee46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1b48a7130caeedc8a33deaa922419c6aea8fee46"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
