)]}'
{"id":"openstack%2Fkayobe~757952","triplet_id":"openstack%2Fkayobe~master~I7670410394b0bb5fc96e4166b6a20714f847250b","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I7670410394b0bb5fc96e4166b6a20714f847250b","subject":"Fix failures in Docker storage driver check","status":"MERGED","created":"2020-10-13 21:00:22.000000000","updated":"2020-10-14 14:46:50.000000000","submitted":"2020-10-14 14:45:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"757952-1602686711652-b16e9fe7","meta_rev_id":"33a3b37e8b4dbab9caa645fddcd491fecef9dc53","_number":757952,"virtual_id_number":757952,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-14 14:45:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-10-14 12:43:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2020-10-14 07:51:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2020-10-14 12:43:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"date":"2020-10-14 12:18:40.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-14 11:11:11.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-14 07:51:50.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-10-14 12:18:40.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-10-14 12:43:16.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-10-14 14:45:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8e400659ba32d417c8f96dff5719ef992be8cb05","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-10-13 21:00:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6b5137bdd178d563f251b97278fd6ab879570b48","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-10-13 21:09:12.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"6f4a20ab1d07fc2b05d34897adb6b405aa83fb7f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-10-13 21:13:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6598871d521ab203ace3f4ff4a82d35161aba593","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 23:13:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cc0f1abfda2c40faac6304404877d79e : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f897a45b66d446db9420f7e54d2af4e6 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/282e9599f389473c9a46458e26f5ec2e : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/92659902138d432aabff0efb59a116ca : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ea294910e9d4198bd31120264f10a7a : SUCCESS in 8m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/27016f1193134e7fa94eecacbc0cb9f3 : SUCCESS in 4m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f11a475d348a46448f60ddd99502d088 : SUCCESS in 9m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7a945159001748088b5f1707b26ab553 : SUCCESS in 20m 08s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/e5fe05fe08bd47c1a0e8a0150903a2bc : SUCCESS in 1h 02m 39s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/82ac574299d048f19477d18a42b3405a : SUCCESS in 1h 10m 16s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/1bb41fa444ed455a9266152ad545bbda : SUCCESS in 15m 00s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/5f10dd239bd84e74b71d8246173bf918 : SUCCESS in 1h 53m 33s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/ecd936da49f144ac9362393097ae8279 : SUCCESS in 17m 11s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/234f5288ddf340388e01eb46abce9c5a : SUCCESS in 26m 48s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/0c2e22b5f4854912ba0b190455690d69 : SUCCESS in 14m 54s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"45ce8d4cf73cc37ebfc49e1c43318169f6a7ecc3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-10-14 07:51:50.000000000","message":"Patch Set 2: Code-Review+2\n\nI was sure I had proposed this fix myself, but it seems I forgot.","accounts_in_message":[],"_revision_number":2},{"id":"c5d9c5abd5e399a88934457cfcbf202417660111","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-10-14 11:11:11.000000000","message":"Patch Set 2: Review-Priority+1\n\nBlocker in AUFN master branch","accounts_in_message":[],"_revision_number":2},{"id":"bbd98b1d0da65784a7a2125862275c159b084819","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-10-14 12:43:16.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3759af06d092e9077539c3ba6d31302ffb4fb1e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 12:43:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"67f8aa2b18a5866733c00a69d6bbec6b5dcd1137","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 14:45:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1443c59c1194caaa90c7cd0d68143bc : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/62080f642d0d4a6f95e688737873831f : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae7118dea62c4657a2c5899a9cbe2da1 : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1f9e30d65aa4d2e81a50735ac340a1b : SUCCESS in 6m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1e725f2269d244da89909932121c4dac : SUCCESS in 4m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fc590a61b5b3407b9a7492d2a6b8d44e : SUCCESS in 8m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/aa4f3d808b0747febbfa7cbb4c557612 : SUCCESS in 20m 38s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/4911204ecca840c18fff057641b3d7e3 : SUCCESS in 1h 34m 15s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/da1d06bbcf174b08afa5b4379eb8da8a : SUCCESS in 1h 03m 43s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/bfec8e5ce5c24719bece92e251daf019 : SUCCESS in 18m 57s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/298be6d5b3944abab484e9687d089434 : SUCCESS in 1h 55m 06s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/89e9e20ddab649fd878b83df86d42294 : SUCCESS in 17m 46s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a4c2ed08b95c4350aef79ce3e1ba2f0f : SUCCESS in 29m 54s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/37c555ad97bb4825bb68978cf7f7fbb2 : SUCCESS in 12m 22s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"6c0bfd93104940e5d268955fb5fd318cd22de203","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 14:45:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"33a3b37e8b4dbab9caa645fddcd491fecef9dc53","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 14:46:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2f4699a75b0479da3f51cc8050ea9f1 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"949ed03de5428c90c728aa8ad1e1c7bd85374622","revisions":{"95c34472ac114227b94683750ac149ff6722bffb":{"kind":"REWORK","_number":1,"created":"2020-10-13 21:00:22.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/52/757952/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/757952/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/757952/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/757952/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/757952/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/757952/1"}}},"commit":{"parents":[{"commit":"1b48a7130caeedc8a33deaa922419c6aea8fee46","subject":"Synchronise global extra variables files with defaults","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1b48a7130caeedc8a33deaa922419c6aea8fee46"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-10-13 20:59:00.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-10-13 20:59:00.000000000","tz":120},"subject":"Fix failures in Docker storage driver check","message":"Fix failures in Docker storage driver check\n\nOccasionally the can_talk_to_docker key is missing from the result of\ndocker_host_info. Assume we cannot talk to Docker if key not present\ninstead of failing.\n\nChange-Id: I7670410394b0bb5fc96e4166b6a20714f847250b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/95c34472ac114227b94683750ac149ff6722bffb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/95c34472ac114227b94683750ac149ff6722bffb"}]},"branch":"refs/heads/master"},"949ed03de5428c90c728aa8ad1e1c7bd85374622":{"kind":"REWORK","_number":2,"created":"2020-10-13 21:13:09.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/52/757952/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/757952/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/757952/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/757952/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/757952/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/757952/2"}}},"commit":{"parents":[{"commit":"1b48a7130caeedc8a33deaa922419c6aea8fee46","subject":"Synchronise global extra variables files with defaults","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1b48a7130caeedc8a33deaa922419c6aea8fee46"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-10-13 21:11:16.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-10-13 21:11:16.000000000","tz":120},"subject":"Fix failures in Docker storage driver check","message":"Fix failures in Docker storage driver check\n\nIn some environments the can_talk_to_docker key is missing from the\nresult of docker_host_info when the call succeeds. Assume we can talk to\nDocker if this key is not present, instead of failing due to a missing\nattribute error.\n\nChange-Id: I7670410394b0bb5fc96e4166b6a20714f847250b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/949ed03de5428c90c728aa8ad1e1c7bd85374622"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/949ed03de5428c90c728aa8ad1e1c7bd85374622"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
