)]}'
{"id":"openstack%2Fkayobe~758073","triplet_id":"openstack%2Fkayobe~stable%2Fussuri~Ib51ac445cc787d7045d72a9ecec14434d432d9e7","project":"openstack/kayobe","branch":"stable/ussuri","hashtags":[],"change_id":"Ib51ac445cc787d7045d72a9ecec14434d432d9e7","subject":"Performance: skip LVM configuration if no groups configured","status":"MERGED","created":"2020-10-14 10:29:20.000000000","updated":"2020-10-14 20:04:37.000000000","submitted":"2020-10-14 20:02:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"758073-1602705733953-760c1410","meta_rev_id":"c4f821844f23d4ef506e3d214910c705ec2d5582","_number":758073,"virtual_id_number":758073,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-14 20:02:13.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2020-10-14 10:41:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-14 13:04:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-14 13:04:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-14 10:41:14.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-10-14 13:04:08.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-10-14 20:02:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5079c58a3c49e0eff0b2a42dcef7218ec0338840","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-10-14 10:29:20.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"3927e0a77dd592695ae0ac2002b4cb526c8b1111","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-10-14 10:41:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"edcf3c9e63be43c93a6ccdcd804adf556be513f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 12:14:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5fc7c20ddcf049debb9809a109f78614 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53f43ce2c267449cae37370d9c5c934f : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/16d6c9ad509540d6922a92162cb4acb6 : SUCCESS in 4m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/672c0339587e4ec0bbb52b5e2d66b70d : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0ac0456707a45e6a06d98b0bd3ea5e7 : SUCCESS in 6m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0e017ae8ed644936bd116e68e7c3f711 : SUCCESS in 4m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e6d023f08db343a990137c24f3baf766 : SUCCESS in 8m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7f1e2316f2ea4265b6f94bf9a3e9c85f : SUCCESS in 21m 07s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/a7cc611a64ee47889eaded5c32656b4c : SUCCESS in 1h 06m 12s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/ad78c3dea1d0404a8a57e7d6288855a6 : SUCCESS in 15m 48s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/8971b7426d4d494090a1e2eb4a945f3a : SUCCESS in 1h 42m 37s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/a0a452b064224e149a5aa915300a6dd1 : SUCCESS in 18m 55s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/d351483147af4263ac9697cc5d91cbe9 : SUCCESS in 30m 36s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/4ba3f95ba49d48218bacf99bf3e8c714 : SUCCESS in 12m 18s","accounts_in_message":[],"_revision_number":1},{"id":"044ebbf338c03fb28949c04e6b36b0437c0cffbf","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-10-14 13:04:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b45438cec866bbb2e46505c93a3d44ee335dc3ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 13:04:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cec58a3f6700f2dc9201780a8b9c23dfd76b9ff3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 15:11:37.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04a5cef7621d46ce9657cab5e8e3dcbf : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eee5ea0a7879471fb8fa6d128ab12345 : SUCCESS in 5m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c568d0acbec34c3589b29e47d13a40ec : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd5daa402ff647e2a8a2d3dbd83a6aaf : SUCCESS in 6m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/581117307f104d108ee9e1be5d85d8ed : SUCCESS in 5m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5b71dec6446b4baeb71a8219bca0b4da : SUCCESS in 8m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c1ebbee5c7b44ddb9a246781bbc35cf0 : SUCCESS in 19m 54s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/8b669cba3335437b99dc9945193ecb54 : TIMED_OUT in 2h 04m 22s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/8dccb293516f4cdaa75d9475900621c3 : SUCCESS in 15m 49s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/3f8c953f9516427fb476ba805cb4efa2 : SUCCESS in 1h 54m 42s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/d473823ee9ae4829a5d3274abcf3dd45 : SUCCESS in 15m 21s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/bbee1db7d70147dba1f021bb5fc0fe5c : SUCCESS in 30m 38s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/7cf995a5dbab4687b5482ae132c8eb83 : SUCCESS in 10m 59s","accounts_in_message":[],"_revision_number":1},{"id":"1c3f1ded41367b3c89aa4f5e9aeb5367777cfbfc","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-10-14 16:07:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"346ce4ce9fc9f56f100042c1325068d712c55217","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 18:05:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b7894ffa30f4d09b289f529eb71cad8 : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6df11b2243d480781371a3ee063a7aa : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/39798e3f31df4e3a81250b1a9bb3d138 : SUCCESS in 4m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c7aa9381f51940428bb95e46dd4f696a : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/beefeaa6592249ecbce7b5b92fd5a586 : SUCCESS in 7m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8b48d9c45d1e4bf79e7e5aeeacf921ef : SUCCESS in 4m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/dbc07fe8f4914cf9b8c45cedcb22af5f : SUCCESS in 9m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f992b5663334477aaed6db72907e88b8 : SUCCESS in 18m 52s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/fae64b0ae4cb40b2a787f1fefca50941 : SUCCESS in 1h 15m 13s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/060eb20372184507bf7ef55862c94ab0 : SUCCESS in 16m 05s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/697e2db1b91d4b129bcc9acaf045ce8a : SUCCESS in 1h 52m 11s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/f7451d93bf4a49cda944c24ab4790188 : SUCCESS in 18m 51s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/5f9031ae0d714d5fae06d2a1f76af9b7 : SUCCESS in 29m 08s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/4e4c69335d4340abaad6c994cb6bb9ed : SUCCESS in 10m 48s","accounts_in_message":[],"_revision_number":1},{"id":"a50c1d11a688ee25855b1364e713f1335753add6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 18:06:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"608ff6f4bcb2e4edcca8ed095733c8ae445b26cf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 20:02:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c0f8a259936dbcb212e78dba10ee65a467270e09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 20:02:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89065ce416484b28b3b601b01fc24508 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c7861c3d8b54580bb4b5c20f0a8f9fb : SUCCESS in 4m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f75ee46ed1c741118e113ac95eae55cb : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9e0600c27ca4e638355d72d07bdaa69 : SUCCESS in 8m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/79e34986f9c844a694647dd443c1d6fb : SUCCESS in 5m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b27f5da9662d430f90d6dbaea7df69ad : SUCCESS in 9m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/17c579a84b8f451f8a46169c68eb117b : SUCCESS in 19m 03s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/079f6b90480047bd9ec1f9e5c9c8c39c : SUCCESS in 1h 04m 58s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/f1d597ae3eee4a5db2129979c22512c8 : SUCCESS in 18m 12s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/4b568137b644443492b858ebe44256d2 : SUCCESS in 1h 49m 19s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/a0e9aaf92a2e488593689f3656a15351 : SUCCESS in 18m 01s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/2b8d11f9db6340caa4773f8334e1445e : SUCCESS in 30m 00s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/939c5b552bb74397af31a1e79ca77f6c : SUCCESS in 11m 54s","accounts_in_message":[],"_revision_number":1},{"id":"c4f821844f23d4ef506e3d214910c705ec2d5582","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 20:04:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de1250db5a77493b921a53d38cb82305 : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7873dd85d2077f35c22d8127757d986e1c1ab79c","revisions":{"7873dd85d2077f35c22d8127757d986e1c1ab79c":{"kind":"REWORK","_number":1,"created":"2020-10-14 10:29:20.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/73/758073/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/73/758073/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/758073/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/758073/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/758073/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/73/758073/1"}}},"commit":{"parents":[{"commit":"1d65c770d65b7148849803fe10de372929d222f4","subject":"Configure bifrost to use firewalld trusted zone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1d65c770d65b7148849803fe10de372929d222f4"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-08-04 16:17:38.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-10-14 10:29:20.000000000","tz":0},"subject":"Performance: skip LVM configuration if no groups configured","message":"Performance: skip LVM configuration if no groups configured\n\nCurrently we import the manage-lvm role with a condition that skips\ntasks when no LVM groups are configured. At large scale this is\ninefficient.\n\nThis change improves the performance for hosts where there are no LVM\ngroups configured, by using include_role rather than a static import.\n\nStory: 2007993\nTask: 40664\n\nChange-Id: Ib51ac445cc787d7045d72a9ecec14434d432d9e7\n(cherry picked from commit 5ff5e6b10011875d1c29d16a8b39498adc0145e0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7873dd85d2077f35c22d8127757d986e1c1ab79c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7873dd85d2077f35c22d8127757d986e1c1ab79c"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
