)]}'
{"id":"openstack%2Fkayobe~758075","triplet_id":"openstack%2Fkayobe~stable%2Fstein~Ib51ac445cc787d7045d72a9ecec14434d432d9e7","project":"openstack/kayobe","branch":"stable/stein","hashtags":[],"change_id":"Ib51ac445cc787d7045d72a9ecec14434d432d9e7","subject":"Performance: skip LVM configuration if no groups configured","status":"MERGED","created":"2020-10-14 10:29:45.000000000","updated":"2020-10-16 15:04:47.000000000","submitted":"2020-10-16 15:03:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"758075-1602860583876-a663f06a","meta_rev_id":"def49e7eed3be042735d9edd6de8d50e3f28a510","_number":758075,"virtual_id_number":758075,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-16 15:03:03.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2020-10-16 10:45:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-16 13:19:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-16 13:19:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-14 10:42:10.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-10-16 13:19:52.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-10-16 15:03:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0657af6aa91e3fdf96b2d2bf3e45d0b8a0623f32","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-10-14 10:29:45.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"32795ba0b05c5213838137665b6aa86dbbe52a8c","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-10-14 10:42:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9d8435c8cc405f12486ff33c709fcf545660adde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 12:27:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1effb763a039404db01c04373477482f : SUCCESS in 8m 16s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3d600effebc24ebcb7236b76d6a50c8f : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f512bf6c19684d408d3c98a784148bfb : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84de8aea71b94c75843f0d81a1a65b2c : SUCCESS in 5m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9849c8a3e52c42b185aff4ffbe4be5e4 : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec4c501f488d421bac86ca295396813f : SUCCESS in 4m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d890269cddfb40e78fde7d1e7759f0cf : SUCCESS in 4m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f539053adfc14f9390b751be434c1076 : SUCCESS in 9m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bb53800dab0640cf89aeb4d4a39c087d : FAILURE in 8m 43s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/b9b0207434ad41c58b3c0faf6af96158 : SUCCESS in 58m 03s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/7158839b5bb5499b957ced6cc4014e77 : SUCCESS in 1h 46m 59s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/0b59de60605d47c98c9e910d16d5863b : SUCCESS in 24m 39s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/105bc2b13c36478d8b694bd4dbccbee6 : FAILURE in 40m 39s","accounts_in_message":[],"_revision_number":1},{"id":"25033103819642f682cd0be1c8d89eaed4efe906","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-10-14 13:05:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"efa5e01e64cedd34c03c69781988bc25f35151bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 13:21:53.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/33f7b55e726f4d0ab0b247fc6f2da4f3 : SUCCESS in 5m 12s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3e11392da9464de68557d4fb16d51b84 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b11c23492d04e00836542de021a0035 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/972d522834b14a688620f4d5fc6a155c : SUCCESS in 4m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/973ec0365ecd491dac762cdbbef4caec : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74c2e8c657d2451ca76efb56946072d1 : SUCCESS in 4m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/60555e023ece4df2885bfa28d9b658ba : SUCCESS in 3m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9821b53e44c147908e0bcf7f449da233 : SUCCESS in 7m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c4a99f1898b74da99f0f909bfe4ec75a : FAILURE in 7m 21s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/c6449673059540f684181b4ffd89b881 : RETRY_LIMIT in 4m 24s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/fd4b8f8edbfb4c01a71ffe50afb6663c : FAILURE in 2m 50s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/f359aa2aceef4f51ad28bc98182641c9 : RETRY_LIMIT in 2m 50s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/5e55127255854927ad00420456dde962 : FAILURE in 3m 09s","accounts_in_message":[],"_revision_number":1},{"id":"95e46a92246ce8477c3130d661a4dc8133f05ac9","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-10-14 14:13:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"dd6605035a337c0b81c477a1424106ba33668b1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 14:29:03.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4a809a66a8dd4126b19a176c84cd0332 : SUCCESS in 4m 33s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6d0b31c13f6747009b82174a9d903ecc : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d82fad613f434fcc8b06dbe79a4b0422 : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f450af1a9b1d44be81f0076c36f3c0f5 : SUCCESS in 5m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f187c7de52854806baa9c88e0ad1f396 : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa7cdd23382c444cbb61dd80af3d1fa4 : SUCCESS in 4m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8533f2da0a5f4cffa3db5ba43b56eba6 : SUCCESS in 4m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/36a6ecbbca5d46eabff0b9995e1c4494 : SUCCESS in 8m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4d3f77e3498f48628458b1a687c2e90d : FAILURE in 8m 06s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/3331794e47484c51ac2109acb32f1468 : RETRY_LIMIT in 4m 35s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/dc1c6a41fe7e43aba3c56c49dbbcd917 : FAILURE in 3m 20s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/c5ed2be9e1564987bb057733c6f7aa7f : RETRY_LIMIT in 2m 44s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/237baafb643541cb832df405db8596a3 : FAILURE in 3m 20s","accounts_in_message":[],"_revision_number":1},{"id":"5147e8a0a1136640c8f37456ab9314e1158b2b92","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-10-16 07:39:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f2741828acc9233a233a83c75a27e6e8c9389edb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 09:19:30.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a892312317514a278ce4182aeee4225f : SUCCESS in 5m 37s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6496e9cb3137492d9ac6017e0b712d6e : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f153968ef39c4c5db6dd58776b3aafa7 : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ea43d91aee44e22a05ca1d16b895713 : SUCCESS in 5m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/aabe2cf36ed44cf09e628998a732b425 : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c954b38bf1045b5aa211c721cae8d45 : SUCCESS in 5m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c6d0cd3655d54f95bb2d98fa97015762 : SUCCESS in 6m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/24e9a53165384f079a453351e683b907 : SUCCESS in 10m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bfe0914445a3417a9f48a49349e39606 : FAILURE in 7m 43s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/d898aee31ad2427ca2c86fa4eb94cfcd : SUCCESS in 1h 05m 00s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/5f70130ff05a4c6cadcac93cd081a014 : SUCCESS in 1h 38m 37s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/7fbde3c643b042eeb959c73512705541 : SUCCESS in 25m 12s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/9de45a4390d84eb9937c2859cbde3d4f : SUCCESS in 23m 33s","accounts_in_message":[],"_revision_number":1},{"id":"cc276a2bcbf9734f6c3a3129cf61c7536c2026a9","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-10-16 10:45:27.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"792cdf07617dfef66565feb2075eeb09b1b59791","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 12:43:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe2d3bcef2e54967886f5073c4f8a92b : SUCCESS in 4m 55s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/cadb85bbfa9f4a548666d520d802079b : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6bdf08f470e94d4198008f4290809a11 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/daae2fb390fd485987a763434d74254e : SUCCESS in 5m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3a75fe3b0a9a40b2bd8a56098d789c3f : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a58785ba27246d981280ac0a291f55f : SUCCESS in 5m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a1663ebd8d864749bb34361730501389 : SUCCESS in 5m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8158b94a9989423fa75eef2a1ab54aca : SUCCESS in 8m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/af2fa6fb58814f0eafb1bff4a2ba0c26 : SUCCESS in 18m 44s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/51afeebab77e46e498a2b135443ae458 : SUCCESS in 1h 07m 44s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/b152fc0c9b9d4c5d871b09cfdcae8beb : SUCCESS in 1h 52m 48s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/17b89a032af246d1b899d557f94bb257 : SUCCESS in 25m 58s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/73f6bb1e8b544eadaed2b7f76d8d992c : SUCCESS in 36m 43s","accounts_in_message":[],"_revision_number":2},{"id":"cd0e34e1aa5fd6bb42327d055e670bb0d133beae","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-10-16 13:19:52.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5fdf5fda2ce72456e6085d782c60890e8da81bb6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 13:20:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"34cdea21ca4f55f75a56b90156780da723dc2ade","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 15:03:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"88eb221048ad2fb12b9b14520b767e859785b76b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 15:03:03.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4c144d7c35504f4eb8fa0ca7442fb818 : SUCCESS in 9m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d7943083a6c14aa5b10dc5eff864615e : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3844c59c6016435ba2eb8541d1961087 : SUCCESS in 7m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/121b5191077848218861297fb532cb6c : SUCCESS in 7m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de2cfe14468b49b18b9660a6e1405036 : SUCCESS in 6m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/244f84ef46764fdba57238eca78ed541 : SUCCESS in 7m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e0ba9dcb8c5f4adf90ac66d2d8be5ebf : SUCCESS in 11m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d903a97d94094ff2b599979d99ada244 : SUCCESS in 23m 19s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/e4901968ce154990ad9d44baafd88126 : SUCCESS in 1h 06m 29s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/34480a959a924f44a55f1b9566b2180d : SUCCESS in 1h 41m 24s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/d593892a62e24f2ebbc4073fb879edbb : SUCCESS in 24m 55s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/98f57ba8dbd9462ebefa66b3369cb49e : SUCCESS in 43m 17s","accounts_in_message":[],"_revision_number":2},{"id":"def49e7eed3be042735d9edd6de8d50e3f28a510","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 15:04:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/132b9d9eed494f2dabdfc55c6cf426c7 : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5eca75c31f4ffe0c875b0bcdcc6245ca174799b3","revisions":{"d143981542da846e8dec9abff985f018b2f4f0b3":{"kind":"REWORK","_number":1,"created":"2020-10-14 10:29:45.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/75/758075/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/75/758075/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/758075/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/758075/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/758075/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/75/758075/1"}}},"commit":{"parents":[{"commit":"f95f9a2c734cc45aa7a87613b8c93a651e9be414","subject":"Make pip_index_url variable conditional","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f95f9a2c734cc45aa7a87613b8c93a651e9be414"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-08-04 16:17:38.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-10-14 10:29:45.000000000","tz":0},"subject":"Performance: skip LVM configuration if no groups configured","message":"Performance: skip LVM configuration if no groups configured\n\nCurrently we import the manage-lvm role with a condition that skips\ntasks when no LVM groups are configured. At large scale this is\ninefficient.\n\nThis change improves the performance for hosts where there are no LVM\ngroups configured, by using include_role rather than a static import.\n\nStory: 2007993\nTask: 40664\n\nChange-Id: Ib51ac445cc787d7045d72a9ecec14434d432d9e7\n(cherry picked from commit 5ff5e6b10011875d1c29d16a8b39498adc0145e0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d143981542da846e8dec9abff985f018b2f4f0b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d143981542da846e8dec9abff985f018b2f4f0b3"}]},"branch":"refs/heads/stable/stein"},"5eca75c31f4ffe0c875b0bcdcc6245ca174799b3":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-10-16 10:45:27.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/75/758075/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/75/758075/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/758075/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/758075/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/758075/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/75/758075/2"}}},"commit":{"parents":[{"commit":"e861aebb2d956374b84ed35814efa4770bf88fdb","subject":"CI: Fix kayobe-tox-molecule job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e861aebb2d956374b84ed35814efa4770bf88fdb"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-08-04 16:17:38.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-10-16 10:45:27.000000000","tz":0},"subject":"Performance: skip LVM configuration if no groups configured","message":"Performance: skip LVM configuration if no groups configured\n\nCurrently we import the manage-lvm role with a condition that skips\ntasks when no LVM groups are configured. At large scale this is\ninefficient.\n\nThis change improves the performance for hosts where there are no LVM\ngroups configured, by using include_role rather than a static import.\n\nStory: 2007993\nTask: 40664\n\nChange-Id: Ib51ac445cc787d7045d72a9ecec14434d432d9e7\n(cherry picked from commit 5ff5e6b10011875d1c29d16a8b39498adc0145e0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5eca75c31f4ffe0c875b0bcdcc6245ca174799b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5eca75c31f4ffe0c875b0bcdcc6245ca174799b3"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
