)]}'
{"id":"openstack%2Fkayobe~758076","triplet_id":"openstack%2Fkayobe~stable%2Fussuri~I3014983f481a5dca7c93e140b3e10caa5d537669","project":"openstack/kayobe","branch":"stable/ussuri","hashtags":[],"change_id":"I3014983f481a5dca7c93e140b3e10caa5d537669","subject":"Add missing barbican.conf support","status":"MERGED","created":"2020-10-14 10:35:06.000000000","updated":"2020-10-16 00:25:26.000000000","submitted":"2020-10-16 00:22:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"758076-1602807761205-11ac1cc1","meta_rev_id":"7f29d18ea82368453380ef2ba4a6d107119d7aa8","_number":758076,"virtual_id_number":758076,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-16 00:22:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2020-10-15 13:37:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2020-10-15 12:03:33.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-14 12:58:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2020-10-15 13:37:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-14 12:58:21.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-10-15 12:03:33.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2020-10-15 13:37:39.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-10-16 00:22:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"79cdfc9907298227b522a9349a2ac178e3850e01","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-10-14 10:35:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"56c5e5c37803c45122c5e4be40fd1bf0de66233b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 12:54:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ce799593feff4bbd9ae0878bf6003574 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a237e0e821b44bca1cee0e6d17f0c42 : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4492b18730d5452bb4da5645255079c2 : SUCCESS in 4m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7830af613d8c4bdea618380dd04ad4e7 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7aba5941b9ad40418cb6b2bf434d37fd : SUCCESS in 6m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/014bacfef5e845019601e181fde52368 : SUCCESS in 4m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/74f17bbeba1f4728bc4a54bef011977e : SUCCESS in 5m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/441f011f26434655bc12cea22ca513e4 : SUCCESS in 8m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4c9023b28f414390a914dbea90985b19 : SUCCESS in 20m 05s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/1e4cbb3933c2475584311fe8027c046c : SUCCESS in 1h 04m 39s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/f798d856a3944c7a86b545e073030c66 : SUCCESS in 17m 04s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/344a501c458248bda8f67d8018f1987b : SUCCESS in 1h 50m 57s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/c00838f3328e4e18a9228b7fd2a5e169 : SUCCESS in 18m 04s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/7fdca1b7c1da4513a96332f8ce822b51 : SUCCESS in 28m 47s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/9ce4c1e6d03844d3b557993dc81c5d49 : SUCCESS in 11m 59s","accounts_in_message":[],"_revision_number":1},{"id":"f2a7c5eb329f2b62634e79c977cde09789e305cb","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-10-14 12:58:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f2cd97eb4f0e703cd10e5c12e91ef60469e6bf7f","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2020-10-15 12:03:33.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"228b46e46cac2aff2c58789ebeab4ccfa917a32b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-10-15 13:37:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"471c85c2ea697cb1014417588c0fca9a3c3bad32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 13:37:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"314d16c66ae9110d921bcf2eedf6dcbd899ed8e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 14:16:50.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e71ba532621f431b83d54a06b1e2758d : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f6a89162c5aa46dca5e161a279b41417 : SUCCESS in 3m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/07d0aff101a44ec39bf358c4e2cfdbbc : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/736857b7f2304ce696548414ebfacc80 : SUCCESS in 7m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9de5ec0c9957422bb1acbd5863da6de7 : SUCCESS in 5m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0967a3fb04204119b4a2b2755bb54786 : SUCCESS in 7m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4c057808f4114ffa8e9888534b48d145 : SUCCESS in 11m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3131dc935e0f46c0b89f529edf511b66 : SUCCESS in 19m 39s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/93201b16aed24661920674b94f112a17 : FAILURE in 28m 45s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/75d0daa4b1e74f418fc12e54d53149c9 : SUCCESS in 17m 35s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/d27c027baf214cc995c252d23970e05f : FAILURE in 30m 17s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/dcd7a80eedab4454bef4737d872cc0ae : FAILURE in 25m 12s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/35f4a3faf79d4d1f902a4bcc94a9b975 : FAILURE in 8m 20s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/fce68f2288244e48b7594570f24619d1 : SUCCESS in 13m 50s","accounts_in_message":[],"_revision_number":1},{"id":"0150a425b321fbea9d72038ad5dcfb8512b9233c","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2020-10-15 14:44:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4a45c301c9449e89f7ff49afcfb1827459b09bc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 16:04:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8370dcc3a6ab4b45a064301901448145 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2635eac9724454b9ee8ec673a91d49b : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f3c39330be541b79b4fc501afdb2317 : SUCCESS in 5m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/28423c0f16134447b72081ed4cf0d92d : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e5790d9fcac41b281eed0ad4c7e5c73 : SUCCESS in 7m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a04015170f0540309eebed07b45909be : SUCCESS in 5m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/49f190fb5c094d90974929ac3a131cf5 : SUCCESS in 7m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6d55563febf047d68a216aa8d91d5f7f : SUCCESS in 9m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b21866b4ba8b40918f29cf0461a8806c : SUCCESS in 21m 34s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/6c1b9c73f09f4700aa7485bb2f748df2 : SUCCESS in 53m 05s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/4f2cfbf0a96441b3901b03369fd68ff1 : SUCCESS in 17m 55s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/c2e4e05396f1489e8635152c778c7610 : FAILURE in 1h 08m 33s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/5cbbde0df8fe4669b22e4acd60ebb134 : SUCCESS in 19m 59s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/6f6f60f1c22b408ca82503fc7d4f3184 : SUCCESS in 37m 48s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/c27caad5dda84514a6612b0432884bb9 : SUCCESS in 14m 24s","accounts_in_message":[],"_revision_number":1},{"id":"a3adb3d1dfebc3556938c6de7e37dab20bdf67c4","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2020-10-15 16:11:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3acc8c59324851a68d18f524ee045376e2da8cfb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 18:19:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a4f191d3dbb84ca1b79a527ea58cb0a6 : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5608f1615c7489d94798cf1f020a943 : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7bfba37f5b7149c8be37a5ab0c8e67af : SUCCESS in 5m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/71a660d3e72b4d428a44af5bcca24db0 : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96cea4a130994518a3b189381bd06408 : SUCCESS in 7m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7670783a745a4ecbbedda9aec92fd212 : SUCCESS in 4m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0e6d8c3b913645e0a46b1f5578c1f33f : SUCCESS in 4m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e345c85b93d2486a90b2557efa6d7df7 : SUCCESS in 9m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e4bd8bd2522643fa96a7492c5ff9ced9 : SUCCESS in 20m 39s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/1714e0610dc84b419ba4f17abe903b66 : SUCCESS in 1h 07m 01s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/79c2a325b3004a739e23ea4acbf00446 : SUCCESS in 18m 25s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/28d606b9b61348628bd349504725d597 : SUCCESS in 1h 56m 28s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/4436ef76f64d4a5fb828bac6cdc7f7dd : SUCCESS in 20m 35s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/49df8fc7bb1a4cfd9ce691542527871e : SUCCESS in 27m 15s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/fc6f4ab6d33d42308f6b90a3f9af4ac9 : SUCCESS in 11m 22s","accounts_in_message":[],"_revision_number":1},{"id":"3db4ce40ecceb2b16287c4b09a33a77e07173170","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 18:19:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"50663070ed5af013ab2f555fa7f8f781a25ba5ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 18:56:41.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b02c0f9e6e946f4ab346951abf6d954 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be237c0675fc4ce583dc957edb852a24 : SUCCESS in 5m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/141e2de9392a485ab0e6da59e5f4b504 : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65bb7e02a95c47c09454278b8d6681b1 : SUCCESS in 9m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b7e1ed86e84b493baab32d829358847f : SUCCESS in 5m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ff8aa2b17aca4b9caba3589d7d5226da : SUCCESS in 5m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/085dd1c964994df1a7689f8a29323386 : SUCCESS in 7m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/367c5db0a3884c1d9f4500c48f2e604b : SUCCESS in 19m 30s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/28ec6d83bc8a4063b00e138be26712be : FAILURE in 25m 12s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/7ad708612e48402e87ddd2be162edbd9 : SUCCESS in 16m 57s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/8a732493a808428eb70198aa2c2d617f : FAILURE in 31m 08s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/cfc0f3722cc443469205731976430ff6 : SUCCESS in 27m 46s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b1b7c14efaa84895984963687a7e9875 : FAILURE in 26m 44s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/e907bb90b4ad4b5f941bc3bcf57f7417 : SUCCESS in 12m 09s","accounts_in_message":[],"_revision_number":1},{"id":"0a9bc819e4ac57bf7be364a656b96ae1aa12cf9f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-10-15 19:56:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e1b51af12ea4931fed4180c5a097fecee7e26e24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 21:59:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f5bc28a22e34ccab8b29109a02def59 : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c74869e898f40d489b30bec047f02df : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/102509208e8b4651aa95d405c4ad5d65 : SUCCESS in 4m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/db1222b62c93486fbb9b6a7e7ef01d35 : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e8dc3870fb24c8fb6fc3b8c14c6c763 : SUCCESS in 9m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cee28ccdcdc746dd8bb0d2f8342830f4 : SUCCESS in 4m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/01e0092edf6042978365261ec0f3b635 : SUCCESS in 6m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cb1f5f7117524307ad8b9e2256f16dab : SUCCESS in 9m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c01e4e212bc243d9932bad7f88e01f23 : SUCCESS in 19m 46s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/efdfa13023cf46f6846ad5c417990c5d : SUCCESS in 1h 22m 17s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/00bc0baa95d2495e946321f64299f1ee : SUCCESS in 16m 45s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/41ac6e5e95344130b8eb3affbe7167f5 : SUCCESS in 1h 55m 42s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/cc457e2c106849f69a1dd27a03aca540 : SUCCESS in 19m 30s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e0cb54d0256c41c0bfb0af08a523f02d : SUCCESS in 30m 04s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/ab8380cb3a1145cfad6f20dc1ca84985 : SUCCESS in 15m 46s","accounts_in_message":[],"_revision_number":1},{"id":"6d5d54c833577e2f3e240e476d9b20bd2e4b6df7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 21:59:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"93cebbb46c12743c7d9755bea512a90b8683c9a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 00:22:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2570c41cd78d4c66ba94a991914e168a : SUCCESS in 7m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9141a02f797f435e8022c709d912346e : SUCCESS in 7m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/508fb09d37714c9192b659f6679d4a66 : SUCCESS in 7m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ce0cfea1ee6403193189adff5ac92e7 : SUCCESS in 8m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c588414c93bb43c485939d9307654a91 : SUCCESS in 7m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/85ae7d0fb8884f67b8ce6aa94887f817 : SUCCESS in 7m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1a1e27d8df4441ec9b181ccd74fd4803 : SUCCESS in 11m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/992b4261566b4e28971a1d8cf88fa23d : SUCCESS in 22m 58s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/e9bcf356b2bd4959950e9fd411a81789 : SUCCESS in 1h 08m 48s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/910edf93dcd34063aebcde89d0838275 : SUCCESS in 20m 22s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/40e7c9166a0646c5ab0f2af47479b1f9 : SUCCESS in 1h 50m 24s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/5606414d66f34b57afde3609ffe4cfe6 : SUCCESS in 18m 52s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/1913edf079f74592a953e58ebcf69c2b : SUCCESS in 29m 35s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/cdcd2776c3364c8b92d7163abbdd7308 : SUCCESS in 13m 35s","accounts_in_message":[],"_revision_number":1},{"id":"d60396a96bbd0de3321759c0fded1d097bc2e9ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 00:22:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7f29d18ea82368453380ef2ba4a6d107119d7aa8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 00:25:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e26548c2a4bf4b4ca815c3ac650f9f34 : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1fac67612f984b0f855384d91924fdeb : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"43b7dc403a06a720d5d6134a3b9bf3065675be74","revisions":{"43b7dc403a06a720d5d6134a3b9bf3065675be74":{"kind":"REWORK","_number":1,"created":"2020-10-14 10:35:06.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/76/758076/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/76/758076/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/758076/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/758076/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/758076/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/76/758076/1"}}},"commit":{"parents":[{"commit":"1d65c770d65b7148849803fe10de372929d222f4","subject":"Configure bifrost to use firewalld trusted zone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1d65c770d65b7148849803fe10de372929d222f4"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-09-21 09:00:35.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-10-14 10:34:58.000000000","tz":60},"subject":"Add missing barbican.conf support","message":"Add missing barbican.conf support\n\nStory: 2008170\nTask: 40925\n\nChange-Id: I3014983f481a5dca7c93e140b3e10caa5d537669\n(cherry picked from commit 0f25900a43895f5154820292d44ec999cd1ec2ed)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/43b7dc403a06a720d5d6134a3b9bf3065675be74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/43b7dc403a06a720d5d6134a3b9bf3065675be74"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
