)]}'
{"id":"openstack%2Fkayobe~758077","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~I3014983f481a5dca7c93e140b3e10caa5d537669","project":"openstack/kayobe","branch":"stable/train","hashtags":[],"change_id":"I3014983f481a5dca7c93e140b3e10caa5d537669","subject":"Add missing barbican.conf support","status":"MERGED","created":"2020-10-14 10:36:09.000000000","updated":"2020-10-15 10:06:20.000000000","submitted":"2020-10-15 10:04:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"758077-1602756279646-52096430","meta_rev_id":"fe31624e0845fb06f6b4489f0d74da67b41ce7dc","_number":758077,"virtual_id_number":758077,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-15 10:04:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-10-15 07:58:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-14 20:23:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2020-10-15 07:58:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2020-10-14 10:36:09.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-14 10:36:09.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-10-14 20:23:07.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-10-15 10:04:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d0fc15dc2c59b63736136395ccbe3bca0973c8c5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-10-14 10:36:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"510b3d35c7703da280f439669816fa7c4902a994","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 12:54:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84fee59ca3214ebb93cd943842f227ec : SUCCESS in 5m 25s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d7bedda5e1964f13a4b86febf47c4a75 : SUCCESS in 8m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/888d9f8ae2034bbebd5d06a0e0cd227c : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5d2d51be6564d68b11b3a1c96ae850a : SUCCESS in 4m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8d545befbb6b45e595b4eba1191776aa : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6eabe5013627477d8afb871dd8bc93c6 : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe56ffb4233d4345825fc397169a6a0c : SUCCESS in 3m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e0ff6129f206414eb3d6c0e899874af5 : SUCCESS in 4m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4b4db78be2cb43cda1691c083191fdcb : SUCCESS in 9m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f49192e855d54e35b88ee08babc23183 : SUCCESS in 19m 22s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/949d1c6a84e747fdb8fc148e1a03a3c0 : FAILURE in 53m 37s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/cce486a1247241f190e2b6930e3d38d0 : SUCCESS in 1h 07m 56s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/897390f0a890445ba312690754b70390 : SUCCESS in 13m 43s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/ddbd6fed8a8a49909bf0b6dad2099e13 : SUCCESS in 16m 41s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/04ccb02b5a804f7bb8d97b082d533b5f : SUCCESS in 1h 44m 51s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/a627ae1718d34ab286c46a9551a00034 : SUCCESS in 23m 52s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/1d34d7011e0b4056adb6e10365c47703 : SUCCESS in 18m 48s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/fbcd3c38720e40a1bbf110311a6bcdd2 : SUCCESS in 51m 36s","accounts_in_message":[],"_revision_number":1},{"id":"170e4f0e1290ecf5e49a110bc5dea10cb1fea6b6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-10-14 13:00:54.000000000","message":"Patch Set 1:\n\nrecheck\n\nDownload error on https://pypi.python.org/simple/pbr/: [SSL: CERTIFICATE_VERIFY_FAILED] certificate verify failed (_ssl.c:618) -- Some packages may not be found!","accounts_in_message":[],"_revision_number":1},{"id":"b6fc324f06e77d1e7af1fd5ea92beb28013a0281","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 14:13:37.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db3bfbfb41f24feaa7c0d89b743ecf81 : SUCCESS in 4m 05s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a483471cf75b4be4be470ad6b5cf96f4 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b1201ef4fbd4dd08afc06cb31550501 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9fc8cf83f334cec92d086ad2b917213 : SUCCESS in 4m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b81477864d1349babb5cd38afc9da81c : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a48653e297ce4a71ba44c955421c0bdb : SUCCESS in 6m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/958f682a4f8a48b480cfff3c6f1501d7 : SUCCESS in 3m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8007f8bbdaa044a982494646a5de47b5 : SUCCESS in 5m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/dc307f684ff54b74b1fba3e05ec6a515 : SUCCESS in 9m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/507230d0769c4f23b9fb61b7ec134bca : SUCCESS in 19m 35s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/b54b5907ada74034b076edeeb15f5ef1 : RETRY_LIMIT in 3m 29s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/1d78ecedc408471581d302b93350b6e9 : SUCCESS in 1h 08m 49s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/f7462dd2382d4181b582ad02d5c29000 : RETRY_LIMIT in 3m 48s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/b7173e69196b4d27a21883f938c04eff : SUCCESS in 15m 15s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/82ae0104ebf64818bcb840a7d8b98e72 : FAILURE in 3m 41s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/23984cc8dde7491786fcc39b0fcefdcd : RETRY_LIMIT in 3m 18s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/7149124148504e7cb396648d4d857db6 : SUCCESS in 17m 28s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/792d6d8c2c004a1da94b890587423ca9 : FAILURE in 3m 06s","accounts_in_message":[],"_revision_number":1},{"id":"41c229c77c0b4fa06d2b6455c626a3a0340ce0a0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-10-14 14:14:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bf4f250e83a4e255cfd333facd537bab449fe572","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 15:29:32.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/adc48029bd88409aa6f4135e245a0c2a : SUCCESS in 4m 48s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3af6631142ed4c559dae1b8a8e0aded7 : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e13406cd08d94125a947d2b6268a78f0 : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5792059fe322434096af1c1729eb4fbe : SUCCESS in 4m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0e13418bc61e46d19dde78b2b4d29756 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b49dd64dbc3439c922397364d99baae : SUCCESS in 6m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/89d2992dcf5142a6982a819026515da1 : SUCCESS in 5m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a34143237cdd42f2b66c38324721f631 : SUCCESS in 5m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/04075c8be5c14c70a588957bcb91161c : SUCCESS in 9m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ca8d7fb9145b4cb09329d0da11749396 : SUCCESS in 19m 31s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/db3945460d5148a1b800548a40b8f5ca : RETRY_LIMIT in 3m 22s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/6b35e16ab22c46c395f638467e97b051 : SUCCESS in 1h 06m 52s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/9c9fcc2523e4409d90833674ad7359c2 : RETRY_LIMIT in 3m 25s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/d6117e40c089495b96a8bf737cfd00ac : SUCCESS in 15m 51s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/198e1a5159b64068a96042007d6e51fe : FAILURE in 4m 03s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/b1897d345f41405a9d0cb4bbb01d618f : RETRY_LIMIT in 4m 04s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/1d398dd710da4c4cbfc0581305562209 : SUCCESS in 17m 50s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/315144c50e804c099d0b284f2a24269c : FAILURE in 3m 02s","accounts_in_message":[],"_revision_number":1},{"id":"f87ec5463b7e0080adaaafc36b129f988d90573d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-10-14 16:41:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"216b7955681080c5717524a60d754c782942d279","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 18:39:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c261e8ded624ddf9bc40f005844951d : SUCCESS in 5m 25s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5eeb66ada21447be9ed74268a4cdc078 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d73414b487845dbaa3f27349c3fbc06 : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9d4d99194c34411841be5e392b8498b : SUCCESS in 5m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e3ccba2cd7fd480ea115a22824ff2419 : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06ed692be1df4890a9f314ce836e1ef1 : SUCCESS in 6m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b878d9d470af454a9b6c261de754cb55 : SUCCESS in 3m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3a52756fc9a94530a53eba7c1446822f : SUCCESS in 5m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/164b5a535ae84b48a0f135e2419a17e6 : SUCCESS in 9m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ee260b3da6ec4636b378c6a5e3187645 : SUCCESS in 20m 48s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/93b0fd2e5fba4900b49bc5a20b30fc15 : SUCCESS in 1h 06m 52s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/b78998002dcd42c0b7afd452f4792acb : SUCCESS in 1h 07m 11s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/97a1c91427704ad799e955dd5ea77ba3 : SUCCESS in 14m 56s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/7a3901ae16ef492c96cb78f302b4d8a6 : SUCCESS in 16m 33s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/6555ae593c3b42d1b03cb072eaa318f1 : SUCCESS in 1h 40m 45s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/db266429bce346a7871ac84b42e0ffa8 : SUCCESS in 21m 16s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/d437b82a68f74b1baf7d0b73276d6929 : SUCCESS in 16m 26s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/1a2a06c6befb4dcab842d50e4015b19a : SUCCESS in 37m 56s","accounts_in_message":[],"_revision_number":1},{"id":"34d046c7dc68ab7a172f83530d586b0781b122d5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-10-14 20:23:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a8b06dbd161dbda0d392039335beb255363c0a0c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-10-15 07:58:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f33926e26570255ece8fc561450d343dfbd1f391","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 07:58:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"47340d527276f0cfa3a5c9bdcfb84712674ed85d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 10:04:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e0ddd98be98d4ca89b34328bf2953d4a : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/35ffadc3a4bc4fcf9c62af2a506f5590 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3237d69ee06c4732aee8597ee4c00676 : SUCCESS in 4m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/00bd0a294ae2436c9974db89e2dd9f59 : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed7f10fdd09e4010a5b2ef220e6a800d : SUCCESS in 6m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4b08c74af46042438c0f40d66f0db835 : SUCCESS in 3m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/99db584f19794e77888173378fd2b454 : SUCCESS in 5m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5996b0ff660f4c61acbf0bcf552fa97c : SUCCESS in 8m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/de9c342da0f04e898b28c988483debfe : SUCCESS in 19m 31s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/0769731c9e8f435394c0b30afbcfe283 : SUCCESS in 1h 03m 11s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/375068e475cf4a2e811a7d9d840e37a8 : SUCCESS in 1h 06m 24s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/e49cf849427347259a73ab0fc5c8c915 : SUCCESS in 17m 19s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/c40748ab37ee4ed49f3f8b9828a8a943 : SUCCESS in 17m 40s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/f47a5b6a02564ab2a15d168055b46d83 : SUCCESS in 2h 04m 20s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/802f09727e2f4eb28c407702f6d3dae2 : SUCCESS in 24m 24s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/79665c5c14054a009731d7a1b6642198 : SUCCESS in 17m 08s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/413c68a05e474369a7ea97f68f5e2600 : SUCCESS in 25m 34s","accounts_in_message":[],"_revision_number":1},{"id":"c6228d3d4f89239cab1cc88d533530a9b1d93787","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 10:04:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fe31624e0845fb06f6b4489f0d74da67b41ce7dc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 10:06:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/879dc111e2b54adb9a7f4e35ae99d7c2 : SUCCESS in 1m 09s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/59579cdb071948ecacd647bdc9c7de79 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"39ef7b5c08dee5362a15dd5a1af2d1be7f040677","revisions":{"39ef7b5c08dee5362a15dd5a1af2d1be7f040677":{"kind":"REWORK","_number":1,"created":"2020-10-14 10:36:09.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/77/758077/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/758077/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/758077/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/758077/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/758077/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/758077/1"}}},"commit":{"parents":[{"commit":"a87ee0267b37fa8bfaa9b7abc683e39e1528653f","subject":"Merge \"When EPEL install flag is not set\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a87ee0267b37fa8bfaa9b7abc683e39e1528653f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-09-21 09:00:35.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-10-14 10:36:00.000000000","tz":60},"subject":"Add missing barbican.conf support","message":"Add missing barbican.conf support\n\nStory: 2008170\nTask: 40925\n\nChange-Id: I3014983f481a5dca7c93e140b3e10caa5d537669\n(cherry picked from commit 0f25900a43895f5154820292d44ec999cd1ec2ed)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/39ef7b5c08dee5362a15dd5a1af2d1be7f040677"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/39ef7b5c08dee5362a15dd5a1af2d1be7f040677"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
