)]}'
{"id":"openstack%2Fkayobe~758202","triplet_id":"openstack%2Fkayobe~master~Ie40a3eed7146e2e2d8b1c08860da395b35ed99d6","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"Ie40a3eed7146e2e2d8b1c08860da395b35ed99d6","subject":"Fix \"Wait for the ironic node to be inspected\" task","status":"MERGED","created":"2020-10-14 15:57:18.000000000","updated":"2020-10-22 00:14:52.000000000","submitted":"2020-10-22 00:13:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"758202-1603325583219-7f647f2f","meta_rev_id":"5802f873a294ed4ada672603dc8111f8fa7a6527","_number":758202,"virtual_id_number":758202,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-22 00:12:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2020-10-21 07:55:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-10-14 16:41:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-10-21 07:46:09.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2020-10-21 07:55:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-14 16:02:28.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-14 16:41:37.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-10-21 07:46:09.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2020-10-21 07:55:04.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-10-22 00:12:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"82d40a7211b5fb6283aac80eb2aca708e64d4da1","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-10-14 15:57:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e1a026bea011795e8480531037ac5ab63001fec0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-10-14 16:02:28.000000000","message":"Patch Set 1: Review-Priority+1\n\nThis is another AUFN@master blocker.","accounts_in_message":[],"_revision_number":1},{"id":"9575c0811b270ee7f61821ef167600a2f8240750","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-10-14 16:41:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ba4d124adb56ee9556cc1184993f833c6e3365d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 18:03:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d36dbefa80c0431899daa9eeb6387c63 : SUCCESS in 6m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a7aee604eec4ce6b770eec54024d17f : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04e1722518344f81b54758ff3afbd208 : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a17d8c72e82f409da02578d2adaa9b17 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a893e7d7e55d41f89dda0d6a04c94f96 : SUCCESS in 6m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4c44e51c801e44eba36b09e810f2d040 : SUCCESS in 6m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bfbc70555a444eefb7cffa029d5d3dda : SUCCESS in 8m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/55ba48af7b954060ac2708ea558b4026 : SUCCESS in 20m 20s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/54268bb21505481ab0ae8e4a57f80af4 : SUCCESS in 1h 02m 25s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/caa5f46533b64203a2329b9c8f3511a8 : SUCCESS in 1h 05m 06s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/1835ab68febc4f10b50b3152029e0c89 : SUCCESS in 15m 43s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/77f01fd05f3b44fb89fb1586e7895d54 : SUCCESS in 1h 58m 34s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/7c12b5e07cd74f2cb19911668b14f4b6 : SUCCESS in 17m 42s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/54be39faeeee41edab9d42909f7d99f3 : SUCCESS in 28m 05s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/71a71d7be86945b391c56778e21fad09 : SUCCESS in 17m 46s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"11095e1c8350da75d297b07d2675d72c968ad41f","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2020-10-21 07:46:09.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7c46bdb2fd3c3612129e593428503eca62f4123e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-10-21 07:55:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6f92ec78ce8a43f9756bc9d8045d8368a2ea1558","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 07:55:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8af8d17d59764e28d11d8db692c68a1d9990c734","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 10:14:53.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4daaef6af71c4ad190f5824a7bd89325 : SUCCESS in 8m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a7f982c5147647c5b2621530e78ddc67 : SUCCESS in 9m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/71e845b5c9b64daab68a381c3e1037dc : SUCCESS in 10m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/240186c7d14a4cab870674f9b2304162 : SUCCESS in 11m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/63fe0626180048309236816149b78302 : SUCCESS in 9m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7082fa5f12044276b3315f45622b57c1 : FAILURE in 12m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/01883c657872497da03d3b9095a525ac : SUCCESS in 24m 37s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/96cb1d07b0ac491eb5d4cca07c9794fe : SUCCESS in 1h 05m 05s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/d60e8999c5df4cb3ba356dd5fad5d549 : SUCCESS in 1h 04m 13s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/9ce9e7966fb74d929fdef5f79b811b65 : POST_FAILURE in 26m 14s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/361a4dab9f4a418787bdbb12797f99b4 : SUCCESS in 1h 54m 00s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/4b1bd5e290344099aedb5a897e2f0548 : SUCCESS in 20m 26s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/3ed5b258a2b04be3bd90efd7c01e220d : SUCCESS in 28m 50s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/dd39c268996440a782074a9c03b8c3de : SUCCESS in 15m 23s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"70a5f55c472f08cd4866fe39420401249d87cc47","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-10-21 11:13:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8aba7da2a58cc8c38e5aa979f8fe5bb901d5da33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 12:31:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/26bd033adbb04731a5d48f64f83a2257 : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/682ae82280ba4fd993d0b1a4a0c9ab9e : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/946bef190749409d8cbc48a629fe6c90 : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bba290635ec544b2937b282d0bda30b4 : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77248f12f7c14860baf0613757ba0a0a : SUCCESS in 8m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5edc11b266b143299a37674e6e689cfd : SUCCESS in 6m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a8f63a1111574100a88fa932d9fe7086 : FAILURE in 11m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7fb9971a5cb94913babccfe241fa1ccf : SUCCESS in 21m 18s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/3a46a21f438d40cc97726ac7c936a476 : SUCCESS in 1h 09m 34s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/57faa23e2141470296e095637ff14c1b : SUCCESS in 1h 10m 50s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/05ede281c7734200b1bbbdaf602c91ec : SUCCESS in 17m 50s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/3075060a74504c8c9eb71748a86e24e2 : RETRY_LIMIT in 21m 59s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/248e18c78c6f4e14b22e6681f5d132db : SUCCESS in 18m 17s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e03306adb26141b4ac8708f2e6b29598 : SUCCESS in 26m 42s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/d1d2847a50074d8eb16aed9bf6ef76e5 : SUCCESS in 13m 13s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"71b1dde26a28d000d9cf45e645f89df064b08a5b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-10-21 13:07:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8639e7a3cfed6669b22f295b08f2a088e98d81ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 15:09:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/644b7646c40347beb6ccbdafc9796f20 : SUCCESS in 6m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41c96d251e7247ffbc25108221216075 : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f70953432cb41958a99dbb0c6e32b34 : SUCCESS in 7m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6452d6e3a79242fe91909ac251ff34f9 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0964396f761413e97a27a676f35e9e0 : SUCCESS in 8m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/27a5a9c9c85046c6b9f5b6a48ce08947 : SUCCESS in 6m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/352fda86151a435db79c13f2b1c28490 : SUCCESS in 9m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9ad7c4a0b2b44581a43caabe79bb828c : SUCCESS in 26m 29s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/80231a86f18a43259adc45d7819af05f : SUCCESS in 1h 02m 38s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/2964c91a7f134ccf83c842aee7232b48 : SUCCESS in 1h 03m 28s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/b77953f333634cffbb761eae8cec9c7f : SUCCESS in 19m 13s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/9c75935452624e199284598956633e95 : SUCCESS in 1h 55m 17s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/81bce6fbfec24b5f87beca97afac96ed : SUCCESS in 18m 41s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/072c354c12204b2ba4b6f57005a85a5d : SUCCESS in 26m 50s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/0bb29cb69bee44f98455d9cb8451479c : SUCCESS in 12m 55s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"736e842d15adaa5fe8c3d5a3b722e13f39431b5b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 15:09:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fa19e46229a6d280ffc4d3c6bfd5b71818e8593e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 17:26:42.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/042b1d3d44b440f787ccc760ce6ff4db : SUCCESS in 7m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d012fe2bb954190b3902e4a1a6fd04a : SUCCESS in 7m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d7221515edda4205b8d0904da51fbea0 : SUCCESS in 7m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b028f09b05d4b4c9623fee106cea8dd : POST_FAILURE in 7m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/202b161cee8f44b68c2517fe99b6592b : SUCCESS in 9m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c4fedb8f5bc843c2a66e1b758e56d5d6 : SUCCESS in 14m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5e20dc7fdbbf4448bcc169208ce34f0e : SUCCESS in 23m 23s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/8f0c04efa66d41f1b9c2c98997483147 : SUCCESS in 1h 12m 52s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/a9381a870d97413cb4912ab60f4dc435 : SUCCESS in 1h 16m 00s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/ed6b2a9f0c9948c5906268bfe956ed93 : SUCCESS in 20m 10s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/132a32074e6344b6a0630cc74b909409 : SUCCESS in 2h 01m 00s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/5870a2132a1249068f455d2df365b0ee : SUCCESS in 18m 54s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/80d2ba44a9714770ad9b782d88d5c686 : SUCCESS in 28m 46s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/92dda9e850224f21b430c40118b9b6de : SUCCESS in 14m 12s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"b4b642655d8f950bd5362102f7946e3e697b3b38","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-10-21 20:11:38.000000000","message":"Patch Set 1:\n\nrecheck\n\nCome on Zuul…","accounts_in_message":[],"_revision_number":1},{"id":"ace1f4a8babb6c2788dd13fd1c40a91546b57b53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 22:14:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c4a516792bce4ac28ea879904bb80b9d : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b6b349251c4438daa04f17490676352 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3069f651438d422b98ca08e4d92cb07e : SUCCESS in 7m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/10000f0502ab4055b7eec4dd5af48d04 : SUCCESS in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7b75ccbb61a4d1da5c4400d6938dc3e : SUCCESS in 9m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/33f61f12926d43a6931b807bad3c1d20 : SUCCESS in 7m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b85e4022a50547ffb306a2df86a8fc2b : SUCCESS in 10m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1864e8de923d4497b19a341dace8e088 : SUCCESS in 21m 29s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/5979c9001850472aad38d82d2ef05aaa : SUCCESS in 1h 03m 06s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/aca7ecc2b27e44c3a1ee7fc64d7b32aa : SUCCESS in 1h 09m 24s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/257441ea33574b428e36810a780a8d8b : SUCCESS in 18m 42s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/fc73ad13364a49e399fb0e8af7eef54e : SUCCESS in 1h 56m 04s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/26ccfcad26aa433db0509730d4851ea7 : SUCCESS in 18m 17s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/1b6cceac4f064680a8e43a34952ba7ad : SUCCESS in 29m 24s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/f97797728a434353b0798246f8fe451b : SUCCESS in 14m 30s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"716b61cf7c151d88709c71388a40f3dd4b289fd9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 22:14:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d41c0af4a6018d6814beb1f87f9ab4604761de87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 00:12:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea7713ef95e04059b37d0bcc919f1939 : SUCCESS in 7m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e3e2c5fe2284e199d4fca091d6eab49 : SUCCESS in 7m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c81eecdad35e4befa946a7402ba31769 : SUCCESS in 7m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72dbc0df44a3483ca5f3fa94a2aa1c96 : SUCCESS in 10m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6fbf72931d684b24af861ea709f76ec0 : SUCCESS in 7m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1cf12c371d2c41bfaee9d8be00dab82c : SUCCESS in 11m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9a128510b5bb4f75a33e16312cd89e49 : SUCCESS in 22m 15s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/a0d3d81f431f4fd99838ae412d0fec40 : SUCCESS in 1h 05m 15s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/4a575e111905486d9ca1988d5b402156 : SUCCESS in 1h 16m 31s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/c4de1164f6be4873bf3b7dcf338d4cb1 : SUCCESS in 19m 23s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/0d08b8c042954a5394835bf77d8720a7 : SUCCESS in 1h 52m 53s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/066d7c564a2f47fdad5a12bee38b573c : SUCCESS in 19m 54s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a620b31d986e4f349289f812a4ea04ce : SUCCESS in 29m 19s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/4612f26ef5214dad8a8cb84ae130faf8 : SUCCESS in 15m 00s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"0b7f037ca909155e5c607013974f934b16a28522","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 00:13:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5802f873a294ed4ada672603dc8111f8fa7a6527","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 00:14:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af0e1e9505c1484a987fb596424e1a7a : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b3c640de7ea4c9801c14000dd78209181edc7049","revisions":{"b3c640de7ea4c9801c14000dd78209181edc7049":{"kind":"REWORK","_number":1,"created":"2020-10-14 15:57:18.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/02/758202/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/758202/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/758202/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/758202/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/758202/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/758202/1"}}},"commit":{"parents":[{"commit":"dd7ff7da488b6a24ec539600819e412bfb637c0f","subject":"Merge \"Fix failures in Docker storage driver check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dd7ff7da488b6a24ec539600819e412bfb637c0f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-10-14 15:54:25.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-10-14 15:54:25.000000000","tz":120},"subject":"Fix \"Wait for the ironic node to be inspected\" task","message":"Fix \"Wait for the ironic node to be inspected\" task\n\nVariables to customise hardware inspection timeouts were introduced in\nchange I2e45876f89121d66cf03b50824bd8258517b88cb. Unfortunately, the\nwait_inspected_timeout variable is turned into a string despite\ninspector_inspection_timeout being originally defined as an integer,\ncausing the following error:\n\nUnexpected templating type error occurred on ({{ wait_inspected_timeout // wait_inspected_interval }}): unsupported operand type(s) for //: \u0027str\u0027 and \u0027int\u0027\n\nChange-Id: Ie40a3eed7146e2e2d8b1c08860da395b35ed99d6\nStory: 2007844\nTask: 40135\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b3c640de7ea4c9801c14000dd78209181edc7049"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b3c640de7ea4c9801c14000dd78209181edc7049"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
