)]}'
{"id":"openstack%2Fkayobe~758459","triplet_id":"openstack%2Fkayobe~master~Ibe42fa372c6fa0c539d2c2b0e238601286dc213d","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"Ibe42fa372c6fa0c539d2c2b0e238601286dc213d","subject":"Fix --limit with commas","status":"MERGED","created":"2020-10-15 16:08:54.000000000","updated":"2021-01-14 15:18:23.000000000","submitted":"2021-01-14 15:16:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"758459","meta_rev_id":"7abb50beefef72fca5fe8f66df43af7010fa74e3","_number":758459,"virtual_id_number":758459,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-14 15:16:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2021-01-14 13:35:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-08 15:17:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2021-01-14 13:35:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2021-01-14 15:18:23.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-15 18:03:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-18 13:14:20.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2021-01-08 11:01:20.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-01-14 13:35:23.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"0b8c3ccbb65985244d09e0ded810cb13f86af99a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-10-15 16:08:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f79909d46df7724465f80d272ffdf403c03bfc79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 18:03:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/43584621803e4d82a565acc6289e2c5b : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/477cc955028c4b10be035ef2f044efd2 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc35724bc3294238ba4fbbf1336128db : SUCCESS in 6m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ef7201638924aecafc1fd4cf3cbb3f8 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ee61da9eb454c5fbf795adc0d810ef5 : SUCCESS in 8m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a71acda87d134c4794b06f500b02c06c : SUCCESS in 4m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9579dcf0519d4577912237530046d6df : SUCCESS in 5m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2315cc08603d4abba20a73282ef3476f : SUCCESS in 9m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e7ac24f6c3194795b51e202c251159b0 : SUCCESS in 17m 01s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/e17e3b71eb994da6a0d4a18934a26067 : SUCCESS in 1h 10m 50s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/7f6fbae01d994f06b45f7e3122cd79d3 : SUCCESS in 1h 09m 53s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/06d736f723de4e05836aa91e32e303a9 : SUCCESS in 15m 32s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/0f44da8a44b94c3ca4247a6d24b05a39 : SUCCESS in 1h 50m 55s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/7b33f2fde9134faa8c00a7fc9b5007d3 : SUCCESS in 15m 41s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/4a2ddd4156df48df83f0d8553b737e04 : SUCCESS in 27m 27s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/35ff10c23b844038a9af75ace19c4fcc : SUCCESS in 15m 06s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"66ce93db30cb086847f7206f95ce2058ba97f7ec","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-09 15:24:29.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"946f975cb32fc242894b32abd2886450d59a94d7","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-12-18 13:14:20.000000000","message":"Patch Set 1: Code-Review+2\n\nhallelujah, the long awaited fix has landed 😊","accounts_in_message":[],"_revision_number":1},{"id":"886154bdeb0242ef1253dfc420867e098190d73d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-08 11:01:20.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"30701f46b1fe1c9fbaa85b92198ae334ffd93fda","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-08 15:08:22.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"11e216426971ab2b0531063d2e1829ccd20101cf","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-08 15:08:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eb98b6d121f628035b3b8ebf66f1603364183127","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-08 15:17:32.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"dbe92a4c0f7096384bb6d55ba5f467268786e23d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 17:15:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca9ac3ebf8624c87a47c8d8f9070af43 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c8931c6c640497f9f949b86a223cccc : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/30a38d2f24d647d893a4cee905b31f18 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70cc3e613e0b401c88fe5f38c06e123c : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a75f335e288c4d328c0efdacb094e659 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18e20211a8e64263bfa5b4dd447fda33 : SUCCESS in 6m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b24aa845ecf416a8209f2246ef3b762 : SUCCESS in 4m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/00d3c76cfe5046cb81c09551331d66ec : SUCCESS in 4m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/127c2ed4a59b497d9b36d479f0bca229 : SUCCESS in 8m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b342da3afa1941d88383a5236d3efc97 : SUCCESS in 18m 05s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/2f8b1a70e640435397284e9bb5c86138 : SUCCESS in 56m 30s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/3840fa617ca44cc091aad61ee82825c9 : FAILURE in 1h 58m 35s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/a71d1b59955444c597c994cf4a228156 : SUCCESS in 14m 57s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/8dd1b5558f574879bdc3a6180ff4a1f3 : SUCCESS in 1h 29m 04s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/3989e35c58424a4e923fe6846fa4f294 : SUCCESS in 17m 23s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/bec8780b3e484a339f514efeebc24f07 : SUCCESS in 24m 16s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/3d7b779cb67e4921a93896706a472adc : SUCCESS in 10m 53s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":2},{"id":"6c2ff1f5a0fbf1cb12a4a3de9ae33d336ae5e649","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-08 18:01:55.000000000","message":"Patch Set 2:\n\nrecheck\n\nbaremetal test deploy callback timeout","accounts_in_message":[],"_revision_number":2},{"id":"5444311c8364b477e08894872a07c6b9ac10c4b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 20:03:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/70cb5fa3cd2f474c9b0accff69d98bbd : SUCCESS in 7m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eece6cff0b4e43bdb5b2c832019549e4 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17504dbdad12477baa813ebbcb3898e3 : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1256e78996894cf2ae524854fdfe45d0 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac98c942cb67494dac2445f5eb37a5cb : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6d4de6e0c5d4ea79d1ee7e62cfb3454 : SUCCESS in 7m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3cb8edeba3534643a0efb1e5fd3cc50a : SUCCESS in 4m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/893cf2ea9f58482daf8605aaf4899b61 : SUCCESS in 5m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9364f36a263e47659c4ab490a09f1775 : SUCCESS in 8m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6453d8f6950d49c99edfb88b4fbfdf4e : SUCCESS in 18m 49s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/380f35385ab949b98ed545c9b123a6d4 : SUCCESS in 1h 06m 59s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/19083beb13ea4172a17e30ccf2377ec5 : SUCCESS in 1h 00m 26s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/98a58236a5404651baad36faf65ac309 : SUCCESS in 14m 47s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/aaf3efac018e4659bab57b2ef642c459 : SUCCESS in 1h 54m 06s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/e9c41b4f516e4ac2a3ca8dd17852dd26 : SUCCESS in 18m 09s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/55d2b5e76324481a930993376eaa687d : SUCCESS in 22m 56s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/cbd559549b604c958cf7e1917b38f2b1 : SUCCESS in 11m 29s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":2},{"id":"a732a2d5e6bd42cfd42e35f285e23a41ae6ea960","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-01-14 13:35:23.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"497a5f870f476d3486760dd237bf5c2bd07624d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 13:35:50.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fc8eaa599538d89f5828b89fa1389c59cd43c873","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 15:16:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb9d4cdd0628484f900e8d2386e61cdd : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aeee147243ce4913b8180df5ce764bd4 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fab6ed7add564aacb60f0d61b3d4fcd1 : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b710c448823d49e28da368bc77b922b9 : SUCCESS in 6m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/32c9a369940d4c8f9994d15c79d430f0 : SUCCESS in 3m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/07935456e0e44433a6fbd3239708ef37 : SUCCESS in 4m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b2cc09d2c1924a10aa7bd9a8e7851b8d : SUCCESS in 7m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2247835864184afd9f1fa5fabf766b27 : SUCCESS in 19m 15s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/7a0ea5b209f54c7b9b2c0f05e89dfdcb : SUCCESS in 56m 43s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/667addce7d3b4e738646267adabf033a : SUCCESS in 1h 14m 38s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/eb1b5d0c470145ee9879d7b8f5e98f5a : SUCCESS in 15m 58s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/c34dac70af304f6193ca2b695bd0740f : SUCCESS in 1h 34m 20s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/119201e711cf4966877945d65d999088 : SUCCESS in 13m 22s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e1eacf217fa946f1b61dd4d4f105a27b : SUCCESS in 29m 07s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/68f44c62e2544a4ba38311643a77564e : SUCCESS in 11m 34s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"6aa29567de1ed8d1675e30720e46fa005e3cc350","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 15:16:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"1f597690113bc9db173ce19da0d7dc6dd60469a5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 15:18:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03adc5eeef814fb4bab25898249e57da : SUCCESS in 1m 12s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/410f1c389faf4ef997f535088978591e : SUCCESS in 1m 36s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"017b092df70706177c542c3a3cab3f9d0a75e5c2","revisions":{"2285e52d354b3b1c1847eb53cd1691ea395b3a07":{"kind":"REWORK","_number":1,"created":"2020-10-15 16:08:54.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/59/758459/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/59/758459/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/758459/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/758459/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/758459/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/59/758459/1"}}},"commit":{"parents":[{"commit":"dd7ff7da488b6a24ec539600819e412bfb637c0f","subject":"Merge \"Fix failures in Docker storage driver check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dd7ff7da488b6a24ec539600819e412bfb637c0f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-10-15 15:35:51.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-10-15 16:02:27.000000000","tz":60},"subject":"Fix --limit with commas","message":"Fix --limit with commas\n\nKayobe allows specifying a --limit argument, which is passed through to\nAnsible. In some cases we wish to add an intersection with a group. This\nallows us to reuse playbooks for the seed, overcloud etc.\n\nFor example, the lvm.yml playbook specifies a host list of\nseed-hypervisor:seed:overcloud. When executed as part of a kayobe\novercloud host configure command, Kayobe passes a limit of overcloud. If\nthe user specifies a --limit argument, this gets intersected with the\novercloud limit: host1:\u0026overcloud.\n\nThe problem happens if the user specifies multiple parts to the host\npattern in their limit using a comma, e.g. host1,host2. This results in\nhost1,host2:\u0026overcloud. Ansible ignores the colon, and treats this as\nhost1 or host2:\u0026overcloud.\n\nThe solution is to use a comma to join the patterns if the user has used\na comma: host1:host2:\u0026overcloud\n\nChange-Id: Ibe42fa372c6fa0c539d2c2b0e238601286dc213d\nStory: 2008255\nTask: 41111\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2285e52d354b3b1c1847eb53cd1691ea395b3a07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2285e52d354b3b1c1847eb53cd1691ea395b3a07"}]},"branch":"refs/heads/master"},"017b092df70706177c542c3a3cab3f9d0a75e5c2":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-01-08 15:08:22.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/59/758459/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/59/758459/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/758459/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/758459/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/758459/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/59/758459/2"}}},"commit":{"parents":[{"commit":"dd7ff7da488b6a24ec539600819e412bfb637c0f","subject":"Merge \"Fix failures in Docker storage driver check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dd7ff7da488b6a24ec539600819e412bfb637c0f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-10-15 15:35:51.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-01-08 15:08:16.000000000","tz":0},"subject":"Fix --limit with commas","message":"Fix --limit with commas\n\nKayobe allows specifying a --limit argument, which is passed through to\nAnsible. In some cases we wish to add an intersection with a group. This\nallows us to reuse playbooks for the seed, overcloud etc.\n\nFor example, the lvm.yml playbook specifies a host list of\nseed-hypervisor:seed:overcloud. When executed as part of a kayobe\novercloud host configure command, Kayobe passes a limit of overcloud. If\nthe user specifies a --limit argument, this gets intersected with the\novercloud limit: host1:\u0026overcloud.\n\nThe problem happens if the user specifies multiple parts to the host\npattern in their limit using a comma, e.g. host1,host2. This results in\nhost1,host2:\u0026overcloud. Ansible ignores the colon, and treats this as\nhost1 or host2:\u0026overcloud.\n\nThe solution is to use a comma to join the patterns if the user has used\na comma: host1,host2,\u0026overcloud\n\nChange-Id: Ibe42fa372c6fa0c539d2c2b0e238601286dc213d\nStory: 2008255\nTask: 41111\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/017b092df70706177c542c3a3cab3f9d0a75e5c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/017b092df70706177c542c3a3cab3f9d0a75e5c2"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
