)]}'
{"id":"openstack%2Fkayobe~758550","triplet_id":"openstack%2Fkayobe~stable%2Fstein~I7ef890a2ae6fd67f53e1b8bca244a1c0d44fea53","project":"openstack/kayobe","branch":"stable/stein","hashtags":[],"change_id":"I7ef890a2ae6fd67f53e1b8bca244a1c0d44fea53","subject":"CI: Fix kayobe-tox-molecule job","status":"MERGED","created":"2020-10-16 10:44:10.000000000","updated":"2020-10-16 12:50:51.000000000","submitted":"2020-10-16 12:49:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"758550-1602852546714-b16e9fe7","meta_rev_id":"079573c119b51360f19414f498ea65ee93bdcc68","_number":758550,"virtual_id_number":758550,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-16 12:49:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2020-10-16 12:27:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2020-10-16 12:25:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2020-10-16 12:27:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-16 12:16:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-16 12:25:11.000000000","updated_by":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"reviewer":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"state":"REVIEWER"},{"updated":"2020-10-16 12:27:40.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-10-16 12:49:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0f3b941501bbb9b4baef419ed676337a823311fb","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-10-16 10:44:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d81edbf9b994a12c9fff96ac1730a76e618e7e89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 11:19:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1037bd82a4744c1789088615bbdee3b9 : SUCCESS in 6m 03s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/203caddf990544fa92cb0eb646480a0d : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2251e67f22a642aca65b9460c863fe49 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1a6cd0ad6674aa68ee483623ac3df5a : SUCCESS in 5m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/580352950e9a457f963be73dccbea721 : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e07d699f17874ed298e0fd6e35dfbd6a : SUCCESS in 5m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e0865189d9be4a4c829c3dd64c726781 : SUCCESS in 5m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/268749bd7b954fbfa9ba900bc340fe72 : SUCCESS in 7m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f08e739f84864c4b9128abc05dadca9d : SUCCESS in 12m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9591f1fe247d4004b24839137ad3375a : SUCCESS in 23m 07s","accounts_in_message":[],"_revision_number":1},{"id":"22d60f068561fd87ef4afe4c7d022d89a6a2c7d5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-10-16 12:16:17.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"19e40b4097f3b91075803691cd874f73ab8b7f5c","author":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"date":"2020-10-16 12:25:11.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4cc1d929a8c9084bde6ad1c71cacd7b2df2f6776","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-10-16 12:27:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6e6a141530b5da115a10b904c49e343ce4df7341","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 12:27:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c0ab0d8b0da73a2f298977310112290362ad07ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 12:49:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/94fe25ab26624f03bf02e9bd09c1b8c2 : SUCCESS in 6m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f828eac239b42e99c52ef990f0701cb : SUCCESS in 6m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/640aa3105b1d4fcea8b879af558db532 : SUCCESS in 6m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ac8b176b776e4ae4b07aa3d26ee6134a : SUCCESS in 6m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3ae6f37d615452bb06261b3f7f23a0f : SUCCESS in 7m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b6f8197e66b1439789926a01b675a16b : SUCCESS in 7m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0e0bf50960cb4ca9a4b76d2721005d32 : SUCCESS in 7m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4d70c0f61c2f4fd39e96bf170688d3f3 : SUCCESS in 11m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fb91489465cc4ffbb1909711267ea0ca : SUCCESS in 20m 56s","accounts_in_message":[],"_revision_number":1},{"id":"01144e5d0e0b184db60e3d935afaf6584a679f7f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 12:49:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"079573c119b51360f19414f498ea65ee93bdcc68","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 12:50:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d37b57f22b37453b969dc1e9eba3712d : SUCCESS in 1m 15s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0bc49ad84b894f6ab59e976278f36d9e : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e861aebb2d956374b84ed35814efa4770bf88fdb","revisions":{"e861aebb2d956374b84ed35814efa4770bf88fdb":{"kind":"REWORK","_number":1,"created":"2020-10-16 10:44:10.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/50/758550/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/50/758550/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/50/758550/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/50/758550/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/50/758550/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/50/758550/1"}}},"commit":{"parents":[{"commit":"f95f9a2c734cc45aa7a87613b8c93a651e9be414","subject":"Make pip_index_url variable conditional","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f95f9a2c734cc45aa7a87613b8c93a651e9be414"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-10-16 10:43:52.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-10-16 10:43:52.000000000","tz":120},"subject":"CI: Fix kayobe-tox-molecule job","message":"CI: Fix kayobe-tox-molecule job\n\nWhen molecule runs, it installs ansible\u003e2.8, which is now version\n2.10.1. This now also installs the ansible-base package, currently at\nversion 2.10.2. Then it installs the version of ansible required by\nkayobe, i.e. ansible\u003e\u003d2.5.0,\u003c2.8.0 (currently 2.7.18).\n\nWe end up with a tox molecule venv containing:\n\nansible\u003d\u003d2.7.18\nansible-base\u003d\u003d2.10.2\n\nThis causes the following issue (hidden by default because of no_log):\n\nAttributeError: \u0027ActionModule\u0027 object has no attribute \u0027get_shell_option\u0027\n\nThis commit adds requirements.txt to the molecule tox deps, which\nensures the correct version of ansible gets installed first.\n\nChange-Id: I7ef890a2ae6fd67f53e1b8bca244a1c0d44fea53\n(cherry picked from commit c58555498836066002dc33620018e9bd1e2eeb6a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e861aebb2d956374b84ed35814efa4770bf88fdb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e861aebb2d956374b84ed35814efa4770bf88fdb"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
