)]}'
{"id":"openstack%2Fkayobe~761332","triplet_id":"openstack%2Fkayobe~stable%2Fstein~Iba16d6232e1a6ef069cd3ccb870e5bfa89cd913c","project":"openstack/kayobe","branch":"stable/stein","hashtags":[],"change_id":"Iba16d6232e1a6ef069cd3ccb870e5bfa89cd913c","subject":"Fix filtering of network names set to an empty string","status":"MERGED","created":"2020-11-04 09:35:52.000000000","updated":"2020-11-19 14:02:24.000000000","submitted":"2020-11-19 14:00:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"761332-1605794459805-76a6aaaf","meta_rev_id":"380e1185826c95edec706bcfe10f81914c14e1c5","_number":761332,"virtual_id_number":761332,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-19 14:00:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-11-10 14:24:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-11-12 13:54:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-11-16 10:55:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-12 13:54:29.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-11-19 14:00:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"578e0f1388c4dfeb44f68143ddff569c1123fecf","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-04 09:35:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b2f4a2555ab021b94639a9c8d6b4d0e85298e97d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 11:30:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39c05c72f8464aa5909ded8ea5cab351 : SUCCESS in 4m 31s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5317a35883cc4b1c834c159831db39d4 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bdb7d4da8d984fc0b92ea646e8835a7f : SUCCESS in 7m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77f1060fa1274f1183c8a125fb06045c : SUCCESS in 3m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/096324071d74444d873894e1dc888947 : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee04e31855c64338aa68995b8f270da0 : SUCCESS in 4m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/23e2cacd912c413982e03dc4c7751d3d : SUCCESS in 3m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/756f8f528d66415c9b3efd8dd0f86b77 : SUCCESS in 7m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6cd2c368402047e7813f2501e072f96b : SUCCESS in 16m 56s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/fc8d4f4e73d94f17990e5f6d42e1e3b4 : SUCCESS in 1h 04m 53s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/eedcd782a20b4f31a05c4b90bdfdf78b : SUCCESS in 1h 40m 06s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/085b22b9035f4a2f8aded82bf9bd10f0 : FAILURE in 53m 19s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/44c25e1d0aef48ca8f015e7921b6fe64 : SUCCESS in 32m 01s","accounts_in_message":[],"_revision_number":1},{"id":"57fff5c439460a90242d32595b848db20452095d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-04 12:19:36.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"570aee441a63e67efdfa809d7a6539673e930722","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 14:06:59.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0c554649c0c0423c88505b6c15acebcb : SUCCESS in 5m 12s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a6e150b407294df296daa1785eca533c : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c44e6e70c2194e8b8afaa9a2bc3bd5f5 : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0444bb93a1a444ce9bd26739a2a7a057 : SUCCESS in 3m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d064f9d142774e25bc980549289d6fc2 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/630255563b0e4475877d657cb7798c31 : SUCCESS in 3m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4d6979ea43474d7bac06d63e709f7dda : SUCCESS in 3m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a18f0194359549baae136de8c7d6e2dd : SUCCESS in 9m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/41ff3cdcd1d34aef9ac19f4f6603bc83 : SUCCESS in 21m 30s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/1ca10fa1f0ff40a98d03b791d2d852c1 : FAILURE in 33m 27s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/0febf1f7e4af4a5aac210422f7d05fe6 : SUCCESS in 1h 41m 07s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/54f0bd93b932478db98c0768c4864bc7 : SUCCESS in 15m 39s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/b1153ef9128643b39f89dff9fdce4a13 : SUCCESS in 38m 57s","accounts_in_message":[],"_revision_number":1},{"id":"38db9978767e3acdc204aa99dd795de9d9268f01","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-05 09:13:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"54ed2a4e97d57d8e24cc402e2c60cabea071ea60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 11:01:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/20f77f19c9104f61a342feca72e991d2 : SUCCESS in 4m 58s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/98cc7449179d4a0898e0056450fd5283 : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5cfe7cf8a735401591e1b841a321044d : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0bd38783c9a40fc8e5151d1db4dc849 : SUCCESS in 3m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1243dd398080403fac8b4021ad897af5 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a9bd4ff76f049e6a47fa80f7d7474ab : SUCCESS in 3m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b2b9f9c4db794e6292631e1f5abf8463 : SUCCESS in 3m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b1d60216c96b43c780b917c1f48dbd21 : SUCCESS in 7m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/71d0104ae99a4f7689fa67238c363c63 : SUCCESS in 17m 43s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/7597402cf7b84a4896591a6e71656508 : SUCCESS in 59m 12s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/11f7eb9a78c644eb846a260a3624b1bd : SUCCESS in 1h 44m 38s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/6ed2d5b76a604d068ec5007a93332599 : FAILURE in 23m 59s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/f0a95147cae7477392e54e739158884d : FAILURE in 44m 04s","accounts_in_message":[],"_revision_number":1},{"id":"6c08c25d3d927b580ed0489f2cbd2871e30e6ba9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-09 13:57:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"859d09eb778e902f3b75ee8d49a37d0f23ea15b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-09 16:53:03.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/351415e165e144c3b22faf99e1dc46da : SUCCESS in 4m 18s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f075626de5f64b6c978fe093881b4163 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/55c1f45f3efe417faba21ac8ae76b104 : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60caea840bbc40a79ef7aa3c25f1a21b : SUCCESS in 3m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/264f20663a054e77b7adbbdec3561ad7 : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/064dda721518442f90be28eb54b5f65a : SUCCESS in 5m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/285444b6ed90479c812887608e3a4bee : SUCCESS in 5m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/19782f9171d44cebbcfb578aa0a5d7a3 : SUCCESS in 7m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7538579991ae48e98cb7f8cd306416bd : SUCCESS in 17m 41s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/faa17c4e1f0b401eb61f43713fd693f1 : FAILURE in 54m 46s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/1c6ce4db99fa4a7db67ca02b6b0f3527 : SUCCESS in 1h 52m 53s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/832ab430c2fe4f2c8cbe4ec45ebd91aa : FAILURE in 53m 20s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/69148665b0cd47cbb1d130e1088b05fb : SUCCESS in 36m 54s","accounts_in_message":[],"_revision_number":1},{"id":"6bdeeb94ff49fe3584a6ab54bd5a7625b9f2d487","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-10 11:56:32.000000000","message":"Patch Set 1:\n\nrecheck\n\nmore mirror issues","accounts_in_message":[],"_revision_number":1},{"id":"f070d7745b0dbf9a994db7a21e27d20752eb6872","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 13:40:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd6b5b4677e749d0b68848948c9d0a87 : SUCCESS in 5m 10s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/32b2d05bb9c4477589346d4b8c1c5f12 : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/65cdbe053f0f42d09cf51a8353c6e6d6 : SUCCESS in 7m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dace4e7934404072a63c5fe4d664762e : SUCCESS in 4m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7ede3dcb5da2483ba41361b9649c811c : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b39edbf0249b49069d082ba85ff12e8d : SUCCESS in 8m 36s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/45353c1761904a1184c2f550ce20343b : SUCCESS in 9m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e13ad6bd53ad499d8d99f97b466fea6a : SUCCESS in 8m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/507ab50c9ddb4d11a8b854b6957608b4 : SUCCESS in 17m 18s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/10d4f6b99a9645fbbee1f83a68d79330 : SUCCESS in 59m 00s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/3f53f59ebaa04f0db0c066c643af50f4 : SUCCESS in 1h 41m 00s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/4900fa9977e845fd98200dda0d8c5f11 : SUCCESS in 20m 19s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/3432968d3a8244df95841afd1010a14d : SUCCESS in 34m 28s","accounts_in_message":[],"_revision_number":1},{"id":"88eea7849cbf82fe1ed90dfd37a57ba674126417","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-10 14:24:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1e1a37246dd11bb431bc256f2156d5930779d080","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-11-12 13:54:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cdc0c1979bf6022e11082633fe7291eb81a915fe","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-16 10:55:48.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ecb09ab0430310cb401906484e755ec79181cde1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 10:56:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c68f2a6030c9d6a73948ffd5ccef630341d66283","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 13:48:48.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/eef186a2cacf4dcb9a478bcdbeb73d16 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/973a337d7ea74b90a8c0763fcce9e1c0 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a343a9d894d447138b35d02d9bc6c295 : SUCCESS in 4m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c956526d2888439695b632f9df37de8a : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b186fbb04dae4656bc305b198fd7fb9e : SUCCESS in 6m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a19cd343c9c4424c9fdc136f97ffc2bc : SUCCESS in 5m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a8c7be5d941040858e1a6dc68e11249d : SUCCESS in 7m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/04fc46c9468b40f299c7625a916dd18a : SUCCESS in 24m 51s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/43df6cb0cb8349d5b3edc3b3e98c0be7 : FAILURE in 57m 05s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/668d6477088845f7a006027db38430b6 : SUCCESS in 1h 51m 35s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/7e93b268cf9f4fe69ffb841cb2ad3e0d : SUCCESS in 21m 08s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/61d611a1cc124a7e9fc3fa65215f8d5a : SUCCESS in 31m 30s","accounts_in_message":[],"_revision_number":1},{"id":"5e1c87e365d495a87283dfadea83ddd709e7a193","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-17 10:50:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3b001d79a1421f404e8589af8f160737d3201e68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 12:38:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7cb58d4d4f2f41a282c1f7e87f046ee2 : SUCCESS in 5m 34s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8c17951b1e80479b8f47df413e55bf75 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d722588572a4887a1c8d9266088e8be : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0b69ecdd6574473be80d836dd036d4c : SUCCESS in 3m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ac83c0b13f3f4b6bb9c35b36251b2988 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3298df3c177940f0b00ec4e74ba6feba : SUCCESS in 4m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3508bc05f76f4134a324bcc3f0551311 : SUCCESS in 3m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a770dd27d13b417296f25b7bbb52cc6e : SUCCESS in 8m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/791ca0e355b94789826ddf97b22702ea : SUCCESS in 19m 21s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/259f3f768a20466a95f70493885fb640 : SUCCESS in 57m 24s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/c6635cfaa1354f1082d8a969ff24ba4a : SUCCESS in 1h 40m 49s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/c30eb81346354f1dbd9dbb9872e07c2b : SUCCESS in 24m 17s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/47bcf3476a7244648dd290eb73722a14 : SUCCESS in 33m 01s","accounts_in_message":[],"_revision_number":1},{"id":"3403a8d5fa7a378b20277330f1d7797c36527edf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 12:38:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"85b80371c5b5216ea5b269838ce37719685db1ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 14:19:05.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9e486efca3444ea19ef401c9e499504d : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3286faa536f14b2b9e02edaf14122c08 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a18a4289d6f47a8a9657ee477a637c9 : SUCCESS in 5m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c1440054f9f940728a9c9a9a915e5799 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0c8fde1680747d085500b7042a3ebc7 : SUCCESS in 6m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6cac5fd056db455dbcfed1c62c2f9fe9 : SUCCESS in 4m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/db1d026dbd204964a0722b5b0cc08992 : SUCCESS in 8m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a9ab7ee7239b4098af1d97c2c3bf5aeb : SUCCESS in 17m 47s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/b28adeb6a0114a88aaae40f10fea18c2 : SUCCESS in 1h 02m 07s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/58118c395c3c450492556ef53eb917de : FAILURE in 24m 55s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/00f447308d6f49198e9adf002ac5824f : FAILURE in 24m 02s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/e13491e4967247e5aea89c008621b6cd : FAILURE in 20m 49s","accounts_in_message":[],"_revision_number":1},{"id":"db57e8563dd6692a66a4e48b4898cbc05c4dd6f0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-19 09:59:29.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"398f87bb0f316cdfeca9f66aec5f28f3c6af8400","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 12:03:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2785d716e2f84d0a9dc38dcf8db6835c : SUCCESS in 5m 09s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9adcf64125d14809b49078ed9d13f5e3 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/839c6a9c311047b0bb90b8a05bac1d3d : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/019ab14e53fd49caad36f1326ab0059a : SUCCESS in 3m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5d5d5cba177e449b8ee83998a06c5e49 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7795fd232b734a369619406a4374562e : SUCCESS in 3m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/511712fff9394067867f1f5fec951094 : SUCCESS in 5m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b0512635288b43f4b162e68f70281f0a : SUCCESS in 9m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/62ac594bf2994effbb302e891f81ae50 : SUCCESS in 16m 28s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/1b1a4db40dd6475cb3260f06f307eaf2 : SUCCESS in 56m 00s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/e74354560aca48619725d6778ec4bf3f : SUCCESS in 1h 54m 06s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/fda869def48e44128f4f1069eca91a53 : SUCCESS in 21m 00s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/e11acbeff2304705b3b1deb385a89655 : SUCCESS in 39m 49s","accounts_in_message":[],"_revision_number":1},{"id":"c44acb2a39fd1afe05ddf845a838495b10a1b9a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 12:03:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c920a820d6386230c009242350b057084bade2b1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 14:00:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"255ac3837226459219e370926284cd8f1440d2c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 14:00:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8c09ee15929c43ecac186498264d78cc : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/63b0faf060db42dbb5be6cf40e447c3d : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41eb2f3ce98343fd86383d0f37ba4b0d : SUCCESS in 4m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9f696fb187a44b17b55f22b1a0baaca1 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a94bc0bde11421996f00fdd70c73cbe : SUCCESS in 4m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ff5fc3a367d64ef1865a4b1e431a255e : SUCCESS in 5m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/749e012718a340f7b7403d859c5a90af : SUCCESS in 8m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/07361e4419c84cd08cc3fa2fc84e2310 : SUCCESS in 17m 54s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/72f35cd62a3e46b2a6dbd27d5470a84d : SUCCESS in 1h 02m 39s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/d4445ff513fc4f63af8225afbd176df7 : SUCCESS in 1h 48m 35s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/3c08aacec6784255ae266697093cb36f : SUCCESS in 21m 51s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/dee803f30e564b91931235e5126658d9 : SUCCESS in 34m 45s","accounts_in_message":[],"_revision_number":1},{"id":"380e1185826c95edec706bcfe10f81914c14e1c5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 14:02:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/695f779de63e4af7b91a981edd0ca675 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"84c463e32d0c7f8234d478543e48589eb46386ba","revisions":{"84c463e32d0c7f8234d478543e48589eb46386ba":{"kind":"REWORK","_number":1,"created":"2020-11-04 09:35:52.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/32/761332/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/32/761332/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/761332/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/761332/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/761332/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/32/761332/1"}}},"commit":{"parents":[{"commit":"5eca75c31f4ffe0c875b0bcdcc6245ca174799b3","subject":"Performance: skip LVM configuration if no groups configured","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5eca75c31f4ffe0c875b0bcdcc6245ca174799b3"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-10-07 19:04:01.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-04 09:35:44.000000000","tz":0},"subject":"Fix filtering of network names set to an empty string","message":"Fix filtering of network names set to an empty string\n\nCurrently we filter out networks that are set to None. However, we\nshould also filter out networks set to an empty string. This affects the\novercloud ironic networks that were disabled in\nI300d0ef136224126f25d2c70a80a42afeea5f586 when ironic is disabled.\n\nStory: 2008207\nTask: 41076\n\nChange-Id: Iba16d6232e1a6ef069cd3ccb870e5bfa89cd913c\n(cherry picked from commit d2109e1c35011fcaeec368eb1b8bf6a9ec75e238)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/84c463e32d0c7f8234d478543e48589eb46386ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/84c463e32d0c7f8234d478543e48589eb46386ba"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
