)]}'
{"id":"openstack%2Fkayobe~763374","triplet_id":"openstack%2Fkayobe~master~Ia5715c6914ac454081a95b6e1788e80f53c72e30","project":"openstack/kayobe","branch":"master","topic":"victoria_release","hashtags":[],"change_id":"Ia5715c6914ac454081a95b6e1788e80f53c72e30","subject":"Revert \"Use OpenStack Victoria release\"","status":"MERGED","created":"2020-11-19 13:58:14.000000000","updated":"2020-11-26 10:36:01.000000000","submitted":"2020-11-26 10:31:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"763374-victoria_release","meta_rev_id":"dbd9cae0ae06ef67ba47b26272ea241002f81c92","_number":763374,"virtual_id_number":763374,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-26 10:31:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2020-11-25 16:52:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2020-11-25 16:52:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-19 13:58:14.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-11-19 13:58:14.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-11-20 11:09:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-25 16:52:15.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"}],"messages":[{"id":"fcb3ef17befcbaa9084df7b5eca206f6bc61fdf4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-19 13:58:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1f5b0ab0f7f235bd41a3b8a29801a620336793b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 15:52:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/89bfd264e9494f408a4aa64f389fb68f : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b8e426e892c42d7bfb0b76cf55dbb5a : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7686eb82fd2c4f988b76abd9ce1e0b35 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/57846ea1c3e34d4493603e3e0a26e4f9 : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dcf27462f63746cb90b18c69d830d729 : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc5b86d5ead146ec99fa79bb4d87b182 : SUCCESS in 6m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df1659a0430a4a7ba85b299ce699a6e4 : SUCCESS in 3m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0e31bb19b3c6497aaf8df1492b8a9422 : SUCCESS in 4m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e4f0221660d540d3bebb175afc16432f : SUCCESS in 9m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/77e8414aba2c444c9e32b4356d16ee40 : SUCCESS in 19m 33s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/c6c9ec94b1b94b9caa1a3bc5a6e75bfc : SUCCESS in 1h 05m 07s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/26ed7306bee142048ee2dc8f80c06012 : SUCCESS in 1h 05m 16s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/5675400ec993418ea0627d684ad9adb1 : SUCCESS in 16m 27s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b1ffb1021cf645f59b9ceb2bb80bc0d9 : FAILURE in 1h 47m 10s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/d6483569a5e841e1ab5e85a3b2ef60a7 : SUCCESS in 16m 31s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/2048523219d14cbf939bdb5cf272e569 : SUCCESS in 24m 47s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/1c775ebfca5042e38e6c6882042572f4 : SUCCESS in 11m 56s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"a330330099a7b6449ecf6e767f802b66e6ed81b7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-20 08:57:11.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"256feb3704b142d8103b900202c9cf48a726fc86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 11:09:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6dca8dc576ea450ebc3b2110db3a02a3 : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bcc019524c74129864b3f6de9f6fef6 : SUCCESS in 7m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a1f57845de764ae4a5aa32b82d32c7a7 : SUCCESS in 6m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/63b5ded3c5bd4f2d958c0dbc186e88bc : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8546f4efabef460ab3d6326360d136b7 : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a375062c985447d5bde08edef541be3c : SUCCESS in 6m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e55dd139d0cf4c1989b425a751ef46d9 : SUCCESS in 4m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/01290e9f8e6f48da9724733da1f0e00b : SUCCESS in 5m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1f58267eec31401fbf31bb14f7c3066e : SUCCESS in 10m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4b353188548449efbde4b417177c3839 : SUCCESS in 22m 35s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/ff75c692f7a04c22b3de0773d2f03115 : SUCCESS in 1h 03m 17s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/44f166f8e10c430d9bfabc8949109c49 : SUCCESS in 1h 10m 51s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/e858493dd5b54b969b81fbffb0d1a97b : SUCCESS in 17m 33s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e1421266e23e4bdd99391e1fdbe63fd0 : SUCCESS in 1h 52m 03s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/d9d2da8324ea4b59aa05c4ea2fe9855b : SUCCESS in 17m 58s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/923aeb7118c64ccdb388f1ffde5f5df1 : FAILURE in 1h 11m 07s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/83b1a5f986d0414a8a37e20a7ac95b2d : SUCCESS in 13m 16s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":2},{"id":"283f4d5dc2cf63f97e6c5b49bce2556e82fe4b10","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-20 11:55:09.000000000","message":"Patch Set 2:\n\nrecheck\n\npip mirror issues","accounts_in_message":[],"_revision_number":2},{"id":"34a2639b68688dc9c57cd8d81e2eb3246fb9a1d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 13:52:00.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e6e34ca11f224967a1daa3887988252b : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d19d438043264227b487b8a05d79a495 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e6c129cbebf44ac897830da1eb79b36 : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/43ae2fd95d4049c4a8659b589801571c : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/459343ebde7047ffbaec2a636980a934 : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b499455ba594c78a98a5bda0eafb66c : SUCCESS in 6m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5007459ed18440eaa9ca323f1d1719da : SUCCESS in 4m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2640e972d7474193863028d9a7242b0f : SUCCESS in 5m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/971b559d3936447fa9199b7153a58120 : SUCCESS in 8m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b3a053719b79436699077c43bb6c5ff2 : SUCCESS in 19m 45s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/f88975fe4ec74391ad9f576b3ff1f94f : SUCCESS in 1h 04m 10s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/de3efbef1c1b459c987e475b236a80fe : SUCCESS in 1h 10m 29s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/6072c0eb10d3488dab026034381c42cb : SUCCESS in 15m 14s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/383080b2afc141b898e7a2663518c58c : SUCCESS in 1h 53m 00s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/c0fdb8b4480e41c7818a17b2d10e786c : SUCCESS in 18m 34s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f7484e9fcbf644a39ccc47b656b426cf : FAILURE in 1h 10m 55s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/44bd8c75d7754341a887b1f17d6ba28b : SUCCESS in 11m 21s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":2},{"id":"4842f6aa667b947ad3b1195e9ba4ebf18f986825","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-20 14:10:37.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f51f1ce8a75031fb248003238bd837ac8f5f0c75","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-23 13:40:20.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"80fe5f4fb621a0c9a80ff48d4ef756332d01bba7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 15:46:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c2e7e0975604042b1271772a287a89f : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd3e4f40f108465f91f7b4b1d45a6600 : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ecb763e198148fbab22dc5d7f4b47d8 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/50b854320f734e759aa0e32258e177a6 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8c2adf46d70c436eb1ebbeeee3b4eb53 : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff7dc56c61fb4f4fb9dd06b11f25a20b : SUCCESS in 7m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d552f736fa744acba5094982d605ee43 : SUCCESS in 5m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/09fc7a170bc348828e5492ed7fef33b9 : SUCCESS in 5m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/45281ad41ad84178b53d602966a0768d : SUCCESS in 8m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/40bfda4b656346c393687d5a6211adc0 : SUCCESS in 20m 24s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/1d122c4d7d334c04b14bb28d9e8d6bd4 : SUCCESS in 1h 06m 42s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/b5ec6c8be8ec488a91b4abfe3e8d3959 : SUCCESS in 1h 05m 18s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/d8d802d369614f71a599973f2b697e22 : SUCCESS in 15m 29s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/327b1cd35b3d4a81a296f4c7a57bc81e : SUCCESS in 1h 53m 30s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/ab7a38d035f845c58cd31fa8d49ad322 : SUCCESS in 15m 49s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/18bdb065c6db42dea142e943e452de89 : SUCCESS in 23m 49s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/6c02f6ff856443a188ddc4004e94d057 : SUCCESS in 9m 58s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":2},{"id":"d84e414586187bdc731cf7305b299edbd684ff17","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-11-23 15:54:17.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"bae335b79dd628d659fa682552c9e1e4324dc6cd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-23 17:00:26.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"5ea256d44b05468960d9da2ce8f3df64501e1c64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 18:47:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d6ebf0dfa930435ab48d88270fbc528a : SUCCESS in 10m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf6973b9546c4b1e8acb4b96a9663a18 : SUCCESS in 10m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a0468b9fcb3c4d9899af653e7e71afa0 : SUCCESS in 10m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/29fb408ac7054a2ca5d1f21893047b1f : SUCCESS in 9m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ddeea50e829d4e84ae1c739b9b29fcf3 : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36dff8465f4e4bab90606a31aa3323e2 : SUCCESS in 6m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2156b8e1c8b44dc8b2d71d67f06fd439 : SUCCESS in 4m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e8cd5c4ab0b9411c9ed812cea86cdaef : SUCCESS in 4m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/65a97b06bbc54feea222d05d550daba0 : SUCCESS in 8m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/898757130bbe4bc4866483e7918ec8de : SUCCESS in 19m 08s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/6dff25943ed24d5ea7187870445cc515 : SUCCESS in 57m 07s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/354cd79ddc88472ab6a4edfa9808466e : SUCCESS in 1h 09m 51s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/0ce46018ab674c858e9c9206dc821c9c : SUCCESS in 22m 34s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/5153e43457824f7a830d309ec40e550d : SUCCESS in 1h 41m 25s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/a465f6ac6efe4d8eac20010f2d961c7b : SUCCESS in 17m 12s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a344b18e6a5144948a67848e42e9a971 : FAILURE in 1h 12m 52s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/e3cc52599049486381c720d7d34a96eb : SUCCESS in 15m 24s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":2},{"id":"d2df937155b5fb692e621309a227c33b077bea65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-24 15:12:13.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"a3f68a6536cfa66bba71805bc6aeecd5a3096ddf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 16:37:24.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ea2abb19b4674d5d898949ca886c0856 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebc67ba1242e4680b1928a54a7a573dd : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e4a7af48aa64e14930859efebc49ad6 : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a80d9fd184f04f9ca20f8e811605c816 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/261e488aad344b7eb8d71eb58f4124c8 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28b9f771478545b0afda8816b4757ad4 : SUCCESS in 6m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb66f5aaa9a044af9da3b2625cc31082 : SUCCESS in 3m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b6eb566ffc904dde9de851e43d1a0d2b : SUCCESS in 4m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/db99afb128dd4328b862b0b8bd92490b : SUCCESS in 8m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c76e24b7d8f44c2ca5f9bb62815ee7be : SUCCESS in 19m 52s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/e215df6e8fa14f49974ea5539dd9313a : SUCCESS in 1h 09m 59s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/5662ff74c8424d48abb2bfc04aaedc6d : SUCCESS in 1h 14m 21s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/ed209ee7c6fb49c2bac37cfd3466619b : SUCCESS in 16m 11s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/2aa1b516b8844d618b419f38238271db : FAILURE in 15m 20s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/caecff350d3245c2b7ee4c00aeae8460 : SUCCESS in 12m 54s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/beb18c6848c343b78cb736519ace8333 : SUCCESS in 25m 32s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/04112790fb2a4e689d9dd7b67a974f1b : SUCCESS in 10m 27s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":3},{"id":"b8a2ca2f2a1acf79d14c7221a9711283b1f57aaa","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-25 14:04:34.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"919550b09827f17c498fa075db150d8ffe3cc816","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 15:51:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a16fdcf556694bca8fb59b73577b3155 : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed5232e945924499ad57f8562fc8214c : SUCCESS in 6m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a4ab3ce252604fd7bdb25e148d6e40be : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/660143c98d984e35ba52cb952cc1b22f : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0457abd611894f7f9e19c075c3052a56 : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bba19069c885491fa31f30c5436e9778 : SUCCESS in 12m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e770a3778d04b8db565fa1a39933a2a : SUCCESS in 5m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cfe2d769743e45b7b59478fd78f39595 : SUCCESS in 6m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6f07847fdefa4f87986c5008b6b0f136 : SUCCESS in 8m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e92e8eef013842128ea7ae9a200c4ddc : SUCCESS in 22m 04s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/9065f4eade2f4afbb6ee756ac84bbf30 : SUCCESS in 1h 05m 56s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/2419403a98ae4f0dacad35f611ceb83e : SUCCESS in 1h 07m 57s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/0af38605a72e46cd899448fa62d5ed47 : SUCCESS in 16m 08s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/9f4e69fb83ef4591aa1267ec4b008bd0 : SUCCESS in 1h 43m 02s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/c2f13304d5c84aaaa734d9458ac83101 : SUCCESS in 17m 16s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/79ffa8d3214245f2bd06895b5fd70fc1 : SUCCESS in 25m 47s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/a566988fa522400ba8b29109190ba1eb : SUCCESS in 10m 31s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":3},{"id":"cfdac3a1153561d1a875c9dfeba50b129c2c86e2","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-11-25 16:52:15.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nOnly the commit message was updated since Pierre\u0027s +2 so W+1","accounts_in_message":[],"_revision_number":3},{"id":"bd39c2bac6211284f9b713182f67ddb41d4fedc1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 08:38:46.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1bd495832fc9cb377e81d078a6ffaa21dbf03f32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 10:31:58.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf9d238e9ba64c28ba7c86058f47820d : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb2fe8c8eb9540e6a66ae1e20d334bd4 : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e0b023af048646609e4ebf59e434cb16 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6176ca4af467479bb8a64a01632b4ab1 : SUCCESS in 7m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ebe5b9c577f649b08b1c3e3f7ca31572 : SUCCESS in 7m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/929e2572260b4d428359b9d53e0159b9 : SUCCESS in 3m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/982ffb7529e34582b740a99b292a19f5 : SUCCESS in 6m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e0351199b31e4db68b619fd04646be43 : SUCCESS in 19m 36s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/1bdd3a318cc04c6e9223ff60cdb9ba29 : SUCCESS in 1h 03m 48s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/3dc077c0018c4df8a7114e411c0639cf : SUCCESS in 1h 11m 14s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/d89858fa773f4a46a0d5445830e7613b : SUCCESS in 14m 53s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/df0bf21fadc741e8b0287a36d6b0bef9 : SUCCESS in 1h 42m 24s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/29492f5f87564f9195fb0ec31acc8937 : SUCCESS in 15m 31s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e6f1ca57a29c44f5872021c397d071de : SUCCESS in 22m 45s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/f8493676110c40f692c0142efad57ab4 : SUCCESS in 10m 41s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":3},{"id":"d24df69ce81caa62360806690cee94d73e4e6cab","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 10:31:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"dbd9cae0ae06ef67ba47b26272ea241002f81c92","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 10:36:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b92804148ec4fa4b67c1a0281933318 : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e78c8acd179141a58b7dc2194e526617 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c1c0f9d018649ba6374d2ed4a3304e8ff04d5a00","revisions":{"9622a81106a95d13c232682d641ec8a88032abc5":{"kind":"REWORK","_number":1,"created":"2020-11-19 13:58:14.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/74/763374/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/763374/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/763374/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/763374/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/763374/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/763374/1"}}},"commit":{"parents":[{"commit":"074024d63f9cb364ca16a7a7f0ac94d77ee9466b","subject":"Use OpenStack Victoria release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/074024d63f9cb364ca16a7a7f0ac94d77ee9466b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-19 13:58:14.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-19 13:58:14.000000000","tz":0},"subject":"Revert \"Use OpenStack Victoria release\"","message":"Revert \"Use OpenStack Victoria release\"\n\nThis reverts commit 074024d63f9cb364ca16a7a7f0ac94d77ee9466b.\n\nChange-Id: Ia5715c6914ac454081a95b6e1788e80f53c72e30\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9622a81106a95d13c232682d641ec8a88032abc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9622a81106a95d13c232682d641ec8a88032abc5"}]},"branch":"refs/heads/master"},"aeb1183de70eddc8bfba0e46f3635bc11793ae56":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-11-20 08:57:11.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/74/763374/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/763374/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/763374/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/763374/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/763374/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/763374/2"}}},"commit":{"parents":[{"commit":"7226a1b8def039525b963ab869ec6fdaec32d2bb","subject":"CI: set previous_release to Victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7226a1b8def039525b963ab869ec6fdaec32d2bb"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-19 13:58:14.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-20 08:57:11.000000000","tz":0},"subject":"Revert \"Use OpenStack Victoria release\"","message":"Revert \"Use OpenStack Victoria release\"\n\nThis reverts commit 074024d63f9cb364ca16a7a7f0ac94d77ee9466b.\n\nChange-Id: Ia5715c6914ac454081a95b6e1788e80f53c72e30\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aeb1183de70eddc8bfba0e46f3635bc11793ae56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aeb1183de70eddc8bfba0e46f3635bc11793ae56"}]},"branch":"refs/heads/master"},"c1c0f9d018649ba6374d2ed4a3304e8ff04d5a00":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-11-24 15:12:13.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/74/763374/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/763374/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/763374/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/763374/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/763374/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/763374/3"}}},"commit":{"parents":[{"commit":"7226a1b8def039525b963ab869ec6fdaec32d2bb","subject":"CI: set previous_release to Victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7226a1b8def039525b963ab869ec6fdaec32d2bb"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-19 13:58:14.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-24 15:12:03.000000000","tz":0},"subject":"Revert \"Use OpenStack Victoria release\"","message":"Revert \"Use OpenStack Victoria release\"\n\nThis reverts commit 074024d63f9cb364ca16a7a7f0ac94d77ee9466b.\n\nDepends-On: https://review.opendev.org/763959\n\nChange-Id: Ia5715c6914ac454081a95b6e1788e80f53c72e30\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c1c0f9d018649ba6374d2ed4a3304e8ff04d5a00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c1c0f9d018649ba6374d2ed4a3304e8ff04d5a00"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
