)]}'
{"id":"openstack%2Fkayobe~763375","triplet_id":"openstack%2Fkayobe~master~I7abb6a25becde1e8855d1d177fafcb618789f0a1","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I7abb6a25becde1e8855d1d177fafcb618789f0a1","subject":"CI: set previous_release to Victoria","status":"MERGED","created":"2020-11-19 14:04:31.000000000","updated":"2020-11-26 10:23:33.000000000","submitted":"2020-11-26 10:19:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"763375","meta_rev_id":"8d5d27dcc874b9f3dc1abf02960a7a97533a8843","_number":763375,"virtual_id_number":763375,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-26 10:19:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2020-11-25 16:52:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-11-26 08:38:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2020-11-20 14:09:32.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-11-26 08:38:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-20 10:40:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-25 16:52:55.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-11-26 08:38:17.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"3abe289ca49e759820b661b031501b825ef52f6f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-19 14:04:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ad7d882ec6c1473ae4121200f3660c59491e7719","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 15:58:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dbf1c049facd46249d994f6f856aa9d0 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a24782c8e354d379f19d2ef3fecf35d : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3633ae52524744b0b465b3d7628cb62e : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc713f1785e8484c86e6acfe69bc08f0 : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/30d393032ea84122af8b03dd2fd71123 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef5fec088f89483197c4c65893d48ebb : SUCCESS in 7m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c907905f8a274b6ea79b4ffa2f51baf7 : SUCCESS in 4m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6566fb18e47f4fd1951e04c774f7433e : SUCCESS in 7m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d1fada6cb4c646d6b3ddea2c5ec6c9ed : SUCCESS in 20m 18s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/d53239584db34690b1a57824d36ad82d : FAILURE in 16m 16s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/eefe6ede13f541a69c177ce788011c58 : FAILURE in 13m 03s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/f74f0c4346554e879e3249c7d2a1c38c : SUCCESS in 16m 46s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/8ab1aec381c942b1aee546661dbb8a93 : SUCCESS in 1h 43m 34s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/26c7079bf4504646bd8137765a26dd14 : SUCCESS in 16m 07s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/fd7ccb7311824b0980029dceccc0723a : FAILURE in 1h 10m 24s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/6020ed4208c8484fb46269773f85c0c2 : SUCCESS in 11m 35s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"633043ea404a158479f999ce6dfc30785fc9292f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-20 08:56:59.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"50c265d906c17dccb8a03e5440a93972ef077584","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 10:40:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/254af1f7f81e41a882dca25dd9b02dd1 : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/327028976e4d48ac9fd6287c2bfe27ff : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61b5c76cdd4e4fb6adfbf3957eebe611 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6f7548e5f194c149941b4269a8be4a3 : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7c45059470a2406fa1419bb121959fbf : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6e0d073842e476bb569f64ff3b1c784 : SUCCESS in 9m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b395551f76584191b86511ffb1607658 : SUCCESS in 5m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b3d6ecc3288749c9902b5b8660653bd3 : SUCCESS in 9m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5665f10e4aef42aa940bf1a8cccce97b : SUCCESS in 21m 08s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/b593c70e9bed4968acd710ea94c0e0ae : SUCCESS in 1h 01m 18s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/e2ed43e6f5164c7fa4c4cc70ec578087 : SUCCESS in 1h 05m 20s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/23e5b12561f44f5d9e22721e1210af31 : SUCCESS in 14m 57s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a5d39d783e21459ea2aec5b516a9ac3d : SUCCESS in 1h 34m 14s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/2ed4144f85074dd2b9c926b0bc0aee4d : SUCCESS in 15m 00s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/2321d9658afd4645a8c0f70346eef72e : SUCCESS in 22m 40s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/c2e77c51fc0843be9c31c85c987bb939 : SUCCESS in 11m 27s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":2},{"id":"74cb6105bdd7a704f9b23d4791a615f4afaa0324","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-20 14:09:32.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"9f73e9b8f1cda0a9f67ddbd2338b288486c98817","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-11-25 16:52:55.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"776c9ca729c1b0f47ebdc3c23442e2875975f177","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-11-26 08:38:17.000000000","message":"Patch Set 2: Workflow+1 Review-Priority+2","accounts_in_message":[],"_revision_number":2},{"id":"253328907b06bc4640aa4335ac7baf9304dadabb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 08:38:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f2fb1b01a0bf76fa11129a8cf22ec044e2c1484c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 10:19:33.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16c0aa8e9b93436b98a01fff19c06d8e : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/95fa481a68e24b5a9e0de40ee66bd8dd : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f55fee8758ec4f19a21a9ac01cb46a9d : SUCCESS in 8m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c94f623f7f1a4241853751f6a85f42b2 : SUCCESS in 19m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e3101bcddc8b4e5183404be895261245 : SUCCESS in 8m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d2e05ec1434f46fb84cc9941585b644a : SUCCESS in 12m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1b50063b914648d4bd33f8b7a428edb1 : SUCCESS in 22m 03s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/5d5cec363ee743299611e7342eddddb6 : SUCCESS in 1h 03m 18s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/6c076f5435c0435faf74d18d3a18e3fe : SUCCESS in 1h 07m 29s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/797f099471b1405cb666dd90245bbfe9 : SUCCESS in 12m 42s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f814d444a8dc4ef9a1c5e0068dc183de : SUCCESS in 1h 35m 43s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/330824a89a7d4f5ea6d58742311bd812 : SUCCESS in 15m 07s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/6abe8a25f67c48eab3c667d3120af7b1 : SUCCESS in 22m 47s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/68735c7b990141819cfa6829b485859a : SUCCESS in 10m 33s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":2},{"id":"b3e2ef843c32c12e04d42d2d41e60286fb183ecd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 10:19:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"8d5d27dcc874b9f3dc1abf02960a7a97533a8843","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 10:23:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/382c04c8412740e89c04eac4c207627b : SUCCESS in 52s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7226a1b8def039525b963ab869ec6fdaec32d2bb","revisions":{"af614093a99561516000d3e390de9873062b879c":{"kind":"REWORK","_number":1,"created":"2020-11-19 14:04:31.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/75/763375/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/75/763375/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/763375/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/763375/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/763375/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/75/763375/1"}}},"commit":{"parents":[{"commit":"9622a81106a95d13c232682d641ec8a88032abc5","subject":"Revert \"Use OpenStack Victoria release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9622a81106a95d13c232682d641ec8a88032abc5"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-19 14:03:51.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-19 14:03:51.000000000","tz":0},"subject":"CI: set previous_release to Victoria","message":"CI: set previous_release to Victoria\n\nThis controls which release we are upgrading from.\n\nChange-Id: I7abb6a25becde1e8855d1d177fafcb618789f0a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/af614093a99561516000d3e390de9873062b879c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/af614093a99561516000d3e390de9873062b879c"}]},"branch":"refs/heads/master"},"7226a1b8def039525b963ab869ec6fdaec32d2bb":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-11-20 08:56:59.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/75/763375/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/75/763375/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/763375/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/763375/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/763375/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/75/763375/2"}}},"commit":{"parents":[{"commit":"ae93b364e55fd8c4d1f426b05b8cecbad632c61f","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ae93b364e55fd8c4d1f426b05b8cecbad632c61f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-19 14:03:51.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-20 08:56:59.000000000","tz":0},"subject":"CI: set previous_release to Victoria","message":"CI: set previous_release to Victoria\n\nThis controls which release we are upgrading from.\n\nChange-Id: I7abb6a25becde1e8855d1d177fafcb618789f0a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7226a1b8def039525b963ab869ec6fdaec32d2bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7226a1b8def039525b963ab869ec6fdaec32d2bb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
