)]}'
{"id":"openstack%2Fkayobe~763734","triplet_id":"openstack%2Fkayobe~stable%2Fvictoria~I1582d06dfa2d7384e1fd33a68570745cde6beaaa","project":"openstack/kayobe","branch":"stable/victoria","hashtags":[],"change_id":"I1582d06dfa2d7384e1fd33a68570745cde6beaaa","subject":"DNM: debug seed-upgrade job","status":"ABANDONED","created":"2020-11-23 19:19:49.000000000","updated":"2020-11-25 16:30:10.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"cherry_pick_of_change":763773,"cherry_pick_of_patch_set":1,"meta_rev_id":"aa1ee237801e71cba9763d63ab33152a227b886d","_number":763734,"virtual_id_number":763734,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-11-23 19:41:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2020-11-23 19:20:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5919e85a11faca29e4c31bd6c77fdd39efc65077","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-23 19:19:49.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* zuul.d/project.yaml","accounts_in_message":[],"_revision_number":1},{"id":"f47e16ed83a17e550fd22b0d99430f206d4b84af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 19:20:11.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kayobe on branch stable/victoria.  The error was:\n\n  while scanning a simple key\n  in \"openstack/kayobe/zuul.d/project.yaml@stable/victoria\", line 3, column 1\ncould not find expected \u0027:\u0027\n  in \"openstack/kayobe/zuul.d/project.yaml@stable/victoria\", line 4, column 14","accounts_in_message":[],"_revision_number":1},{"id":"53be32c05fc79e00c2efedbb1c8190e798b77e93","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-23 19:20:31.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"c3cb7ed1db950c369ab79815f5cdecf391e74653","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 19:41:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/91d804a26e9845849f80099c052c646e : FAILURE in 18m 53s","accounts_in_message":[],"_revision_number":2},{"id":"aa1ee237801e71cba9763d63ab33152a227b886d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-25 16:30:10.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"55dc9272e3ed0ed41df3a22a49930d0f337a062a","revisions":{"3635371ef1f13c62edc1649eb7fd2775bed54561":{"kind":"REWORK","_number":1,"created":"2020-11-23 19:19:49.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/34/763734/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/34/763734/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/763734/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/763734/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/763734/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/34/763734/1"}}},"commit":{"parents":[{"commit":"b1be865d33fd52e9c969619f27aaa04a2c2a2a0c","subject":"Update TOX_CONSTRAINTS_FILE for stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b1be865d33fd52e9c969619f27aaa04a2c2a2a0c"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-23 11:47:47.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-23 19:19:49.000000000","tz":0},"subject":"DNM: debug seed-upgrade job","message":"DNM: debug seed-upgrade job\n\nChange-Id: I1582d06dfa2d7384e1fd33a68570745cde6beaaa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3635371ef1f13c62edc1649eb7fd2775bed54561"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3635371ef1f13c62edc1649eb7fd2775bed54561"}]},"branch":"refs/heads/stable/victoria"},"55dc9272e3ed0ed41df3a22a49930d0f337a062a":{"kind":"REWORK","_number":2,"created":"2020-11-23 19:20:31.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/34/763734/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/34/763734/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/763734/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/763734/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/763734/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/34/763734/2"}}},"commit":{"parents":[{"commit":"b1be865d33fd52e9c969619f27aaa04a2c2a2a0c","subject":"Update TOX_CONSTRAINTS_FILE for stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b1be865d33fd52e9c969619f27aaa04a2c2a2a0c"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-23 11:47:47.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-23 19:20:25.000000000","tz":0},"subject":"DNM: debug seed-upgrade job","message":"DNM: debug seed-upgrade job\n\nChange-Id: I1582d06dfa2d7384e1fd33a68570745cde6beaaa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/55dc9272e3ed0ed41df3a22a49930d0f337a062a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/55dc9272e3ed0ed41df3a22a49930d0f337a062a"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
