)]}'
{"id":"openstack%2Fkayobe~763773","triplet_id":"openstack%2Fkayobe~master~I1582d06dfa2d7384e1fd33a68570745cde6beaaa","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I1582d06dfa2d7384e1fd33a68570745cde6beaaa","subject":"DNM: debug seed-upgrade job","status":"ABANDONED","created":"2020-11-23 11:48:23.000000000","updated":"2020-11-25 16:30:06.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"933dc9e99d1468ea5e7ee29537158b91da787149","_number":763773,"virtual_id_number":763773,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-11-24 11:09:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-23 12:18:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4b257f9a5404e0f71df6b7bdd60e12d918ca6cad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-23 11:48:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"29ff413d54d21d370b3f90ab09604c1ca4a683f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 12:18:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/ffe2b13bec754aaea6a23817cbbbe7f4 : SUCCESS in 23m 46s","accounts_in_message":[],"_revision_number":1},{"id":"794c376a4ff51aecce030d03a4261760ade501ca","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-23 12:38:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7b13365a728a67f4371014cbd2b6cf8bba762a17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 13:09:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/610e3901e0744ee694279cc31b1f4275 : SUCCESS in 24m 38s","accounts_in_message":[],"_revision_number":1},{"id":"96c7a4aac01ae0ed1fabbdf50ae166a36a49de73","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-23 14:09:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6cc2857ea80bb792d795506c67b9738cbff58848","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 14:33:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/17c52f2140cf4499a1dd73d6e4383a4d : SUCCESS in 21m 35s","accounts_in_message":[],"_revision_number":1},{"id":"ef37503b631d580dff0c531e0a52038545a6148a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-23 19:21:31.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"703d521c021ad3e3c05c3b06842f52f70dcdbd38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 20:50:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/2f6b6f56205647c7b9afaf564e334b79 : FAILURE in 1h 22m 30s","accounts_in_message":[],"_revision_number":2},{"id":"d6c3d5bcf7e3f30e02525a92468be0203a698664","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-24 10:46:31.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"e7b088bf9e61c2280138959926d168a650ebaf7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 11:09:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/dc6c76ce8f9a4f9281a215527bdd895c : SUCCESS in 18m 37s","accounts_in_message":[],"_revision_number":3},{"id":"2d5f0b6a7fa228868b7f9ab2e73f2b711f11fa3d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-24 11:10:32.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"b3bfa1a1b930abd8d023bcaa25fd9184c3513f0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 11:41:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/73dbe058f5fe44d59cdf475c9bd85638 : SUCCESS in 25m 40s","accounts_in_message":[],"_revision_number":3},{"id":"3ee11db15bdad38f978cae213c0e4833b4a16d81","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-24 11:56:09.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"4512d48871bb7dc6141fea8210fb2044580ae9c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 12:21:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/6d1e8169997b49dd81e3df77b6b0734b : SUCCESS in 23m 20s","accounts_in_message":[],"_revision_number":3},{"id":"53a4cc95e549f849510c7f42f1aefb28ef3e7a39","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-24 12:25:12.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"45f48b846685bebd9ae199d1bd6668cd6ade5d7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 12:52:25.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/19e36ef1752141559300a0b27ffd913e : SUCCESS in 24m 28s","accounts_in_message":[],"_revision_number":3},{"id":"d828965a9816570fe6534b22681fff50ec5f8546","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-24 12:53:57.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"d9d75cd38764733f9363240f1910a6e7d0571364","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 13:29:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/adc90d2c2dc64faf94e17b96f41b99ad : SUCCESS in 26m 14s","accounts_in_message":[],"_revision_number":3},{"id":"933dc9e99d1468ea5e7ee29537158b91da787149","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-25 16:30:06.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3e739171ebdde7bfe2c3facfc294c077f92ac3ca","revisions":{"593324e4a36b6d80f9b00b79e13bc89c40a42f39":{"kind":"REWORK","_number":1,"created":"2020-11-23 11:48:23.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/73/763773/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/73/763773/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/763773/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/763773/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/763773/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/73/763773/1"}}},"commit":{"parents":[{"commit":"aeb1183de70eddc8bfba0e46f3635bc11793ae56","subject":"Revert \"Use OpenStack Victoria release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aeb1183de70eddc8bfba0e46f3635bc11793ae56"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-23 11:47:47.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-23 11:47:47.000000000","tz":0},"subject":"DNM: debug seed-upgrade job","message":"DNM: debug seed-upgrade job\n\nChange-Id: I1582d06dfa2d7384e1fd33a68570745cde6beaaa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/593324e4a36b6d80f9b00b79e13bc89c40a42f39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/593324e4a36b6d80f9b00b79e13bc89c40a42f39"}]},"branch":"refs/heads/master"},"88dc95b1ecc1f3abf081bfaebf4aa8db356b3979":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-11-23 19:21:31.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/73/763773/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/73/763773/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/763773/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/763773/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/763773/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/73/763773/2"}}},"commit":{"parents":[{"commit":"aeb1183de70eddc8bfba0e46f3635bc11793ae56","subject":"Revert \"Use OpenStack Victoria release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aeb1183de70eddc8bfba0e46f3635bc11793ae56"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-23 11:47:47.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-23 19:21:22.000000000","tz":0},"subject":"DNM: debug seed-upgrade job","message":"DNM: debug seed-upgrade job\n\nDepends-On: https://review.opendev.org/763734\n\nChange-Id: I1582d06dfa2d7384e1fd33a68570745cde6beaaa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/88dc95b1ecc1f3abf081bfaebf4aa8db356b3979"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/88dc95b1ecc1f3abf081bfaebf4aa8db356b3979"}]},"branch":"refs/heads/master","description":"Edit commit message"},"3e739171ebdde7bfe2c3facfc294c077f92ac3ca":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-11-24 10:46:31.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/73/763773/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/73/763773/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/763773/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/763773/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/763773/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/73/763773/3"}}},"commit":{"parents":[{"commit":"aeb1183de70eddc8bfba0e46f3635bc11793ae56","subject":"Revert \"Use OpenStack Victoria release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aeb1183de70eddc8bfba0e46f3635bc11793ae56"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-23 11:47:47.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-24 10:46:25.000000000","tz":0},"subject":"DNM: debug seed-upgrade job","message":"DNM: debug seed-upgrade job\n\nDepends-On: https://review.opendev.org/763734\nDepends-On: https://review.opendev.org/763959\n\nChange-Id: I1582d06dfa2d7384e1fd33a68570745cde6beaaa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3e739171ebdde7bfe2c3facfc294c077f92ac3ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3e739171ebdde7bfe2c3facfc294c077f92ac3ca"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
