)]}'
{"id":"openstack%2Fkayobe~763959","triplet_id":"openstack%2Fkayobe~master~I47115e5d8d1fb448cb9e46dc43b56c64987e428b","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I47115e5d8d1fb448cb9e46dc43b56c64987e428b","subject":"CI: add dummy1 port for seed upgrade and VM jobs","status":"MERGED","created":"2020-11-24 10:45:49.000000000","updated":"2020-11-25 13:47:34.000000000","submitted":"2020-11-25 13:45:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"763959","meta_rev_id":"ae33ae7c1bb63860b9b2186fe9fc99066a754994","_number":763959,"virtual_id_number":763959,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-25 13:45:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2020-11-24 15:11:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-11-24 19:34:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-11-24 11:15:30.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2020-11-24 19:34:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2020-11-24 15:09:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-24 11:13:09.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2020-11-24 12:44:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-24 15:11:43.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-11-24 19:34:54.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"}],"messages":[{"id":"de838c34135c51432b24ae88bd4dc1e3feab9ca7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-24 10:45:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7f4672a837c808852d12216b99b3afc9b33efd66","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2020-11-24 11:15:30.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b910b842a220883fa2e4e3dbf498123947ac6dac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 12:44:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b2cc91db7f74049ae631728b083661b : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd6c0f97545a44fe821eec6806e68ee7 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b2a4c8a843c48d3afa4fadf40bd5991 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/82e01091f6f14d4db33771c42843dc02 : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d36346f62a4a4d39b1c8cb0363d15e6c : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c3282a2e3eb4a5381a9f9571b6bf577 : SUCCESS in 7m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/333bf0bcd2ce4167b040561a23d14140 : SUCCESS in 3m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bd2964cfd4c943bd95f3f87b543cd3bc : SUCCESS in 6m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5f6457b5cc0f4be1a5ebe0c8d674b43e : SUCCESS in 21m 14s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/d5451db864b24890b26a5d63cb290982 : SUCCESS in 1h 01m 35s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/10a614429bc74905ac175eccbf046b99 : SUCCESS in 1h 08m 08s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/9803ad25de924f84b784b15472e30f39 : SUCCESS in 16m 22s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/1c7ed3f09eca489faf503b0d7c13d21f : SUCCESS in 1h 51m 25s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/817c4b6b20d64d44a8a273c11ab307c9 : SUCCESS in 15m 24s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/bedca6b0a3764a4993662fa38ec85e3c : FAILURE in 18m 11s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/73050242f82b49199de29312a21760f0 : SUCCESS in 10m 52s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"cb5dd3e876ae720c0112896323f006e5fb9c8451","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-24 12:49:26.000000000","message":"Patch Set 1:\n\nrecheck\n\nDifferent error at least this time...\n\nFailed to download metadata for repo \u0027AppStream\u0027: Cannot download repomd.xml: Cannot download repodata/repomd.xml: All mirrors were tried","accounts_in_message":[],"_revision_number":1},{"id":"605916c18d067f55011c3810585a3f59b7e6c530","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 14:50:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3aa4d4dab470401f865ae8fa1bcc062e : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99330e8368574d9fb4e59c4225a1de31 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3fde7db5a4a45e6a7fdbce99b30dd31 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db1fbb753872446b98179437d3cfaa01 : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dae15aa66ed34911a1b2cd7b19fd0470 : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89d023e374b34c42bdee3b75322a82dc : SUCCESS in 6m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8562f3fd0c224240887be3cfffdb25ef : SUCCESS in 4m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/070c85f8d19e49af8ec6059027fe74c2 : SUCCESS in 8m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1a3a88d1f8754f4598a60b03f4081e3c : SUCCESS in 18m 47s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/5fda1bea05a14e6d95b1dbf14349efaa : SUCCESS in 48m 35s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/2eaf96583a41411fabf80f50a6fec450 : SUCCESS in 1h 07m 26s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/52eed2fecc5e435388cf35391af8b7a9 : SUCCESS in 17m 32s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/230aa2c8bc1746a7a405035e01826b89 : SUCCESS in 1h 55m 20s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/ba7625ca8d7743bf8969c432330a9eaf : SUCCESS in 16m 06s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/7265dd28470a474e9dc66a447ec5f71f : SUCCESS in 28m 03s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/50a970b5d46547b3ac00620c4de8ee7a : SUCCESS in 10m 44s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"3fd6cfd259e6862fab0d25dd1d1853d542d8871d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-24 15:09:49.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"5c019d2e60ea68a7b90eabc96e26002bcb812573","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-11-24 15:11:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2b47a521b0cf31c78dfd60baaf121ea88961959c","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-11-24 19:34:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"434a254d01bd71a0256fe292b61d906d4cc719d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 19:35:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"725c944b39e9546a1008bc4ca57b6757945549e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 21:34:21.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e3dde19a72f43708fb1cc591f141b7a : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9683a42147d42e1b3bd03adc1a33412 : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/24a8b5f917f646b491a1b697fad30737 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e53b01b86eac43909c91e5bfed2c3c27 : SUCCESS in 6m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9f3fee4a09dc472cb76fbf54310d6122 : SUCCESS in 4m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/51cf39e6d19e4e689ec66fef151d1d02 : SUCCESS in 8m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b0181c650ed54202a82dbda4c81761d2 : SUCCESS in 19m 33s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/93f71d15cdfd41d2836ec8f4f9ab07b6 : SUCCESS in 1h 11m 32s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/582aacac71aa4fe98fa24b13ee1eeebd : FAILURE in 1h 14m 19s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/2db92ea96f284048b6b61caa89dc40fb : SUCCESS in 14m 47s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/fbd4d13722b54cf7a6f76a4b5efe2eb7 : SUCCESS in 1h 52m 58s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/bec2a413d09e40ea9c82498638469888 : SUCCESS in 15m 04s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/653bf6d721e646feb20cecce46ace664 : SUCCESS in 41m 50s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/c9be2707b73a42e18233e88e0341d48d : SUCCESS in 11m 59s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"d48652d83cda35a3e25b4f1b20b8a82ea9530965","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-25 09:33:14.000000000","message":"Patch Set 1:\n\nrecheck\n\nfailed to ping baremetal instance, although it looks like it had just about booted, with a metadata failure.","accounts_in_message":[],"_revision_number":1},{"id":"998521ca4246f3d2ded03db1ef8eb79ca79b0816","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 11:30:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f036491af1940eba98c0665d3720366 : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd5d15dffcaf4a039908092618f0b685 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c7addee52a8549c887b9e217880345b4 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e4d953953ef4717b83c61ee91e8adc5 : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fa677d593190497d8763e9a8c5d7fa0a : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c536484a8214f4c85ea4b8f9f391bb7 : SUCCESS in 6m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/76ca5f948ea04a0a8d0a1a79c6042104 : SUCCESS in 4m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/dd1558f374dc422b955dea10a644ea2c : SUCCESS in 8m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/066a661269d64d86a2937e56f1996b6b : SUCCESS in 21m 09s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/52c887fa4ad041f59853c9c1079882ff : SUCCESS in 1h 00m 43s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/222e5e30618d42069ab18aa892c1bbc9 : SUCCESS in 59m 44s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/e782b816094547489ca6f9a32d14ef1e : SUCCESS in 18m 32s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/4c321009c6e543cdb23c06a6fc5e80d6 : SUCCESS in 1h 51m 40s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/6b5d24230d2b480595303ce7fd81b9fe : SUCCESS in 16m 43s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/be2553f11bb74692b51e7be6b19a6524 : SUCCESS in 24m 20s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/7395c0c25d4c4f18b2219b646716e936 : SUCCESS in 10m 16s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"6295fdaca6e01c6d5203a925f0ffe0c06e58ba5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 11:30:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3aec110368b5c92a8f874678408821000cda2274","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 13:45:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/200dd9a780de49c9a119145e687e8517 : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/91f5c3ae5314446d8316e097e0462c3c : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/31b06f569066464687b7ffc7066bb816 : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/475c6c376f724cd5affa01d6842340fa : SUCCESS in 7m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d074e095d393415bba3ce954855500fc : SUCCESS in 4m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/37259ed1146848cf8b527d7cf86b3280 : SUCCESS in 9m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1875e6689e0a4b93a49f18ab47ab86af : SUCCESS in 20m 17s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/3129ac703d274cb6a42e1d235b04431c : SUCCESS in 1h 04m 50s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/9447c4612a7046bfa82f72f059d079ef : SUCCESS in 1h 09m 50s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/f890cbfe2a3e4964aaafcdcec9ad8090 : SUCCESS in 16m 20s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/874dac18558243e4aa8d9c8f43804590 : SUCCESS in 2h 08m 34s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/9597fc7228d945bca4bce843390f6f9b : SUCCESS in 16m 48s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/d9f9e36a53da4c17ba232b23ea276322 : SUCCESS in 25m 04s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/6922395e96db499eb1db335024c40b96 : SUCCESS in 10m 36s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"a6d24b818e01eeb1f20327de83dc754d0aea0c24","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 13:45:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ae33ae7c1bb63860b9b2186fe9fc99066a754994","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 13:47:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76b58b1f4c894bf88873ca95326c2575 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0766df1673da0cc8c3a3d3d03e357ec5836e2d94","revisions":{"0766df1673da0cc8c3a3d3d03e357ec5836e2d94":{"kind":"REWORK","_number":1,"created":"2020-11-24 10:45:49.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/59/763959/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/59/763959/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/763959/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/763959/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/763959/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/59/763959/1"}}},"commit":{"parents":[{"commit":"ae93b364e55fd8c4d1f426b05b8cecbad632c61f","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ae93b364e55fd8c4d1f426b05b8cecbad632c61f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-24 10:35:00.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-24 10:45:35.000000000","tz":0},"subject":"CI: add dummy1 port for seed upgrade and VM jobs","message":"CI: add dummy1 port for seed upgrade and VM jobs\n\nFor some reason this configuration was omitted or incomplete in these\njobs. Without this configuration, Kayobe development configuration adds\neth1 to the breth1 bridge if it exists, or leaves the bridge without a\nport otherwise. This can lead to network connectivity issues, if the IP\naddress of eth1 becomes inaccessible when it is added to breth1.\n\nThis change fixes the issue by ensuring that the dummy1 interface\nexists, and configuring the aio network to add it to breth1 as a port,\ninstead of eth1.\n\nChange-Id: I47115e5d8d1fb448cb9e46dc43b56c64987e428b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0766df1673da0cc8c3a3d3d03e357ec5836e2d94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0766df1673da0cc8c3a3d3d03e357ec5836e2d94"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
