)]}'
{"id":"openstack%2Fkayobe~764183","triplet_id":"openstack%2Fkayobe~stable%2Fussuri~I47115e5d8d1fb448cb9e46dc43b56c64987e428b","project":"openstack/kayobe","branch":"stable/ussuri","hashtags":[],"change_id":"I47115e5d8d1fb448cb9e46dc43b56c64987e428b","subject":"CI: add dummy1 port for seed upgrade and VM jobs","status":"ABANDONED","created":"2020-11-25 16:28:24.000000000","updated":"2020-12-22 12:25:09.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":763959,"cherry_pick_of_patch_set":1,"meta_rev_id":"8e32197ddd39534fa69377230351dd01b399a397","_number":764183,"virtual_id_number":764183,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"rejected":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":-2,"date":"2020-12-18 15:33:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-12-18 12:54:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2020-11-26 14:08:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2020-12-18 12:54:21.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-25 18:48:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-18 12:54:21.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"2796a1945f663e7295f701f7c41da897d753e4c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-25 16:28:24.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"13d6dd2b4cd88d553184070ae3ac786626a17040","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 18:48:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec8ab6b48b374c09ba707a6e49a20fd3 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e97d59ef8215409d9ac9c78deb5f2adb : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/758a6673c32e4e22aec8dfe28b3b7646 : SUCCESS in 4m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8faaaaaa598b4cbf8d207ae10a411516 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7af74e0f22a14aea95303589e064b9d3 : SUCCESS in 9m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/82a13ce4e09b4e579b743dc017f5100a : SUCCESS in 5m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fb7019a0008e4e5499aa3391747e4855 : SUCCESS in 8m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/da3849a84adf471f9410ed76a7d8aa35 : SUCCESS in 22m 24s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/8944c253ef694af2acc7dec4913b2b2d : SUCCESS in 1h 07m 40s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/e64a43ff3ef64c2586424070e2564e58 : SUCCESS in 16m 41s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/245963a28e7649c994068e45c9acd741 : SUCCESS in 1h 59m 05s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/56647bfd23164d068969831f0925722b : SUCCESS in 19m 06s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/39a4e1a1f16547fdbbf8190b0e8b8537 : SUCCESS in 30m 02s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/c7d17343fe9d45fe84af5aea3034d23f : SUCCESS in 11m 54s","accounts_in_message":[],"_revision_number":1},{"id":"eb7fad80302f92501317d1c366694f440d35a7b5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-26 14:08:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fa81601f2864443f72b0560f1196a2742d1953ef","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-12-18 12:54:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0a4b76eb1a4d456743718154fcf11de5fe8c752d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 12:54:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ed060517377c569a95d81cdf2fbc87bc6baef666","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 15:33:29.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb6a747287084768986b8d3ba854ac1b : FAILURE in 2m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe72255f1d5742d9a9f1f88e0caae024 : FAILURE in 2m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8b19bb55dcd648db8bffa4c3c421cb55 : FAILURE in 2m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59e38399967d4ab68db319ead1a91a8a : SUCCESS in 5m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fadffd5081794ea4ac079c3f9e7ad6ae : FAILURE in 2m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/04e5119b7d4a4259a1de002e55732320 : FAILURE in 2m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/45095e2593b44697abbf5c1c7c844cc2 : FAILURE in 3m 07s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/6e99f99062e746bd8ed0e79b47578949 : FAILURE in 47m 28s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/3ea491e201f64ca09b41d38e8d46d4e6 : SUCCESS in 15m 02s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/28551dece49e4f48aa7a05c1a2a6468d : FAILURE in 47m 13s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/b605294b25b04719b18beb5051f92fd0 : SUCCESS in 16m 43s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a3a0afc76eff4905bf3e07255c95ca9e : SUCCESS in 24m 26s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/3a7a17cdcbd64586a557e4a6fc382758 : SUCCESS in 10m 25s","accounts_in_message":[],"_revision_number":1},{"id":"8e32197ddd39534fa69377230351dd01b399a397","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-22 12:25:09.000000000","message":"Abandoned\n\nRolled into https://review.opendev.org/c/openstack/kayobe/+/767599","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8f50c72a7bb5b5d50f0d610a90aaae320d14a252","revisions":{"8f50c72a7bb5b5d50f0d610a90aaae320d14a252":{"kind":"REWORK","_number":1,"created":"2020-11-25 16:28:24.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/83/764183/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/83/764183/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/764183/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/764183/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/764183/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/83/764183/1"}}},"commit":{"parents":[{"commit":"a8d5cc0ba9caf4ab348b16873c36e065ed244d68","subject":"Merge \"docs: Add link back to Train CentOS 8 migration page\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a8d5cc0ba9caf4ab348b16873c36e065ed244d68"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-24 10:35:00.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-25 16:28:24.000000000","tz":0},"subject":"CI: add dummy1 port for seed upgrade and VM jobs","message":"CI: add dummy1 port for seed upgrade and VM jobs\n\nFor some reason this configuration was omitted or incomplete in these\njobs. Without this configuration, Kayobe development configuration adds\neth1 to the breth1 bridge if it exists, or leaves the bridge without a\nport otherwise. This can lead to network connectivity issues, if the IP\naddress of eth1 becomes inaccessible when it is added to breth1.\n\nThis change fixes the issue by ensuring that the dummy1 interface\nexists, and configuring the aio network to add it to breth1 as a port,\ninstead of eth1.\n\nChange-Id: I47115e5d8d1fb448cb9e46dc43b56c64987e428b\n(cherry picked from commit 0766df1673da0cc8c3a3d3d03e357ec5836e2d94)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8f50c72a7bb5b5d50f0d610a90aaae320d14a252"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8f50c72a7bb5b5d50f0d610a90aaae320d14a252"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
