)]}'
{"id":"openstack%2Fkayobe~764184","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~I47115e5d8d1fb448cb9e46dc43b56c64987e428b","project":"openstack/kayobe","branch":"stable/train","hashtags":[],"change_id":"I47115e5d8d1fb448cb9e46dc43b56c64987e428b","subject":"CI: add dummy1 port for seed upgrade and VM jobs","status":"MERGED","created":"2020-11-25 16:29:53.000000000","updated":"2021-01-26 20:13:30.000000000","submitted":"2021-01-26 20:11:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"764184","cherry_pick_of_change":763959,"cherry_pick_of_patch_set":1,"meta_rev_id":"059d7fa6ec839ec0b95402069660a5443c4aca88","_number":764184,"virtual_id_number":764184,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-26 20:11:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-01-26 20:13:30.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-01-26 20:13:30.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-26 18:06:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-26 18:06:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-25 18:43:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-14 13:36:28.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-01-26 10:53:39.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2021-01-26 18:06:02.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"8818b62c81847b99473dbf18928750305401c6cb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-25 16:29:53.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* playbooks/kayobe-seed-vm-base/overrides.yml.j2","accounts_in_message":[],"_revision_number":1},{"id":"59d3e614c514745d94e6c2115f01f08114e24673","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-25 16:31:11.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"56bedfd70d251838e16626df2b7a66fc8267bbde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 18:43:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b14f8dd150543a5a6fae974a01f2a44 : SUCCESS in 6m 00s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/bb4d0de41481485ea938a740c7aa007b : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a97b0a4265174a9486d87d3425321194 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eae675c745164692856a56f687a31017 : SUCCESS in 4m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8bf5ca9373924d62b06b6a0204b25d9b : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49b9352668624eecaed4e969784d94bb : SUCCESS in 5m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/05866e5b1a004d6fb839a37afd252a51 : SUCCESS in 4m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c40903b1263a41b5948cbeb4150e498c : SUCCESS in 8m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b103aa41588240189e5597a71b6bb6a2 : SUCCESS in 19m 22s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/a61f0d6cacd645dfb6c4fe3e06d5767f : SUCCESS in 1h 05m 37s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/80d550ba7221498087c1e2abdc044868 : SUCCESS in 1h 06m 03s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/11f7ba96fa50424d95d2940277065182 : SUCCESS in 15m 43s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/3f30ad945f594c9ebbe78eb33c446fa8 : SUCCESS in 17m 16s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/4c26d58b4266445e953aa4e7ee544c3f : SUCCESS in 1h 43m 13s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/f4506c20846448889f28523bdf3e6095 : SUCCESS in 24m 28s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/b2bcc9d116bc461d87211932c28add15 : SUCCESS in 18m 08s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/e557feb02e8044b792f0871aa55e7fad : FAILURE in 23m 35s","accounts_in_message":[],"_revision_number":2},{"id":"42865f4755823d9ca353eada915d3310ddfde09d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-05 11:05:39.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"043b37ae9303eab035c223f5ac9cf47b3c49a046","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 12:49:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/749604ebdfe4482fbba2d0481a459c50 : SUCCESS in 4m 25s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7816cc25dd434434a6b1be105e49cca8 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c98b4cb834de483aac5c2cac3b62c3cf : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc1643114e924c4293627d8b2335bac9 : SUCCESS in 3m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1a0ab202ab6b4fe59335dfeb99497df9 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb8957c7063f43a39dd813acf0655c6e : SUCCESS in 9m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d4563923b2504004b6383286e655f611 : SUCCESS in 5m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/135dceffd4a2480b9bf2fc52babd7600 : SUCCESS in 9m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9675fd86b3c04945adb65385d404d3c7 : SUCCESS in 20m 38s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/b5714da24cfd48709de931da9a5d20a1 : SUCCESS in 1h 07m 05s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/cad4f1cac2444210adb0215bd4c1688d : SUCCESS in 58m 41s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/0c0dc2a2625f418581838bc1445a518f : SUCCESS in 14m 47s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/4cf3398f33a84498a42a658b68219d83 : SUCCESS in 16m 33s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/ba14069d94ed4a5a83a9b5ba3ad89637 : SUCCESS in 1h 41m 40s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/b8c43cab2e1743d88f2a80dd788a1d1c : SUCCESS in 19m 24s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/e4948df4229f4c1081f3895d64b06dad : SUCCESS in 17m 41s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/5af80563e43946229bbabaf26fab6877 : SUCCESS in 31m 52s","accounts_in_message":[],"_revision_number":2},{"id":"a4be7df990ed8ef2ecbb7cabffc24423c0bc463d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-05 13:37:03.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1d7044ae77475afed2b223ff133cc706bf37cb3e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-01-14 13:36:28.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9aaa3375ca771235731f22e96447d9440b31afbb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 13:36:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"31af56abc0fcdf087a0b2211f8032efcda5924eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 15:16:27.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1fd9cc59f953450b98ac19269b0c5d8b : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58864d6ff4d347e5baa5fff2f5d2b1bd : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0fa46b7d4d34f55ba47f2ee54f21aab : SUCCESS in 5m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fc11c59d2e6f442ca8df7c7773b9050d : SUCCESS in 7m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5d24a6015a84a8b9acb3720c4111494 : SUCCESS in 5m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/aec4c0326dab416eb3d8f00fdc9fd57b : SUCCESS in 4m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ea96db30b82b4114a348eaff84606e39 : SUCCESS in 8m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/49ac4844921a41999ad0ddbe5ebbb61b : SUCCESS in 16m 41s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/e73129b8f4f546dca5571b90cb39d9a5 : SUCCESS in 59m 21s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/28e8c31aef6a4a77aa708afc98d17895 : SUCCESS in 1h 22m 39s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/e84de3b5385543a1bbbba1a246fb34bb : SUCCESS in 11m 21s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/b4ba8d0d5ae04cf6aab063237969d3a9 : SUCCESS in 18m 07s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/282b0354132c41d89fa0a3a315603ddf : FAILURE in 53m 34s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/6a29a24dfb2044b29fc2391166254179 : SUCCESS in 15m 49s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/709ab2e707b0425684d144d21985aaa1 : SUCCESS in 16m 30s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/3543e8f4c7d14b0c9cff0b5e868c7b81 : SUCCESS in 30m 26s","accounts_in_message":[],"_revision_number":2},{"id":"e6eca0855d001947494c3dd9f81c015dab5f2be1","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-26 10:53:39.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f20e0655f14ead91b80022500a40a59dcff10605","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 12:55:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d2e1d3f05114cc384e245afab07d801 : SUCCESS in 3m 49s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5c443425177a4fc2b651488ec6b805e7 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a8788fb2c584eeab94c091873db30f0 : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a13bf9822f94dd8b68ba1109cfdb531 : SUCCESS in 4m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/164c3eb40c9a47c4b2328301f1a04ba4 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf5f6d3e4e074d168d45b97eddc9ef8d : SUCCESS in 5m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/78c49725da94411a8fd9eb457f4273b8 : SUCCESS in 3m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e25f3b3b7d684df097f73ed4676d4c3a : SUCCESS in 7m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/aaac342a775b458186591bb7bd9276b8 : SUCCESS in 20m 18s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/0792f775138544fe866ac07bbd37ce48 : SUCCESS in 59m 00s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/3ac63f5ad2d0412f8ee3e02222f7094c : FAILURE in 1h 58m 59s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/be09756e4656424ab7bf4b7ae838adb6 : SUCCESS in 13m 04s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/0a318a1e936f4804834b6ccfaf1372b6 : SUCCESS in 12m 33s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/b8c371c3495448699be8de7e383acf54 : SUCCESS in 1h 44m 23s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/0a45110f517c4dff96d9fc2ecd7a2e02 : SUCCESS in 21m 07s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/26d71f3aacd04658b3081a8ee569fe2b : SUCCESS in 19m 28s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/4ea99bbf200a4132b56baabbadafdd7d : SUCCESS in 31m 07s","accounts_in_message":[],"_revision_number":2},{"id":"9ca7f70423bc66f024f79063ce026e21dd12280d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-26 12:58:27.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"09b5652841a7f76b8c83647a97647832523aac43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 14:06:00.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b0e1d86fb51941be85a742cdba42583b : SUCCESS in 5m 28s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3c94a185960e46f991c86c88badfcda5 : SUCCESS in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b85d1446e56449aa1ebad4e48df2751 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65145d38c2f247658bb255d299ca8c19 : SUCCESS in 3m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f376e7bc89ed451baff0db89f4c25d3f : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/873a9fcd3b82402e9a8b638bfbf64bb1 : SUCCESS in 18m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/52a6af38729042bdae2a832310c14aa8 : SUCCESS in 3m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/608090c433114777b17ca9a02fcf353c : SUCCESS in 9m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2a9f52919a014c16a8db8639097d3497 : SUCCESS in 18m 37s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/2ccfc984f33f4105be9522fcb8d9ff08 : SUCCESS in 1h 01m 17s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/acb65ccc202544d6a47309cf22ac5ac6 : SUCCESS in 55m 30s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/2b103941daa54a9d9f5edd563cd1e43e : SUCCESS in 15m 19s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/31c2c67bf3264600998f8c313aee634c : SUCCESS in 14m 01s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/d7f036f675b94f818a87b3752567d97e : FAILURE in 52m 51s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/8faac770203f470e8f62a21b1fc7f291 : SUCCESS in 23m 32s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/4600431866ce45e9a71964e188906584 : SUCCESS in 13m 29s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/704c2a4199d34a31be9fa41b2f7b6db2 : SUCCESS in 31m 37s","accounts_in_message":[],"_revision_number":3},{"id":"4441a343c285198b703779ebc3deb9d0e24de274","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-26 15:54:05.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"af2153f4fd1f2928abad8daa71c8e183adb8a71b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 17:57:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/09e6167bedd247628cb18b801cdd8f40 : SUCCESS in 4m 30s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/693c4b713bc644cba3cca6046854818a : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c589acc7bd9455a92f9f1dd819967e9 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47ac9195d6ec454babb30f52c62b77f9 : SUCCESS in 5m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e9f540a2289a444b8c906af21852d80b : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/769fe91f27084177809b88892e11dca6 : SUCCESS in 6m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/718436ae41584a58a1d4947da94274da : SUCCESS in 4m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f093f0868a554fdca77f00fae71a805f : SUCCESS in 9m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3f6a1e2f53ed4ef78d14acf8e993b286 : SUCCESS in 16m 52s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/20357f7fb75640d1b4bfbef44448ca9e : SUCCESS in 1h 53m 04s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/a074560cf99643caaec3b9084bfc82b8 : SUCCESS in 1h 05m 27s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/8b4ce391381c4ec084068748954a4010 : SUCCESS in 20m 22s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/aac47def5e744d9898ff08f1f7ca7c24 : SUCCESS in 23m 57s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/1e67f4a2f51b48789a75a378a4a38a47 : SUCCESS in 1h 56m 53s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/dc08c1504d3549cb86fd4ef9ba32a1dc : SUCCESS in 20m 11s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/3becb6590a5c41b3b8040c42fa187f42 : SUCCESS in 19m 28s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/f2a51aaaea3c445e8d24a81c851c1bcd : SUCCESS in 41m 00s","accounts_in_message":[],"_revision_number":3},{"id":"874f9bff9542d1c086cd83ab2b32bb0c900b3fcb","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-26 18:06:02.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nCrossing fingers!","accounts_in_message":[],"_revision_number":3},{"id":"c82b8a048c93491ac971be4e86153b2e8cc3d4ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 18:13:55.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"27d785dec7a1a9f0d79fcc0c3d46b60b2f57611b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 20:11:37.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d2d47dd51c3d431380e9aa11a442594b : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4cfff5dc9dd04492a934b63126c7f7a5 : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb031eaae32f43dfb0684d863ea702f9 : SUCCESS in 3m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6c801af5cbc043c699a02ebe1dd9574d : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26daa0071e364ebda7ba4794caa5d5f8 : SUCCESS in 6m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/74029cbccf644125a7d681c455afe204 : SUCCESS in 4m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/848dfcb2c6d44091b97e707eaac5aba4 : SUCCESS in 8m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7e6cfa50d29c47ffa27fe6adb59c7ef1 : SUCCESS in 16m 40s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/c27814ffcacd47c582faeefdc8d909e6 : SUCCESS in 1h 00m 10s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/18046341a9e34139898630b59dd97f5b : SUCCESS in 58m 55s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/01472f2fefce4015b12a306b3fd82fc9 : SUCCESS in 12m 28s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/5c213e16fe984fb4bc9fd6193141deee : SUCCESS in 13m 37s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/aec4079435de48cab8cd2ccc8188125a : SUCCESS in 1h 44m 34s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/ec525debd3064b6abb94440088bacff0 : SUCCESS in 22m 33s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/f2e07b41a34d47b8b7a60d854ba1c66a : SUCCESS in 18m 17s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/9d3cb5e6182f43679c51c37c01bb2559 : SUCCESS in 38m 57s","accounts_in_message":[],"_revision_number":3},{"id":"c10868adb5879592f9c66816c90d177a21242d3f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 20:11:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"5c9a1e83538a82447494f078c7111d204c807875","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 20:13:30.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acbf7ea1959a49acbbacd0f4bc3d93e7 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9ff90dc8195dd14beb37eed0e8541ca1f3b50ff3","revisions":{"1e6dfcdd4b0605c9c4b4a5ef27353ad57a8e78e3":{"kind":"REWORK","_number":1,"created":"2020-11-25 16:29:53.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/84/764184/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/84/764184/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/764184/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/764184/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/764184/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/84/764184/1"}}},"commit":{"parents":[{"commit":"5f48f0199bce1e6bea0e67787457789b66de2e7b","subject":"Fix filtering of network names set to an empty string","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5f48f0199bce1e6bea0e67787457789b66de2e7b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-24 10:35:00.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-25 16:29:53.000000000","tz":0},"subject":"CI: add dummy1 port for seed upgrade and VM jobs","message":"CI: add dummy1 port for seed upgrade and VM jobs\n\nFor some reason this configuration was omitted or incomplete in these\njobs. Without this configuration, Kayobe development configuration adds\neth1 to the breth1 bridge if it exists, or leaves the bridge without a\nport otherwise. This can lead to network connectivity issues, if the IP\naddress of eth1 becomes inaccessible when it is added to breth1.\n\nThis change fixes the issue by ensuring that the dummy1 interface\nexists, and configuring the aio network to add it to breth1 as a port,\ninstead of eth1.\n\nChange-Id: I47115e5d8d1fb448cb9e46dc43b56c64987e428b\n(cherry picked from commit 0766df1673da0cc8c3a3d3d03e357ec5836e2d94)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1e6dfcdd4b0605c9c4b4a5ef27353ad57a8e78e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1e6dfcdd4b0605c9c4b4a5ef27353ad57a8e78e3"}]},"branch":"refs/heads/stable/train"},"8f20c51a21ab9e7c6ebc5aab55b31ab8cef390b9":{"kind":"REWORK","_number":2,"created":"2020-11-25 16:31:11.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/84/764184/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/84/764184/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/764184/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/764184/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/764184/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/84/764184/2"}}},"commit":{"parents":[{"commit":"5f48f0199bce1e6bea0e67787457789b66de2e7b","subject":"Fix filtering of network names set to an empty string","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5f48f0199bce1e6bea0e67787457789b66de2e7b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-24 10:35:00.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-25 16:31:04.000000000","tz":0},"subject":"CI: add dummy1 port for seed upgrade and VM jobs","message":"CI: add dummy1 port for seed upgrade and VM jobs\n\nFor some reason this configuration was omitted or incomplete in these\njobs. Without this configuration, Kayobe development configuration adds\neth1 to the breth1 bridge if it exists, or leaves the bridge without a\nport otherwise. This can lead to network connectivity issues, if the IP\naddress of eth1 becomes inaccessible when it is added to breth1.\n\nThis change fixes the issue by ensuring that the dummy1 interface\nexists, and configuring the aio network to add it to breth1 as a port,\ninstead of eth1.\n\nChange-Id: I47115e5d8d1fb448cb9e46dc43b56c64987e428b\n(cherry picked from commit 0766df1673da0cc8c3a3d3d03e357ec5836e2d94)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8f20c51a21ab9e7c6ebc5aab55b31ab8cef390b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8f20c51a21ab9e7c6ebc5aab55b31ab8cef390b9"}]},"branch":"refs/heads/stable/train"},"9ff90dc8195dd14beb37eed0e8541ca1f3b50ff3":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-01-26 12:58:27.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/84/764184/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/84/764184/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/764184/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/764184/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/764184/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/84/764184/3"}}},"commit":{"parents":[{"commit":"0e1e1c4448cc5d65ea81365010fb9ee934e5fff0","subject":"Fix pygments style","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0e1e1c4448cc5d65ea81365010fb9ee934e5fff0"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-24 10:35:00.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-01-26 12:58:27.000000000","tz":0},"subject":"CI: add dummy1 port for seed upgrade and VM jobs","message":"CI: add dummy1 port for seed upgrade and VM jobs\n\nFor some reason this configuration was omitted or incomplete in these\njobs. Without this configuration, Kayobe development configuration adds\neth1 to the breth1 bridge if it exists, or leaves the bridge without a\nport otherwise. This can lead to network connectivity issues, if the IP\naddress of eth1 becomes inaccessible when it is added to breth1.\n\nThis change fixes the issue by ensuring that the dummy1 interface\nexists, and configuring the aio network to add it to breth1 as a port,\ninstead of eth1.\n\nChange-Id: I47115e5d8d1fb448cb9e46dc43b56c64987e428b\n(cherry picked from commit 0766df1673da0cc8c3a3d3d03e357ec5836e2d94)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9ff90dc8195dd14beb37eed0e8541ca1f3b50ff3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9ff90dc8195dd14beb37eed0e8541ca1f3b50ff3"}]},"branch":"refs/heads/stable/train","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
