)]}'
{"id":"openstack%2Fkayobe~764362","triplet_id":"openstack%2Fkayobe~master~I1264832398744c48a6f706efe91f084890a37772","project":"openstack/kayobe","branch":"master","topic":"story/2008388","hashtags":[],"change_id":"I1264832398744c48a6f706efe91f084890a37772","subject":"Fix overcloud provisioning","status":"MERGED","created":"2020-11-26 15:45:22.000000000","updated":"2020-12-21 17:16:55.000000000","submitted":"2020-12-21 17:11:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"764362-story/2008388","meta_rev_id":"55a8234c7198dd76ef15d5e68d6a990768047f7f","_number":764362,"virtual_id_number":764362,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-21 17:11:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2020-12-21 15:08:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2020-12-21 15:08:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2020-12-21 17:16:55.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-26 17:56:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-27 14:11:00.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-12-18 12:48:22.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2020-12-21 10:37:18.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"bc9479a7bd39b8551721cbdd70b03509a4fba60c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-26 15:45:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bf6f3b7401007415bd73fed3bddfbb8c86711c74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 17:56:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/61c22555caa74f3392d7040cd267fb2a : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc662775f374407c8a433410d6b89d2b : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aef8e083790f429fb07a439344e61810 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b95e21029f74724bbf95d6277949483 : SUCCESS in 5m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b0022b4bf76245018238c98948b348a7 : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/254982da7dbd43b8801eb12bbfe7cf96 : SUCCESS in 7m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/26e1f39488064fe9b13ddef29a34f53e : SUCCESS in 4m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/805bc4ce6c784136bb6d3574c0fc66e1 : SUCCESS in 8m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f073dd26bfce4a3093fa241df9169020 : SUCCESS in 18m 01s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/680f5e57ebfd420aa65cf2c93244a510 : SUCCESS in 1h 35m 18s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/c1b397b60d6841cca73a4707f3d47068 : SUCCESS in 1h 37m 51s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/5b928bbf1ae64892b95c17724cc3b811 : SUCCESS in 14m 54s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/d2c694e83de94bbfbfca4102ba42cde8 : SUCCESS in 1h 58m 51s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/5e6ffc220f2f4060b31862210c2b1d49 : SUCCESS in 16m 02s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/973dbfdecd7e43558c748da264eb993e : SUCCESS in 23m 38s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/e476688449a7432aa2cca050f0d27249 : SUCCESS in 11m 16s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"5175921894b9aca4ded588c2c5612d40dc4b9339","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-27 12:22:55.000000000","message":"Patch Set 1: Backport-Candidate+2","accounts_in_message":[],"_revision_number":1},{"id":"fe87395bddcedfbd0c9e713074e6f7232ce2ea10","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-11-27 14:11:00.000000000","message":"Patch Set 1: Code-Review+2\n\nMakes sense","accounts_in_message":[],"_revision_number":1},{"id":"edc6005396c9ecfa5e7930a0feee6d0fc126c4dd","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-12-18 12:48:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4774eb15c18f0176045c382cb2cc2d7e2878fe6d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 12:49:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5b60ee294bfc92a5020b27f3b5439942a0744d60","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 14:42:19.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/526301bab40c46beaa4a52ccb172c9ce : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3faf2708f1e9451cbcf24e497799b7c4 : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a22a8555d79046068089b71cd63a38e7 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2de02faf4c94f72abc40bb629db6690 : SUCCESS in 23m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6206112042b14829a14427738c66c905 : SUCCESS in 6m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/05c6862ac3604e5aadcffd24878f0fdb : SUCCESS in 7m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e43d45cb9560488d9ba0a5019243bd7c : SUCCESS in 14m 50s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/4ee760c5422341d29e3d6dfb5f8c16a4 : FAILURE in 1h 47m 23s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/9c73e13df6bc48d69433314d4b00b905 : SUCCESS in 1h 12m 48s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/2de9253a07af45acb12f4135c5edb720 : SUCCESS in 21m 33s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/95cbe0f2e38749c3b147f58c8efa4c8b : SUCCESS in 1h 36m 43s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/ca949889f79b44eabbe67faf1ba63bac : SUCCESS in 20m 02s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f609b9f5ef6d40d68062b13c5aa07b80 : SUCCESS in 24m 15s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/50e614744e2f445cb7095dcaef97b63b : SUCCESS in 10m 11s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"007e22b1140a03ef9f153d58c3dbc74a9b5d7339","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-12-21 10:37:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0837df5fbc0d2142ea3e7ae2eb37891da22b403f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-21 12:49:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f7ea7ef25fd0468eada939de4c91978f : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ae7c94ae69744068a5faeec1873fa53 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/20fc9c6ffdac4dd49394110dfcba3452 : SUCCESS in 7m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a5001bb57a1640508c18b3be88060cfc : SUCCESS in 6m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/93f3d5a92e7f40a88105e2680b73c569 : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34224799a6854d8396831c5723060b8a : SUCCESS in 6m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b6e1380d62b8475e985059bc7b60a80b : SUCCESS in 4m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1e0415e7b81d4f9599e125844d9da1ea : SUCCESS in 7m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/de9618d8408f4cdfb5a15a9c383d43fc : SUCCESS in 17m 08s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/c41d7c9a04f14f4ba5e648328da565e3 : TIMED_OUT in 2h 03m 14s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/f799ad00645c427d954d2d9abec48549 : SUCCESS in 1h 08m 37s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/f7b34c040ec44d80b9614cdd4aa27d94 : SUCCESS in 16m 56s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e74987b5e4a64afeb11cda40b8a27091 : SUCCESS in 1h 37m 06s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/c97116932850469e90be606960c4f735 : SUCCESS in 15m 25s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e0ca95e973dd490b8e152b44454b4f12 : SUCCESS in 22m 42s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/b90612b54d964130b62e89f5adff1459 : SUCCESS in 10m 20s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":2},{"id":"e26461ace4675126ee65e8458065d97184ee7fbb","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-12-21 13:13:54.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"064d95550c1735ff530ce21a59fa8d398e2f3edc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-21 15:03:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9857ddd126b740018a0eb3aeae192e4e : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebe7c9ff390c4f308dd4704acc20469f : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/46e14b7543504c34b118ceb910a320ce : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/89e12470b3d2460fbf69d57f0578345e : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e5f6576c058743398eb6678a75d88f3f : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70d9ec415bed4293884f6d198423bf4a : SUCCESS in 7m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ee831ac41ebe49409c89e84a88ad1782 : SUCCESS in 4m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/372d7fd1c59b46019c403fb06ddd321c : SUCCESS in 9m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/52bf2e7c25b04e3e96665c9c10f4b2df : SUCCESS in 16m 38s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/f1ae7792510d43ee9904e14c5f5386eb : SUCCESS in 59m 31s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/42f060efca704453b9ef7e4dbedf790e : SUCCESS in 1h 15m 53s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/bdf3f65be5354af8976d4b1ccf67b57b : SUCCESS in 16m 58s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/6e7be4034b16451191b0627fc9681e76 : SUCCESS in 1h 36m 37s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/9fd2bc43253a493c830d50138d8d8bd8 : SUCCESS in 13m 54s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/5f7fcd402cd44547a72906aae210cb54 : SUCCESS in 25m 29s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/d6367d1e77f14383810ccb7dd3d5a4ee : SUCCESS in 11m 29s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":2},{"id":"03811a81f2ecb38b8bc28b5d66dbe213c6cd8056","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-21 15:08:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nPreviously approved","accounts_in_message":[],"_revision_number":2},{"id":"1b619c3c7bc80666393c1fa884584daa5428bb81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-21 15:08:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"254e2c6750004c02256ac015241cd20dc27e9735","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-21 17:11:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0af7327e8436421bb3656a6521dd1be1 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/187b2c9b0c4e4d608ba8c54a5b1c4ef6 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b33614050444fd495801c40f8adfe8c : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6737ec1f62844076a77209163140d657 : SUCCESS in 6m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6220a873986e42dcae19dfc5251860ba : SUCCESS in 4m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ae6a44b240f24b19bf34fa9113493e88 : SUCCESS in 8m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1b71bdb4868c42bba532df48c07d01b5 : SUCCESS in 18m 26s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/20d88d2308a8457a92fb1bd704c338ee : SUCCESS in 1h 04m 17s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/1887dca85829466891cdd8dba1cf5856 : SUCCESS in 59m 55s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/17155ecf68c842ddbf8fe052a4b92c2e : SUCCESS in 15m 45s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/7235759ec5014b519e7d66a77d8bc5da : SUCCESS in 1h 56m 14s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/add526b2d5d1423ab608412623288703 : SUCCESS in 15m 44s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/7f847236a8a847328213900d75de52ef : SUCCESS in 23m 17s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/5d2c205627b54cd2901e5e30a010ddec : SUCCESS in 10m 52s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":2},{"id":"c574338d77c2a46eb0d679787e9d8f0c4b183d53","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-21 17:11:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"6eeb2c4c7d8be8339b8d3c479b0799711f61e89e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-21 17:16:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8ca49d80c85468d96e41d30ba38e48a : SUCCESS in 50s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1ad073f0d0b181fc3eaf9b3523c29d96302a356a","revisions":{"1fdd64d38e15c460c0254acfe7bbabf8d5cf5f7a":{"kind":"REWORK","_number":1,"created":"2020-11-26 15:45:22.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/62/764362/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/62/764362/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/764362/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/764362/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/764362/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/62/764362/1"}}},"commit":{"parents":[{"commit":"0766df1673da0cc8c3a3d3d03e357ec5836e2d94","subject":"CI: add dummy1 port for seed upgrade and VM jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0766df1673da0cc8c3a3d3d03e357ec5836e2d94"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-26 15:41:37.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-26 15:45:08.000000000","tz":0},"subject":"Fix overcloud provisioning","message":"Fix overcloud provisioning\n\nAffects master \u0026 Victoria only.\n\nSince Victoria, Bifrost added a play to gather facts for localhost at\nthe beginning of deploy-dynamic.yaml. Kayobe executes the playbook with\na limit, which does not include localhost. This causes fact gatehring to\nbe skipped for localhost:\n\n    PLAY [Gather facts on localhost] ***********************************\n      skipping: no hosts matched\n\nThis change fixes the issue by adding localhost to the limit.\n\nChange-Id: I1264832398744c48a6f706efe91f084890a37772\nStory: 2008388\nTask: 41311\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1fdd64d38e15c460c0254acfe7bbabf8d5cf5f7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1fdd64d38e15c460c0254acfe7bbabf8d5cf5f7a"}]},"branch":"refs/heads/master"},"1ad073f0d0b181fc3eaf9b3523c29d96302a356a":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2020-12-21 10:37:18.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/62/764362/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/62/764362/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/764362/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/764362/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/764362/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/62/764362/2"}}},"commit":{"parents":[{"commit":"e569036a05d06840148886cafd13a96b78e1df09","subject":"Merge \"Remove retired Qinling support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e569036a05d06840148886cafd13a96b78e1df09"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-26 15:41:37.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-12-21 10:36:51.000000000","tz":60},"subject":"Fix overcloud provisioning","message":"Fix overcloud provisioning\n\nAffects master \u0026 Victoria only.\n\nSince Victoria, Bifrost added a play to gather facts for localhost at\nthe beginning of deploy-dynamic.yaml. Kayobe executes the playbook with\na limit, which does not include localhost. This causes fact gathering to\nbe skipped for localhost:\n\n    PLAY [Gather facts on localhost] ***********************************\n      skipping: no hosts matched\n\nThis change fixes the issue by adding localhost to the limit.\n\nChange-Id: I1264832398744c48a6f706efe91f084890a37772\nStory: 2008388\nTask: 41311\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1ad073f0d0b181fc3eaf9b3523c29d96302a356a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1ad073f0d0b181fc3eaf9b3523c29d96302a356a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
