)]}'
{"id":"openstack%2Fkayobe~764510","triplet_id":"openstack%2Fkayobe~stable%2Fussuri~Ibe44583cb93b7dca8f5091c893386a15288af915","project":"openstack/kayobe","branch":"stable/ussuri","topic":"enforce_list-stable/ussuri","hashtags":[],"change_id":"Ibe44583cb93b7dca8f5091c893386a15288af915","subject":"Fix Python setup when venv is not used","status":"MERGED","created":"2020-11-30 09:46:19.000000000","updated":"2021-01-15 20:31:10.000000000","submitted":"2021-01-15 20:26:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"764510-enforce_list-stable/ussuri","cherry_pick_of_change":763659,"cherry_pick_of_patch_set":2,"meta_rev_id":"221043ac9cf240224242e8fc39741d5da5e46623","_number":764510,"virtual_id_number":764510,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-15 20:26:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"all":[{"value":1,"date":"2021-01-15 20:31:10.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":2,"date":"2021-01-13 14:02:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-13 14:06:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-13 14:06:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-30 09:46:19.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"state":"REVIEWER"},{"updated":"2020-11-30 12:15:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-13 14:06:03.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"6ec0fa8bbc5348fe73af29eb08af3d834a8e3f94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-30 09:46:19.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"4ec64bca4888f4744aed3fa2f2dfd8a31b615758","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2020-11-30 10:40:42.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"92e5a44a5480299b44e867b28527881c62a983dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 12:15:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d9baa1f01f17464aa654a4dc7a10030b : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e19540c86a44ed8815f51d1aba09d84 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce8da42a2a354d2abfcd05a3043c4b93 : SUCCESS in 4m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/16438a6e270b4f25a4827c96dccc3329 : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6041c22eac8046a699f9239f57120375 : SUCCESS in 5m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e7c458fe160745429ccae335ce2d7e84 : SUCCESS in 3m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/eb3f4feb036c4ed78ddcfe6b4c7d1098 : SUCCESS in 3m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4cf537376aa34c229f8f61a94d5a9b33 : SUCCESS in 8m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2ca2339192774dceac05fde99291b3d6 : SUCCESS in 19m 16s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/0a5262f68f9f4e1197743bf8ad0d48b5 : SUCCESS in 1h 02m 24s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/705bbaac798c46bf8f733e1a4ed5de26 : SUCCESS in 14m 31s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a1cebb540567484f9e76561db961be55 : SUCCESS in 2h 18m 35s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/4d56e241978740c88b95e94bab3cb7c3 : SUCCESS in 22m 49s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/7e34eacb799f4c369260242e31d0295b : FAILURE in 12m 19s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/dac83356003a418a901fc8089f5de8f2 : SUCCESS in 12m 26s","accounts_in_message":[],"_revision_number":1},{"id":"13046b0b597e701e5ed545386977f097191f9dd4","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2020-12-01 14:59:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5d0663d919672e2ee38438fa3499a600d90ad6ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 16:28:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc7990715d354783bfaae478cf58a8d4 : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63cb627832914124be2e2944f2c16945 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c8bf139e9210481a9691ae8220ec10ff : SUCCESS in 4m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0c6b8e0083924969a78065e4abb777d5 : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7dca4b9ad5464b3fb7172aad6de32351 : SUCCESS in 7m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f8ca98b505a49e690ed108ead9585ab : SUCCESS in 4m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3856c6669401464a96df94a57cd1bfdc : SUCCESS in 5m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/54a8475fa1a44b17ad4d36a14bce5f61 : SUCCESS in 9m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/502518bab4f4489abe59ef51a7936304 : SUCCESS in 20m 05s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/83443fad26d44d7194798f966c6a9964 : SUCCESS in 1h 04m 12s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/fb65d491c3a4477eb70505c4cea5b1b9 : SUCCESS in 18m 24s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/4632c407db37463b93aec8f320bb36c5 : FAILURE in 56m 04s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/27cf53482f5842baa1699b7b6f7233c7 : SUCCESS in 18m 09s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b8a7f00123e548189cc7792f39e63498 : SUCCESS in 27m 34s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/41cf224097b845a48fb578f66b6aa32f : SUCCESS in 11m 36s","accounts_in_message":[],"_revision_number":1},{"id":"21d712bb613f293a3d50f3f3eaeeba7122197041","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-05 11:05:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8fbdb666d142ac81cfa8fed4e7ae85fa664faaaa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 12:43:54.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f4af11e2b2e4d68b2cace4c68cdb3a5 : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5003b3f787fd4c4384ac7f649743aa57 : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c835c43ee6340b2b2e8b9ffd00a05ca : SUCCESS in 5m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/afb239824af14b58a13b67dba4d7edf7 : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef7984c0c09449e9a1b6dad5ffa452d9 : SUCCESS in 9m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0713e43e4a6f4faeb926e3f144350b0c : SUCCESS in 3m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fd670020630c4dad89b8514c95b3b0e7 : SUCCESS in 4m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f925e06214264e50beaa73cd15f82f7c : SUCCESS in 10m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0c8beb00d0c34fb79b5e8fdb542156ac : SUCCESS in 18m 25s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/71db630512b74ccfb855ae3ae3f88788 : SUCCESS in 1h 00m 02s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/e383d66b143842fe9e3dc9b3426a129c : SUCCESS in 15m 25s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/ab845f9dfd604b428990a903e96dedfd : SUCCESS in 1h 37m 01s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/7395f38c740443229d10d318ec2b6a35 : FAILURE in 17m 57s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/88a67f51125d46bab4d68784d8f2809e : FAILURE in 26m 35s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/85ddff1b798148eda220e43eafc67a9b : SUCCESS in 10m 18s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"578a020cf06826ef8c13cf2a76a11b3ad83a8001","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-06 09:13:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"997bfb347e535dd4d513a54769dfdc5105ea2f2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-06 10:53:14.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35da9fc1d3b3471ebd29f1a51e17845f : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09d6843ee9db4badb179c88104918465 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f6e4ef6defa94131985415201646e29b : SUCCESS in 5m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b86a54d5e9a6466282dbf0880d5b5dba : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c9a78cfa99f42088d46f1fb6b1525b3 : SUCCESS in 17m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c83f2f624a84efcbe2f7647bd7a8942 : SUCCESS in 6m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3c52b0a29af44de5847d8bd86df88c62 : SUCCESS in 10m 00s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ba11073be6894712849961a52c98bcf3 : SUCCESS in 14m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4513db6b49834696837d1b6eea086853 : SUCCESS in 23m 24s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/058353c34a3340568801b5580106021b : SUCCESS in 1h 11m 09s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/0295d4ffc48341c893b795dd72f71285 : SUCCESS in 17m 21s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/6cb45c647368401ca87292aa12b9ff2d : SUCCESS in 1h 35m 55s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/3abf88cadb544e3d8fe5e3ac120ffcb2 : FAILURE in 19m 08s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/78a7017bcc3e4955b29dfd266d193d89 : FAILURE in 20m 43s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/f5927127db2f4402a2309fad8f216877 : SUCCESS in 17m 53s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"0aa34673062e8ac3e813ecd51fe89941017af1e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-13 11:35:08.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"637862207f6bda65a16c7e65460f2ffc01b428b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 14:00:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7bdffe8cc3bb46ea8e5f5107d804a380 : SUCCESS in 9m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad2f6caf88784e2ea6091ed5a127143f : SUCCESS in 8m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a6abab0061b6434dac8cfbebfc3ad41a : SUCCESS in 5m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d5972183e502458295354303d2e14c95 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5c04a51ba22477dbf2c40d08c4138c5 : SUCCESS in 5m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2620a3a7fcd242fb9167748cece60166 : SUCCESS in 3m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/abeea19c77fa425d9d4615999fafeeb8 : SUCCESS in 5m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cb6d2c1cbafb4de3ad3fe5d7475034fc : SUCCESS in 9m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/43c836beb2474bbd9f116e195f434f02 : SUCCESS in 17m 50s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/a9b8edac1ab3420ea7273c52c0c88967 : SUCCESS in 1h 09m 08s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/127abb8c09264b0d9c5bb9580ef18360 : SUCCESS in 16m 02s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/67b3ad1311424e81b520df39a284b9ad : SUCCESS in 2h 13m 13s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/b98cdc25703b4609afdd7b4629a4d434 : SUCCESS in 20m 23s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/86c51396554f4ff4840fbb1776951100 : SUCCESS in 29m 14s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/5fb49c88644a4499902dab02617eaa63 : SUCCESS in 15m 25s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":2},{"id":"e3fada8c38a475a2023483cd11ef4d962f0f4de1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-13 14:02:55.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"64148408b91c621ca2f55295bc9547fc8be584e3","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-13 14:06:03.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"03e08a84c4ccbb53d62e3eb9406ff67da4f41cac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 14:06:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b388fe6e3686f91ff335a3c202bdb928aabd9387","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 15:56:16.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/709b67e59e34447b929e65013168ca2a : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58d3c823b67f451383d777b7928b7315 : SUCCESS in 4m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/062c7b930c7840cd8251012cebffcdb8 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba5165acd71443a8b2bc96bdb43296a7 : SUCCESS in 19m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/142051fdec76481ab13fe895c9adfa35 : SUCCESS in 3m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3b61227fbba740afa04f3fc4ba882481 : SUCCESS in 4m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bef84659a84e4617908a6a601be017a9 : SUCCESS in 10m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bc6229374b33427cae1f06ea7c175ad5 : SUCCESS in 25m 48s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/559ce2327b9747a3afeba0b77c520b51 : SUCCESS in 1h 06m 43s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/cd2bbc0aeb4840409d2c3a82532bdc45 : SUCCESS in 23m 16s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/5a7b0f65742843d999c6c4d418bcb864 : SUCCESS in 1h 43m 22s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/6c7d0df8fdea44b0a951280bde1a8c9a : SUCCESS in 15m 54s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/bb5e23cc775d4abc8556909ab665ebd3 : SUCCESS in 32m 36s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/ff63601c0d8347b797ea5d69faca03a2 : FAILURE in 15m 24s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":2},{"id":"be81b234d52e906190fa6165030c638faac5bcc0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-13 17:00:50.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8437edd022797f86f7359dfb8668e1ff817391d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 19:39:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/45e2daaa49fe465bb77feb87cecc28ab : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7acc2c4d4384345aca778db1fcc10aa : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a834c24a8d094c4ba01979777d387915 : SUCCESS in 5m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4ca5c6d959354b199d8ed03c75b1be9e : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94acf58cc7c54bd68b9624746e59d672 : SUCCESS in 6m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52a0139e108a4f67b032a2234e7b6464 : SUCCESS in 3m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6c7f4855a3f54d0badacadbe48d81b38 : SUCCESS in 4m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b3c9b32f1996400d8b020ca168077ed6 : SUCCESS in 7m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5a05921293b246d58dd04b32ce805351 : SUCCESS in 20m 01s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/d5249335684546918a2026afce58b98b : TIMED_OUT in 2h 01m 55s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/eb86e6aa837e4218b12fa976da2891e8 : SUCCESS in 14m 40s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/700f6716a04e4dc6bfa70524faf3ef38 : TIMED_OUT in 2h 31m 14s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/ff3799ea085444078d6356161635cce0 : SUCCESS in 54m 16s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/5a23159c4f5b4b6da762b9650b181786 : SUCCESS in 1h 14m 46s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/7a18c43954c34121a523bc016532a2c2 : SUCCESS in 24m 13s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":2},{"id":"7de37cc66e01e2f0ffa6224ca29403f7945262ff","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-14 08:45:55.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"96eb4d91f05b6944210517415593051138b5200d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 10:54:19.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/07fa5e1f01fd468a8a24b5b103a41567 : SUCCESS in 12m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79ea395baf98499280def636511797a2 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1cf0811d968448cb1e5f23602e3ee7a : SUCCESS in 5m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/652b885d49084a2cb0dcf468dcf544d0 : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1f3aee67c6b4291b30c259f126290a2 : SUCCESS in 7m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b634f3093f294c669ef697b4df6f5c6b : SUCCESS in 4m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b44268a50f554f7b80ef7cc39e8d8ecf : SUCCESS in 5m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1ac3fd68e86a4f26b028884c94eff5ab : SUCCESS in 8m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b9be29dd4d3d48d092fa3ba8ac9333b7 : SUCCESS in 18m 56s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/303e8ccb7c29417c93afa32f5fb12d1b : FAILURE in 1h 49m 52s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/840c99821cd64ccd9971556dc156b17f : SUCCESS in 23m 12s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/121f144f99584bc89297e6792b106bae : SUCCESS in 1h 54m 44s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/3f1ea744665e4f4d8ff998d4e7f7c584 : SUCCESS in 16m 29s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/1b56319c787045abb82085c0e96215b3 : SUCCESS in 27m 28s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/6bb28ec9703141038a9f6b2c08feefe7 : SUCCESS in 12m 02s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":2},{"id":"9ea6fbd86ba310655c5a1a56eeeab9430ee121a9","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-15 11:53:26.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"b7fb4d0fa77c8c4642ba6942cd4482581c284fa2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 14:10:01.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f94c4f862481497ca7507f436779be8e : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd2d2dc1ffd14f75a436ffaabff41ec1 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d696e33bff094ab1bdb2fb3f6c189108 : SUCCESS in 7m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3ab33fbf64a14747a1b5b2139dc0581d : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83e0351e69364e4dbd9336ee624767a2 : SUCCESS in 6m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4260c3d8f58c494d9c6912346a99e4a7 : SUCCESS in 4m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6b2832ebc4b140d9808c410b8f5ec378 : SUCCESS in 6m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/16ea4a7f44404f3ebcb6e7e0028f7713 : SUCCESS in 8m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1a090abe162e4912842d07563d61f971 : SUCCESS in 17m 55s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/e67202895faf412f93a51b552565c24b : TIMED_OUT in 2h 02m 31s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/8eca90fa15154976bb295230957a73c1 : SUCCESS in 20m 21s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/ca0acd9c197e4d9ba9bfde759036c3e9 : SUCCESS in 1h 38m 58s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/62bc40b0114340ce8503657a415493f4 : SUCCESS in 16m 45s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/bcf78e9f396a43a9a9be1800e81a21e4 : SUCCESS in 27m 59s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/0cceb48b6cf6463ebf66ddb114b150cd : SUCCESS in 9m 31s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"ae2f4836728a88914a07a85cec3c1041866994be","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-15 16:45:43.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a76723f5a25f97a0be30cbaa4b68b575b44933fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 18:36:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4576fec85d784695a8a1bd6ce0acf2a5 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c212cc5ea8ab4ec393af70007a3a4933 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c7ce76d1521f44da938efa19c4707bd2 : SUCCESS in 4m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2b95f499744544b9b512266f26a0fac2 : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/428aecc93fb2454ba85d07f35c4d0222 : SUCCESS in 6m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f648ae53fcf7463d88093d47d52686cf : SUCCESS in 3m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/747d6c45181c42ff9f595332817c76e7 : SUCCESS in 5m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0032df07030e41d682f31b1ac9c610d4 : SUCCESS in 8m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d3acc11f8a294da489a7efc39ccb235b : SUCCESS in 18m 03s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/97dc18ff980240eea813936194bda894 : SUCCESS in 58m 50s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/09b681e75497460dba3b8de2c8d39d99 : SUCCESS in 16m 46s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/855e4d96e209450384aeb76a1e78caaa : SUCCESS in 1h 47m 34s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/777dec910523414a8a861a5c0160a8bd : SUCCESS in 18m 31s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/ad8c607b78584a5e9c659133dffce667 : SUCCESS in 25m 14s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/4a16bb549e72426bb0c263288a97bcdb : SUCCESS in 12m 32s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"1631fcdd27089429c5df431ff359127726f747e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 18:36:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f72d070b79e1059cf95a31af1643ae84c3598920","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 20:26:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5a4a99f4d2a47bd8357b3560355ba94 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc4c1c492cc847eb880b02645474a1c8 : SUCCESS in 5m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/edcce13e52c24bb2810a8ea2f7fb33d6 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a87938b136094753bc25d25b5b40dc9e : SUCCESS in 8m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/95a40d7373564ef2b94089aa4c55b4ee : SUCCESS in 4m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ca9018640339451391e875473ac1d95e : SUCCESS in 5m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c88602c9fdb84d5da43223d6c6ec1f33 : SUCCESS in 8m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5e03d440803e4f6cb02592092dfb1ecc : SUCCESS in 14m 12s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/92bac23bab6f4717b03c5235382b1714 : SUCCESS in 1h 03m 51s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/21294b2f615c493fb9dc8183ada3ec61 : SUCCESS in 16m 42s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e91f56e6634149589a34757f4bbbed46 : SUCCESS in 1h 47m 53s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/bd580c2445864df390a4974ecf7ca792 : SUCCESS in 17m 58s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/d426e0adcc5048f280716d921dc29122 : SUCCESS in 28m 45s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/301bbcfaf89448d6b63cb42501d476c3 : SUCCESS in 8m 59s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":2},{"id":"dbeb57605cc4cd9e15161ffd4917fba6e60afb23","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 20:26:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"d4003ad970092074b94a403971946b965f248f2e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 20:31:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02ebeb4a061b43d190592b33ad02506c : SUCCESS in 1m 28s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bcaa8ccea97442f6bcf6faa52b10be15 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6ee7333e7e6ebe6862e84df7fb69ddde431b0d4a","revisions":{"5aded02f162fdc076d4fab264e1d8e13ee1f22b0":{"kind":"REWORK","_number":1,"created":"2020-11-30 09:46:19.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/10/764510/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/10/764510/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/764510/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/764510/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/764510/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/10/764510/1"}}},"commit":{"parents":[{"commit":"a8d5cc0ba9caf4ab348b16873c36e065ed244d68","subject":"Merge \"docs: Add link back to Train CentOS 8 migration page\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a8d5cc0ba9caf4ab348b16873c36e065ed244d68"}]}],"author":{"name":"Piotr Parczewski","email":"piotr.parczewski@outlook.com","date":"2020-11-20 12:08:49.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-30 09:46:19.000000000","tz":0},"subject":"Fix Python setup when venv is not used","message":"Fix Python setup when venv is not used\n\nSet packages argument explicitly as a list to avoid \"No package matching\n\u0027\u003cgenerator object select_or_reject at 0x7ff25f341750\u003e\u0027 is available\" error.\n\nStory: 2008378\nTask: 41298\nChange-Id: Ibe44583cb93b7dca8f5091c893386a15288af915\n(cherry picked from commit e748faea00346aeb1a05a1810a386b1f3d21fa75)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5aded02f162fdc076d4fab264e1d8e13ee1f22b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5aded02f162fdc076d4fab264e1d8e13ee1f22b0"}]},"branch":"refs/heads/stable/ussuri"},"6ee7333e7e6ebe6862e84df7fb69ddde431b0d4a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-01-13 11:35:08.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/10/764510/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/10/764510/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/764510/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/764510/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/764510/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/10/764510/2"}}},"commit":{"parents":[{"commit":"ea073b8cb4960356cd3318f89f0ea7476b21aa03","subject":"Merge \"Fix implementation of kayobe seed hypervisor host package update\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ea073b8cb4960356cd3318f89f0ea7476b21aa03"}]}],"author":{"name":"Piotr Parczewski","email":"piotr.parczewski@outlook.com","date":"2020-11-20 12:08:49.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-01-13 11:35:08.000000000","tz":0},"subject":"Fix Python setup when venv is not used","message":"Fix Python setup when venv is not used\n\nSet packages argument explicitly as a list to avoid \"No package matching\n\u0027\u003cgenerator object select_or_reject at 0x7ff25f341750\u003e\u0027 is available\" error.\n\nStory: 2008378\nTask: 41298\nChange-Id: Ibe44583cb93b7dca8f5091c893386a15288af915\n(cherry picked from commit e748faea00346aeb1a05a1810a386b1f3d21fa75)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6ee7333e7e6ebe6862e84df7fb69ddde431b0d4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6ee7333e7e6ebe6862e84df7fb69ddde431b0d4a"}]},"branch":"refs/heads/stable/ussuri","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
