)]}'
{"id":"openstack%2Fkayobe~764511","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~Ibe44583cb93b7dca8f5091c893386a15288af915","project":"openstack/kayobe","branch":"stable/train","topic":"enforce_list-stable/train","hashtags":[],"change_id":"Ibe44583cb93b7dca8f5091c893386a15288af915","subject":"Fix Python setup when venv is not used","status":"MERGED","created":"2020-11-30 09:48:06.000000000","updated":"2021-01-27 12:49:26.000000000","submitted":"2021-01-27 12:47:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"764511-enforce_list-stable/train","cherry_pick_of_change":763659,"cherry_pick_of_patch_set":2,"meta_rev_id":"ad379ab621e2481642d06dc38d6852a9034f615b","_number":764511,"virtual_id_number":764511,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-27 12:47:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":2,"date":"2021-01-27 12:49:26.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-27 12:49:26.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":2,"date":"2021-01-27 12:49:26.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":1,"date":"2021-01-27 11:01:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-30 09:48:06.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"state":"REVIEWER"},{"updated":"2020-11-30 11:59:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-24 06:09:05.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2021-01-13 14:06:46.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"f322d35f6370cc5f1ec3a5b4a0e8123ca21b8576","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-30 09:48:06.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"df03bb39755ab62979011149cbe0c16906a42813","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2020-11-30 10:43:36.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e7b19024afd08601c88483a4e73704fe7bd7ab17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 11:59:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c73bbe3ae3da47f099ea904500d61446 : SUCCESS in 5m 21s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/74a8bcfc453c473c9a6ea2bd622626a3 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/32535fac754c44afad2982fb67d5a818 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a70f5c2806bb4ba5a4036bdcc08bf958 : SUCCESS in 3m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6acafccc786b4d928d9644f68750587e : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/882183d65175437a8dfc37502d48abdc : SUCCESS in 5m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/13f8225f15cc40a1992f7005bab1e9d7 : SUCCESS in 4m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ce72166ed4774646884ef19a82100ca2 : SUCCESS in 4m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0a35622f659845f4bab20024fd38533c : SUCCESS in 7m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5ecddf77c73e45dabbca2d7bc2965c3d : SUCCESS in 17m 36s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/7a24f6700e774175a90a5a9fd75aabb1 : SUCCESS in 1h 01m 34s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/1415f14a4b6345289366c94538e2c378 : SUCCESS in 1h 11m 24s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/96f2a9ba692544c2988a55314bce670e : SUCCESS in 17m 11s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/9f5b537ecb144249960edb9f046fcb62 : SUCCESS in 19m 01s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/2b8e64565f8d4c36b7e9de77bffb798a : SUCCESS in 1h 55m 48s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/a499207d3b7642fb973c0f63743ebf40 : FAILURE in 9m 49s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/3278da13e6ef4b6d9a7de83794b9f402 : SUCCESS in 16m 12s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/39ca775400c846a9a1a6f0c860b4ba18 : SUCCESS in 42m 17s","accounts_in_message":[],"_revision_number":1},{"id":"8f99332c901d1e36bfcaf68c8dcc06942369d71d","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2020-11-30 16:38:41.000000000","message":"Patch Set 1:\n\nrecehck","accounts_in_message":[],"_revision_number":1},{"id":"4b95962c356904b25e207fbfd382b5597438b313","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2020-12-01 08:06:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"966a7af6234392689e0f648306d8e7baccb83746","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 10:09:52.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d6af1766d5fc43b084ae25c259a6d7c7 : SUCCESS in 3m 42s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/993fa2c086a34b31a347b1b2e925d6d5 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/90a31c03a1b14ea29ce427911ed674cf : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bec8ada759c44121905c7e953ef0ee85 : SUCCESS in 3m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1a06ec6a062042938e4ea848673419f5 : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c68d188159c4e379fac850a3e95093d : SUCCESS in 5m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bbfb785ed36e49ad897f68077e91678d : SUCCESS in 3m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c1406a647ba644e5b426c5c0f4ed4d18 : SUCCESS in 4m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bdcf982e73d44a6086d5b724f9e26b78 : SUCCESS in 8m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4db45b3973d9473e8d6a805f4f1bf071 : SUCCESS in 19m 17s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/544e98ac85374961a92d43dda745786b : SUCCESS in 1h 11m 15s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/be37b509afe04b74a162920f95b20ba6 : FAILURE in 1h 03m 47s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/2c8c00aa747e45e3afdafded6e7aaa3e : SUCCESS in 15m 26s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/1792af0a437c4907b01182f4fa75c1ba : SUCCESS in 16m 10s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/a8b8490cba164b16b4b00a7ccc5a498d : SUCCESS in 1h 58m 10s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/755f52b7c5674a7a81a910e64fef45d9 : SUCCESS in 24m 10s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/0bd950ce0d6e4090a5dd181fc06d7100 : SUCCESS in 17m 06s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/b790096c69554e848c3d3c0e7a701985 : SUCCESS in 40m 22s","accounts_in_message":[],"_revision_number":1},{"id":"85e370cd99461b628a18981f3fdd71306fdff0c4","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2020-12-01 19:08:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f29f8d196d20e45a812aa27bc88cd5bfd399036c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 21:16:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9bb5bbf2f41d47bd886c285395977b7e : SUCCESS in 3m 48s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/56609c7266814b31a8e4d13734129b74 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f965f6bde5c425c999d387bda0acc75 : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/514361849ffa488ab35d69e1d8ed4105 : SUCCESS in 4m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8d3ab82eac0848759a74d7147f0702bc : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c29690f1ca0b4c43b4fedc65beb1a1d7 : SUCCESS in 5m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06da2f57f54141cc8fd8fc70a134d073 : SUCCESS in 3m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2b21100b7d234ee4b8713bb2084a29ee : SUCCESS in 4m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0c01d56d6af34d1899aa01a920ced5a0 : SUCCESS in 10m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6e062993c65b4b90b30b521653f53412 : SUCCESS in 19m 15s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/e0a9e6e0fca04000a1d45839048bffc3 : SUCCESS in 1h 03m 22s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/c6aa5e6388664353a030362363ea1dec : FAILURE in 55m 41s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/2792de882e5e4a7a9004d061a673489d : SUCCESS in 13m 47s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/8d6b2b77f2cc4325ab7415aea2f70bd3 : SUCCESS in 15m 51s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/429b24f5237f422b8933e7b1afbcd662 : SUCCESS in 1h 57m 41s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/4f308f654904481aa640e9d08e1b164a : SUCCESS in 45m 00s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/a7f8465d981040b08aa57190502e5c69 : SUCCESS in 37m 16s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/12afbf6f3d234a98adae2c8546ccfa92 : FAILURE in 56m 18s","accounts_in_message":[],"_revision_number":1},{"id":"de1a694e121a703b89fb0056ea0e1964a73a8cd2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-22 13:16:01.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"18f8a012b340d957652045198f991694c4b76a46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 14:29:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b32a6fe37e384b8dab95574eebed3acb : SUCCESS in 3m 58s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/cfc58097db6b425abd960f4ade879f47 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/937f6b021528454a9ebc480d6b714c25 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fc1e5a951244ff1b2c1277af3f4b018 : SUCCESS in 3m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d606b594f8fa4354a4563dcd828f5e08 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1407746e157348839de349376b1170bf : SUCCESS in 5m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e8e212061484cbbb731ea1d81f65cee : SUCCESS in 3m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7189b3396f034927a64c28df67e35df6 : SUCCESS in 4m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8b80a614cfae4495b3175e6a9b5723ce : SUCCESS in 8m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/929e828232f8483e94fddef4e20ad251 : SUCCESS in 16m 07s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/803786496fe844468d7ed254bb38e255 : SUCCESS in 1h 08m 12s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/a6964113c11e4cf586d3e64bebc706b0 : SUCCESS in 1h 04m 03s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/425ffd29290f41fd8322eccde64342ad : SUCCESS in 14m 12s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/ee8d91cd81c7419d8f9599ca8aa75e7d : SUCCESS in 14m 11s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/ce1f9db68af4402898f76bc47199f521 : FAILURE in 53m 14s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/557f12d732fb41a3a2dc7292562bbd21 : SUCCESS in 19m 31s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/5d58cd25d5f646fabe39a9c01c821af1 : SUCCESS in 16m 40s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/f1f7ae32a0de440d86a70abebb02fbbe : SUCCESS in 42m 36s","accounts_in_message":[],"_revision_number":2},{"id":"37796d6e4c1f6f4fbe70858c8ac8f188e90da05e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-22 17:22:21.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"217033b405157a10c4a18141213e74dd82527fcd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 19:16:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9b6f7a9f1ed4764b99db4e71cacc77b : SUCCESS in 3m 47s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c6b99c9dd9e941ee991cd7a407add791 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6af003e89cd64997b1308d5aaf11615e : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca3135cc132e40f2906e65fec6f2fbe9 : SUCCESS in 3m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0684de991cc74d08afef384a8310519e : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa8d0a728277427fb0791eeba016cfbb : SUCCESS in 5m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e29541470e143d5ba028a1178ef05fb : SUCCESS in 3m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/10484ae0a0fb4babab199981ff53e096 : SUCCESS in 3m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a068e10483194cd5b4ac2b8a48eee1e9 : SUCCESS in 8m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/284f897a63034b999afaf64a49f8cd67 : SUCCESS in 17m 22s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/12e9cbe66016474090faf0992704e745 : SUCCESS in 1h 01m 32s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/efcdfdde41554886ad6004dfa38c36bc : SUCCESS in 1h 02m 06s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/466bb06030094e8fa396ccbdbd125da3 : SUCCESS in 15m 03s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/99425fbbf990422a9243a8f4079ae4e8 : SUCCESS in 14m 14s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/5eeb0b9d929d4bc4bd7873690099ab53 : SUCCESS in 1h 49m 50s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/3bb1176cca344c4bb4b08cc05dc590d6 : SUCCESS in 22m 39s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/4ea5a84ab78d4acb823bd62f04e67f9e : SUCCESS in 15m 46s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/f487f908763c4295a83e4ae15134985c : SUCCESS in 35m 08s","accounts_in_message":[],"_revision_number":2},{"id":"0d99614c9b203a5088ad2c99fc2f2a936e2d60ce","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-23 10:08:02.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c189ee7b1cd92d658d4529edf0279e598254cd68","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-12-24 06:09:05.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"4616aa458a5d26cda937d29089d6ad6652328471","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-13 14:06:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7cef73eb0c7042db713c026056316e52b14d0272","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 14:07:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e003bd29d2d150d41dc55ff4e3dd176166be2da8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 15:55:39.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/682ca9b62b774c40a069c022f6408d02 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d807c128b4564b78934501ffd8544b03 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4973697365194226b1fbe7fc60c9a2f4 : SUCCESS in 5m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/abaf6cea231442c393b5ee0b28446a31 : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d72d37b3cf746978860068bf43b4d29 : SUCCESS in 6m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2948a7ffb9794e0386cd0a4d6499d6b2 : SUCCESS in 3m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/871758a694484290ad31a2841569cdd2 : SUCCESS in 3m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e78478c50a724456b489765c6304e510 : SUCCESS in 8m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/03729ca6f52147e39e6e12c3097a246e : SUCCESS in 20m 54s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/ba8c653c67a4490f802347a7f585a51f : SUCCESS in 1h 11m 24s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/56be17310e694aa1adf8c87302ea25b2 : SUCCESS in 1h 21m 31s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/e2b56a311b4e430c9202513815becf1b : SUCCESS in 15m 50s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/4bdeed244bd748c78a00354f6eee2200 : SUCCESS in 17m 15s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/88f6b2a9d34345eda96ba78ed280b919 : FAILURE in 55m 27s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/2800a3fd41fd4204b6b4981eae1b14d4 : SUCCESS in 27m 14s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/295185e9c89d4b958e6f4a028f32b7d6 : SUCCESS in 19m 19s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/9c03f818867d4e31adc79d36143453dd : SUCCESS in 34m 42s","accounts_in_message":[],"_revision_number":2},{"id":"90dbdaa7ee9296007a62e6795fe481f7da8ec323","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-13 17:00:58.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"30c1e402870cb9c99f31b4a47c9b3a70172997fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 19:11:14.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cd0a48832917462eaea29e13b4877f78 : SUCCESS in 3m 45s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c70c11e61a404addae9c2d21eb59b447 : SUCCESS in 13m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c0a6a8d39504faf87e45a96b5d66d9f : SUCCESS in 11m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a095300e248451b80b7261eac904470 : SUCCESS in 6m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b59ad6816aba4e2491d3d5cceeec201b : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/055adbd0c6e74c1c9245627806341c4f : SUCCESS in 6m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9df73eda6c9f4df89e6283b7ce311c3e : SUCCESS in 2m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/aac5518daab5410893b9992396ce9a4d : SUCCESS in 11m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5695b07dc92346c6a6b081ae05238b8c : SUCCESS in 16m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4efb7ef5202347df817e370da9d1d8af : SUCCESS in 27m 56s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/eacb8835da78401e8822fa190694bae8 : TIMED_OUT in 2h 01m 07s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/2968c4eadedb4ed0ad1161de5f585b85 : TIMED_OUT in 2h 01m 36s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/3f01f75280f7441489b27f32bbdf6496 : SUCCESS in 15m 02s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/84630525ceea4d24996d510dbc3303d8 : SUCCESS in 20m 49s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/0addcf943d354afaa909733c94e7611b : FAILURE in 53m 19s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/87c021d436334b37a6d29593ec8e7537 : SUCCESS in 47m 33s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/d68074a29feb4847a687b249c9d3674a : SUCCESS in 37m 18s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/b37d878477d4418d9aa0720e5cea97e2 : FAILURE in 57m 37s","accounts_in_message":[],"_revision_number":2},{"id":"52c9a211a56be6f77c8f9415b1addfbbb0111eb5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-14 08:46:03.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7d1700ff338a2fe459d3c280f7611e4a3179dd5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 12:46:38.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2e39e140a2242a0b0f3f9d6270589ce : SUCCESS in 5m 27s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d2b1463b163f437596f708792e36f744 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac5e5fb1118c4866a927815d456ec2f1 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9483bfabe88b4b489532755d033df57d : SUCCESS in 5m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/97a421e2d66e4ba289b0a010f88b5bae : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce52e8cc89b346bb91a3cfeb52a7ac82 : SUCCESS in 7m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/266e4111dad84f509923380665a85cad : SUCCESS in 4m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9accee31386347458d6800ff29354463 : SUCCESS in 5m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fc07b1fab51c4f2cb0003d982b2acea7 : SUCCESS in 10m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fee374b0d655450083fa3a6dd5b4a94d : SUCCESS in 18m 40s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/b92b7fdf1ad843f2bda113279d635e1a : TIMED_OUT in 2h 01m 34s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/e5deb15477c4461889fb36f02a1726f5 : TIMED_OUT in 2h 01m 22s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/a73cf58280184aa0a30a266c3fb7f0cf : SUCCESS in 16m 36s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/e469857b18234b0997e57d91e0d32231 : SUCCESS in 17m 24s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/252d76f307a84fda8273696377bee957 : FAILURE in 54m 59s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/68a6f71221eb4c28a833bfa099c4b231 : SUCCESS in 22m 39s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/78293e0d910a45e7a4984e8718fa5dac : SUCCESS in 18m 16s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/f42ff413d45b4307acd81fde2bc65897 : FAILURE in 54m 26s","accounts_in_message":[],"_revision_number":2},{"id":"3dbdafc0ed486eadb26e3dde576f6554de3dfb93","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-18 08:44:36.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"029e72c3f85cdea1a447527b3a52f823bad67651","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 10:24:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/40e049fe16514fc28debdcfb1a058003 : SUCCESS in 3m 57s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/df95a3c3297044709ae307fc4f37c744 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/03048a83bb124904aac0dd296ffe3e94 : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e2c7f70a9004a7f89e9060abdb3cc8e : SUCCESS in 4m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e4b145dd23134e999623e0c4d1b3b602 : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ae401d76e7d409f944c3d20433df77c : SUCCESS in 6m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/83b468e677f041b185ef92660bb5afb3 : SUCCESS in 3m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/54d594c7342240e998d0e52383712ba7 : SUCCESS in 4m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5e3bfd2785b0420b8bce34d406219fb1 : SUCCESS in 7m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/52ed345a574c4bbd9ff5c4017b7d938c : SUCCESS in 17m 56s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/1ed5d0e4c8c644209fb9695eba490864 : SUCCESS in 1h 00m 08s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/504ba9d7ea974fbdba88d4d2ca92cf70 : SUCCESS in 59m 04s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/6a715e781772474db7a66e869a345ca5 : SUCCESS in 12m 58s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/bb7bff6cb16c46e7a2884c1a1e9ad71d : SUCCESS in 16m 10s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/41ecb8d9502a41d19c678b25d43d6493 : SUCCESS in 1h 35m 37s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/6c71c29ae9fb44bca3655f4c23244c87 : SUCCESS in 20m 14s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/c0096fd99a1f423db42f7eeb97e79372 : SUCCESS in 14m 30s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/25142cc6f8b145e6bfaea2c9c9903590 : SUCCESS in 33m 11s","accounts_in_message":[],"_revision_number":2},{"id":"93c1673ecf5cf4c39cc29d5989771d317b31e803","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 10:24:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7723a9ff9e4556c3203f8cbab99ca9f4fd25c168","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 10:47:23.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b135052e318447f38f0a3b6ab1c3c4e5 : RETRY_LIMIT in 2m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4825a3a0a3a04fd0b1ef01c2af229d35 : RETRY_LIMIT in 2m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/287d77f2600d4fa683d6cb80a92e048a : RETRY_LIMIT in 2m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f4953e3ea9b443c39c7c6d011ac104a2 : RETRY_LIMIT in 2m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80f22012bda14883bb4776f257a6544c : RETRY_LIMIT in 2m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b21f8e52e9c6414487a8ebfaea1ded2a : RETRY_LIMIT in 1m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/650dafbb8835420ebf30fa5a726c69af : RETRY_LIMIT in 2m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5b8183b0590b465292360ff57a9ca37f : RETRY_LIMIT in 2m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9c7dfa7e70c04581845e3cd6f4ecbcf5 : RETRY_LIMIT in 2m 56s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/cfce0e41fc6f4265b624deee2320e149 : RETRY_LIMIT in 2m 04s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/45593dec1fd34c8cbc03f62534b90849 : RETRY_LIMIT in 2m 29s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/913da4fa9ff742da9757edac80c6014c : RETRY_LIMIT in 2m 24s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/85a42e23622c4c35bc91e1b0d149f78f : RETRY_LIMIT in 3m 05s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/7a45681907d846ac875fa8dac71fd5fc : RETRY_LIMIT in 2m 33s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/6525249be14f409b824fc2baa43eae87 : RETRY_LIMIT in 2m 11s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/4f8aee17a1c74364ac2a3209deb3902e : RETRY_LIMIT in 3m 34s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/f25a92f0a2284fa6a9ee9b5c1d546ecc : RETRY_LIMIT in 2m 59s","accounts_in_message":[],"_revision_number":2},{"id":"d86a36c6efaae010e78d89e4f3eb1b097a790c40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-27 09:20:41.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"71102f609846956362ced486797a7a7abcd86c7a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-27 09:57:24.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"d7472098691f4c458e10265e1d1c233a2c8d494b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 10:57:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ed08247d33e4369ab3c27a018536d56 : SUCCESS in 4m 18s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/37e8f818cd5d4642ba290b743c4f77f7 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df74f481cf2b4cc0a08f7b16977ed345 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/660a8590a22042399ac7c35c93f5149b : SUCCESS in 3m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f2d1a0dbeb5d4e5796bf5fb953a290df : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9249cbf08a14fe291ef1d6970647353 : SUCCESS in 5m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e617475d80149b09b7b0fde8b35b430 : SUCCESS in 3m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/58b55c9291fd478d870897a1857bb7ac : SUCCESS in 3m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cd6c14564c444a6a917bb443af35690a : SUCCESS in 7m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/66e034071d8445a0aa1d32c2d0746b10 : SUCCESS in 16m 42s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/cd15d47d50b940d7a35cde37d04105c9 : SUCCESS in 1h 02m 55s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/1d594008138143958bd5922a0fb3c066 : SUCCESS in 55m 45s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/2554462dcb734b1791d66f727b3eb6df : SUCCESS in 13m 47s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/90dda4fcee744912a9ef6c8dcc9b3680 : SUCCESS in 15m 28s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/20e146bcac084104b83bfb1f3b31fd8d : SUCCESS in 1h 32m 43s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/8d3c0d6669cc475b960d4631145c9c5e : SUCCESS in 19m 38s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/2a348946ab5546069611e02fece509dd : SUCCESS in 16m 46s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/ef9627a907cd48eeae05aa5705333ec6 : SUCCESS in 36m 04s","accounts_in_message":[],"_revision_number":3},{"id":"e3cb0d81401850e061c924cc934905ff7400159c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-27 11:01:10.000000000","message":"Patch Set 3: Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"c8799cbbf4ca830adfe03ebbe685b7580d740949","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 11:01:25.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1841255d20b289ca15af58f1bf7dec483ce49fb4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 12:47:10.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d75206177d6840f99e9672ae649aa7b9 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c247eccb15074107b43aade88e8051b4 : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4e11cd2746848598f5f659e437df2d8 : SUCCESS in 5m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/393e6c04d0f0464487c54e24d76c8b4f : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8834437f4b644489884676e13bfcc3f4 : SUCCESS in 7m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bca76a7312604407b23f76d1ee3c0ee9 : SUCCESS in 6m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/65d6849f1c75483e8d0e40497161abf9 : SUCCESS in 4m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/936ba38713ad4acbb8ab8e6f9b218f65 : SUCCESS in 7m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/604985ebf9c142f498ccd587cca024f5 : SUCCESS in 15m 46s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/2a1b57da1f30412e947dac3164190cb6 : SUCCESS in 1h 02m 31s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/ef00e6b653d14234803c599e92764a43 : SUCCESS in 59m 35s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/5adff7b612f24835829d017b346f72e3 : SUCCESS in 13m 33s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/61c656d5b9e240a0b83956152571fb12 : SUCCESS in 14m 24s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/6ea0aa7c045e414aacba4a40b1ff1e17 : SUCCESS in 1h 41m 45s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/03f1ccbc11a14adfb15118d94d26e7c2 : SUCCESS in 18m 25s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/0c47d9e666ee464ebca6d92548a949a6 : SUCCESS in 15m 12s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/74fa0024d38b4dfca9144a5b164ca667 : SUCCESS in 33m 48s","accounts_in_message":[],"_revision_number":3},{"id":"f8ff7e86ddfe49d7a43527c0937e76f44f52c609","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 12:47:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"193c23db84b7e0364a0a79e42ffd6561a7804706","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 12:49:26.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aeeee571187a4115a30475febdc29e5f : SUCCESS in 1m 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a71926dc80448d69c53e70e62a3b46f : SUCCESS in 1m 35s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e0e0db459d3c99fbc1cb266739fa7e6e17f0c00e","revisions":{"4c66868f738a5acdfe6da985106c685b047008d3":{"kind":"REWORK","_number":1,"created":"2020-11-30 09:48:06.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/11/764511/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/11/764511/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/764511/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/764511/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/764511/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/11/764511/1"}}},"commit":{"parents":[{"commit":"5f48f0199bce1e6bea0e67787457789b66de2e7b","subject":"Fix filtering of network names set to an empty string","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5f48f0199bce1e6bea0e67787457789b66de2e7b"}]}],"author":{"name":"Piotr Parczewski","email":"piotr.parczewski@outlook.com","date":"2020-11-20 12:08:49.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-30 09:48:06.000000000","tz":0},"subject":"Fix Python setup when venv is not used","message":"Fix Python setup when venv is not used\n\nSet packages argument explicitly as a list to avoid \"No package matching\n\u0027\u003cgenerator object select_or_reject at 0x7ff25f341750\u003e\u0027 is available\" error.\n\nStory: 2008378\nTask: 41298\nChange-Id: Ibe44583cb93b7dca8f5091c893386a15288af915\n(cherry picked from commit e748faea00346aeb1a05a1810a386b1f3d21fa75)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4c66868f738a5acdfe6da985106c685b047008d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4c66868f738a5acdfe6da985106c685b047008d3"}]},"branch":"refs/heads/stable/train"},"cc639a88be9f7ac5ca5d770e602bf40bb76ef3ee":{"kind":"REWORK","_number":2,"created":"2020-12-22 13:16:01.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/11/764511/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/11/764511/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/764511/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/764511/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/764511/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/11/764511/2"}}},"commit":{"parents":[{"commit":"5f48f0199bce1e6bea0e67787457789b66de2e7b","subject":"Fix filtering of network names set to an empty string","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5f48f0199bce1e6bea0e67787457789b66de2e7b"}]}],"author":{"name":"Piotr Parczewski","email":"piotr.parczewski@outlook.com","date":"2020-11-20 12:08:49.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-22 13:15:55.000000000","tz":0},"subject":"Fix Python setup when venv is not used","message":"Fix Python setup when venv is not used\n\nSet packages argument explicitly as a list to avoid \"No package matching\n\u0027\u003cgenerator object select_or_reject at 0x7ff25f341750\u003e\u0027 is available\" error.\n\nStory: 2008378\nTask: 41298\nChange-Id: Ibe44583cb93b7dca8f5091c893386a15288af915\n(cherry picked from commit e748faea00346aeb1a05a1810a386b1f3d21fa75)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cc639a88be9f7ac5ca5d770e602bf40bb76ef3ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cc639a88be9f7ac5ca5d770e602bf40bb76ef3ee"}]},"branch":"refs/heads/stable/train"},"e0e0db459d3c99fbc1cb266739fa7e6e17f0c00e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-01-27 09:20:41.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/11/764511/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/11/764511/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/764511/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/764511/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/764511/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/11/764511/3"}}},"commit":{"parents":[{"commit":"1073fcfb3eb06c0b581a16aca9ea2a198ad1ddc7","subject":"Fix overcloud host image resolv workaround on CentOS 8.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1073fcfb3eb06c0b581a16aca9ea2a198ad1ddc7"}]}],"author":{"name":"Piotr Parczewski","email":"piotr.parczewski@outlook.com","date":"2020-11-20 12:08:49.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-01-27 09:20:41.000000000","tz":0},"subject":"Fix Python setup when venv is not used","message":"Fix Python setup when venv is not used\n\nSet packages argument explicitly as a list to avoid \"No package matching\n\u0027\u003cgenerator object select_or_reject at 0x7ff25f341750\u003e\u0027 is available\" error.\n\nStory: 2008378\nTask: 41298\nChange-Id: Ibe44583cb93b7dca8f5091c893386a15288af915\n(cherry picked from commit e748faea00346aeb1a05a1810a386b1f3d21fa75)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e0e0db459d3c99fbc1cb266739fa7e6e17f0c00e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e0e0db459d3c99fbc1cb266739fa7e6e17f0c00e"}]},"branch":"refs/heads/stable/train","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
