)]}'
{"id":"openstack%2Fkayobe~764541","triplet_id":"openstack%2Fkayobe~master~I59a61dda68e66eeaa0526e2b91de9316fd04dee8","project":"openstack/kayobe","branch":"master","topic":"retire-qinling","hashtags":[],"change_id":"I59a61dda68e66eeaa0526e2b91de9316fd04dee8","subject":"Remove retired Qinling support","status":"MERGED","created":"2020-11-28 06:05:23.000000000","updated":"2020-12-19 03:38:36.000000000","submitted":"2020-12-19 03:37:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"764541-retire-qinling","meta_rev_id":"c846aa1f95d460467a7320843754961ff7493f21","_number":764541,"virtual_id_number":764541,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-19 03:37:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2020-12-15 10:22:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2020-12-08 10:36:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2020-12-18 08:40:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-28 08:06:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-08 10:36:17.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-12-13 18:50:28.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-12-13 18:51:35.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"df83052429a6dbefd477a96efaebcc21f8bbcb2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-11-28 06:05:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"919e563bc73e46e1ace925ef3ee02d197fd635b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-28 08:06:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/528c87beed6842b39ed5a0b02aff08d5 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebca968052314678a1eb77b8ff041a9d : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e26da5e027d4428392dcbf2dafbe3fd6 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd7556a3f71544faba784074d6ab1aea : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ded656f7a6bb4e9b981aae7138289def : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b717fcfdf49547f6ad21caf2e341f5e9 : SUCCESS in 6m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a428128b70d4d7589319861b0e868ed : SUCCESS in 3m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7ad353a4f1644ade96fcdb86cab5a739 : SUCCESS in 8m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/281254d8bd36482283e3dcb6ff1a14d9 : SUCCESS in 6m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/80c9a07cc9d24c87b1674d26c24fb163 : SUCCESS in 21m 51s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/e3eb355e39c0439598257ff7c885fb49 : SUCCESS in 58m 01s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/aa1e340af4ea430298092f7b96e9fbe5 : SUCCESS in 1h 19m 45s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/5926cf64e87c42cc9501904bd85841db : SUCCESS in 16m 00s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/15a65469c6f447d783c04d15cc0fc34e : SUCCESS in 1h 37m 02s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/e9a8d3db479f4a18a412ec9f29e7f750 : SUCCESS in 16m 10s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/8118d71503f047cba03d826b0e097636 : SUCCESS in 24m 29s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/62419962de88463580740fe1d5eabfe8 : SUCCESS in 11m 19s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"b99e61240dca8a46ba4376fe6f68139175f28fef","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-08 10:36:17.000000000","message":"Patch Set 1: Code-Review+2\n\nThanks","accounts_in_message":[],"_revision_number":1},{"id":"e193c482b8e580252677411b054b4728444a3ef9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-14 09:03:30.000000000","message":"Patch Set 1:\n\nCI currently broken, will merge when fixed.","accounts_in_message":[],"_revision_number":1},{"id":"e55b5d9d10de2e527c437777e71d771447d6ea4d","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-12-15 10:22:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0f960f45584553578293ed049934191d5bdb8a58","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-17 21:48:45.000000000","message":"Patch Set 1:\n\nrecheck CI is green now","accounts_in_message":[],"_revision_number":1},{"id":"12b281f4938d271385b504b98ad9c25e59e4c163","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 23:38:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d6398a85989b4f689b773f686ee09ab9 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d38376a455548ee9c6464f540423142 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0956c2c76d4849b5883fae70c74ef4fc : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c740126a032414dbf478e1b4b0ded12 : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f72199b7718a47978d4b0f2310f81351 : SUCCESS in 4m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d26c6561e4da48c5b67b910fc9726da3 : SUCCESS in 6m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/66a38a5cea0e4d4492909724c871d6be : SUCCESS in 4m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ba4fb8320c7244b2aaf6984ab3a53979 : SUCCESS in 5m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/13cbf2d958f3482c8dbcf8ae6b6816f5 : SUCCESS in 9m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f021082c9c8a4d06b3a1506f8062e7a0 : SUCCESS in 18m 43s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/d05d0e82862642bb9538c74330f2daaf : SUCCESS in 57m 23s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/f5eea3f9dcfb46e98bd1b5306c5e10c8 : SUCCESS in 1h 04m 08s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/f6704a9d3dc84eedbfe4373ff4d0902d : SUCCESS in 15m 30s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/c20716d633224077b24057c2ba134cda : SUCCESS in 1h 43m 09s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/5842ee5674304b1a89316a6f8e0c5583 : SUCCESS in 17m 06s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a62b790602d040d69cfb43546db9709a : SUCCESS in 22m 27s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/ee45fa31197842059d19b9a51c20f4ca : SUCCESS in 10m 14s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"674de30eb5dc42ffcac1f4874a996560a37de19a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-18 08:40:56.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"071620d0def725e0d12040e5e7bf3a11317974ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 08:41:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"54c260891fed19ddbed7d1ecc9b35b97bbe520a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 10:29:37.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e488a4c106b146098f1215540d0e819f : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/326bf65063234274bde02275499dfbb2 : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d7358bf12f94a7e8406f4e4ff69cd49 : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c3c5b97a8d741279e87baeb39302099 : SUCCESS in 5m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6016012144274d38a94fbef8dcb77036 : SUCCESS in 3m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bce906dc17814a4dba753e24e14eb2e6 : SUCCESS in 4m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8af3502736a34c69bbf1637d6e136845 : SUCCESS in 9m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/451a50faf370453db64f4b0beac14e27 : SUCCESS in 20m 41s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/7971400bce8b48d4bd0f766da934a405 : SUCCESS in 1h 01m 30s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/d24847f75254421080c26b3ea86f2696 : FAILURE in 8m 31s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/22263971104746a4872aba930e7f8d29 : SUCCESS in 16m 34s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/888b1a8c127048439011e8a52a20211c : SUCCESS in 1h 45m 58s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/0e6bef6d9070430fa2596c41193b00a0 : SUCCESS in 17m 02s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/500b0a08ffb649ba9daf078c87e5f702 : SUCCESS in 23m 56s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/a88ce4d480f74652b8734b1accbfefb9 : SUCCESS in 10m 16s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"cfb171ee6244c1b63c6d2d6631324fbc13e777d7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-18 11:24:32.000000000","message":"Patch Set 1:\n\nGate failure: Timeout (12s) waiting for privilege escalation prompt: \n\nSecond time I\u0027ve seen that recently. Possibly we need to bump the timeout.","accounts_in_message":[],"_revision_number":1},{"id":"a1018521e83def5979842a93bfe48a697c01d323","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-18 11:24:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bb2e6e5b59ba3be41a748908d2fe132057aaa848","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 14:01:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/378c1d36aee84f6fa3ff4c5ba27788d1 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fc990d733c0483aa23b31578ecf1c2b : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c33f3711e8474f85b2b97bbe6ebe4961 : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/291ab03374a946458b70dfda040a758e : SUCCESS in 5m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ff2426b3c3484be2917a1a0daa7a6352 : SUCCESS in 4m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73420b3b9a9841a9b58d0036ab05f6c7 : SUCCESS in 6m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d0bbc6e863894a7c84cd69b381e5f45c : SUCCESS in 3m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6c7ba3b97e1a4dd1bd6f785bfca38573 : SUCCESS in 4m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/009526d757114b8ab9101f3d5b1f4c63 : FAILURE in 7m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a6063181f5974a9fae96328bcaab1ed1 : SUCCESS in 17m 37s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/59797597bb734812af03d84cf17c557d : FAILURE in 1h 59m 55s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/653850d6658d4d1c86bbdfa855da2e6e : SUCCESS in 56m 37s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/3c13723127944e41b7283e03bb4d8e61 : SUCCESS in 16m 20s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/46a7fcd88780460ba68b09fa926cb060 : TIMED_OUT in 2h 31m 55s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/ce711bf2385b45e6aa52fdd6d3e354dc : SUCCESS in 21m 35s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a2cff8ef01d046248079a2fc240e1e43 : SUCCESS in 31m 34s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/0045b8e22f25401cb8e4584512e14296 : SUCCESS in 12m 41s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"dccf435a15f4d148cf1c07a351e46738f66a3b18","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-19 00:07:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"59bbc368ed2f086696d7a25ab1f31a141c54c2dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-19 01:47:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6372fcfd415e45a2ada197052b0b294b : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c22c61016b94b67b43ef573b26c384c : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0485b5df3073499193c1ba9a5343f9e4 : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f4112f7f0a764a3fb843c86584e3cdfb : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/66acc879d8254e2eb9659599e552eea7 : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e03c7a1b3144618ba7e4f786b92ba94 : SUCCESS in 5m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9b841599f0a4486b66dc945329ef373 : SUCCESS in 4m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0cf4088327b94b07af435647abffe650 : SUCCESS in 3m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6e53af69e959401985f3dd88cbb07a75 : SUCCESS in 8m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/91c518c763104721b534491482c44616 : SUCCESS in 19m 52s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/7acbf3d5ceae4301aabadead782b0142 : SUCCESS in 53m 19s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/b17e680f9ee848948ddb5dd7116ee1ab : SUCCESS in 55m 21s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/6c2855b4155f485b84008a2cc4d20159 : SUCCESS in 13m 28s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/81ed4ad85f4840aba881b5a20e62d21b : SUCCESS in 1h 38m 46s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/e3edfc7a02fa4f68a55d3b4be8397dd1 : SUCCESS in 17m 41s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/4aa42a0e2afc404a9213dc7bc0b1e331 : SUCCESS in 22m 28s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/e842e19a4c2a4af69efc71e3a379351c : SUCCESS in 9m 45s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"e620f113f4d9567dea2105ffe25f07d70897f05e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-19 01:47:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9e1073d75a56243d4a2df251a46cb3d94d102b52","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-19 03:37:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cf87f24bc48485b8587263a4aa33aed : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f35d7e41ec64ff78b4807ba0d61e843 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/23b76ec97e994a71b97fa3aaad8fdcf9 : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9eea092773a8479395185134a7c772c6 : SUCCESS in 6m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/269814049ae144cfa45d241a4a6634a9 : SUCCESS in 3m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7d579c55bc254a999e10327bccd51ce9 : SUCCESS in 4m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ba39e66e58384725a97a118e6ca1a21f : SUCCESS in 7m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/69742f09fea048bda3c3eafce2c5c5d7 : SUCCESS in 17m 12s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/1e2ec8a8f598482fbf35b43e2ad41823 : SUCCESS in 1h 08m 19s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/fe33a67b393043be8348c4dca17b60db : SUCCESS in 59m 26s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/0913eda510034ea89235729459b31019 : SUCCESS in 15m 55s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/096d018426ac4fb49cdf78e5898c3aaa : SUCCESS in 1h 44m 51s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/b76c4d28d8704743ac2a4be2cd8d0732 : SUCCESS in 15m 29s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/7dc32625d980412d896a8310f5cfac31 : SUCCESS in 24m 39s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/9a5ade6e46f949358be3bc657dbde275 : SUCCESS in 10m 16s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"6e4bc02fd9125c5e93e3eb6bd1cf758c7959a0d4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-19 03:37:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c846aa1f95d460467a7320843754961ff7493f21","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-19 03:38:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37ac5ea514ae4cfaaaafb721735945fb : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1339527fd6a14213b05020d0540ae116 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e5505b8b344162b379ef90342a07c90a80acf622","revisions":{"e5505b8b344162b379ef90342a07c90a80acf622":{"kind":"REWORK","_number":1,"created":"2020-11-28 06:05:23.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/41/764541/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/41/764541/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/764541/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/764541/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/764541/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/41/764541/1"}}},"commit":{"parents":[{"commit":"d169b4f5c74df4e36745d35e4ba2bc4e8d8584d7","subject":"Merge \"Fix Python setup when venv is not used\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d169b4f5c74df4e36745d35e4ba2bc4e8d8584d7"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-11-28 06:04:56.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-11-28 06:05:22.000000000","tz":-360},"subject":"Remove retired Qinling support","message":"Remove retired Qinling support\n\nQinling project is retiring in Wallaby cycle[1].\nThis commit removes the usages of Qinling project\nbefore its code is removed.\n\nNeeded-By: https://review.opendev.org/c/openstack/qinling/+/764521\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-November/018638.html\n\nChange-Id: I59a61dda68e66eeaa0526e2b91de9316fd04dee8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e5505b8b344162b379ef90342a07c90a80acf622"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e5505b8b344162b379ef90342a07c90a80acf622"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
