)]}'
{"id":"openstack%2Fkayobe~764600","triplet_id":"openstack%2Fkayobe~master~I7fa88a203274c3030b3d8e5208f94b2230d43c68","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I7fa88a203274c3030b3d8e5208f94b2230d43c68","subject":"Fix /tmp/swift-rings/backups/ files deletion","status":"MERGED","created":"2020-11-30 05:10:51.000000000","updated":"2021-01-21 12:27:33.000000000","submitted":"2021-01-21 12:26:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"764600","meta_rev_id":"ed4d31a4ca445df42923a45e69f2c57c47b7ce00","_number":764600,"virtual_id_number":764600,"owner":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-21 12:26:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-01-14 13:39:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2020-12-21 18:02:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":1,"date":"2020-12-19 23:44:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2020-12-21 10:09:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-12-20 02:37:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-01-14 13:39:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2020-12-21 18:02:36.000000000","_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":1,"date":"2020-12-19 23:44:36.000000000","_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-30 07:01:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-30 09:53:54.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2020-12-14 09:04:59.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2020-12-20 02:37:39.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2020-12-21 10:09:09.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-12-21 18:02:36.000000000","updated_by":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"reviewer":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"state":"REVIEWER"},{"updated":"2021-01-14 13:39:19.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"e39671925cc992e8f7c6d2d6cafa737aec1a1792","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"date":"2020-11-30 05:10:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"620e9210e8f3cf53acf0f6c4b8f6c06441ea3b40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 07:01:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9a98773b3f446e995c59addd9bfd985 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97661df9f28544d2ad681998d92f344b : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f3a504ed84eb452489a79d72b94a595f : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4e603601d39e43daaa4b961c8c26db0f : SUCCESS in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0a9d78bdbecd4829a9e441e8a5aa4cb3 : SUCCESS in 3m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4625613e0ef43d5bf0ad9b536a8a419 : SUCCESS in 6m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/457cac6f98cd4fa1a79e4590553ad2f2 : SUCCESS in 3m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b6cd5412fca146b89cbec7a8b69b3306 : FAILURE in 7m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d778f6b5201f4536993ef66ff5c1fb94 : SUCCESS in 19m 02s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/efa56fc71dcf4e079ab70b01423dfa5b : SUCCESS in 1h 05m 47s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/f0a0ed4a2b5542fda4033ab6ae164f10 : SUCCESS in 1h 08m 27s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/c910893ea8a04b2aabd4e057d7224d9b : SUCCESS in 15m 03s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a37e6e72b2604207ba824a64d8e6c4f8 : SUCCESS in 1h 45m 41s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/9e299a2e50034ebeac66fe434fe02bc5 : SUCCESS in 16m 17s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/9e0b1e6894f243a58b4e7d22819be00f : SUCCESS in 26m 18s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/f896ba51d5e0407db55a7c9a980860af : SUCCESS in 10m 25s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"f9afc7f57d3b9b5d1727f6f176a63ec296aacabe","author":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"date":"2020-11-30 07:28:22.000000000","message":"Patch Set 1:\n\n/recheck","accounts_in_message":[],"_revision_number":1},{"id":"84ee1ab7a68fbe85d0d897a6ba1d11661cf0e09e","author":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"date":"2020-12-11 06:38:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"70d9cd8adf077085b5e612aeb0d9653bde3e766e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 07:34:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fdeafc51a23d4dd89dfe90a411b03b82 : FAILURE in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63ba4adfa16641f2beedff19666cb787 : FAILURE in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d7e0cac215a641c084b6a958d5536b29 : FAILURE in 2m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa65bc23a61e4389a530e446cb7240b8 : FAILURE in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c65cc9bc3d740928c8093deee11105a : FAILURE in 2m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f367dead929948be84bfd7f0df0810a3 : SUCCESS in 6m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5f1090e6625447d4ad0d903062141ca1 : FAILURE in 2m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f5175730dec44b2d84e6b4e58b74754a : FAILURE in 2m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/52924f8b4a2d4bea9f453ba43a6d9b67 : FAILURE in 7m 18s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/de9e742a00a04078b9b5774b56a1df35 : FAILURE in 19m 26s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/2145e8cd18e644ce83a720363e367642 : FAILURE in 24m 19s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/100116802dec49d0bc1d148568cbd6c4 : SUCCESS in 18m 04s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/1affa726691d40558afe69ed00398a68 : FAILURE in 53m 25s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/7b168d1889104488992b3f3f755acaf9 : SUCCESS in 16m 54s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/67e7a247e31c46aeaa50041a8c43710d : SUCCESS in 25m 29s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/4136fdb31964460ba04c44693bfa2e19 : FAILURE in 9m 33s","accounts_in_message":[],"_revision_number":1},{"id":"be8cbd1d3d3deefbf75273c1b64fa21df68dc23a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"date":"2020-12-13 15:13:36.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent, and commit message as Patch Set 1.","accounts_in_message":[],"_revision_number":2},{"id":"460a307941a69c8bb13d7c669b1c37cf5786bb65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"date":"2020-12-13 15:15:50.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6804dc56900dac45dc134d9200a7b9843330bb06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-13 16:10:29.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c5810a56d4d549249ac76cd11313de98 : FAILURE in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/337a6ebdcdbd4ca295bc0eee4ff3c87f : FAILURE in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/559ce30b2bcb42a5996473322dcc6e58 : FAILURE in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a0a1bb8c5122483480d7a9f86ecacefe : FAILURE in 3m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0c9f9f77eeae4b87ab026c9c35a60b38 : FAILURE in 2m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b23cfd49df942f5ac92b3f4db419774 : SUCCESS in 7m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/13477e9d97274fddaaf1110e4b255a45 : SUCCESS in 4m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f81b3533f1124a20b98c330c9a4d6d61 : FAILURE in 3m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/050e5d033ab84316b97343b89b0a2908 : FAILURE in 2m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f9864b2028504bc5bb02418d7d2203ce : FAILURE in 3m 00s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/8ee97fde9d0c427b9cd2143bd84e1a74 : FAILURE in 46m 23s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/30a62dc40fbb44898cf091e90611b687 : FAILURE in 53m 28s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/68386bc0eb4a47c3bb7e8219e29c7090 : SUCCESS in 14m 23s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/7fe4151714ac4a64b2b60a3095d225e9 : FAILURE in 45m 50s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/fb0f0b08c4794bdc908c8ce70fed848b : SUCCESS in 15m 00s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f9c081ac99b54793b0568af678a67c54 : SUCCESS in 22m 57s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/e1d894da260f47cc80dfd85a45447227 : FAILURE in 9m 24s","accounts_in_message":[],"_revision_number":3},{"id":"768145f855b24266752545b77089dfb4c57b9b63","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-14 09:04:59.000000000","message":"Patch Set 3:\n\nCI currently broken, will review when fixed.","accounts_in_message":[],"_revision_number":3},{"id":"9e4bff1ea1646e5720bd6d5bfe9c6c2fa5cc26d1","author":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"date":"2020-12-18 12:57:28.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"dfbeb8d537a716d88791adb34bfc1388550ff775","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 14:42:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/45c88ffd2b004f32867eceb7f83c0193 : SUCCESS in 13m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1259e2d436de4c999a5c57cc6e4ecbff : SUCCESS in 12m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/60ecd717b6a744399a6ccf8c1f6658b2 : SUCCESS in 13m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/46dc38bf69d54ffb8a5b5faff0cafddd : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8c91b85415ed490ba895a100e9399fbe : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41005be27ff241aebd01cf5f4b204376 : SUCCESS in 7m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/17a2ab5d829a41e2af7af7079023ad50 : SUCCESS in 4m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d9b91138a04b4452a1c3e636808f9cdd : SUCCESS in 4m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6c374507c1b7412aa3b3d5aaf584dc87 : SUCCESS in 7m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4c93b6d8067746648df296bd73d72d0f : SUCCESS in 20m 29s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/9ba17397da7b4e49b931d679b2e31a2e : SUCCESS in 57m 57s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/3fc33e39a7654b239af7eaa9b459f21d : SUCCESS in 1h 04m 19s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/9e106936d29644db985333858ab02742 : SUCCESS in 17m 04s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/5a84769c5bed4681884b13bc1d1862e1 : SUCCESS in 1h 35m 46s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/1bd0139971954928bf400f5ac80c9678 : SUCCESS in 40m 21s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/1d52ee25677f4537aed674555bd8a726 : SUCCESS in 1h 04m 14s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/78ed7b65b8b245559b00e9d78a604795 : SUCCESS in 11m 14s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":3},{"id":"68f2ad7c2ee4c578b7029217a0dc849ebbb671e7","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2020-12-19 23:44:36.000000000","message":"Patch Set 3: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"3c87634a1cb60488531f1a1d78150bbb28e1b78c","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-12-20 02:37:39.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"d9fa761ea89209a03fbc17743cf9918c46deb61d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-21 10:09:09.000000000","message":"Patch Set 3: Code-Review+2\n\nI think if the other patch is used, then this is not required. However, this patch makes it possible to clean up after trying with the old broken code.","accounts_in_message":[],"_revision_number":3},{"id":"682392c3332d8d5d83fd3cf6f01ce4c54ec390d9","author":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"date":"2020-12-21 18:02:36.000000000","message":"Patch Set 3: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"ab72b61461d066401472f72a00c8caa8fe5fa986","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-01-14 13:39:19.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e7842aa12f876da19cd7086c207b4b8fa550de85","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 13:39:35.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"eb52f007a93bd318bbaeedd542e459bf59ad683e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 17:56:32.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5666d8366ff4409b21132a8e22e46e5 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86fdc3285f984783aa3806042258fcd0 : SUCCESS in 7m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a21a7ee092e44f1587f79a03fe852560 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df5a52d96b4346748d9629c51e8bdce7 : SUCCESS in 6m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b5ffccde2d74a93a783664d9841ea4c : SUCCESS in 3m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/528d1d4488164a09ab993d966beec88f : SUCCESS in 4m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cb29a25a8766480f94c9b7002c49b561 : SUCCESS in 8m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/106e17c97b6f49aa80297fdb0b3a12a7 : SUCCESS in 18m 44s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/e57a61ee993f40f88718291003fd4279 : SUCCESS in 1h 08m 22s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/9d549a9eff46446ba66cb1b9e4918bbc : SUCCESS in 59m 59s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/971c89ebd37e4945bcfe2fc3d3216ebe : SUCCESS in 20m 24s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b0bcbccdd48d4e32a1c8a1e0cb353dae : FAILURE in 14m 32s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/d1f24e0631384126ad58b1e7dad60778 : SUCCESS in 14m 54s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/864dde0636d1496aaf7f8c54bfee93f7 : SUCCESS in 23m 48s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/0f537440ef0d476abcf1846a5ea2fe0c : SUCCESS in 11m 45s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":3},{"id":"55c115abbca1cb1fdfb7b06b687ed4f4d0b58100","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-15 09:03:53.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"f298c91f7d91739740272fb40ab79da2511ae27c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 11:03:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/05288b7c26824d459355af2f80b9d00d : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6fb39c0272543dabb7506e1af029006 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c4663909f56a48df8b1b1b78efd2155b : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fd5b4a3405d9458795ca8d7f9fd77aea : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/91de116911504a4eaac5f93617786026 : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/149755fa36a04640979baf4edbf88a13 : SUCCESS in 6m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/038d0299517249db80b7aef20607ba64 : SUCCESS in 4m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a81b96716719470f920db235ebb0d524 : SUCCESS in 4m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c437cea344f84a9fa7c9d24bab3dbf2b : SUCCESS in 9m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dca81a0590954ccb9f329e3e801788fd : SUCCESS in 18m 37s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/a28f1f6f1c4d4dc9a97ec57b28531998 : SUCCESS in 1h 03m 01s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/1ba8b1858c454aadbb98adf14da344b2 : SUCCESS in 1h 05m 05s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/ed581346d08b49b580a1d3978a5bf4a6 : SUCCESS in 18m 24s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/58c1e7619ccc48b6890b33ea7814c434 : SUCCESS in 1h 54m 35s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/c3b2524be8a14b068ee1c6e22988c9f6 : SUCCESS in 16m 43s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/736e6321cde04782aeffe8b0f7ec9c1c : SUCCESS in 23m 22s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/44f1cc408ecf40c0af53cdade3163c5c : SUCCESS in 13m 37s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":3},{"id":"96ddda308f7ca7ebe350ee328e740af0a5a9765a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 11:03:52.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c8375a4a60a3f788bfee128f88b3f93983591dc8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 12:32:57.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ddae01f89034ac78715a0ae927a570d : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0bf4fcf8d5134c1cb0df2862ff3e3eb6 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40098825da064d11af8312d92d078e9b : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4a73d0c374e402f859bea15c1301e5e : SUCCESS in 6m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b3f3a6fc61754d8f8a8bc415961910ed : SUCCESS in 4m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0fa2683212354b27a80488c70e22394f : SUCCESS in 5m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6062c8d20efc4b57b193063e3907f289 : SUCCESS in 8m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/73f9da1f75c9402da96fe2f3947d08e3 : SUCCESS in 14m 44s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/9c45d416cd714dfda663ffff35420b65 : SUCCESS in 57m 41s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/43c6805860e74b1ca532317c6ab491fe : SUCCESS in 1h 02m 12s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/80576f6ef1474d758c539400cdbccef4 : SUCCESS in 14m 35s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/88b957fff35546e9afce5b79b27f5b1f : FAILURE in 1h 24m 44s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/1a208a1170ab4b50b2833e1e7a0aab86 : SUCCESS in 15m 08s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/66f7ac025dae446d9a5c2cc07faad4ef : SUCCESS in 25m 35s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/a744fadda6804a0eaecb47033b63cf7c : SUCCESS in 11m 45s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":3},{"id":"f9a544ee158063062b3a6e5e4827e8d9bab204de","author":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"date":"2021-01-21 07:01:48.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"3e3fe9bf4285cf96b1a788801a047ee02769e525","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 09:05:06.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6fd9ff6c228c4730b10337d09679820e : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b77c2544689d4e1f804e7ae174d0d953 : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a519f04b279413aa60b68aaa805a3f4 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d3dd20ba33d402dab7621984d7b82d4 : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/81b91970afc9494b8825dbf9a95e98bc : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d4157244af34708aa82e2dd43a12f65 : SUCCESS in 6m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6bb3b0ecac5740799886c6a6fa128d0b : SUCCESS in 4m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9dfb1cb3795d405eb7af013929c5d494 : SUCCESS in 5m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e597474453dd42af92aba7d76863e780 : SUCCESS in 8m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8963e29fc0e14ce6b26bf4b03e1af0a0 : SUCCESS in 18m 41s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/36e65b875aa14f4484ebf26fb475db89 : FAILURE in 1h 59m 25s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/07447ecb13744bdf9dcc692e65663c66 : FAILURE in 1h 52m 00s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/9dec14aaa59a4798868d0d6509f48944 : SUCCESS in 15m 40s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/95048071992c45d6909aba276ddd44a8 : FAILURE in 1h 53m 37s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/df763c7199d74daeb8c4ddcd57f14b1f : SUCCESS in 16m 32s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/8e653c376b2e43d9a25eed6a14ebfd8f : SUCCESS in 23m 59s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/7ca2fc21207c44c29c9c5d4ed377ba57 : SUCCESS in 10m 52s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":3},{"id":"c117b737cf6aadc73accc8d22e9107bede5ef02b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-21 09:06:17.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"cce3196bd61baeec350da785100232bca53d6b88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 10:52:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a51885775fec4379afc06c5bb00d3f98 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee69e339a71c440489f62325458f04ba : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a9a4f9173ff64877ac3f840d10a44a6d : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7d3d8194237e4e59b3578452bb982c4e : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c82477e6cb74afe8be3e1e1dac4fc7c : SUCCESS in 4m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cba7157c613432283effd1a6a747296 : SUCCESS in 6m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb305c3155cd46309a7e184376a3c706 : SUCCESS in 3m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d750f405b9794b82ab27099150fd5fff : SUCCESS in 4m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/497f01e769204c8d93e9664e744ed4c7 : SUCCESS in 8m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f4bc5c6219f64b31ab0ecad57aa913c9 : SUCCESS in 19m 30s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/7d5d2db1be0f42d8964db7d093276590 : SUCCESS in 1h 01m 54s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/42dc391ec83242ba85acb5e407026483 : SUCCESS in 1h 05m 50s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/138219ba8a4748768f1f0b45ab8bb339 : SUCCESS in 14m 14s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/023166cd5da94e01a2bfef8059218958 : SUCCESS in 1h 38m 14s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/66943e6795414c72bdf06f69af03ab37 : SUCCESS in 14m 31s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/43dee1deecd9491bba9d2a5d37b354d3 : SUCCESS in 27m 05s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/07fafa210b63447986dbc981c814473a : SUCCESS in 12m 45s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":3},{"id":"a496ef20efb7fc9ea2c771acfc6b9b1eb750e48c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 10:53:10.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7626a05f9ff5cec2b97819d752d481ed91b6bba9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 12:26:02.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c17584bc89d543cd9d62aaa8bb3b5564 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb3f6a45d20b45b68e313fa23a6d88fa : SUCCESS in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2ba8086bf8cc4544a1b07e087f33326c : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a12eb15f417f4e78a6f988247d5d927b : SUCCESS in 6m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ebac1ca8849e48f2b7af355455c5fdfb : SUCCESS in 4m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ea11c5e46b304ed98f7a5cc78c1e3fb4 : SUCCESS in 5m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0e9f943b284740978dba36c1616e24fd : SUCCESS in 7m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/119d8d677add463d8e06055ae8227165 : SUCCESS in 20m 12s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/fc70119a7a52415d933d54d023cae2e2 : SUCCESS in 57m 55s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/101fda1ff1474b1c9e86450af759bb20 : SUCCESS in 1h 00m 39s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/5b36d46e2385475f9e8910e6537ef824 : SUCCESS in 15m 49s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f4ab5e4b476741b397e15361b3700c99 : SUCCESS in 1h 20m 29s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/6918fcdd9b1d48d4a27ad29d4c39eca2 : SUCCESS in 14m 41s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/fbd72a0d9d2c428f8521eec74f73a48f : SUCCESS in 24m 58s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/239eabfb26b846c288380074eb418495 : SUCCESS in 10m 18s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":3},{"id":"a1196b7c718231034358c6e0c7629bc8b5515c47","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 12:26:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"ed4d31a4ca445df42923a45e69f2c57c47b7ce00","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 12:27:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ad59ac81e9845a7b1c08ca889b3a2c4 : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1882011f4e0a4b0faa08545374742256 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f811d3148ac542f32b476146ffc791aa4ea72693","revisions":{"c79cdae7fcf03bde9b183db4e57bf00848e9dc14":{"kind":"REWORK","_number":1,"created":"2020-11-30 05:10:51.000000000","uploader":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"ref":"refs/changes/00/764600/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/00/764600/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/764600/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/764600/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/764600/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/00/764600/1"}}},"commit":{"parents":[{"commit":"d169b4f5c74df4e36745d35e4ba2bc4e8d8584d7","subject":"Merge \"Fix Python setup when venv is not used\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d169b4f5c74df4e36745d35e4ba2bc4e8d8584d7"}]}],"author":{"name":"Margarita Shakhova","email":"shakhova.margarita@gmail.com","date":"2020-11-30 04:53:46.000000000","tz":-300},"committer":{"name":"Margarita Shakhova","email":"shakhova.margarita@gmail.com","date":"2020-11-30 05:10:40.000000000","tz":-300},"subject":"Fix /tmp/swift-rings/backups/ files deletion","message":"Fix /tmp/swift-rings/backups/ files deletion\n\nAdd permission escalation to the swift ring generation playbook\n\nTask: 41256\nStory: 2008354\nChange-Id: I7fa88a203274c3030b3d8e5208f94b2230d43c68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c79cdae7fcf03bde9b183db4e57bf00848e9dc14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c79cdae7fcf03bde9b183db4e57bf00848e9dc14"}]},"branch":"refs/heads/master"},"c03b67fe74e030e0ec2f36139d2d3aefc6accb42":{"kind":"NO_CHANGE","_number":2,"created":"2020-12-13 15:13:36.000000000","uploader":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"ref":"refs/changes/00/764600/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/00/764600/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/764600/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/764600/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/764600/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/00/764600/2"}}},"commit":{"parents":[{"commit":"d169b4f5c74df4e36745d35e4ba2bc4e8d8584d7","subject":"Merge \"Fix Python setup when venv is not used\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d169b4f5c74df4e36745d35e4ba2bc4e8d8584d7"}]}],"author":{"name":"Margarita Shakhova","email":"shakhova.margarita@gmail.com","date":"2020-11-30 04:53:46.000000000","tz":-300},"committer":{"name":"Margarita Shakhova","email":"shakhova.margarita@gmail.com","date":"2020-12-13 15:13:34.000000000","tz":-300},"subject":"Fix /tmp/swift-rings/backups/ files deletion","message":"Fix /tmp/swift-rings/backups/ files deletion\n\nAdd permission escalation to the swift ring generation playbook\n\nTask: 41256\nStory: 2008354\nChange-Id: I7fa88a203274c3030b3d8e5208f94b2230d43c68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c03b67fe74e030e0ec2f36139d2d3aefc6accb42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c03b67fe74e030e0ec2f36139d2d3aefc6accb42"}]},"branch":"refs/heads/master"},"f811d3148ac542f32b476146ffc791aa4ea72693":{"kind":"REWORK","_number":3,"created":"2020-12-13 15:15:50.000000000","uploader":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"ref":"refs/changes/00/764600/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/00/764600/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/764600/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/764600/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/764600/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/00/764600/3"}}},"commit":{"parents":[{"commit":"d169b4f5c74df4e36745d35e4ba2bc4e8d8584d7","subject":"Merge \"Fix Python setup when venv is not used\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d169b4f5c74df4e36745d35e4ba2bc4e8d8584d7"}]}],"author":{"name":"Margarita Shakhova","email":"shakhova.margarita@gmail.com","date":"2020-11-30 04:53:46.000000000","tz":-300},"committer":{"name":"Margarita Shakhova","email":"shakhova.margarita@gmail.com","date":"2020-12-13 15:15:36.000000000","tz":-300},"subject":"Fix /tmp/swift-rings/backups/ files deletion","message":"Fix /tmp/swift-rings/backups/ files deletion\n\nAdd permission escalation to the swift ring generation playbook\n\nTask: 41256\nStory: 2008354\nChange-Id: I7fa88a203274c3030b3d8e5208f94b2230d43c68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f811d3148ac542f32b476146ffc791aa4ea72693"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f811d3148ac542f32b476146ffc791aa4ea72693"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
