)]}'
{"id":"openstack%2Fkayobe~766059","triplet_id":"openstack%2Fkayobe~master~I9c6b0d9d5e13e8b82ebb8e8a3a0f432efb865e28","project":"openstack/kayobe","branch":"master","topic":"story/2007297","hashtags":[],"change_id":"I9c6b0d9d5e13e8b82ebb8e8a3a0f432efb865e28","subject":"Fix copying Swift ring files","status":"MERGED","created":"2020-12-08 17:35:46.000000000","updated":"2021-02-03 13:46:00.000000000","submitted":"2021-02-03 13:36:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"766059-story/2007297","meta_rev_id":"d969714d9580c4ffaee447dfc0ed545a220304b3","_number":766059,"virtual_id_number":766059,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-03 13:36:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2020-12-18 12:47:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":1,"date":"2020-12-18 13:45:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-03 11:55:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2020-12-18 20:40:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-03 11:55:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2021-02-03 13:46:00.000000000","post_submit":true,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":1,"date":"2020-12-18 13:45:17.000000000","_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2020-12-18 20:40:23.000000000","_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-08 19:01:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-10 06:15:55.000000000","updated_by":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"reviewer":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"state":"REVIEWER"},{"updated":"2020-12-10 06:44:42.000000000","updated_by":{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},"reviewer":{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},"state":"REVIEWER"},{"updated":"2020-12-10 07:04:12.000000000","updated_by":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"reviewer":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"state":"REVIEWER"},{"updated":"2020-12-10 09:37:43.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2020-12-18 12:47:03.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2021-02-03 10:58:07.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"b3fb2267193eb2124d2f176f361f04f11357f2c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-08 17:35:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"07a25a64ddbbb7f5fb783aaeb3b0f813f2862c01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 19:01:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/676463fb6afb4cc0a1ac8c44f40384eb : FAILURE in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a89e8a42d124d2c9649bda666cce466 : FAILURE in 3m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/49b0f369e27542c09a18b60b3d08b087 : FAILURE in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e23d97a75baf4363bef118979e9cdcf5 : FAILURE in 2m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7bb024f3181a42d0ae86c8b3248da9b9 : FAILURE in 2m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1401aea61dc141ecbd56ffe3bafefd11 : SUCCESS in 6m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a03439edce9e420fb300952b2b4ad192 : SUCCESS in 3m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/620cbf4a9ee74102a2a7d3e197707c3a : FAILURE in 3m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2bc7b374781743dd99f4586e43d47f7c : FAILURE in 4m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a475efa3c8374163961ea6072423655f : FAILURE in 4m 57s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/eed71f4cbf50450a99853709a246f5cb : FAILURE in 59m 31s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/744f028ad04d47c2857b5d2248659ec7 : FAILURE in 56m 01s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/3514fd39d5414d4d8b76d9e49751e2db : SUCCESS in 15m 51s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/03d0bcd6ab3441d5a020d38593d0b867 : FAILURE in 1h 08m 10s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/3ac0804064044ea1bea301df2f444dbb : SUCCESS in 18m 43s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/6c0c1fa54ffc4da6a4842da918d5010b : SUCCESS in 27m 56s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/e3829a96481e4aa3bcfab575c14d069e : FAILURE in 11m 48s","accounts_in_message":[],"_revision_number":1},{"id":"7c721b2d3e2901c1390631649db04d3521ac59b3","author":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"date":"2020-12-10 06:15:55.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"bdd094c682f428addf3d638ee0775d08b933d0d5","author":{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},"date":"2020-12-10 06:44:42.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ce6573ffdc0274b26d98899e8429be1263a24ef4","author":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"date":"2020-12-10 07:04:12.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+1\n\nNeeded patch","accounts_in_message":[],"_revision_number":1},{"id":"d5f051bcc6e697f010f2b4f5b1092a36eb950b06","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2020-12-10 09:37:43.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0222d999d89dfb3db97ded2ee09f2a642fccdfa4","author":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"date":"2020-12-11 06:37:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1bdbb5e219a3f0c7a86862b0e1bd15161276cc3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 07:31:25.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cfb6035d643f4e2ab97fdb99b94227bf : FAILURE in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b4273fb6a1d453ebc9790c9a20364de : FAILURE in 3m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aaf2dd03454a4c7eac9b4049accb39a2 : FAILURE in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc7fa674034e4364bc9fb8409684e2d8 : FAILURE in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/efdc576c7b69467692a9630dbfa2217b : FAILURE in 4m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e8bce33e3a04a75ae319ddd5e8c3c31 : SUCCESS in 6m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26b773d2a1044bfb9cb9ed776bd47c96 : SUCCESS in 3m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f1373adac8b149928245d2c4630ed940 : FAILURE in 4m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ece322f5078241c388f9e1bf6a05fb42 : FAILURE in 2m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/afa6e8fd256a4ecbab6b3d590065e593 : FAILURE in 3m 49s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/7de15f7abe23490cb6712bee23460960 : FAILURE in 21m 14s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/332c57a618fa4b999abf3c32cd7a17f1 : FAILURE in 20m 11s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/8e99ce0267d241f3bc65b3aae380aa47 : SUCCESS in 16m 28s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e4903ab75f5f448885a91497e4c42964 : FAILURE in 49m 01s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/8bb411ea10fa4162b8c41282b7063af3 : SUCCESS in 16m 43s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/9c1db90c36f04f3589d72b1925afb7c4 : SUCCESS in 23m 43s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/5fd2feda741f495c936e8ad99deebf05 : FAILURE in 11m 49s","accounts_in_message":[],"_revision_number":1},{"id":"5ddeb12330332f8be4b30ede953ec984aa7d0d76","author":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"date":"2020-12-13 14:59:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0b8bc3667abb97e74155598c4c28b0cd176e4b3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-13 15:49:36.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6ae6fd0b919e49e69f9051277b800c2c : FAILURE in 2m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edf8e30f33654f6f8977cd3e2be56c49 : FAILURE in 3m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8eb10a07fc38474ea5cde5f7f6add1ef : FAILURE in 2m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18fcb28eddf947d08f981284b10e8be9 : FAILURE in 2m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/33c935ddedc34c2485f8479ef8b02c73 : FAILURE in 2m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdea8ed0669f412dacd617ee3bf1d398 : SUCCESS in 5m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/94de2a5ec6314e9e87d978e24fefe0bc : SUCCESS in 3m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2464166863734b3facb6633e813d88c8 : FAILURE in 2m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5f509b2977e14c49a576021dbbf57d7b : FAILURE in 2m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/024bc49d457f4a0d87b96b6c7391c9ad : FAILURE in 2m 50s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/f90386c6714e4a2ebeb254373a58d73a : FAILURE in 44m 42s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/3d2793d255a947f2abc859c45a0782dc : FAILURE in 49m 51s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/aa3fc8ca562b42949b6e53902accb714 : SUCCESS in 13m 44s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/5cc80fde858e4e06becd3be20a89c65e : FAILURE in 44m 19s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/19c93183fefd4300b99b9b773355311f : SUCCESS in 16m 51s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/3a933b09b7ab4c80b9a0e016d78bbc20 : SUCCESS in 18m 21s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/47455b63cc4e4cd397fc056a5283dc09 : FAILURE in 10m 32s","accounts_in_message":[],"_revision_number":1},{"id":"4e01eaf43df7b0b320a4d9c9a7f21b01465f66cf","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-14 09:04:40.000000000","message":"Patch Set 1:\n\nCI currently broken, will merge when fixed.","accounts_in_message":[],"_revision_number":1},{"id":"61541833f5ebd71a1c96f7d1ed17ad1a2cb3dab0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-15 16:56:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7ecbd163698fe422a62ca18b0c25ed9e9ec77a45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 17:53:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1fa44cb6de324c26a865129b34b42c90 : FAILURE in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92553393153541c891880ed979fb2a58 : FAILURE in 2m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c90fad18add4de1a24323385ab2f320 : FAILURE in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b6b82254f7bf40108b173a04258ec55a : FAILURE in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/953466cf38064e8790cb573f073bcf2b : FAILURE in 5m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/600d399173cb48bd90b7e441d1f5611e : SUCCESS in 5m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52d407a59022492ba66099a65b3d55dd : SUCCESS in 4m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/011e203887b441af8c8e6203ab205791 : FAILURE in 3m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/194da7c80e1c4a419f60cc6c80a9757d : FAILURE in 3m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f0350722dc6543e0a6e4999176973c3a : FAILURE in 3m 27s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/06ec9af43af9456d958f1aac1eb8bdf0 : FAILURE in 46m 30s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/610aea7d809a4a959c5726120e441bac : FAILURE in 39m 24s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/7c96031a134f4d06af8a46648cf9b903 : SUCCESS in 14m 38s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/d89e065436cd4386b808fffeb1567624 : FAILURE in 50m 31s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/993fec5e2a14439d9b5478f9baa7f2da : SUCCESS in 16m 42s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/48d0ffae36574f4189b6985a8e7dc36a : SUCCESS in 23m 26s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/b96e3e1fb4b6410982fa0425a99234e4 : FAILURE in 10m 43s","accounts_in_message":[],"_revision_number":2},{"id":"c3398c2e8f878f68e7676c65b02291a2da1b49d3","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-15 19:49:12.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"144600528ef1577cf45bc532c2101263de87ebd3","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-17 19:27:21.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"aefb7752ed42c02678ed4ccc495b8489023b0e0b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-17 19:27:29.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e2d061cc8ac5531c53d044af7e67829fa66b75b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 21:31:58.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/40541e1f3be24e6793732c628da311a0 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6acbbfd61a5544b98dabee0123481e34 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b94e69db55664953a53034247b207dc8 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4e06d6ca36144e18b148deef1940025a : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/65934c61ecb64dd7be70e34ae2bb5f0e : SUCCESS in 8m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b88b849085bc4bac876606d2a07ffba4 : SUCCESS in 6m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac96a7dc8132463b9a7cc8c263e76c24 : SUCCESS in 3m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2197bc3ff94b42a88883763685e428a7 : FAILURE in 6m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4041c77543af46b4bde31e901995e0f6 : SUCCESS in 8m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2056409cbccb42d2bf1db749e00c45ca : SUCCESS in 17m 51s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/528c07f691cc4db7ba20dba600323ee2 : FAILURE in 46m 28s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/306f06bc982e4bd492dc3126f8563fcf : SUCCESS in 1h 09m 09s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/b63759ed475b493f82950df4aa87328f : SUCCESS in 14m 10s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/445bce48d16c4c7aa60b1033ac94fd5f : SUCCESS in 1h 51m 42s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/2d7ffd5b4916414196b2d84f83d929a8 : SUCCESS in 14m 58s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/7646df62e3814eb6890944e327c778be : SUCCESS in 23m 42s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/c1edb159a3b3404e9729f5963ee3bb65 : SUCCESS in 11m 04s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"af9080614e48bf19512c15b844c3d5e5778f5d42","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-18 08:50:17.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"121f2685165b3e611d5f26f6499155b415284c4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 10:43:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a6ea2bb062946d2ad790eae63bbf1e7 : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/befd322063ac4fe2bd386961b31eadce : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f56a499087db414b81086b72025d5aec : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce4afbc159a543febb1d2324375aa139 : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e8013ed6a244a3a92e0c56ddf7f6793 : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dabf8da1fb8446c9f4d472713e8ce49 : SUCCESS in 6m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7fd843124ff546ad86d5cb43654c2078 : SUCCESS in 3m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a100ef208beb4c759129e67d9985b19a : SUCCESS in 4m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c6699ec674644d71b5e250975f07e3b6 : SUCCESS in 8m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/df556ad2a1bf4a34bcd327ea094c4994 : SUCCESS in 17m 05s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/d7f51dfe78814f43b09772f962b409ac : SUCCESS in 55m 50s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/ca9e31a867144c83bee712b04fe10c55 : SUCCESS in 58m 23s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/5a9a922661ef4462b1435d91a8bc503a : SUCCESS in 18m 13s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/9d25596758074685b1ed0585c5fd8485 : SUCCESS in 1h 47m 45s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/0a51258c9e604c92ae43ca76d1218d01 : SUCCESS in 16m 29s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/1672319ffa4649f8a5786c79aa8436a7 : SUCCESS in 49m 04s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/1664336c982a4b868eb02832ee723c99 : SUCCESS in 13m 10s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"9d347fb50ff2b91372f85e1207924cc16c0761ea","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-12-18 12:47:03.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"25f1a31371c9f1dc8e765be0d6d4d52c632cf94e","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2020-12-18 13:45:17.000000000","message":"Patch Set 2: Backport-Candidate+1","accounts_in_message":[],"_revision_number":2},{"id":"292157ad5bf7547adcc46ea586e71e30669341de","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2020-12-18 13:45:27.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"3ec9e154acf6c453519b55b5a5f6f6c2fe562a81","author":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"date":"2020-12-18 20:40:23.000000000","message":"Patch Set 2: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"6881eb6e90a092fbbd78ae5b6bcc3982b65ee77d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-21 10:12:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1c8cb1a26248a62d47843f9771a51ced4e51f5b2","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-03 11:55:45.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7b9ab890507bce295ba836b659725f13c747f81f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 11:56:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8122dee97aa8c1ade3099c0f3d406dabf894da5e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 13:36:37.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4ada8d344194df6b332ebe8fe56d51f : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/935b28b062494f04b5cac3baf59d80aa : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ba24266401c436792805a880f578e06 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c5abda53b804c0b8293f161ed6e3393 : SUCCESS in 6m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ebbda8b7fe649a8a5fd5ae3c6d02446 : SUCCESS in 4m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a0c34be744894eeda6f7aa72496bf1c9 : SUCCESS in 4m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c2f447a807974a36bf81c583a80cb71e : SUCCESS in 7m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7545b8627268476d83193cde475ad2fc : SUCCESS in 17m 01s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/f5e159c29d6b4008bc45ab89c83f759b : SUCCESS in 1h 00m 38s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/cec10116cb54470094ba88aafabd9f35 : SUCCESS in 1h 03m 25s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/438d5ac5a0d34ce9908b9e9ed695997c : SUCCESS in 14m 06s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/450158ed4aac49aebd3f3721bba423b3 : SUCCESS in 1h 33m 50s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/11787f4b6d804ff99cff165e5ecd7ff2 : SUCCESS in 14m 30s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/cc29e7c4133040e79aa22c534365b2e6 : SUCCESS in 21m 10s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/fb0c3072bdf54fdfb131f94eb6713beb : SUCCESS in 9m 21s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"0ca43ae86d33dfad6cddfcfeee4dc6066f2ddbdf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 13:36:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"201918699005403575b5e4c83154c7636d9f3ac0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 13:46:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d7ea5af20d24bdf9bb68c2f4f821ad9 : SUCCESS in 1m 25s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eef11ad77ee34338b0007e867df049a4 : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c847ef91572afa87a6d76b265c88db2d9e50ccb8","revisions":{"46db8e8a34699c830a6b605ee4b087a3ff930166":{"kind":"REWORK","_number":1,"created":"2020-12-08 17:35:46.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/59/766059/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/59/766059/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/766059/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/766059/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/766059/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/59/766059/1"}}},"commit":{"parents":[{"commit":"d169b4f5c74df4e36745d35e4ba2bc4e8d8584d7","subject":"Merge \"Fix Python setup when venv is not used\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d169b4f5c74df4e36745d35e4ba2bc4e8d8584d7"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-08 17:33:51.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-08 17:33:51.000000000","tz":0},"subject":"Fix copying Swift ring files","message":"Fix copying Swift ring files\n\nRing files are binary formatted and should not be templated. Attempting\nto template them causes an error such as the following:\n\n    Template source files must be utf-8 encoded\n\nThis change fixes the issue by implementing support for configuration\nfiles that are listed as \u0027untemplated\u0027.\n\nChange-Id: I9c6b0d9d5e13e8b82ebb8e8a3a0f432efb865e28\nStory: 2007297\nTask: 38774\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/46db8e8a34699c830a6b605ee4b087a3ff930166"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/46db8e8a34699c830a6b605ee4b087a3ff930166"}]},"branch":"refs/heads/master"},"c847ef91572afa87a6d76b265c88db2d9e50ccb8":{"kind":"REWORK","_number":2,"created":"2020-12-15 16:56:25.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/59/766059/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/59/766059/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/766059/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/766059/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/766059/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/59/766059/2"}}},"commit":{"parents":[{"commit":"d169b4f5c74df4e36745d35e4ba2bc4e8d8584d7","subject":"Merge \"Fix Python setup when venv is not used\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d169b4f5c74df4e36745d35e4ba2bc4e8d8584d7"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-08 17:33:51.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-15 16:55:31.000000000","tz":0},"subject":"Fix copying Swift ring files","message":"Fix copying Swift ring files\n\nRing files are binary formatted and should not be templated. Attempting\nto template them causes an error such as the following:\n\n    Template source files must be utf-8 encoded\n\nThis change fixes the issue by implementing support for configuration\nfiles that are listed as \u0027untemplated\u0027.\n\nChange-Id: I9c6b0d9d5e13e8b82ebb8e8a3a0f432efb865e28\nStory: 2007297\nTask: 38774\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c847ef91572afa87a6d76b265c88db2d9e50ccb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c847ef91572afa87a6d76b265c88db2d9e50ccb8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
