)]}'
{"id":"openstack%2Fkayobe~766203","triplet_id":"openstack%2Fkayobe~master~Ie8fd965165e8d347d27528a2c16d0647e412ccdc","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"Ie8fd965165e8d347d27528a2c16d0647e412ccdc","subject":"Fixes for CentOS 8.3","status":"MERGED","created":"2020-12-09 11:33:38.000000000","updated":"2020-12-17 17:25:15.000000000","submitted":"2020-12-17 17:23:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"766203","meta_rev_id":"a3e78850857ae41d709584cda53999093c3c3c2d","_number":766203,"virtual_id_number":766203,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-17 17:23:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2020-12-17 15:48:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2020-12-17 15:47:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2020-12-17 15:48:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-12-17 17:25:15.000000000","post_submit":true,"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-09 11:34:54.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-12-09 12:28:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-11 10:48:36.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2020-12-15 09:22:03.000000000","updated_by":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"reviewer":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"state":"REVIEWER"},{"updated":"2020-12-17 15:48:33.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"74c19e9ba5f4036aeff06d265118cbe1e8dae3cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-09 11:33:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8244648eb968d457e991dfe0c5f1bcf823f52d1b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-12-09 11:34:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3f63cf11980e2b2b99468b4e163e5d9dd9ff358d","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-09 11:36:27.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"910ff7fa2fc4cd34720703fcec645ebc801ed9cf","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-09 11:48:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"04c9e99910e9fc533a035913cac191093a1b4280","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-09 11:49:18.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"d9e4c48ab5db3f0554925c1005803799074e4ab0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-09 11:57:40.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1ff3f6789b32af7a1af7270dc482b974b0e43c5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 12:28:01.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1bc51b2f95814d10bc499b00a2e4d641 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd2b309c4b2b4d0c8a1cb0e747019711 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca49cd99495e44a99f067a73e9afb01a : SUCCESS in 6m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/061c3728a9a94dbaa888f33e58dc7e84 : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/af2bea6906df4d94a51ed45ff7d5f85a : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/198a126f88d64efc923f40ccdea5c598 : SUCCESS in 6m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c436a6df430247b38604c709d09a780d : SUCCESS in 3m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7d8fbadf09294e2c8dfa38f8cac65558 : SUCCESS in 4m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/25acc887f4c246af985e34d719b7f4a1 : SUCCESS in 8m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3d67a31f81e44ba6b780c67c7f448ce1 : SUCCESS in 18m 04s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/0b4bbdb2a9d74c6ab5eea1ec971bb59a : FAILURE in 20m 42s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/f652c8114dc243bd8bbc53a58a8365ee : FAILURE in 24m 08s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/c857ae72fbc14dd297caa50526b2c759 : SUCCESS in 15m 46s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/28624834f4cd43e0a6720f76f8a65306 : FAILURE in 18m 11s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/42d4bc1f7e744a7b8083fa8d038cf56f : SUCCESS in 16m 56s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/fe58f6c5d04246fba9ee0cd446b76086 : SUCCESS in 25m 49s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/564e49a34f2442599878dff4e970d83b : SUCCESS in 10m 59s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":4},{"id":"72a45f934b5e7060ebffac7786bdc4d51428a803","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-09 12:30:09.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"64883a17351bbf31042dcac9e4eb3306897d1010","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 14:13:17.000000000","message":"Patch Set 5: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"1f833c07065aaf77602ab2ae7a93e01c3eb47cbd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-09 14:15:50.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"ef8ef8ab481eba2d8fc594e40ae8eee6df8c2c9c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-09 14:29:34.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"16bbc6843c967edeee5c6c2bf49d9d166d209d8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 15:24:27.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d72a0b8173d4735aa4c1fd06bab420c : SUCCESS in 6m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/276bdee67d7b4d49a649a808381c482a : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0f9a18d8dfd4ca28689f2b71d572b20 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a541d75445314a52a2a6e07a01ae9c04 : SUCCESS in 5m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a951a0ab0fc5453c8c67c81443b8c9eb : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e06906c00bc4416b93044da22034484 : SUCCESS in 7m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38c4a401b22a4554981a13841e3f84ee : SUCCESS in 3m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2481a96a8648417fb8cf1444d09f36fe : SUCCESS in 4m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/efa53dac4ea840f7a3550d3c6c34afad : SUCCESS in 9m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e8ac5045057149fc9e25fa22353ee287 : SUCCESS in 21m 16s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/6e8645e0a31041e0a7192ca8fae1f22e : FAILURE in 57m 20s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/ef3966c7a5f449bc900f77760951b8d7 : FAILURE in 47m 16s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/8ba27e7bc9294cd281195520bd050d0f : SUCCESS in 20m 44s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/836465203e7d4f8a90b9cc4f6ffc98a0 : FAILURE in 24m 39s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/e72750b012854bd8a62725fe4232874c : SUCCESS in 13m 46s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/0d2c454159bd4b08957ecb840b246c2e : SUCCESS in 20m 43s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/ad9326e6ec7440c7abaf57b8f1cbe053 : SUCCESS in 12m 26s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":5},{"id":"a3b3bbdff6bde1bb9380b8e4c44153e2f2d464d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-09 19:07:43.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"eaf28d3ce660fce96c4191b73f0fa28b900c5e6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 19:18:30.000000000","message":"Patch Set 6: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":6},{"id":"33a7a9b69ee528b01b2e52c4245d3ab9a8cfba1e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-10 09:18:17.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"e812cce1a8af29ee59c459819e51706832dcab99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 10:14:00.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/10c76233dc46477c9971cd4143a95d75 : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80089dcb3ddb4f7daa996dd0b0f352db : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7243ffd63b024721b0c4104ab294027a : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dbce47114c324eba84a0fde5fc721350 : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a32d7d1274a84fb0904627a3885600fd : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08002f1e949643a4a7b937b1eb39ac23 : SUCCESS in 7m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c5074fca4264a50b36d5f7db69c90c1 : SUCCESS in 3m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6b4ff87a9996446ba4627e9410c2b10e : SUCCESS in 6m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cebca704d1204f958b0c974c264ab4aa : SUCCESS in 10m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ed035853ddc24a818ff65c422037a4e8 : SUCCESS in 19m 26s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/693602f132fd4af48ce0b6479657ba06 : FAILURE in 49m 59s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/7964b2200cb444f496239ac1276a017c : FAILURE in 52m 46s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/a761b6df316b4171af2d03f92fb95fd2 : SUCCESS in 17m 50s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/dccdb51dc9774de0af05701f5f0bb132 : FAILURE in 19m 22s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/970793871a904ab58b9fbe307d4d6e89 : SUCCESS in 16m 18s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f7d541f93dfd4153a837257d64a30daa : SUCCESS in 29m 27s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/7f242d0f74b646c3a05cdf3267cb3381 : SUCCESS in 11m 26s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":6},{"id":"416a238e7cd96b8e5d673993d9d4f90fc9306979","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-10 20:31:16.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"e3048e2c2502f40b9ffbf5a4baaccef0ef79be82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 21:45:15.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/03d51b39213945f7acbc313ecf8bdfdd : SUCCESS in 9m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ec43641c7b64b0f8691ba8c19612d3e : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/efe9cc79a08648f1a51bc48f34423ee6 : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/43e4aa31353d4ae39e8e081a0418fdb4 : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9776cf6d51384e09a490e8d7b65243b3 : SUCCESS in 9m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/409f380f78db4ef89eb918c68d3fc03c : SUCCESS in 6m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e20be6ef1b9a4170b692550e31d18ef7 : SUCCESS in 6m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2581ddc59f744dc1806522a0575a22f2 : SUCCESS in 4m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5521dc0c0f22414387eea183cf771e11 : SUCCESS in 14m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6610e4817cb74fe3aa08f48845086c09 : SUCCESS in 18m 24s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/0455432286384d0d83404a5ccd10e639 : FAILURE in 1h 06m 52s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/2cb9ed5498c749a2b74c74a288f6bb63 : FAILURE in 51m 20s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/1b41c6f11dd7407197d6fe5bf4ea0bac : SUCCESS in 24m 13s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b8e079c6e6dd4608bb1d4f36e9d6725c : FAILURE in 21m 40s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/a6eae3fab55d44d083dcac2b335d5217 : SUCCESS in 30m 53s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/1608f2be51f64547be0ff70bfa878fb5 : SUCCESS in 25m 45s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/16d49caf28f94cf5860b5abbae91e520 : SUCCESS in 18m 30s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":7},{"id":"81757de02fa2bb19a0d77a1e289c0592494cc005","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-11 10:01:11.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"3dacd3cfce55b9ea5ef74ac30a17e27c28ad28bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 11:20:23.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/892bb7a1ecb240fb8f81dc40d83a3304 : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2aa50f80194d4e57b66f31b25a9d7494 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/80214d4cb0614175a2b3cbe193065373 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b32ae00685541d09570f9a696357614 : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a0b431a49eb1435aa1c739d26a6107c7 : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edcde31b846146779e9950f1ebe73e23 : SUCCESS in 7m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b78e0a8ca8bf4401a393519ea358261a : SUCCESS in 4m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1ddfcd2ceb2d4cc689202d2f0d1944e6 : SUCCESS in 4m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/11a950577e8d468a96b8367be4d097be : SUCCESS in 7m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/444ac6d26cec47e89a7d155405544056 : SUCCESS in 15m 13s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/786eab0259d14b29ae98296223ccfcbb : SUCCESS in 55m 48s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/fbda9d972bc34df7845afc2d27c11404 : SUCCESS in 1h 10m 44s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/73462acd0db7401bb6705ed78935f1ef : SUCCESS in 17m 31s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/8932ab07f7cc436da198f0ed0c668e85 : FAILURE in 59m 43s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/66780d4c8a2c4536bce17696497f300b : SUCCESS in 15m 55s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/18ca2bd169b9482a90b6c4a255063ff2 : SUCCESS in 23m 31s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/fcaa8357ec154525b1d68cce556ca02a : SUCCESS in 12m 17s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":8},{"id":"f775fcb1b571685bc24331d9ad0d5a4841b30b6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-11 12:46:00.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"543aac456c8fa9f8393ffaef8ea99eeb21c401a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 14:34:46.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b1849009561f4df99472d4267d357027 : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67061b633d74437d9c3009269ba0f375 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eaf25b7bfe464ee192de639589cf8e09 : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/17bbb807c08a4e0fbefb90ec36b22de4 : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/75179e00af934e088ea0261c8f6baee1 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0de7247d89a044048d2172732265357b : SUCCESS in 7m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/65b8a61b68e442069700fa05d0b741b3 : SUCCESS in 3m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1f008a6bafaf47fab22566d0faef8fd4 : SUCCESS in 5m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/12bedb8fed674d8b8f35825af3229dd7 : SUCCESS in 9m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c457a6fe1df946db8338f91fbeb7c6fc : SUCCESS in 19m 15s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/e21e30a8619044db9c1127857ff9322f : SUCCESS in 1h 34m 50s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/59d9a3494ee943a89faafde906c686c2 : SUCCESS in 1h 04m 53s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/cf4d716c61f3441c955ca436af79f0b2 : SUCCESS in 18m 48s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/bd7d820373444315a2a7c54b301c55bf : SUCCESS in 1h 33m 57s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/09d5c2072bdd436684a7508c02039a33 : SUCCESS in 18m 16s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a6e16ebf093d4ebea065f4db82982f8b : SUCCESS in 18m 32s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/751f1a2a19114645b903fbcf19ccc4e9 : SUCCESS in 8m 38s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":9},{"id":"f6bf0ec15c23294016178a95d9d173c4940d7d76","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-11 17:45:07.000000000","message":"Patch Set 9:\n\nTODO: update references to tenks in docs","accounts_in_message":[],"_revision_number":9},{"id":"6167e37b6db90121e00217599b5d3120b9536a04","author":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"date":"2020-12-15 09:22:03.000000000","message":"Patch Set 9: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"87f25ce4a3fbe613ffcc44a79abed02dc03168f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-15 16:37:16.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"b6747ec35671eb06fc7b534383afa8147705d575","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 18:21:04.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7845445b1c8e4f5a972fc069abccefe7 : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a110a995c0543a08e6d55b80dd437bf : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a7e01f8b0474f93989ae8b5c87a7a3a : SUCCESS in 8m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8a3fc7e657945e1b01d2b8c6643cb3c : SUCCESS in 7m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5ee6ec324a5b4ee0b9d20d84f1bb3e39 : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b44b5659836a48ab8d286afd722a8f33 : SUCCESS in 6m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e51f536d59d40f9876ada2e1d1e4766 : SUCCESS in 3m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/311b3cd93209412daea75316c4438706 : SUCCESS in 4m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f76aa889e64d4dd9bbcead6f1eb4cbf8 : SUCCESS in 10m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/01f993b9889f4838b150fd8d343053cc : SUCCESS in 26m 53s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/60bfdb1f7cf74b0c9298552decc6eb95 : SUCCESS in 58m 56s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/5893bfed3e154699811616b830fb4b17 : SUCCESS in 1h 35m 35s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/c583cb6eeb8c489cab242d01631329da : SUCCESS in 18m 16s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b33ef64f013a4350a66121289138d128 : SUCCESS in 1h 37m 05s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/92d237717d8047e7822a522094a9ccc5 : SUCCESS in 18m 13s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/68acf4e9a8594629bfcf5371f57ad1ea : SUCCESS in 24m 58s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/bd37ca7ed28a494aac48d7be6b82edc4 : SUCCESS in 11m 37s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":10},{"id":"09241fc15d48fcaf2af7a5f58875f33fab2b4e6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-16 08:59:41.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"815b49d12ad5aaa9fad5508afa89f5c0245cf550","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-16 09:25:48.000000000","message":"Uploaded patch set 12: Commit message was updated.","accounts_in_message":[],"_revision_number":12},{"id":"01d98ed87dbd75a72501a2c8686eda1048740480","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-16 11:12:57.000000000","message":"Patch Set 12:\n\nArgh, more failures.\n* Bifrost now enables authentication.\n* Tenks failing when waiting for placement.","accounts_in_message":[],"_revision_number":12},{"id":"0c2713e586b5f38dd77304696b0b22f75aff627b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 11:13:15.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0587ef7b993d468c821c0f9840b901f8 : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9e795f427f4420fafd88ab788e8df8f : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ed58f64605543df970ff9c013c48201 : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/25980a3bc6474eb080ef040d3beec162 : SUCCESS in 12m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b0026fd0393c42f4b797542e28f7d13f : SUCCESS in 9m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6e3d15a8ad14b0fa5cfccda755baf93 : SUCCESS in 23m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a51ea3959f664884b109ada7a378c6d0 : SUCCESS in 7m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/86f2cc4037a24740a39549e8203689e3 : SUCCESS in 3m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7a5a75b3b2c14b2780f516ac15a90299 : SUCCESS in 8m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/13935af44a6549a99f250b6cf9ad58bd : SUCCESS in 17m 08s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/55021418e6bd475a86884b740fdb1219 : FAILURE in 51m 44s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/5617791926bc434bbcd003c7eb630bfd : FAILURE in 55m 22s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/d1c115be887f467c9854e21fbc30c658 : SUCCESS in 14m 45s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b8dfc047d83b4c9a8d102a1445e2cce5 : FAILURE in 1h 34m 13s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/fea76483053a456f87ec98fc393db4f4 : FAILURE in 15m 54s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/230c156e4a5c4685ad7ddcc1d0dcb362 : FAILURE in 24m 49s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/750c73d9ee5642979c912e85cfe21b87 : SUCCESS in 10m 12s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":12},{"id":"f6ea4261b2de6c832c98e95d530607c929deab74","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-16 11:14:22.000000000","message":"Patch Set 12:\n\nNode 543e0a5e-ae27-4ea7-9f8b-363249c12b10 is not ready for a deployment, reporting resources as reserved for it. Node\u0027s provision state is available, power state is None and maintenance is True.","accounts_in_message":[],"_revision_number":12},{"id":"cf97be34430171bf1f1afb7a7955ee6780db9a45","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-16 11:22:32.000000000","message":"Patch Set 12:\n\n2020-12-16 10:23:08.851 8 WARNING ironic.conductor.manager [req-b82dbd88-cd0a-48e8-9c9d-0f4428af08d7 - - - - -] During sync_power_state, could not get power state for node 543e0a5e-ae27-4ea7-9f8b-363249c12b10, attempt 1 of 3. Error: An exclusive lock is required, but the current context has a shared lock..: ironic.common.exception.ExclusiveLockRequired: An exclusive lock is required, but the current context has a shared lock.\n2020-12-16 10:23:08.851 8 DEBUG ironic.conductor.task_manager [req-b82dbd88-cd0a-48e8-9c9d-0f4428af08d7 - - - - -] Successfully released shared lock for power state sync on node 543e0a5e-ae27-4ea7-9f8b-363249c12b10 (lock was held 0.03 sec) release_resources /usr/lib/python3.6/site-packages/ironic/conductor/task_manager.py:387","accounts_in_message":[],"_revision_number":12},{"id":"e728202da5c001eca7f3f059ae6844235bd57de9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-17 10:22:14.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"b6129f794b2dc3add3ba3b5d1d191c64673fd1e6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-17 12:04:31.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"3e5973b570a49bf99bf5606b93a68e16770d01dd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-17 12:04:51.000000000","message":"Patch Set 13:\n\n\u003e Patch Set 12:\n\u003e \n\u003e recheck\n\nWeird ansible escalation prompt failure","accounts_in_message":[],"_revision_number":13},{"id":"4090ec3b577c9bbe106c055f81664d7ec8fde071","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 12:05:18.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c1714d98c364360a3c692b45710289f : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db197198d8d547c1862e3dfec22ea5f5 : SUCCESS in 9m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b128d0e4cba45b3a71691ef9e058061 : SUCCESS in 8m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a0cb645e1b8b4f32ac2cf39f015738cf : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bcad57ec39614813a29d8e7600cdf59e : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b3aea172abb4a2a8c1646c75fda7373 : SUCCESS in 6m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f35313dbff17427d8842be7029e0c01f : SUCCESS in 6m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9dc05805c9d34ebb8334543a1392a8ea : SUCCESS in 4m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/92a05ea5fa174733baa2682c6e5dee95 : SUCCESS in 9m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/63886b5c6a5b4d4a848959629d59ce62 : SUCCESS in 18m 15s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/8c9570c3024c43c3946f29852542c5f7 : FAILURE in 29m 45s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/61d4b84035864ed8b86764f8c9dab52c : SUCCESS in 50m 17s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/0f5f81f64d3540e3bff90336a86d5ad8 : SUCCESS in 14m 26s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f647ef4799234fd58a37daac514cd53a : SUCCESS in 1h 34m 30s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/da9792e5cf2b4cb6afd90f06ead2e93b : SUCCESS in 16m 35s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/c377326efcf04b97bb7f882746fb6808 : SUCCESS in 43m 18s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/c8e56f863ca7491f8481098c633904a1 : SUCCESS in 17m 45s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":13},{"id":"e2bee3edb634638d9ec37eb9baa86b469739840d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-17 13:48:04.000000000","message":"Patch Set 13:\n\nSame issue again. Affects overcloud deploy, but only with TLS disabled.\n\nhttp://paste.openstack.org/show/801126/","accounts_in_message":[],"_revision_number":13},{"id":"475b19a130e74dff85a68a7c27aea71ac2beafa2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-17 13:49:09.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"a6aa40d2614257239df524150907953d4bd701a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 15:30:32.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/afdedec4d77e4ea3951c2207bb405344 : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92ebb9893bfb492492cf3a0e7e5de512 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4dc2edd92c904556b57c2ca09d910e4c : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41278558e60b43069e41e01badfd0485 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/17a21e300dbf4520b861033c685fcc64 : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44861897c2d749bb95a62cda5c10780c : SUCCESS in 7m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e9533f0827d405fa4a1706c5d19c0a3 : SUCCESS in 3m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f64a833df9294538bdb8ed53c5981069 : SUCCESS in 5m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a995d675e83c4f31978ba37df446bc92 : SUCCESS in 11m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/56bf961611c64677947f91ad510232c4 : SUCCESS in 20m 19s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/6c8f858148864d7dacedcc957497bab0 : SUCCESS in 58m 04s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/8b4666dbd3474c519d6178de2d9c9185 : SUCCESS in 1h 02m 37s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/7b397ed9f96b415f944e2afc52dad470 : SUCCESS in 13m 57s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/906cbcc8b93f4271a3d97f39063eb947 : SUCCESS in 1h 36m 38s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/f12177c2106d4f2cb0e69d9b31a701f3 : SUCCESS in 14m 29s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/c663151bf32747378e6b1ff49ce5f774 : SUCCESS in 21m 58s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/ebe8fb1b660c4e5d817bb78452f5fe58 : SUCCESS in 10m 12s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":13},{"id":"0f1abf0be31e70b763946426adbd5f261f0120ed","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-12-17 15:47:45.000000000","message":"Patch Set 13: Code-Review+2","accounts_in_message":[],"_revision_number":13},{"id":"997b26825f75ace93c18760e086583bae653d858","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-12-17 15:48:33.000000000","message":"Patch Set 13: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":13},{"id":"f7fa8b6c54023eea35e8ab512f16044c717b10db","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-12-17 15:48:44.000000000","message":"Patch Set 13: Workflow+1","accounts_in_message":[],"_revision_number":13},{"id":"6287f76110cf9ac2961f2866a4b567021b009f65","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 15:49:02.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"ca9e7fe919efb2f769e8fc0061219d1072d374be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 17:23:33.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c53f529380f4c29b2338391edaec35a : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce505a94588a492abedbed6550ced405 : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d59281a9068942c98e780e2b7d455cc1 : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72a24fdf51cc4908aebf775de94fc4fe : SUCCESS in 6m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/793ed2cbe96a478cb57fa31fafecb835 : SUCCESS in 3m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/762217ca91f644a6a9ce9533631ab425 : SUCCESS in 4m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/48421c5616084549ba8e96e81f4b1e9b : SUCCESS in 9m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/18af26f0469d4f08ace0336de55c3c6b : SUCCESS in 19m 18s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/270f5b7550634fe6ad364d773fc7a3ee : SUCCESS in 55m 12s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/41f054c177fb4746a302b0e8da28469c : SUCCESS in 1h 00m 31s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/3ef93d1f5df6468382ba84de8c10ae49 : SUCCESS in 16m 48s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/7d701c8c7ba54c8d8530032bd50c25ea : SUCCESS in 1h 30m 24s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/446dcfc060714b80acc0cc0e9cfb0ae0 : SUCCESS in 15m 17s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/becab1e5dc424f1d9aeb576a15ab5754 : SUCCESS in 26m 11s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/70c6f0f575b3479ba61f48ce9c435ddc : SUCCESS in 13m 07s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":13},{"id":"625e4ed47ba8aa76108467e4a7c20e98a6c14c23","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 17:23:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":13},{"id":"5fbe7d63b4d8791e392f740a0e6d623261f3fb2c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 17:25:15.000000000","message":"Patch Set 13:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df4e085c507e48128fc1dfd34da806ea : SUCCESS in 1m 10s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/466b310de4ed4eb6b958c9d201084118 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"4398856ec80ae765e805df9aab7d6f2105a34009","revisions":{"a32ff5cb9284e628a0f204d187d1cece91fd8906":{"kind":"REWORK","_number":1,"created":"2020-12-09 11:33:38.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/03/766203/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/03/766203/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/766203/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/766203/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/766203/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/03/766203/1"}}},"commit":{"parents":[{"commit":"d169b4f5c74df4e36745d35e4ba2bc4e8d8584d7","subject":"Merge \"Fix Python setup when venv is not used\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d169b4f5c74df4e36745d35e4ba2bc4e8d8584d7"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-09 11:26:40.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-09 11:33:31.000000000","tz":0},"subject":"Fixes for CentOS 8.3","message":"Fixes for CentOS 8.3\n\n* Bump stackhpc.libvirt-host to v1.7.1. On seed-hypervisors installed\n  using CentOS 8.2 or earlier, interaction with libvirt may fail due to\n  libgcrypt being incompatible. See\n  https://github.com/stackhpc/ansible-role-libvirt-host/issues/42\n\n* Bump MichaelRigart.interfaces to v1.9.1. The CentOS 8.3 cloud image\n  includes an ifcfg-ens3-1 file. See\n  https://github.com/michaelrigart/ansible-role-interfaces/pull/93\n\nStory: 2008429\nTask: 41378\n\nChange-Id: Ie8fd965165e8d347d27528a2c16d0647e412ccdc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a32ff5cb9284e628a0f204d187d1cece91fd8906"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a32ff5cb9284e628a0f204d187d1cece91fd8906"}]},"branch":"refs/heads/master"},"7d143d60a525baa560460e75cb32dbcd4d856ba9":{"kind":"REWORK","_number":2,"created":"2020-12-09 11:48:28.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/03/766203/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/03/766203/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/766203/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/766203/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/766203/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/03/766203/2"}}},"commit":{"parents":[{"commit":"d169b4f5c74df4e36745d35e4ba2bc4e8d8584d7","subject":"Merge \"Fix Python setup when venv is not used\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d169b4f5c74df4e36745d35e4ba2bc4e8d8584d7"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-09 11:26:40.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-09 11:48:20.000000000","tz":0},"subject":"Fixes for CentOS 8.3","message":"Fixes for CentOS 8.3\n\n* Bump stackhpc.libvirt-host to v1.7.1. On seed-hypervisors installed\n  using CentOS 8.2 or earlier, interaction with libvirt may fail due to\n  libgcrypt being incompatible. See\n  https://github.com/stackhpc/ansible-role-libvirt-host/issues/42\n\n* Bump MichaelRigart.interfaces to v1.9.1. The CentOS 8.3 cloud image\n  includes an ifcfg-ens3-1 file. See\n  https://github.com/michaelrigart/ansible-role-interfaces/pull/93\n\nStory: 2008429\nTask: 41378\n\nChange-Id: Ie8fd965165e8d347d27528a2c16d0647e412ccdc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7d143d60a525baa560460e75cb32dbcd4d856ba9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7d143d60a525baa560460e75cb32dbcd4d856ba9"}]},"branch":"refs/heads/master"},"3aaab7b022be4832fbd9af793cc4da59f7884b39":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-12-09 11:49:18.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/03/766203/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/03/766203/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/766203/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/766203/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/766203/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/03/766203/3"}}},"commit":{"parents":[{"commit":"d169b4f5c74df4e36745d35e4ba2bc4e8d8584d7","subject":"Merge \"Fix Python setup when venv is not used\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d169b4f5c74df4e36745d35e4ba2bc4e8d8584d7"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-09 11:26:40.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-09 11:48:33.000000000","tz":0},"subject":"Fixes for CentOS 8.3","message":"Fixes for CentOS 8.3\n\n* Bump stackhpc.libvirt-host to v1.7.1. On seed-hypervisors installed\n  using CentOS 8.2 or earlier, interaction with libvirt may fail due to\n  libgcrypt being incompatible. See\n  https://github.com/stackhpc/ansible-role-libvirt-host/issues/42\n\n* Bump MichaelRigart.interfaces to v1.9.1. The CentOS 8.3 cloud image\n  includes an ifcfg-ens3-1 file. See\n  https://github.com/michaelrigart/ansible-role-interfaces/pull/93\n\n* Not CentOS 8.3 related, but tox jobs are now failing on installing\n  docker python package. Try removing the \u003c3 constraint\n\nStory: 2008429\nTask: 41378\n\nChange-Id: Ie8fd965165e8d347d27528a2c16d0647e412ccdc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3aaab7b022be4832fbd9af793cc4da59f7884b39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3aaab7b022be4832fbd9af793cc4da59f7884b39"}]},"branch":"refs/heads/master"},"7371509a9c052162c725a134168291554dd1edd8":{"kind":"REWORK","_number":4,"created":"2020-12-09 11:57:40.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/03/766203/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/03/766203/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/766203/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/766203/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/766203/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/03/766203/4"}}},"commit":{"parents":[{"commit":"d169b4f5c74df4e36745d35e4ba2bc4e8d8584d7","subject":"Merge \"Fix Python setup when venv is not used\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d169b4f5c74df4e36745d35e4ba2bc4e8d8584d7"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-09 11:26:40.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-09 11:56:57.000000000","tz":0},"subject":"Fixes for CentOS 8.3","message":"Fixes for CentOS 8.3\n\n* Bump stackhpc.libvirt-host to v1.7.1. On seed-hypervisors installed\n  using CentOS 8.2 or earlier, interaction with libvirt may fail due to\n  libgcrypt being incompatible. See\n  https://github.com/stackhpc/ansible-role-libvirt-host/issues/42\n\n* Bump MichaelRigart.interfaces to v1.9.1. The CentOS 8.3 cloud image\n  includes an ifcfg-ens3-1 file. See\n  https://github.com/michaelrigart/ansible-role-interfaces/pull/93\n\n* Not CentOS 8.3 related, but tox jobs are now failing on python\n  dependencies. Remove upper limits from docker and paramiko.\n\nStory: 2008429\nTask: 41378\n\nChange-Id: Ie8fd965165e8d347d27528a2c16d0647e412ccdc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7371509a9c052162c725a134168291554dd1edd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7371509a9c052162c725a134168291554dd1edd8"}]},"branch":"refs/heads/master"},"51b26ad3868d4d12209950c78cf021c58a4b22e2":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2020-12-09 12:30:09.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/03/766203/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/03/766203/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/766203/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/766203/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/766203/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/03/766203/5"}}},"commit":{"parents":[{"commit":"d169b4f5c74df4e36745d35e4ba2bc4e8d8584d7","subject":"Merge \"Fix Python setup when venv is not used\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d169b4f5c74df4e36745d35e4ba2bc4e8d8584d7"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-09 11:26:40.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-09 12:29:08.000000000","tz":0},"subject":"Fixes for CentOS 8.3","message":"Fixes for CentOS 8.3\n\n* Bump stackhpc.libvirt-host to v1.7.1. On seed-hypervisors installed\n  using CentOS 8.2 or earlier, interaction with libvirt may fail due to\n  libgcrypt being incompatible. See\n  https://github.com/stackhpc/ansible-role-libvirt-host/issues/42\n\n* Bump MichaelRigart.interfaces to v1.9.1. The CentOS 8.3 cloud image\n  includes an ifcfg-ens3-1 file. See\n  https://github.com/michaelrigart/ansible-role-interfaces/pull/93\n\n* Not CentOS 8.3 related, but tox jobs are now failing on python\n  dependencies. Remove upper limits from docker and paramiko.\n\nDepends-On: https://review.opendev.org/766211\n\nStory: 2008429\nTask: 41378\n\nChange-Id: Ie8fd965165e8d347d27528a2c16d0647e412ccdc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/51b26ad3868d4d12209950c78cf021c58a4b22e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/51b26ad3868d4d12209950c78cf021c58a4b22e2"}]},"branch":"refs/heads/master","description":"Edit commit message"},"f73f50fe93e466591821c3fd5550c6c5b982ab09":{"kind":"REWORK","_number":6,"created":"2020-12-09 19:07:43.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/03/766203/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/03/766203/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/766203/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/766203/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/766203/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/03/766203/6"}}},"commit":{"parents":[{"commit":"d169b4f5c74df4e36745d35e4ba2bc4e8d8584d7","subject":"Merge \"Fix Python setup when venv is not used\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d169b4f5c74df4e36745d35e4ba2bc4e8d8584d7"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-09 11:26:40.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-09 19:06:20.000000000","tz":0},"subject":"Fixes for CentOS 8.3","message":"Fixes for CentOS 8.3\n\n* Bump stackhpc.libvirt-host to v1.7.1. On seed-hypervisors installed\n  using CentOS 8.2 or earlier, interaction with libvirt may fail due to\n  libgcrypt being incompatible. See\n  https://github.com/stackhpc/ansible-role-libvirt-host/issues/42\n\n* Bump MichaelRigart.interfaces to v1.9.1. The CentOS 8.3 cloud image\n  includes an ifcfg-ens3-1 file. See\n  https://github.com/michaelrigart/ansible-role-interfaces/pull/93\n\n* Not CentOS 8.3 related, but tox jobs are now failing on python\n  dependencies. Remove upper limits from docker and paramiko.\n\n* Update Tenks config to use Nova libvirt daemon\n\nDepends-On: https://review.opendev.org/766211\nDepends-On: https://review.opendev.org/766279\n\nStory: 2008429\nTask: 41378\n\nChange-Id: Ie8fd965165e8d347d27528a2c16d0647e412ccdc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f73f50fe93e466591821c3fd5550c6c5b982ab09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f73f50fe93e466591821c3fd5550c6c5b982ab09"}]},"branch":"refs/heads/master"},"d48629ec5671f4742549c17e4b5fbb61d4d12cee":{"kind":"REWORK","_number":7,"created":"2020-12-10 20:31:16.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/03/766203/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/03/766203/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/766203/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/766203/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/766203/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/03/766203/7"}}},"commit":{"parents":[{"commit":"d169b4f5c74df4e36745d35e4ba2bc4e8d8584d7","subject":"Merge \"Fix Python setup when venv is not used\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d169b4f5c74df4e36745d35e4ba2bc4e8d8584d7"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-09 11:26:40.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-10 20:31:07.000000000","tz":0},"subject":"Fixes for CentOS 8.3","message":"Fixes for CentOS 8.3\n\n* Bump stackhpc.libvirt-host to v1.7.1. On seed-hypervisors installed\n  using CentOS 8.2 or earlier, interaction with libvirt may fail due to\n  libgcrypt being incompatible. See\n  https://github.com/stackhpc/ansible-role-libvirt-host/issues/42\n\n* Bump MichaelRigart.interfaces to v1.9.1. The CentOS 8.3 cloud image\n  includes an ifcfg-ens3-1 file. See\n  https://github.com/michaelrigart/ansible-role-interfaces/pull/93\n\n* Not CentOS 8.3 related, but tox jobs are now failing on python\n  dependencies. Remove upper limits from docker and paramiko.\n\n* Update Tenks config to use Nova libvirt daemon\n\nDepends-On: https://review.opendev.org/766211\nDepends-On: https://review.opendev.org/766279\n\nStory: 2008429\nTask: 41378\n\nChange-Id: Ie8fd965165e8d347d27528a2c16d0647e412ccdc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d48629ec5671f4742549c17e4b5fbb61d4d12cee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d48629ec5671f4742549c17e4b5fbb61d4d12cee"}]},"branch":"refs/heads/master"},"53939947ac1482403189c9664ab0c840490a7f22":{"kind":"REWORK","_number":8,"created":"2020-12-11 10:01:11.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/03/766203/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/03/766203/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/766203/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/766203/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/766203/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/03/766203/8"}}},"commit":{"parents":[{"commit":"d169b4f5c74df4e36745d35e4ba2bc4e8d8584d7","subject":"Merge \"Fix Python setup when venv is not used\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d169b4f5c74df4e36745d35e4ba2bc4e8d8584d7"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-09 11:26:40.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-11 09:34:30.000000000","tz":0},"subject":"Fixes for CentOS 8.3","message":"Fixes for CentOS 8.3\n\n* Bump stackhpc.libvirt-host to v1.7.1. On seed-hypervisors installed\n  using CentOS 8.2 or earlier, interaction with libvirt may fail due to\n  libgcrypt being incompatible. See\n  https://github.com/stackhpc/ansible-role-libvirt-host/issues/42\n\n* Bump MichaelRigart.interfaces to v1.9.1. The CentOS 8.3 cloud image\n  includes an ifcfg-ens3-1 file. See\n  https://github.com/michaelrigart/ansible-role-interfaces/pull/93\n\n* Not CentOS 8.3 related, but tox jobs are now failing on python\n  dependencies. Remove upper limits from docker and paramiko.\n\n* Update Tenks config to use Nova libvirt daemon\n\nDepends-On: https://review.opendev.org/766211\nDepends-On: https://review.opendev.org/766279\n\nStory: 2008429\nTask: 41378\n\nChange-Id: Ie8fd965165e8d347d27528a2c16d0647e412ccdc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/53939947ac1482403189c9664ab0c840490a7f22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/53939947ac1482403189c9664ab0c840490a7f22"}]},"branch":"refs/heads/master"},"df0fd7c1d354ee69da85d8bbc0313dce190136ad":{"kind":"REWORK","_number":9,"created":"2020-12-11 12:46:00.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/03/766203/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/03/766203/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/766203/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/766203/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/766203/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/03/766203/9"}}},"commit":{"parents":[{"commit":"d169b4f5c74df4e36745d35e4ba2bc4e8d8584d7","subject":"Merge \"Fix Python setup when venv is not used\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d169b4f5c74df4e36745d35e4ba2bc4e8d8584d7"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-09 11:26:40.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-11 12:45:53.000000000","tz":0},"subject":"Fixes for CentOS 8.3","message":"Fixes for CentOS 8.3\n\n* Bump stackhpc.libvirt-host to v1.7.1. On seed-hypervisors installed\n  using CentOS 8.2 or earlier, interaction with libvirt may fail due to\n  libgcrypt being incompatible. See\n  https://github.com/stackhpc/ansible-role-libvirt-host/issues/42\n\n* Bump MichaelRigart.interfaces to v1.9.1. The CentOS 8.3 cloud image\n  includes an ifcfg-ens3-1 file. See\n  https://github.com/michaelrigart/ansible-role-interfaces/pull/93\n\n* Not CentOS 8.3 related, but tox jobs are now failing on python\n  dependencies. Remove upper limits from docker and paramiko.\n\n* Update Tenks config to use Nova libvirt daemon\n\nDepends-On: https://review.opendev.org/766211\nDepends-On: https://review.opendev.org/766279\n\nStory: 2008429\nTask: 41378\n\nChange-Id: Ie8fd965165e8d347d27528a2c16d0647e412ccdc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/df0fd7c1d354ee69da85d8bbc0313dce190136ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/df0fd7c1d354ee69da85d8bbc0313dce190136ad"}]},"branch":"refs/heads/master"},"709cc141f44ccaf851ffbf0e2ba8de5e7bc4025e":{"kind":"REWORK","_number":10,"created":"2020-12-15 16:37:16.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/03/766203/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/03/766203/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/766203/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/766203/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/766203/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/03/766203/10"}}},"commit":{"parents":[{"commit":"d169b4f5c74df4e36745d35e4ba2bc4e8d8584d7","subject":"Merge \"Fix Python setup when venv is not used\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d169b4f5c74df4e36745d35e4ba2bc4e8d8584d7"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-09 11:26:40.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-15 16:36:29.000000000","tz":0},"subject":"Fixes for CentOS 8.3","message":"Fixes for CentOS 8.3\n\n* Bump stackhpc.libvirt-host to v1.7.1. On seed-hypervisors installed\n  using CentOS 8.2 or earlier, interaction with libvirt may fail due to\n  libgcrypt being incompatible. See\n  https://github.com/stackhpc/ansible-role-libvirt-host/issues/42\n\n* Bump MichaelRigart.interfaces to v1.9.2. The CentOS 8.3 cloud image\n  includes an ifcfg-ens3-1 file. See\n  https://github.com/michaelrigart/ansible-role-interfaces/pull/93\n\n* Not CentOS 8.3 related, but tox jobs are now failing on python\n  dependencies. Remove upper limits from docker and paramiko.\n\n* Update Tenks config to use Nova libvirt daemon\n\nDepends-On: https://review.opendev.org/766211\nDepends-On: https://review.opendev.org/766279\n\nStory: 2008429\nTask: 41378\n\nChange-Id: Ie8fd965165e8d347d27528a2c16d0647e412ccdc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/709cc141f44ccaf851ffbf0e2ba8de5e7bc4025e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/709cc141f44ccaf851ffbf0e2ba8de5e7bc4025e"}]},"branch":"refs/heads/master"},"dc6a12d660c858fa90c9185675d6b57eb027aafc":{"kind":"REWORK","_number":11,"created":"2020-12-16 08:59:41.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/03/766203/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/03/766203/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/766203/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/766203/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/766203/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/03/766203/11"}}},"commit":{"parents":[{"commit":"d169b4f5c74df4e36745d35e4ba2bc4e8d8584d7","subject":"Merge \"Fix Python setup when venv is not used\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d169b4f5c74df4e36745d35e4ba2bc4e8d8584d7"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-09 11:26:40.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-16 08:59:20.000000000","tz":0},"subject":"Fixes for CentOS 8.3","message":"Fixes for CentOS 8.3\n\n* Bump stackhpc.libvirt-host to v1.7.1. On seed-hypervisors installed\n  using CentOS 8.2 or earlier, interaction with libvirt may fail due to\n  libgcrypt being incompatible. See\n  https://github.com/stackhpc/ansible-role-libvirt-host/issues/42\n\n* Bump MichaelRigart.interfaces to v1.9.2. The CentOS 8.3 cloud image\n  includes an ifcfg-ens3-1 file. See\n  https://github.com/michaelrigart/ansible-role-interfaces/pull/93\n\n* Previously a second libvirt daemon was installed by Tenks on the host,\n  however changes in libvirt 6.0.0 to separate libvirtd into multiple\n  daemons do not allow for customisation of the PID files used by the\n  new daemons. This leads to a conflict between the container and host\n  daemons. Update the Tenks config to use the containerised Nova libvirt\n  daemon. This depends on a change to the stackhpc.libvirt-host role:\n  https://github.com/stackhpc/ansible-role-libvirt-host/pull/44\n\n* Not CentOS 8.3 related, but tox jobs are now failing on python\n  dependencies. Remove upper limits from docker and paramiko.\n\nDepends-On: https://review.opendev.org/766211\nDepends-On: https://review.opendev.org/766279\n\nStory: 2008429\nTask: 41378\n\nChange-Id: Ie8fd965165e8d347d27528a2c16d0647e412ccdc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dc6a12d660c858fa90c9185675d6b57eb027aafc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dc6a12d660c858fa90c9185675d6b57eb027aafc"}]},"branch":"refs/heads/master"},"e6c70c690c30c3021a7a96e5bbd17b30e39d94b0":{"kind":"NO_CODE_CHANGE","_number":12,"created":"2020-12-16 09:25:48.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/03/766203/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/03/766203/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/766203/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/766203/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/766203/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/03/766203/12"}}},"commit":{"parents":[{"commit":"d169b4f5c74df4e36745d35e4ba2bc4e8d8584d7","subject":"Merge \"Fix Python setup when venv is not used\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d169b4f5c74df4e36745d35e4ba2bc4e8d8584d7"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-09 11:26:40.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-16 09:25:35.000000000","tz":0},"subject":"Fixes for CentOS 8.3","message":"Fixes for CentOS 8.3\n\n* Bump stackhpc.libvirt-host to v1.7.1. On seed-hypervisors installed\n  using CentOS 8.2 or earlier, interaction with libvirt may fail due to\n  libgcrypt being incompatible. See\n  https://github.com/stackhpc/ansible-role-libvirt-host/issues/42\n\n* Bump MichaelRigart.interfaces to v1.9.2. The CentOS 8.3 cloud image\n  includes an ifcfg-ens3-1 file. See\n  https://github.com/michaelrigart/ansible-role-interfaces/pull/93\n\n* Previously a second libvirt daemon was installed by Tenks on the host,\n  however changes in libvirt 6.0.0 to separate libvirtd into multiple\n  daemons do not allow for customisation of the PID files used by the\n  new daemons. This leads to a conflict between the container and host\n  daemons. Update the Tenks config to use the containerised Nova libvirt\n  daemon. This depends on a change to the stackhpc.libvirt-host role:\n  https://github.com/stackhpc/ansible-role-libvirt-host/pull/44\n\n* Not CentOS 8.3 related, but tox jobs are now failing on python\n  dependencies. Remove upper limits from docker and paramiko.\n\nStory: 2008429\nTask: 41378\n\nChange-Id: Ie8fd965165e8d347d27528a2c16d0647e412ccdc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e6c70c690c30c3021a7a96e5bbd17b30e39d94b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e6c70c690c30c3021a7a96e5bbd17b30e39d94b0"}]},"branch":"refs/heads/master"},"4398856ec80ae765e805df9aab7d6f2105a34009":{"kind":"REWORK","_number":13,"created":"2020-12-17 10:22:14.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/03/766203/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/03/766203/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/766203/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/766203/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/766203/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/03/766203/13"}}},"commit":{"parents":[{"commit":"d169b4f5c74df4e36745d35e4ba2bc4e8d8584d7","subject":"Merge \"Fix Python setup when venv is not used\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d169b4f5c74df4e36745d35e4ba2bc4e8d8584d7"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-09 11:26:40.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-16 11:04:48.000000000","tz":0},"subject":"Fixes for CentOS 8.3","message":"Fixes for CentOS 8.3\n\n* Bump stackhpc.libvirt-host to v1.7.1. On seed-hypervisors installed\n  using CentOS 8.2 or earlier, interaction with libvirt may fail due to\n  libgcrypt being incompatible. See\n  https://github.com/stackhpc/ansible-role-libvirt-host/issues/42\n\n* Bump MichaelRigart.interfaces to v1.9.2. The CentOS 8.3 cloud image\n  includes an ifcfg-ens3-1 file. See\n  https://github.com/michaelrigart/ansible-role-interfaces/pull/93\n\n* Previously a second libvirt daemon was installed by Tenks on the host,\n  however changes in libvirt 6.0.0 to separate libvirtd into multiple\n  daemons do not allow for customisation of the PID files used by the\n  new daemons. This leads to a conflict between the container and host\n  daemons. Update the Tenks config to use the containerised Nova libvirt\n  daemon. This depends on a change to the stackhpc.libvirt-host role:\n  https://github.com/stackhpc/ansible-role-libvirt-host/pull/44\n\n* Not CentOS 8.3 related, but tox jobs are now failing on python\n  dependencies. Remove upper limits from docker and paramiko.\n\n* Not CentOS 8.3 related, but Bifrost has enabled authentication by\n  default. We are not ready to support this, so override it.\n\nStory: 2008429\nTask: 41378\n\nChange-Id: Ie8fd965165e8d347d27528a2c16d0647e412ccdc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4398856ec80ae765e805df9aab7d6f2105a34009"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4398856ec80ae765e805df9aab7d6f2105a34009"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
