)]}'
{"id":"openstack%2Fkayobe~767597","triplet_id":"openstack%2Fkayobe~stable%2Fvictoria~Ie8fd965165e8d347d27528a2c16d0647e412ccdc","project":"openstack/kayobe","branch":"stable/victoria","topic":"change-767597-2","hashtags":[],"change_id":"Ie8fd965165e8d347d27528a2c16d0647e412ccdc","subject":"Fixes for CentOS 8.3","status":"MERGED","created":"2020-12-17 17:54:12.000000000","updated":"2020-12-21 20:40:08.000000000","submitted":"2020-12-21 20:38:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"767597-change-767597-2","meta_rev_id":"b6bef8fbdf568313e0fe27fc2b4cede8677da2d8","_number":767597,"virtual_id_number":767597,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-21 20:38:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2020-12-21 14:39:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2020-12-21 14:39:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}],"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-17 19:44:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-17 22:09:54.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2020-12-18 10:00:08.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"9ca1bc91d5c8bab0adb36a1b32da9eeedc40e15a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-17 17:54:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"54106bc907c6f1b39663fc6ca32ecaafbd8c064a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-17 17:55:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"79d8489691d57befde34e527a5936fe3070869e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 19:44:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c702a0494dce44ea829151d7d70bcf6e : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fab3c87f8644d98aa35ecc5beaf1f49 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2277be76bca44932a1ef2b1f6526d8ac : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7928e750010341bf89f0068469d104b5 : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4592d8246ec425db3255e88474c95f1 : SUCCESS in 7m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/02103bbea0e54934bde11440e90786cd : SUCCESS in 3m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/13c779ee9f2e4e459f4e9cec4ae5ea80 : SUCCESS in 5m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/234698cc861c42fa97a0b946fda2b99d : SUCCESS in 9m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6a7a4709899347f7ae3072725d403eeb : SUCCESS in 20m 04s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/075abf9648894c41b12ea2c088826400 : SUCCESS in 55m 53s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/20d0376c1a20406b97aca51e1ad16a8d : SUCCESS in 1h 08m 02s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/2143900ca9da42d9b0f190fe60a4e17b : SUCCESS in 16m 42s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/5dd1d1e396714f51be87104773189521 : FAILURE in 1h 46m 05s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/2bceb9bad88e46998afa6b2a14878617 : SUCCESS in 17m 36s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/83ca163b3d884a1681912026fa78c5b5 : SUCCESS in 36m 27s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/f0ba29982eb84889b3c57b689029e23b : SUCCESS in 9m 43s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"f0e7c4df7fdf796f81f8034121c634169d270930","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-17 20:29:30.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"16ac1c9c5fe582be8e5c18791575f69f6c3678ab","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-12-17 22:09:54.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"1f9c93125a6e2f0ac79fc6e29bd7faa9674e1443","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 22:33:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b497b54778134c93b155e6cf79587166 : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1510bb1d2c714b81a422ebf4f65efa32 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa271319ce59450d8bfc20e58ecec248 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d9949933e4e24d7788c4e1998597909f : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/282b369cb8a14bc29855c017f4307181 : SUCCESS in 7m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c960ddc6156e4964ba035033f2cc66fb : SUCCESS in 3m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/22700c7474c7497e8859364a84ec07be : SUCCESS in 4m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7e38e9b3015746949b443e70c1c57b51 : SUCCESS in 7m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/18099c95f8df4a25997abae19b88c5d1 : SUCCESS in 21m 28s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/5b3481c8824843ebaa8da2db9ccd9023 : SUCCESS in 54m 01s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/2616be56f0994642b0047df50ee3fe65 : SUCCESS in 56m 23s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/e6473ddd8680405280ec2d388d0293da : SUCCESS in 14m 45s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/7566bdd1f9f84cffa3fca1a9069d5ccf : SUCCESS in 1h 58m 36s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/eaf2d06209974db4a3040cc0aef8a8b2 : SUCCESS in 16m 18s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/7781d66cbb114df3bf10147dad511624 : SUCCESS in 27m 18s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/9a07064141604c939749b24a7ebf7138 : SUCCESS in 10m 07s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"41d3b7d0976d0f729689ac0133d22c91f682b7b8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-18 09:07:24.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"27091426ea94490151b96cfa8e210b857cdd1db5","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-12-18 10:00:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d39c14dadb0441441a0bec0ea291d9ef6ed885d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 10:00:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"434f60c3e07f09f798a2dc848f1a6508f44896b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 12:28:40.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91b66605a1d04e2ab94e6b18633fe7a0 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/56e5af61c6d74edb87a890570c693b03 : SUCCESS in 6m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/14104000403249999580df3667ffb96e : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d0a8d4964f54d1ca389dfa111690123 : SUCCESS in 6m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ffca341d5869420d988d3a1ef15f8567 : SUCCESS in 4m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cef3ac6039cb47358e26d62a1ee13002 : SUCCESS in 4m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/81b64976cd39474fb98e66ca1c40ab0e : SUCCESS in 7m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e543288b58334c63a08854a2ee454a12 : SUCCESS in 18m 27s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/f06fd61d421d47d7a81949970652c33b : SUCCESS in 1h 15m 49s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/b2f4abbcde7146c7ae2fc6cdcaa9e945 : SUCCESS in 58m 28s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/dc18e5e4c32342a097a6bb13c305609e : SUCCESS in 16m 23s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/14f56ec827d84f8c9f9a060166f456c3 : SUCCESS in 2h 20m 55s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/866f4ed781b84a769c8c7a556229712e : SUCCESS in 16m 06s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/519b462d56da4a7bb2b4fc4bd197f4a4 : FAILURE in 1h 12m 01s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/7344bd2ce25b445ba1210980839c7b9f : SUCCESS in 9m 50s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"5aeb4179bd1a3c2960a05fe90a91c8c332369a56","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-18 12:35:49.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c5a69df3bbd281b6df51890263143992ad9e7d0d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-18 12:36:27.000000000","message":"Patch Set 2:\n\nNeeds https://review.opendev.org/c/openstack/kayobe/+/764182. Roll into this patch if it keeps failing.","accounts_in_message":[],"_revision_number":2},{"id":"ef3c14d411edaa949ba9d42ea796f87d15360c95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 14:22:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/601341febe1445cdb2f868490abe477d : SUCCESS in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c78cd1e98f84abe847ceed339b637b5 : SUCCESS in 7m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ba17192713d4e54889f82ef643ff2a9 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bbd6e374f55e4573aaa9d76d7ebbae93 : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74a130e3f8324c15ab1e6912d5b021e8 : SUCCESS in 5m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/23ea64346e044dcaa87ba53ccc5f3726 : SUCCESS in 4m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bc54e34208b740958848bc3c3e5d492e : SUCCESS in 4m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/71f7c5cbd2724a42a7033831fed0c431 : SUCCESS in 8m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2078b0858c4c4f5392bc6cf4e047af34 : SUCCESS in 20m 23s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/5ba245755d564f10b279d0640bc26787 : SUCCESS in 59m 19s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/64cab01cac6f43839a6831ddb034e1e0 : SUCCESS in 1h 06m 46s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/2471a869de194c9a9c9bd00c3ec83d0b : SUCCESS in 15m 49s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/3a6f012f75134f60917c0179a0830253 : SUCCESS in 1h 44m 51s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/fede54030c5444edab28e25c61112667 : SUCCESS in 15m 45s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/53346159394247e99d345b54b2627788 : SUCCESS in 25m 01s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/9daa8432b1774c349d34be0df097a6b5 : SUCCESS in 10m 23s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":2},{"id":"d5364878716867734af63e88555594a3fc5ce5a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 14:22:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"796b335d665fb07c9f43d77917de122943ac9a4d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 16:53:51.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ebd3123d9b44873b6f96bc3e2388c8e : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5a0f32e95f74bcf8b5577755b72c452 : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f14e1b737cf4b92b91f5cb8cd7bbfc1 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f14947f632b412c8209c53818d801c9 : SUCCESS in 6m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a4327f16bab4b11920e2e51f097866a : SUCCESS in 3m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/de1937f8ac794d88a90c75592b04f49d : SUCCESS in 4m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/68ad6ac7369a4bdc87fd5b193dca9410 : FAILURE in 7m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ef1f8ac5096e4a09b46d3a7c68cc30d4 : SUCCESS in 16m 54s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/bbdde08ccb484b618fe69be2ec07c931 : SUCCESS in 58m 23s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/f88da824cea24980b5593356a6a1f488 : SUCCESS in 1h 00m 27s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/3752f465ed2f4f61bd08dfa5a2164808 : SUCCESS in 15m 01s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/fb09154bc37249a8bf9eb4040492a789 : SUCCESS in 1h 46m 00s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/3dedb98c61ae4ef3ac1b9ef68bec5e3a : SUCCESS in 16m 03s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/28fbca4c022b4ac0b0e4d17268cbf4cf : SUCCESS in 23m 55s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/62155faf31c2465bb9f030100beb6574 : SUCCESS in 9m 59s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":2},{"id":"d7152053518d002a134786dd7121be1b8cca9478","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-18 17:17:54.000000000","message":"Patch Set 2:\n\nFailure in kayobe-tox-ansible looks like a race condition in the test. Fix proposed: https://review.opendev.org/c/openstack/kayobe/+/767919","accounts_in_message":[],"_revision_number":2},{"id":"94d9cade089bebe176652aea41dd2b407a8ebae6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-18 17:17:59.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"b40d9cb1850595a9f4a0a65d69f70f05cb65137a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 19:30:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dcfbbcf8321b45379030d91c4cac1f69 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2df99e4362174b9eb983b273e6c3afd5 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2eeecfa3a3354ed38c010011db380658 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/68ad777a8ef949a5a6a91d3bc21e9820 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b46c871d1d504ab7b35a4e92d0f22342 : SUCCESS in 8m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/958cf12cc39249fc8b728912338fe737 : SUCCESS in 5m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c68df78a80e745daaf1bc8dd1f4581e2 : SUCCESS in 5m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1768fc6adb6b4d6aa5e738416e81addc : SUCCESS in 10m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f6f6ebb2f9ac48229cecb7edef7fb689 : SUCCESS in 23m 28s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/7c6f536e80bb425eaac50dde23d7a7c1 : SUCCESS in 1h 05m 03s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/bf5c5395a102451fbfe11d41ec00ddbb : SUCCESS in 1h 11m 14s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/ed39ae861fdf43bcac517bdfe321aab3 : SUCCESS in 14m 41s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/fba56510830745a69287f02dffdd046a : SUCCESS in 2h 02m 43s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/0dedcecc7d124ad382c2ad91f7547b51 : SUCCESS in 17m 18s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/87474dcc255b40d8b803c9a437435b10 : SUCCESS in 23m 02s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/0ca598ccab644bdcba38154dcfe0a980 : SUCCESS in 10m 14s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":2},{"id":"2ccdc45e75a7776f0345d42be3550ccafd49fddf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 19:31:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f54edbe263fd3d9073bd2594f8029cc1d5644b1a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 22:00:33.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d4785ae99504ca887edb091f1f73c32 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/78d2d1d251164f53a340a512e135f1b4 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/81c0dedfc37f46838788cb0912748423 : SUCCESS in 8m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10464958deea4805bcf575b3a4713e96 : SUCCESS in 18m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4e976ee426a342de99737f283ed24ca6 : SUCCESS in 3m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e3a2b3124c304a79925ea70e951ba095 : SUCCESS in 4m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8ffb6c92a29b4879ace30e3b365fff37 : SUCCESS in 10m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4fbbaed34cfa42cb8f086846a333f1a6 : SUCCESS in 17m 50s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/bec1b7c625c04c3aa66c0d197eea94bd : SUCCESS in 55m 30s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/35a03583dad744e2956910994d3edd4f : FAILURE in 1h 49m 00s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/d685118ce0744e568008a50667088841 : SUCCESS in 11m 53s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/2564d95ec01e4b8789e2d359a75c911d : SUCCESS in 2h 22m 08s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/44885337bbed4224989d0a815b76c0fa : SUCCESS in 16m 00s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f0c476fcf8714ebcad53acdd79df7168 : SUCCESS in 26m 18s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/a5e3988efa754f0c833b8e6758cdbc05 : SUCCESS in 10m 11s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":2},{"id":"a48a83ead527506c2fa0363d4432a4da6db86e7c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-21 10:15:34.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"cbd3869a1a568278b868fd16098e80a254fadddb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-21 12:12:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/96078cf4ce5044f6924f1d8f8eca60c1 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5026cad8cc045bea3bcff6ba9f0a8a3 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f22941cb5e2415794dbc35d8b81cdca : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/175c802c35d04d8088dde798f15d091e : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b70a6212d0e413c913ebf4f4684f2f5 : SUCCESS in 6m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52a6d88fe82c4c7b9455b9b6538a0035 : SUCCESS in 3m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/29069f7179ca4650bef75863a58b9d10 : SUCCESS in 4m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d67f0b62350a4d1ba4d6245dd3f283da : SUCCESS in 14m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/87a56247102c48d9a8247185907c33ea : SUCCESS in 27m 00s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/4951053543534f768d8f4fb0f33098a3 : SUCCESS in 50m 25s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/b7dd6f5a3f2b47afb638b6f7831eab6b : SUCCESS in 1h 13m 11s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/216398ab543b4f9c86d17618d4fb57c6 : SUCCESS in 21m 34s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/cca8d0d1e9fa438595161eb45e503e54 : SUCCESS in 1h 49m 54s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/906e13c2539f46f6a2255777260f5849 : SUCCESS in 14m 39s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e5f041dfbd9a4c9ea28da93c25999ed7 : SUCCESS in 26m 13s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/6357a9aed0a347acb2e78bd5e13dee11 : SUCCESS in 10m 14s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":2},{"id":"c6d0e64b5aabc9302920e58683bfb03a41cd505d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-21 12:13:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b7cda9327a9cfa5b075bc741d404390e2427fd59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-21 14:20:06.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdd3473af5404abfb8ef594a124f81c0 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc1320d0740f4a6fbb05233a3131a477 : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa21f7c44dae440883b478bdc645f2e8 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2587ec6db94f4ed7870ce200c72ccfc7 : SUCCESS in 7m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0fc970282d7a4368ba8b6515010c1577 : SUCCESS in 3m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cd3093facee1431c91ede35367c4e787 : SUCCESS in 4m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e25574dce47b4bb1b49f67f8bcdf6d8f : SUCCESS in 8m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7ddbac5dbc6d4617b94b67f15a91d3cf : SUCCESS in 18m 01s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/a0dba387caae47ea9161775ece7a3539 : SUCCESS in 55m 57s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/57b1c38ddfaf4748b9a6031f9d05a8f5 : SUCCESS in 1h 04m 12s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/dfac7774ceb3495ea92f0441ca7f0d1f : SUCCESS in 15m 01s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b650f8362fdf405cb740d71fe6bb150e : SUCCESS in 2h 01m 31s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/a976c5e4143c419a8a24e2eb1f533a3e : SUCCESS in 16m 34s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/011c0f954f5f4fd1aeabc9167319248e : FAILURE in 1h 10m 27s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/a09d7c663c8e4165b3f0de4d4df0c9e6 : SUCCESS in 11m 01s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":2},{"id":"38c519fde33e2631201e356364159794bef6f43b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-21 14:23:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a7ca1aba5d934ecc6c90adee775b012c41189e17","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-21 14:39:19.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"1b79bca85343a2ec68ae5520fab5a1d2edee9c1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-21 16:28:43.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c6b0af5f985442fb831bcc84fd6023b6 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88e54e9f66e04c7595324761cbefe5b6 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d08dff983ca49669333060b8609441a : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49e2e0a424f445688248e2853720dcd9 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8524c296fd5e44e3b81517428b8ca8b8 : SUCCESS in 6m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8252c22aaad2424b9a6237bdca026c47 : SUCCESS in 3m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/279aba48dfd547b5aa9b5132662b8bc1 : SUCCESS in 4m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f4bd0aa94ed54bfa8548319afcb60888 : SUCCESS in 7m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/75189ffb0219417aa97ee37dc5e18787 : SUCCESS in 19m 07s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/1926de93a83847148fb8420f5c944888 : SUCCESS in 1h 01m 59s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/ccdc08b76ad242d2896a093020fb9227 : SUCCESS in 1h 52m 11s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/891dafbd4ce247c08c9a6720f91b20bf : SUCCESS in 12m 17s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/3689e19cbb944629b9baa97a27d90967 : SUCCESS in 2h 02m 26s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/64857bddc1be4f59ac4dd6b2b324f96a : SUCCESS in 20m 17s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/689929bd2b054da3b6f6a5303a8af5a4 : SUCCESS in 23m 59s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/9ffa7150aade41908d420788b20a9f29 : FAILURE in 20m 56s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":3},{"id":"61cf0e52b1aed7639f03855c1f1be9c9aaf78900","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-21 16:46:34.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"5ed7c06fb9d759b9b84931535287b133e36985d9","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-12-21 16:46:45.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"d3d103c99388df53d14ebf29f517d14125c438a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-21 18:50:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/48b9b85e75544dc2a3e5c7b658a66747 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27fccdac232247ec8478794753bd8193 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b81d33502b774066b604d02e46be0e7e : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0858acc8c1ee4d42b38c3ec65fb85557 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e085c733a9f46c99f84048aed2fcd4a : SUCCESS in 11m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f0e97df6d8f64a82b977b38d73bfe77c : SUCCESS in 3m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/00580e797c8a4f26b1242809679bc914 : SUCCESS in 5m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f46727f37d7948ac8b44d9b1638c3333 : SUCCESS in 8m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ff06ebe1baf04e3baf78248dd38db185 : SUCCESS in 19m 13s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/d83409f839be44479f57f5a295ec929c : SUCCESS in 59m 43s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/3f871f55728944d0a4ff79a06e0c3aad : SUCCESS in 1h 00m 59s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/e542053e5c574367906af1eb9322d86f : SUCCESS in 16m 08s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/37cf94da419a4a7b83e7776b9e2dc0dc : SUCCESS in 1h 59m 17s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/b171a9133c7b4c75a206f986efcc7e93 : SUCCESS in 15m 01s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/c7debad8f50d418eaf0cf47e91558fd4 : SUCCESS in 31m 03s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/386aea0b937c4825b2cf337dea21b135 : SUCCESS in 12m 11s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":3},{"id":"26efcc3d745b8455999b2cb58a207ecec5927754","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-21 18:50:56.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3251b26bb66c877c0b945b03ee99fb2e6645a32f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-21 20:38:51.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/953d5cca6a0c43b0a45b806f476594c9 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/25e80e6ade204e6c83a6313c9f188dd9 : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/08f3960f64b847e09b2caffd03e00c54 : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0586fb72e5e4406db5ef21eb6e2879ee : SUCCESS in 6m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aefec893de05457aa921cd8a4847ac67 : SUCCESS in 2m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/165b34cbd9bb49d199ed2d618c5c1b4b : SUCCESS in 4m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d6db51714c7a4899afa3fe2373ea662a : SUCCESS in 8m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1b257b44f193485f9771bfcc4c4c88a3 : SUCCESS in 17m 08s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/6d010af13f284188beb6dcbb7874981e : SUCCESS in 1h 03m 32s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/ddd0b68894ca48d1b079df7b101ee43f : SUCCESS in 1h 05m 06s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/67aa8012e0a24a9ebec73bbd6eb53576 : SUCCESS in 14m 26s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/d8e69ec5d5ab4448a2b92daedccc3327 : SUCCESS in 1h 44m 35s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/c514e233c80b47b08e0e775066ba9b89 : SUCCESS in 17m 04s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a53f7cf647674dfc8fabb72c7d146922 : SUCCESS in 23m 59s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/2f6a06b90348449ea38aad2e72639273 : SUCCESS in 11m 52s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":3},{"id":"42fdf8c0dcf7ce796916422ac8e15f3abf321cd0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-21 20:38:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"b6bef8fbdf568313e0fe27fc2b4cede8677da2d8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-21 20:40:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e497f5070ba4cd086aa67fc2b35af17 : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ccc753894dea42eea8654238ac53764d : SUCCESS in 52s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"885dd064f48bbe5c72b921c972cc11ff85ac7858","revisions":{"c3cd975ebcb6bbabf7d1c89baba9ce6f2c4653ef":{"kind":"REWORK","_number":1,"created":"2020-12-17 17:54:12.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/97/767597/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/767597/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/767597/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/767597/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/767597/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/767597/1"}}},"commit":{"parents":[{"commit":"b1be865d33fd52e9c969619f27aaa04a2c2a2a0c","subject":"Update TOX_CONSTRAINTS_FILE for stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b1be865d33fd52e9c969619f27aaa04a2c2a2a0c"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-09 11:26:40.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-17 17:54:01.000000000","tz":0},"subject":"Fixes for CentOS 8.3","message":"Fixes for CentOS 8.3\n\n* Bump stackhpc.libvirt-host to v1.7.1. On seed-hypervisors installed\n  using CentOS 8.2 or earlier, interaction with libvirt may fail due to\n  libgcrypt being incompatible. See\n  https://github.com/stackhpc/ansible-role-libvirt-host/issues/42\n\n* Bump MichaelRigart.interfaces to v1.9.2. The CentOS 8.3 cloud image\n  includes an ifcfg-ens3-1 file. See\n  https://github.com/michaelrigart/ansible-role-interfaces/pull/93\n\n* Previously a second libvirt daemon was installed by Tenks on the host,\n  however changes in libvirt 6.0.0 to separate libvirtd into multiple\n  daemons do not allow for customisation of the PID files used by the\n  new daemons. This leads to a conflict between the container and host\n  daemons. Update the Tenks config to use the containerised Nova libvirt\n  daemon. This depends on a change to the stackhpc.libvirt-host role:\n  https://github.com/stackhpc/ansible-role-libvirt-host/pull/44\n\n* Not CentOS 8.3 related, but tox jobs are now failing on python\n  dependencies. Remove upper limits from docker and paramiko.\n\n* Not CentOS 8.3 related, but Bifrost has enabled authentication by\n  default. We are not ready to support this, so override it.\n\nStory: 2008429\nTask: 41378\n\nChange-Id: Ie8fd965165e8d347d27528a2c16d0647e412ccdc\n(cherry picked from commit 4398856ec80ae765e805df9aab7d6f2105a34009)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c3cd975ebcb6bbabf7d1c89baba9ce6f2c4653ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c3cd975ebcb6bbabf7d1c89baba9ce6f2c4653ef"}]},"branch":"refs/heads/stable/victoria"},"5a360509b205dfdcc92fca8c437ba10248bb3406":{"kind":"REWORK","_number":2,"created":"2020-12-17 17:55:36.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/97/767597/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/767597/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/767597/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/767597/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/767597/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/767597/2"}}},"commit":{"parents":[{"commit":"b1be865d33fd52e9c969619f27aaa04a2c2a2a0c","subject":"Update TOX_CONSTRAINTS_FILE for stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b1be865d33fd52e9c969619f27aaa04a2c2a2a0c"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-09 11:26:40.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-17 17:55:19.000000000","tz":0},"subject":"Fixes for CentOS 8.3","message":"Fixes for CentOS 8.3\n\n* Bump stackhpc.libvirt-host to v1.7.1. On seed-hypervisors installed\n  using CentOS 8.2 or earlier, interaction with libvirt may fail due to\n  libgcrypt being incompatible. See\n  https://github.com/stackhpc/ansible-role-libvirt-host/issues/42\n\n* Bump MichaelRigart.interfaces to v1.9.2. The CentOS 8.3 cloud image\n  includes an ifcfg-ens3-1 file. See\n  https://github.com/michaelrigart/ansible-role-interfaces/pull/93\n\n* Previously a second libvirt daemon was installed by Tenks on the host,\n  however changes in libvirt 6.0.0 to separate libvirtd into multiple\n  daemons do not allow for customisation of the PID files used by the\n  new daemons. This leads to a conflict between the container and host\n  daemons. Update the Tenks config to use the containerised Nova libvirt\n  daemon. This depends on a change to the stackhpc.libvirt-host role:\n  https://github.com/stackhpc/ansible-role-libvirt-host/pull/44\n\n* Not CentOS 8.3 related, but tox jobs are now failing on python\n  dependencies. Remove upper limits from docker and paramiko.\n\nStory: 2008429\nTask: 41378\n\nChange-Id: Ie8fd965165e8d347d27528a2c16d0647e412ccdc\n(cherry picked from commit 4398856ec80ae765e805df9aab7d6f2105a34009)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5a360509b205dfdcc92fca8c437ba10248bb3406"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5a360509b205dfdcc92fca8c437ba10248bb3406"}]},"branch":"refs/heads/stable/victoria"},"885dd064f48bbe5c72b921c972cc11ff85ac7858":{"kind":"REWORK","_number":3,"created":"2020-12-21 14:23:45.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/97/767597/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/767597/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/767597/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/767597/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/767597/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/767597/3"}}},"commit":{"parents":[{"commit":"b1be865d33fd52e9c969619f27aaa04a2c2a2a0c","subject":"Update TOX_CONSTRAINTS_FILE for stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b1be865d33fd52e9c969619f27aaa04a2c2a2a0c"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-09 11:26:40.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-21 14:22:51.000000000","tz":0},"subject":"Fixes for CentOS 8.3","message":"Fixes for CentOS 8.3\n\n* Bump stackhpc.libvirt-host to v1.7.1. On seed-hypervisors installed\n  using CentOS 8.2 or earlier, interaction with libvirt may fail due to\n  libgcrypt being incompatible. See\n  https://github.com/stackhpc/ansible-role-libvirt-host/issues/42\n\n* Bump MichaelRigart.interfaces to v1.9.2. The CentOS 8.3 cloud image\n  includes an ifcfg-ens3-1 file. See\n  https://github.com/michaelrigart/ansible-role-interfaces/pull/93\n\n* Previously a second libvirt daemon was installed by Tenks on the host,\n  however changes in libvirt 6.0.0 to separate libvirtd into multiple\n  daemons do not allow for customisation of the PID files used by the\n  new daemons. This leads to a conflict between the container and host\n  daemons. Update the Tenks config to use the containerised Nova libvirt\n  daemon. This depends on a change to the stackhpc.libvirt-host role:\n  https://github.com/stackhpc/ansible-role-libvirt-host/pull/44\n\n* Not CentOS 8.3 related, but tox jobs are now failing on python\n  dependencies. Remove upper limits from docker and paramiko.\n\nThis change also includes the following backport:\n\nCI: add dummy1 port for seed upgrade and VM jobs\n\nFor some reason this configuration was omitted or incomplete in these\njobs. Without this configuration, Kayobe development configuration adds\neth1 to the breth1 bridge if it exists, or leaves the bridge without a\nport otherwise. This can lead to network connectivity issues, if the IP\naddress of eth1 becomes inaccessible when it is added to breth1.\n\nThis change fixes the issue by ensuring that the dummy1 interface\nexists, and configuring the aio network to add it to breth1 as a port,\ninstead of eth1.\n\nBackport Change-Id: I47115e5d8d1fb448cb9e46dc43b56c64987e428b\n(cherry picked from commit 0766df1673da0cc8c3a3d3d03e357ec5836e2d94)\n\nStory: 2008429\nTask: 41378\n\nChange-Id: Ie8fd965165e8d347d27528a2c16d0647e412ccdc\n(cherry picked from commit 4398856ec80ae765e805df9aab7d6f2105a34009)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/885dd064f48bbe5c72b921c972cc11ff85ac7858"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/885dd064f48bbe5c72b921c972cc11ff85ac7858"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
