)]}'
{"id":"openstack%2Fkayobe~767599","triplet_id":"openstack%2Fkayobe~stable%2Fussuri~Ie8fd965165e8d347d27528a2c16d0647e412ccdc","project":"openstack/kayobe","branch":"stable/ussuri","hashtags":[],"change_id":"Ie8fd965165e8d347d27528a2c16d0647e412ccdc","subject":"Fixes for CentOS 8.3","status":"MERGED","created":"2020-12-17 18:00:39.000000000","updated":"2020-12-23 01:27:04.000000000","submitted":"2020-12-22 13:56:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"767599","meta_rev_id":"eac89fe7acf2ea0b3d281314e8f95077b3968d03","_number":767599,"virtual_id_number":767599,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-22 13:56:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-12-22 12:14:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-12-23 01:27:04.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-12-22 12:14:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-17 19:05:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-18 06:36:31.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-12-23 01:27:04.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"}],"messages":[{"id":"142b5240c4b77b9f5f35500827d6dcdd7751d645","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-17 18:00:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a19e34ea9933be8fc6970c0e768312988941643e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-17 18:03:19.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"6a584a0c1641431381fecb02f4781e3cf0bc54d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 19:05:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c5b62f580b0d48e8833d4b1ca4d06a5e : FAILURE in 2m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/823efbddafd54f0d89cf9cba0c0f0a62 : FAILURE in 2m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/831738e8609a4c0185712d4abda5fc86 : FAILURE in 2m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3f6a1a39bee947c7b764760b62cc926c : FAILURE in 2m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fae456b9ba54d36b6627df8bc78d08d : SUCCESS in 6m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4f05f188bf454d5ba758a83011899da2 : SUCCESS in 3m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e610c6dae9324881b458cb187ec3830c : FAILURE in 2m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ed2a89f495ff4db2a23677940b1700bc : FAILURE in 3m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9c90f8ae10fc412796c69049db8ebd47 : FAILURE in 4m 52s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/62ae4a2710cd4000beeda7719c6edb21 : SUCCESS in 1h 00m 50s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/848cd335924a4779a060d1684a6b8049 : SUCCESS in 17m 37s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/701b4fcf750e47dbbcf4cd227f76f563 : FAILURE in 55m 11s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/24cb10bcdad8406bbf2bf7e3cf6dcb84 : SUCCESS in 17m 57s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/3fd6d86b8fdf4821b605a6d5a1f5c54d : SUCCESS in 28m 09s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/fb0c1a3b7f404f4cb350b96337bd4bff : SUCCESS in 10m 19s","accounts_in_message":[],"_revision_number":2},{"id":"3c0d4597f6fff8c2c4e736edf4914be29632aad7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-17 19:23:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"66bc574b0ed2059b21b69b99993fe3ea53354029","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 20:38:44.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1dd6e0d5575740daa0c7cffa05a405e0 : FAILURE in 8m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de332db6928a47d094074eaf47612e19 : FAILURE in 8m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ea61217acb549dab8fd76c9f63c122d : FAILURE in 9m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cdca011c3cae4664ae4227ab1dab3259 : FAILURE in 8m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b78455ddb044496baedef724a59cbdd : SUCCESS in 5m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e693b3ba1498462e9b70f15dc1eba9ce : SUCCESS in 3m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7d801b12da704756961f065ac3870f8a : FAILURE in 17m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4588e81c1aaa44a8bb12fa814e752763 : FAILURE in 8m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a6e9b6acbfd6473b9f791c3a10b5838e : FAILURE in 8m 28s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/6d92c1cab6314ab9b48a5258440fde93 : SUCCESS in 1h 04m 59s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/2f10aba9546c4d3a898d8d8a8f55426a : SUCCESS in 17m 24s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/af00286159e841c49a95b798ecc7cbe7 : FAILURE in 50m 37s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/0b8228ead9634a2fa7b6a20b26bbe307 : SUCCESS in 19m 31s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/725d6785ff5a41d1a9bb91cfda15b183 : SUCCESS in 28m 22s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/6eafc5d4b10146fb8a159f048d7e90c3 : SUCCESS in 10m 52s","accounts_in_message":[],"_revision_number":3},{"id":"8f7dcc40a759a9f25afcf2a2f20c0ca80e582e91","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-12-18 06:36:31.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ef6644b337444b13c5494c17abbfc44880029474","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 07:51:31.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/48c9449b03ad4b7190c80e7e70cf1c6a : SUCCESS in 7m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5ea0bec0d2a425cb19a8dd9c2854580 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ceb29e3f330c41d3915ec0db7e9ad938 : SUCCESS in 4m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/60c9b8671f144d2fbce996c91aff7a27 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18173e725dd54ef3a10088a1703bad1a : SUCCESS in 18m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/854a06e035444d0791b2d2604d656d8a : SUCCESS in 5m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e0e9a5cd1c71451eb6d5ea812fca892b : SUCCESS in 7m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ad803cad67e44cd8964b5ba378416139 : SUCCESS in 8m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d7dcc91a58ea41748a799895e3814a80 : SUCCESS in 18m 28s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/bd361a9dd034475985698ced21b85f01 : SUCCESS in 1h 05m 54s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/e384aa3bda5a4ffd8a5371874a9ec924 : SUCCESS in 21m 29s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/653cf179c768491babdd5bcde7f1c171 : FAILURE in 1h 07m 33s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/a8302477024642aeb4a173ba773d69d7 : SUCCESS in 26m 24s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/c519f71c0d4942679be4c23e1358eadd : FAILURE in 14m 18s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/5b11a4b1da1c44fa9fa573bbd5a7f7b7 : SUCCESS in 16m 20s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":4},{"id":"8410c5218663a0d9b28b8629d7c7997e2a6ece7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-12-18 08:54:46.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"77242bc1e7d5e5a77e26be97eb8ba55b787e50b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 10:05:26.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/17931331f08b4faabb66128c386dec79 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a7695737dc94cb389e60d60bce1fa93 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/45196d371ee04e0f9455ac9d5c5aedf7 : SUCCESS in 6m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ca791e96f43c4465b56e996286c6e484 : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b01632df8764a0096fbc743cb215829 : SUCCESS in 7m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b28744869b34224a0ab15b99ac5dd5a : SUCCESS in 3m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/eec6ea62821f480181dbfe433868aeac : SUCCESS in 4m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b6ce8645ce7145b3bae491b0aa52d7f6 : SUCCESS in 8m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8a0c662219a74feeb84e4f06d9a9c433 : SUCCESS in 16m 32s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/821d396799704a278c3edee9034cbc66 : SUCCESS in 1h 02m 04s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/dbbf120d598b4fd68222cef31a0662fd : SUCCESS in 17m 42s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/85b64125f9b14a20b89d6a661a5bcf5d : FAILURE in 48m 56s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/eae99994937f4f59bf5bfe7c886d2737 : SUCCESS in 15m 43s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/fc6a1ba089ce4740a4b56b7ac11accce : SUCCESS in 25m 13s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/4509b9acd85645368e1f8d0d5994c32a : SUCCESS in 10m 42s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":5},{"id":"13cf9c1137fccfe3fc529d20f36f936a28884e94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-18 10:21:16.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"56aade99d6b5106944c7a04339b882d79a33feb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 11:38:39.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/976e80dc53ea4b06bca8657a90746c69 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8adf39cfb4674b6c80cc2e0846936d5f : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/871b2257a8c4462eab93a4d8b01ddf12 : SUCCESS in 4m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e4374c63a3734da7b91a36796a446007 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2115327738a4f52a6e40350f4ae2438 : SUCCESS in 6m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/465daec85561411d891c4759e17f5190 : SUCCESS in 3m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2e8f66c331bb458ba0f45af1a9c1edf2 : SUCCESS in 4m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cda0d7643187432a94599f465079ec2f : SUCCESS in 9m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d910beb747de4f2fada5cdd8b5d31c6e : SUCCESS in 18m 00s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/3745bf5a58614ce6ad76c9fedf9afdcf : SUCCESS in 1h 06m 55s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/ff9e7632fe9248baa8ce862c9d30af7a : SUCCESS in 14m 05s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/965bc36fcb15407da5d67b8fe4dd7d6e : FAILURE in 50m 08s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/7594dcb8d1fd43baac98646337a999c9 : SUCCESS in 17m 15s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/72341dfaf22d413ea3a88a8234cd3a88 : SUCCESS in 28m 07s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/d03935dd73c140048413911fcc17db4c : SUCCESS in 10m 00s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":6},{"id":"8ed557e64055c73e183906ab10a79f7d40eb10c2","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-12-18 12:02:52.000000000","message":"Patch Set 6:\n\nFailed to connect socket to \u0027/var/run/libvirt-tenks/libvirt-sock\u0027: No such file or directory","accounts_in_message":[],"_revision_number":6},{"id":"5e41dd802074972344990e296fe6acf4240a0bac","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-18 12:23:59.000000000","message":"Patch Set 6:\n\n\u003e Patch Set 6:\n\u003e \n\u003e Failed to connect socket to \u0027/var/run/libvirt-tenks/libvirt-sock\u0027: No such file or directory\n\nDoh. Of course.","accounts_in_message":[],"_revision_number":6},{"id":"d869104e1ae701bdc56f655f6d67f600c85da6e9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-18 12:24:40.000000000","message":"Patch Set 6:\n\n\u003e Patch Set 6:\n\u003e \n\u003e \u003e Patch Set 6:\n\u003e \u003e \n\u003e \u003e Failed to connect socket to \u0027/var/run/libvirt-tenks/libvirt-sock\u0027: No such file or directory\n\u003e \n\u003e Doh. Of course.\n\nPossibly we need to merge train first then add a dependency.","accounts_in_message":[],"_revision_number":6},{"id":"5709640b18064874d758b2d8daa3d22a6bd67e2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-22 09:05:48.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"8a98bc7c584c7802193a7ec5502dc1c6f1553f2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 11:10:09.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b283694012ff40e89b11104ab0ec0085 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eba251f8189942678b6e0229459532de : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/667fde1242d14c74bfddcbe1c4d6dfb8 : SUCCESS in 4m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/956d547f09b04b81b2d688e2fe4b26a7 : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb6dbf4e60264648b1b861db727958d5 : SUCCESS in 6m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d3e3289e5d5426eb0b58d3a66bc3389 : SUCCESS in 3m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e472e34610b24003b8eaae126fdccba4 : SUCCESS in 4m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/db631fe14e9b45de84763cece942a5be : SUCCESS in 9m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/809033c5df424b1c984282779314eb8e : SUCCESS in 18m 22s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/8dab04249fe744e6a42e85ab5b51302f : SUCCESS in 1h 12m 07s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/fb8c9670d08e4bdba92c161196fc3fcf : SUCCESS in 15m 05s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/318d28b78d8c40d7beb1c5f29e063546 : SUCCESS in 1h 56m 54s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/19fd4a54f19a4b88bcba67754b7420db : SUCCESS in 15m 28s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b559153686794827b5c7f0e69b762b00 : SUCCESS in 28m 07s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/5b10a8ad688b4958967df675bf914bf0 : SUCCESS in 10m 10s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":7},{"id":"2356e8917d46382b5195c8b11168c09b026346f3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-22 12:14:02.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\nApproving for green CI","accounts_in_message":[],"_revision_number":7},{"id":"a9591c129d7072c08a9fb467b5ea80705425a391","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 12:14:16.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"7db57b9b25ccc8ad951a70fe05160ab03de48095","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 13:56:07.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/573b309e393a4d068595bbde59d85279 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2116620a043245a1a5b6e0370d21b1de : SUCCESS in 5m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/12e491b2756848dea16e40a25c76bbfe : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea203dd3a21c4d5fbeda4e8549e8d58f : SUCCESS in 5m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a0b9eafe5089431a8f4504cd8a7bfc52 : SUCCESS in 3m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/16da48c5daf043ce93f385c7e8c3d3c3 : SUCCESS in 5m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/904f9483542845e38acb524f319843fc : SUCCESS in 9m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0aeee6861f084d48ab5673b066f46cb7 : SUCCESS in 18m 26s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/c969c27d5551491f8b367c5c340d62cb : SUCCESS in 1h 13m 42s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/ec37088f00674b1eb15601bd42043893 : SUCCESS in 19m 09s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/6541020a466c4287a508cd7e5443c4b9 : SUCCESS in 1h 40m 58s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/73e89da46ca743f5b270ee78c560f1ce : SUCCESS in 16m 51s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/54157aec56ee408e8fb622b7f94f47a6 : SUCCESS in 21m 58s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/ad40f4dc3cdc487a8e93ba56c45c79a6 : SUCCESS in 10m 32s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":7},{"id":"cc1fa201b4b34d2dcb7f2dbc44c3d96de4b15c08","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 13:56:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"bde9f5226b90d65ff73399f9428a468e613dd935","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 13:57:29.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ee30ee974964b308b448974d911cf61 : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5aeb57d80b044d6883e73771d8c9f380 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":7},{"id":"eac89fe7acf2ea0b3d281314e8f95077b3968d03","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-12-23 01:27:04.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"5cfd2ba5aef07e31efc6695a8f987a1b7391ba66","revisions":{"d2d9135e5ca2e23688cf999e88608e8b53f1903d":{"kind":"REWORK","_number":1,"created":"2020-12-17 18:00:39.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/99/767599/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/99/767599/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/767599/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/767599/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/767599/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/99/767599/1"}}},"commit":{"parents":[{"commit":"a8d5cc0ba9caf4ab348b16873c36e065ed244d68","subject":"Merge \"docs: Add link back to Train CentOS 8 migration page\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a8d5cc0ba9caf4ab348b16873c36e065ed244d68"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-09 11:26:40.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-17 18:00:21.000000000","tz":0},"subject":"Fixes for CentOS 8.3","message":"Fixes for CentOS 8.3\n\n* Bump stackhpc.libvirt-host to v1.7.1. On seed-hypervisors installed\n  using CentOS 8.2 or earlier, interaction with libvirt may fail due to\n  libgcrypt being incompatible. See\n  https://github.com/stackhpc/ansible-role-libvirt-host/issues/42\n\n* Bump MichaelRigart.interfaces to v1.9.2. The CentOS 8.3 cloud image\n  includes an ifcfg-ens3-1 file. See\n  https://github.com/michaelrigart/ansible-role-interfaces/pull/93\n\n* Previously a second libvirt daemon was installed by Tenks on the host,\n  however changes in libvirt 6.0.0 to separate libvirtd into multiple\n  daemons do not allow for customisation of the PID files used by the\n  new daemons. This leads to a conflict between the container and host\n  daemons. Update the Tenks config to use the containerised Nova libvirt\n  daemon. This depends on a change to the stackhpc.libvirt-host role:\n  https://github.com/stackhpc/ansible-role-libvirt-host/pull/44\n\nStory: 2008429\nTask: 41378\n\nChange-Id: Ie8fd965165e8d347d27528a2c16d0647e412ccdc\n(cherry picked from commit 4398856ec80ae765e805df9aab7d6f2105a34009)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d2d9135e5ca2e23688cf999e88608e8b53f1903d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d2d9135e5ca2e23688cf999e88608e8b53f1903d"}]},"branch":"refs/heads/stable/ussuri"},"c4feb1d95547889640868036e3681661876b4692":{"kind":"REWORK","_number":2,"created":"2020-12-17 18:03:19.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/99/767599/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/99/767599/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/767599/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/767599/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/767599/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/99/767599/2"}}},"commit":{"parents":[{"commit":"a8d5cc0ba9caf4ab348b16873c36e065ed244d68","subject":"Merge \"docs: Add link back to Train CentOS 8 migration page\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a8d5cc0ba9caf4ab348b16873c36e065ed244d68"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-09 11:26:40.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-17 18:03:14.000000000","tz":0},"subject":"Fixes for CentOS 8.3","message":"Fixes for CentOS 8.3\n\n* Bump stackhpc.libvirt-host to v1.7.1. On seed-hypervisors installed\n  using CentOS 8.2 or earlier, interaction with libvirt may fail due to\n  libgcrypt being incompatible. See\n  https://github.com/stackhpc/ansible-role-libvirt-host/issues/42\n\n* Bump MichaelRigart.interfaces to v1.9.2. The CentOS 8.3 cloud image\n  includes an ifcfg-ens3-1 file. See\n  https://github.com/michaelrigart/ansible-role-interfaces/pull/93\n\n* Previously a second libvirt daemon was installed by Tenks on the host,\n  however changes in libvirt 6.0.0 to separate libvirtd into multiple\n  daemons do not allow for customisation of the PID files used by the\n  new daemons. This leads to a conflict between the container and host\n  daemons. Update the Tenks config to use the containerised Nova libvirt\n  daemon. This depends on a change to the stackhpc.libvirt-host role:\n  https://github.com/stackhpc/ansible-role-libvirt-host/pull/44\n\nStory: 2008429\nTask: 41378\n\nChange-Id: Ie8fd965165e8d347d27528a2c16d0647e412ccdc\n(cherry picked from commit 4398856ec80ae765e805df9aab7d6f2105a34009)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c4feb1d95547889640868036e3681661876b4692"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c4feb1d95547889640868036e3681661876b4692"}]},"branch":"refs/heads/stable/ussuri"},"a00ac47225b558fdeb9c09210508c5e2fd13c6d9":{"kind":"REWORK","_number":3,"created":"2020-12-17 19:23:19.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/99/767599/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/99/767599/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/767599/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/767599/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/767599/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/99/767599/3"}}},"commit":{"parents":[{"commit":"a8d5cc0ba9caf4ab348b16873c36e065ed244d68","subject":"Merge \"docs: Add link back to Train CentOS 8 migration page\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a8d5cc0ba9caf4ab348b16873c36e065ed244d68"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-09 11:26:40.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-17 19:22:44.000000000","tz":0},"subject":"Fixes for CentOS 8.3","message":"Fixes for CentOS 8.3\n\n* Bump stackhpc.libvirt-host to v1.7.1. On seed-hypervisors installed\n  using CentOS 8.2 or earlier, interaction with libvirt may fail due to\n  libgcrypt being incompatible. See\n  https://github.com/stackhpc/ansible-role-libvirt-host/issues/42\n\n* Bump MichaelRigart.interfaces to v1.9.2. The CentOS 8.3 cloud image\n  includes an ifcfg-ens3-1 file. See\n  https://github.com/michaelrigart/ansible-role-interfaces/pull/93\n\n* Previously a second libvirt daemon was installed by Tenks on the host,\n  however changes in libvirt 6.0.0 to separate libvirtd into multiple\n  daemons do not allow for customisation of the PID files used by the\n  new daemons. This leads to a conflict between the container and host\n  daemons. Update the Tenks config to use the containerised Nova libvirt\n  daemon. This depends on a change to the stackhpc.libvirt-host role:\n  https://github.com/stackhpc/ansible-role-libvirt-host/pull/44\n\n* Not CentOS 8.3 related, but tox jobs are now failing on python\n  dependencies. Remove upper limits from docker and paramiko.\n\nStory: 2008429\nTask: 41378\n\nChange-Id: Ie8fd965165e8d347d27528a2c16d0647e412ccdc\n(cherry picked from commit 4398856ec80ae765e805df9aab7d6f2105a34009)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a00ac47225b558fdeb9c09210508c5e2fd13c6d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a00ac47225b558fdeb9c09210508c5e2fd13c6d9"}]},"branch":"refs/heads/stable/ussuri"},"17c52cdc76e9102069b57f362f12ad13184d3e85":{"kind":"REWORK","_number":4,"created":"2020-12-18 06:36:31.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/99/767599/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/99/767599/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/767599/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/767599/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/767599/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/99/767599/4"}}},"commit":{"parents":[{"commit":"a8d5cc0ba9caf4ab348b16873c36e065ed244d68","subject":"Merge \"docs: Add link back to Train CentOS 8 migration page\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a8d5cc0ba9caf4ab348b16873c36e065ed244d68"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-09 11:26:40.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-12-18 06:34:30.000000000","tz":60},"subject":"Fixes for CentOS 8.3","message":"Fixes for CentOS 8.3\n\n* Bump stackhpc.libvirt-host to v1.7.1. On seed-hypervisors installed\n  using CentOS 8.2 or earlier, interaction with libvirt may fail due to\n  libgcrypt being incompatible. See\n  https://github.com/stackhpc/ansible-role-libvirt-host/issues/42\n\n* Bump MichaelRigart.interfaces to v1.9.2. The CentOS 8.3 cloud image\n  includes an ifcfg-ens3-1 file. See\n  https://github.com/michaelrigart/ansible-role-interfaces/pull/93\n\n* Previously a second libvirt daemon was installed by Tenks on the host,\n  however changes in libvirt 6.0.0 to separate libvirtd into multiple\n  daemons do not allow for customisation of the PID files used by the\n  new daemons. This leads to a conflict between the container and host\n  daemons. Update the Tenks config to use the containerised Nova libvirt\n  daemon. This depends on a change to the stackhpc.libvirt-host role:\n  https://github.com/stackhpc/ansible-role-libvirt-host/pull/44\n\n* Not CentOS 8.3 related, but tox jobs are now failing on python\n  dependencies. Remove upper limits from docker and paramiko. Update\n  hacking which is requiring an old version of ansible-lint through the\n  new pip dependency resolver (cherry picked from commit\n  7b605851349a8c469d11d0f9daacc0f56474d23f).\n\nStory: 2008429\nTask: 41378\n\nChange-Id: Ie8fd965165e8d347d27528a2c16d0647e412ccdc\n(cherry picked from commit 4398856ec80ae765e805df9aab7d6f2105a34009)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/17c52cdc76e9102069b57f362f12ad13184d3e85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/17c52cdc76e9102069b57f362f12ad13184d3e85"}]},"branch":"refs/heads/stable/ussuri"},"89825e3c66fa2aa4c53270b4e336216c44aae403":{"kind":"REWORK","_number":5,"created":"2020-12-18 08:54:46.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/99/767599/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/99/767599/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/767599/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/767599/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/767599/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/99/767599/5"}}},"commit":{"parents":[{"commit":"a8d5cc0ba9caf4ab348b16873c36e065ed244d68","subject":"Merge \"docs: Add link back to Train CentOS 8 migration page\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a8d5cc0ba9caf4ab348b16873c36e065ed244d68"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-12-18 08:54:35.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-12-18 08:54:35.000000000","tz":60},"subject":"Fixes for CentOS 8.3","message":"Fixes for CentOS 8.3\n\n* Bump stackhpc.libvirt-host to v1.7.1. On seed-hypervisors installed\n  using CentOS 8.2 or earlier, interaction with libvirt may fail due to\n  libgcrypt being incompatible. See\n  https://github.com/stackhpc/ansible-role-libvirt-host/issues/42\n\n* Bump MichaelRigart.interfaces to v1.9.2. The CentOS 8.3 cloud image\n  includes an ifcfg-ens3-1 file. See\n  https://github.com/michaelrigart/ansible-role-interfaces/pull/93\n\n* Previously a second libvirt daemon was installed by Tenks on the host,\n  however changes in libvirt 6.0.0 to separate libvirtd into multiple\n  daemons do not allow for customisation of the PID files used by the\n  new daemons. This leads to a conflict between the container and host\n  daemons. Update the Tenks config to use the containerised Nova libvirt\n  daemon. This depends on a change to the stackhpc.libvirt-host role:\n  https://github.com/stackhpc/ansible-role-libvirt-host/pull/44\n\n* Not CentOS 8.3 related, but tox jobs are now failing on python\n  dependencies. Remove upper limits from docker and paramiko. Update\n  hacking which is requiring an old version of ansible-lint through the\n  new pip dependency resolver (cherry picked from commit\n  7b605851349a8c469d11d0f9daacc0f56474d23f).\n\nStory: 2008429\nTask: 41378\n\nChange-Id: Ie8fd965165e8d347d27528a2c16d0647e412ccdc\n(cherry picked from commit 4398856ec80ae765e805df9aab7d6f2105a34009)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/89825e3c66fa2aa4c53270b4e336216c44aae403"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/89825e3c66fa2aa4c53270b4e336216c44aae403"}]},"branch":"refs/heads/stable/ussuri"},"b86099c7ae9756c6bd3f51cb647d7d70583e778d":{"kind":"REWORK","_number":6,"created":"2020-12-18 10:21:16.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/99/767599/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/99/767599/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/767599/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/767599/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/767599/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/99/767599/6"}}},"commit":{"parents":[{"commit":"a8d5cc0ba9caf4ab348b16873c36e065ed244d68","subject":"Merge \"docs: Add link back to Train CentOS 8 migration page\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a8d5cc0ba9caf4ab348b16873c36e065ed244d68"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-12-18 08:54:35.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-18 10:21:13.000000000","tz":0},"subject":"Fixes for CentOS 8.3","message":"Fixes for CentOS 8.3\n\n* Bump stackhpc.libvirt-host to v1.7.1. On seed-hypervisors installed\n  using CentOS 8.2 or earlier, interaction with libvirt may fail due to\n  libgcrypt being incompatible. See\n  https://github.com/stackhpc/ansible-role-libvirt-host/issues/42\n\n* Bump MichaelRigart.interfaces to v1.9.2. The CentOS 8.3 cloud image\n  includes an ifcfg-ens3-1 file. See\n  https://github.com/michaelrigart/ansible-role-interfaces/pull/93\n\n* Previously a second libvirt daemon was installed by Tenks on the host,\n  however changes in libvirt 6.0.0 to separate libvirtd into multiple\n  daemons do not allow for customisation of the PID files used by the\n  new daemons. This leads to a conflict between the container and host\n  daemons. Update the Tenks config to use the containerised Nova libvirt\n  daemon. This depends on a change to the stackhpc.libvirt-host role:\n  https://github.com/stackhpc/ansible-role-libvirt-host/pull/44\n\n* Not CentOS 8.3 related, but tox jobs are now failing on python\n  dependencies. Remove upper limits from docker and paramiko. Update\n  hacking which is requiring an old version of ansible-lint through the\n  new pip dependency resolver (cherry picked from commit\n  7b605851349a8c469d11d0f9daacc0f56474d23f).\n\nStory: 2008429\nTask: 41378\n\nChange-Id: Ie8fd965165e8d347d27528a2c16d0647e412ccdc\n(cherry picked from commit 4398856ec80ae765e805df9aab7d6f2105a34009)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b86099c7ae9756c6bd3f51cb647d7d70583e778d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b86099c7ae9756c6bd3f51cb647d7d70583e778d"}]},"branch":"refs/heads/stable/ussuri"},"5cfd2ba5aef07e31efc6695a8f987a1b7391ba66":{"kind":"REWORK","_number":7,"created":"2020-12-22 09:05:48.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/99/767599/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/99/767599/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/767599/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/767599/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/767599/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/99/767599/7"}}},"commit":{"parents":[{"commit":"a8d5cc0ba9caf4ab348b16873c36e065ed244d68","subject":"Merge \"docs: Add link back to Train CentOS 8 migration page\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a8d5cc0ba9caf4ab348b16873c36e065ed244d68"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-12-18 08:54:35.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-22 09:00:28.000000000","tz":0},"subject":"Fixes for CentOS 8.3","message":"Fixes for CentOS 8.3\n\n* Bump stackhpc.libvirt-host to v1.7.1. On seed-hypervisors installed\n  using CentOS 8.2 or earlier, interaction with libvirt may fail due to\n  libgcrypt being incompatible. See\n  https://github.com/stackhpc/ansible-role-libvirt-host/issues/42\n\n* Bump MichaelRigart.interfaces to v1.9.2. The CentOS 8.3 cloud image\n  includes an ifcfg-ens3-1 file. See\n  https://github.com/michaelrigart/ansible-role-interfaces/pull/93\n\n* Previously a second libvirt daemon was installed by Tenks on the host,\n  however changes in libvirt 6.0.0 to separate libvirtd into multiple\n  daemons do not allow for customisation of the PID files used by the\n  new daemons. This leads to a conflict between the container and host\n  daemons. Update the Tenks config to use the containerised Nova libvirt\n  daemon. This depends on a change to the stackhpc.libvirt-host role:\n  https://github.com/stackhpc/ansible-role-libvirt-host/pull/44\n\n* Not CentOS 8.3 related, but tox jobs are now failing on python\n  dependencies. Remove upper limits from docker and paramiko. Update\n  hacking which is requiring an old version of ansible-lint through the\n  new pip dependency resolver (cherry picked from commit\n  7b605851349a8c469d11d0f9daacc0f56474d23f).\n\nThis also includes a backport of the following change:\n\nCI: add dummy1 port for seed upgrade and VM jobs\n\nFor some reason this configuration was omitted or incomplete in these\njobs. Without this configuration, Kayobe development configuration adds\neth1 to the breth1 bridge if it exists, or leaves the bridge without a\nport otherwise. This can lead to network connectivity issues, if the IP\naddress of eth1 becomes inaccessible when it is added to breth1.\n\nThis change fixes the issue by ensuring that the dummy1 interface\nexists, and configuring the aio network to add it to breth1 as a port,\ninstead of eth1.\n\nBackported Change-Id: I47115e5d8d1fb448cb9e46dc43b56c64987e428b\n(cherry picked from commit 0766df1673da0cc8c3a3d3d03e357ec5836e2d94)\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe/+/767600\n\nStory: 2008429\nTask: 41378\n\nChange-Id: Ie8fd965165e8d347d27528a2c16d0647e412ccdc\n(cherry picked from commit 4398856ec80ae765e805df9aab7d6f2105a34009)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5cfd2ba5aef07e31efc6695a8f987a1b7391ba66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5cfd2ba5aef07e31efc6695a8f987a1b7391ba66"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
