)]}'
{"id":"openstack%2Fkayobe~767705","triplet_id":"openstack%2Fkayobe~master~I1985efae7c18f55c3ff7c27c17d6242523904f3e","project":"openstack/kayobe","branch":"master","topic":"story/2004960","hashtags":[],"change_id":"I1985efae7c18f55c3ff7c27c17d6242523904f3e","subject":"CI: add Ubuntu overcloud deploy job","status":"MERGED","created":"2020-12-18 12:18:44.000000000","updated":"2021-03-04 14:30:47.000000000","submitted":"2021-03-04 14:22:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"767705-story/2004960","meta_rev_id":"e5de9b918d7e2365cbf36e7733b552f43458d386","_number":767705,"virtual_id_number":767705,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-04 14:22:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-03-04 11:12:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-04 14:30:47.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-03-04 11:12:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-18 12:19:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-18 12:32:50.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2020-12-21 16:48:17.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2021-02-19 14:19:23.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-03-04 11:12:11.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"885934a39600d8d1b17bce155e7643dd8e4860b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-18 12:18:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"afe4d60f0474b33360d1d6bb50cb553af4e36a09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 12:19:05.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kayobe on branch master.  The error was:\n\n  Job kayobe-overcloud-ubuntu not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      check:\n        jobs:\n          - kayobe-overcloud-ubuntu\n  \n      gate:\n        queue: kayobe\n        jobs:\n          - kayobe-tox-ansible-syntax\n          - kayobe-tox-ansible\n          - kayobe-tox-molecule\n          - kayobe-overcloud-centos8\n          - kayobe-overcloud-tls-centos8\n          - kayobe-overcloud-host-configure-centos8\n          - kayobe-overcloud-upgrade-centos8\n          - kayobe-seed-centos8\n          - kayobe-seed-upgrade-centos8\n          - kayobe-seed-vm-centos8\n\n  in \"openstack/kayobe/zuul.d/project.yaml@master\", line 2, column 3","accounts_in_message":[],"_revision_number":1},{"id":"d59e7f15cb6ce00133409fd6461ca6aa1e44bea3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-18 12:22:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"22d9edb959fcafb77fad21458f09aa691829c68f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-18 12:32:52.000000000","message":"Patch Set 2:\n\n2020-12-18 12:31:49.756219 | TASK [Configure the firewall]\n2020-12-18 12:31:50.291424 | primary | Using Kayobe config from /home/zuul/src/opendev.org/openstack/kayobe-config-dev\n2020-12-18 12:31:50.356413 | primary | net.bridge.bridge-nf-call-iptables \u003d 1\n2020-12-18 12:31:50.366529 | primary | net.bridge.bridge-nf-call-ip6tables \u003d 1\n2020-12-18 12:31:50.434797 | primary | /home/zuul/src/opendev.org/openstack/kayobe/dev/functions: line 223: kayobe: command not found\n2020-12-18 12:31:50.435747 | primary | [Call Trace]\n2020-12-18 12:31:50.435808 | primary | dev/configure-firewall.sh:19:main\n2020-12-18 12:31:50.435858 | primary | dev/configure-firewall.sh:16:configure_iptables\n2020-12-18 12:31:50.435931 | primary | /home/zuul/src/opendev.org/openstack/kayobe/dev/functions:806:is_ipxe_enabled\n2020-12-18 12:31:50.435985 | primary | /home/zuul/src/opendev.org/openstack/kayobe/dev/functions:750:to_bool\n2020-12-18 12:31:50.436058 | primary | /home/zuul/src/opendev.org/openstack/kayobe/dev/functions:744:die\n2020-12-18 12:31:50.439967 | primary | [ERROR] /home/zuul/src/opendev.org/openstack/kayobe/dev/functions:744  was not a valid yaml boolean\n2020-12-18 12:31:51.816766 | primary | ERROR\n2020-12-18 12:31:51.817324 | primary | {\n2020-12-18 12:31:51.817492 | primary |   \"delta\": \"0:00:01.180300\",\n2020-12-18 12:31:51.818411 | primary |   \"end\": \"2020-12-18 12:31:51.442513\",\n2020-12-18 12:31:51.818598 | primary |   \"msg\": \"non-zero return code\",\n2020-12-18 12:31:51.818749 | primary |   \"rc\": 1,\n2020-12-18 12:31:51.818897 | primary |   \"start\": \"2020-12-18 12:31:50.262213\"\n2020-12-18 12:31:51.819001 | primary | }","accounts_in_message":[],"_revision_number":2},{"id":"adc83dcd9ea9fcad477e5e972f02cc2193751c72","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-18 12:32:56.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"8f43dba26a0a36ae0b803fcdb0c547dcb1558f02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 12:53:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-overcloud-ubuntu https://zuul.opendev.org/t/openstack/build/4e570fd79d6d4081958e6468377f4d5f : RETRY_LIMIT in 2m 41s","accounts_in_message":[],"_revision_number":2},{"id":"5a2c8a68a72939b40b2bf685d98c2c128563a969","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-18 12:56:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4ec7fff8dd5c9ad6dbf772882dff45086007aeeb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 13:01:11.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-overcloud-ubuntu https://zuul.opendev.org/t/openstack/build/1c164d4fe7df4c72900a67f734186383 : RETRY_LIMIT in 2m 37s","accounts_in_message":[],"_revision_number":3},{"id":"6a55f8ab5e1fd495726c634c94ddf69638dc0aff","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-18 13:58:38.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e1b94a275a3db5aa9080ce34c9092aa00e19beef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 14:11:00.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kayobe-overcloud-ubuntu https://zuul.opendev.org/t/openstack/build/cbd845832f754ae5998735535ac11e2e : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":4},{"id":"f38b6fc44911e643976d9e27997e635273dae86e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-18 14:16:35.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"0c67cc6b3f2f404a05d555a8307a6974c82d4b99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 14:35:47.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-overcloud-ubuntu https://zuul.opendev.org/t/openstack/build/ff8a52192a764913b0f52db60b469dcc : FAILURE in 7m 08s","accounts_in_message":[],"_revision_number":5},{"id":"b59e9dd60a3f800419c90ddcd39de0b502da1b10","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-18 14:45:18.000000000","message":"Patch Set 5:\n\nRUNNING HANDLER [MichaelRigart.interfaces : Bounce network devices] ************\ntask path: /home/zuul/kayobe-venv/share/kayobe/ansible/roles/MichaelRigart.interfaces/handlers/main.yml:110\n\u003c192.168.33.3\u003e ESTABLISH SSH CONNECTION FOR USER: stack\n\u003c192.168.33.3\u003e SSH: EXEC ssh -C -o ControlMaster\u003dauto -o ControlPersist\u003d60s -o KbdInteractiveAuthentication\u003dno -o PreferredAuthentications\u003dgssapi-with-mic,gssapi-keyex,hostbased,publickey -o PasswordAuthentication\u003dno -o \u0027User\u003d\"stack\"\u0027 -o ConnectTimeout\u003d10 -o ControlPath\u003d/home/zuul/.ansible/cp/7b927b2174 192.168.33.3 \u0027/bin/sh -c \u0027\"\u0027\"\u0027echo ~stack \u0026\u0026 sleep 0\u0027\"\u0027\"\u0027\u0027\n\u003c192.168.33.3\u003e (0, b\u0027/home/stack\\n\u0027, b\u0027\u0027)\n\u003c192.168.33.3\u003e ESTABLISH SSH CONNECTION FOR USER: stack\n\u003c192.168.33.3\u003e SSH: EXEC ssh -C -o ControlMaster\u003dauto -o ControlPersist\u003d60s -o KbdInteractiveAuthentication\u003dno -o PreferredAuthentications\u003dgssapi-with-mic,gssapi-keyex,hostbased,publickey -o PasswordAuthentication\u003dno -o \u0027User\u003d\"stack\"\u0027 -o ConnectTimeout\u003d10 -o ControlPath\u003d/home/zuul/.ansible/cp/7b927b2174 192.168.33.3 \u0027/bin/sh -c \u0027\"\u0027\"\u0027( umask 77 \u0026\u0026 mkdir -p \"` echo /home/stack/.ansible/tmp `\"\u0026\u0026 mkdir \"` echo /home/stack/.ansible/tmp/ansible-tmp-1608302076.6059563-6604-9612638971575 `\" \u0026\u0026 echo ansible-tmp-1608302076.6059563-6604-9612638971575\u003d\"` echo /home/stack/.ansible/tmp/ansible-tmp-1608302076.6059563-6604-9612638971575 `\" ) \u0026\u0026 sleep 0\u0027\"\u0027\"\u0027\u0027\n\u003c192.168.33.3\u003e (0, b\u0027ansible-tmp-1608302076.6059563-6604-9612638971575\u003d/home/stack/.ansible/tmp/ansible-tmp-1608302076.6059563-6604-9612638971575\\n\u0027, b\u0027\u0027)\nUsing module file /home/zuul/kayobe-venv/lib/python3.8/site-packages/ansible/modules/commands/command.py\n\u003c192.168.33.3\u003e PUT /home/zuul/.ansible/tmp/ansible-local-3344s28_kvn8/tmpmed3gvcm TO /home/stack/.ansible/tmp/ansible-tmp-1608302076.6059563-6604-9612638971575/AnsiballZ_command.py\n\u003c192.168.33.3\u003e SSH: EXEC sftp -b - -C -o ControlMaster\u003dauto -o ControlPersist\u003d60s -o KbdInteractiveAuthentication\u003dno -o PreferredAuthentications\u003dgssapi-with-mic,gssapi-keyex,hostbased,publickey -o PasswordAuthentication\u003dno -o \u0027User\u003d\"stack\"\u0027 -o ConnectTimeout\u003d10 -o ControlPath\u003d/home/zuul/.ansible/cp/7b927b2174 \u0027[192.168.33.3]\u0027\n\u003c192.168.33.3\u003e (0, b\u0027sftp\u003e put /home/zuul/.ansible/tmp/ansible-local-3344s28_kvn8/tmpmed3gvcm /home/stack/.ansible/tmp/ansible-tmp-1608302076.6059563-6604-9612638971575/AnsiballZ_command.py\\n\u0027, b\u0027\u0027)\n\u003c192.168.33.3\u003e ESTABLISH SSH CONNECTION FOR USER: stack\n\u003c192.168.33.3\u003e SSH: EXEC ssh -C -o ControlMaster\u003dauto -o ControlPersist\u003d60s -o KbdInteractiveAuthentication\u003dno -o PreferredAuthentications\u003dgssapi-with-mic,gssapi-keyex,hostbased,publickey -o PasswordAuthentication\u003dno -o \u0027User\u003d\"stack\"\u0027 -o ConnectTimeout\u003d10 -o ControlPath\u003d/home/zuul/.ansible/cp/7b927b2174 192.168.33.3 \u0027/bin/sh -c \u0027\"\u0027\"\u0027chmod u+x /home/stack/.ansible/tmp/ansible-tmp-1608302076.6059563-6604-9612638971575/ /home/stack/.ansible/tmp/ansible-tmp-1608302076.6059563-6604-9612638971575/AnsiballZ_command.py \u0026\u0026 sleep 0\u0027\"\u0027\"\u0027\u0027\n\u003c192.168.33.3\u003e (0, b\u0027\u0027, b\u0027\u0027)\n\u003c192.168.33.3\u003e ESTABLISH SSH CONNECTION FOR USER: stack\n\u003c192.168.33.3\u003e SSH: EXEC ssh -C -o ControlMaster\u003dauto -o ControlPersist\u003d60s -o KbdInteractiveAuthentication\u003dno -o PreferredAuthentications\u003dgssapi-with-mic,gssapi-keyex,hostbased,publickey -o PasswordAuthentication\u003dno -o \u0027User\u003d\"stack\"\u0027 -o ConnectTimeout\u003d10 -o ControlPath\u003d/home/zuul/.ansible/cp/7b927b2174 -tt 192.168.33.3 \u0027/bin/sh -c \u0027\"\u0027\"\u0027sudo -H -S -n  -u root /bin/sh -c \u0027\"\u0027\"\u0027\"\u0027\"\u0027\"\u0027\"\u0027\"\u0027\"\u0027echo BECOME-SUCCESS-emuxvckwbigicvjuxxenttymfjhkvvre ; /opt/kayobe/venvs/kayobe/bin/python /home/stack/.ansible/tmp/ansible-tmp-1608302076.6059563-6604-9612638971575/AnsiballZ_command.py\u0027\"\u0027\"\u0027\"\u0027\"\u0027\"\u0027\"\u0027\"\u0027\"\u0027 \u0026\u0026 sleep 0\u0027\"\u0027\"\u0027\u0027\nEscalation succeeded\n\u003c192.168.33.3\u003e (1, b\u0027\\r\\n{\"msg\": \"non-zero return code\", \"cmd\": [\"nohup\", \"bash\", \"-c\", \" returncode\u003d0  ifdown --allow auto breth1;  ifdown --allow auto dummy1;  if ! ifup --allow auto dummy1; then\\\\n  echo \\\\\"Failed to bring up interface dummy1\\\\\";\\\\n  returncode\u003d1\\\\nfi;  if ! ifup --allow auto breth1; then\\\\n  echo \\\\\"Failed to bring up interface breth1\\\\\";\\\\n  returncode\u003d1\\\\nfi; exit $returncode\"], \"stdout\": \"\\\\nWaiting for breth1 to get ready (MAXWAIT is 32 seconds).\\\\nFailed to bring up interface breth1\", \"stderr\": \"nohup: ignoring input\\\\nifdown: interface breth1 not configured\\\\nifdown: interface dummy1 not configured\\\\nRTNETLINK answers: File exists\\\\nifup: failed to bring up breth1\", \"rc\": 1, \"start\": \"2020-12-18 14:34:36.887398\", \"end\": \"2020-12-18 14:34:37.046680\", \"delta\": \"0:00:00.159282\", \"changed\": true, \"failed\": true, \"invocation\": {\"module_args\": {\"_raw_params\": \"nohup bash -c \\\\\" returncode\u003d0  ifdown --allow auto breth1;  ifdown --allow auto dummy1;  if ! ifup --allow auto dummy1; then\\\\n  echo \\\\\\\\\\\\\"Failed to bring up interface dummy1\\\\\\\\\\\\\";\\\\n  returncode\u003d1\\\\nfi;  if ! ifup --allow auto breth1; then\\\\n  echo \\\\\\\\\\\\\"Failed to bring up interface breth1\\\\\\\\\\\\\";\\\\n  returncode\u003d1\\\\nfi; exit $returncode\\\\\"\\\\n\", \"warn\": true, \"_uses_shell\": false, \"stdin_add_newline\": true, \"strip_empty_ends\": true, \"argv\": null, \"chdir\": null, \"executable\": null, \"creates\": null, \"removes\": null, \"stdin\": null}}}\\r\\n\u0027, b\u0027Shared connection to 192.168.33.3 closed.\\r\\n\u0027)\n\u003c192.168.33.3\u003e Failed to connect to the host via ssh: Shared connection to 192.168.33.3 closed.\n\u003c192.168.33.3\u003e ESTABLISH SSH CONNECTION FOR USER: stack\n\u003c192.168.33.3\u003e SSH: EXEC ssh -C -o ControlMaster\u003dauto -o ControlPersist\u003d60s -o KbdInteractiveAuthentication\u003dno -o PreferredAuthentications\u003dgssapi-with-mic,gssapi-keyex,hostbased,publickey -o PasswordAuthentication\u003dno -o \u0027User\u003d\"stack\"\u0027 -o ConnectTimeout\u003d10 -o ControlPath\u003d/home/zuul/.ansible/cp/7b927b2174 192.168.33.3 \u0027/bin/sh -c \u0027\"\u0027\"\u0027rm -f -r /home/stack/.ansible/tmp/ansible-tmp-1608302076.6059563-6604-9612638971575/ \u003e /dev/null 2\u003e\u00261 \u0026\u0026 sleep 0\u0027\"\u0027\"\u0027\u0027\n\u003c192.168.33.3\u003e (0, b\u0027\u0027, b\u0027\u0027)\nfatal: [controller0]: FAILED! \u003d\u003e {\n    \"changed\": true,\n    \"cmd\": [\n        \"nohup\",\n        \"bash\",\n        \"-c\",\n        \" returncode\u003d0  ifdown --allow auto breth1;  ifdown --allow auto dummy1;  if ! ifup --allow auto dummy1; then\\n  echo \\\"Failed to bring up interface dummy1\\\";\\n  returncode\u003d1\\nfi;  if ! ifup --allow auto breth1; then\\n  echo \\\"Failed to bring up interface breth1\\\";\\n  returncode\u003d1\\nfi; exit $returncode\"\n    ],\n    \"delta\": \"0:00:00.159282\",\n    \"end\": \"2020-12-18 14:34:37.046680\",\n    \"invocation\": {\n        \"module_args\": {\n            \"_raw_params\": \"nohup bash -c \\\" returncode\u003d0  ifdown --allow auto breth1;  ifdown --allow auto dummy1;  if ! ifup --allow auto dummy1; then\\n  echo \\\\\\\"Failed to bring up interface dummy1\\\\\\\";\\n  returncode\u003d1\\nfi;  if ! ifup --allow auto breth1; then\\n  echo \\\\\\\"Failed to bring up interface breth1\\\\\\\";\\n  returncode\u003d1\\nfi; exit $returncode\\\"\\n\",\n            \"_uses_shell\": false,\n            \"argv\": null,\n            \"chdir\": null,\n            \"creates\": null,\n            \"executable\": null,\n            \"removes\": null,\n            \"stdin\": null,\n            \"stdin_add_newline\": true,\n            \"strip_empty_ends\": true,\n            \"warn\": true\n        }\n    },\n    \"msg\": \"non-zero return code\",\n    \"rc\": 1,\n    \"start\": \"2020-12-18 14:34:36.887398\",\n    \"stderr\": \"nohup: ignoring input\\nifdown: interface breth1 not configured\\nifdown: interface dummy1 not configured\\nRTNETLINK answers: File exists\\nifup: failed to bring up breth1\",\n    \"stderr_lines\": [\n        \"nohup: ignoring input\",\n        \"ifdown: interface breth1 not configured\",\n        \"ifdown: interface dummy1 not configured\",\n        \"RTNETLINK answers: File exists\",\n        \"ifup: failed to bring up breth1\"\n    ],\n    \"stdout\": \"\\nWaiting for breth1 to get ready (MAXWAIT is 32 seconds).\\nFailed to bring up interface breth1\",\n    \"stdout_lines\": [\n        \"\",\n        \"Waiting for breth1 to get ready (MAXWAIT is 32 seconds).\",\n        \"Failed to bring up interface breth1\"\n    ]\n}\nMETA: ran handlers","accounts_in_message":[],"_revision_number":5},{"id":"85307f5bef85f354733422e730005ed6632b5c13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-18 16:39:24.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"0c67be4e10eff410fc61dea6fef72bb0e92e0d2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 16:48:00.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-overcloud-ubuntu https://zuul.opendev.org/t/openstack/build/17922ac00e654b9f9a1f02b0dfd531a1 : RETRY_LIMIT in 2m 30s","accounts_in_message":[],"_revision_number":6},{"id":"51f823c9bf7b31f3f23dd57eb43008291bfc1818","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-18 17:09:00.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"5ac46bbc934683083d6cd201a72c62f01b4479b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 17:16:47.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-overcloud-ubuntu https://zuul.opendev.org/t/openstack/build/847b78765a074e018486ddb2975cba85 : RETRY_LIMIT in 2m 01s","accounts_in_message":[],"_revision_number":7},{"id":"49d742c7c9fe20a271a3f534b7df83ea186f09be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-18 17:21:18.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"5f3a481b7b716b34554acbe868a4304da8d13ba5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 17:31:11.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-overcloud-ubuntu https://zuul.opendev.org/t/openstack/build/41f5852403424b25b2d66ffd7ade0f6a : RETRY_LIMIT in 2m 43s","accounts_in_message":[],"_revision_number":8},{"id":"8a8e920aef0ab06f00f894a422d6238676ba9dbe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-18 17:41:01.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"310bc6c75cbecf30f49529d76895dbdd2a9204c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 17:48:44.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-overcloud-ubuntu https://zuul.opendev.org/t/openstack/build/daec2b9f062a47438b7e41b67afa73c2 : RETRY_LIMIT in 3m 32s","accounts_in_message":[],"_revision_number":9},{"id":"76dcfc026d08f1d02af0fcb81b1626331fbfc113","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-18 18:06:04.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"220857e5a5632549980cc615e65e4021c014aedb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 18:12:30.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-overcloud-ubuntu https://zuul.opendev.org/t/openstack/build/b77cd1ecd6fe408ba03d296b6044633d : RETRY_LIMIT in 2m 34s","accounts_in_message":[],"_revision_number":10},{"id":"eda758020f25949e4864085109d22dac57407033","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-18 18:36:27.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"5e71833fe494ff7aef5e482a01fa5af0558d3edf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 18:44:01.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-overcloud-ubuntu https://zuul.opendev.org/t/openstack/build/6016d22cc1554873a33dcd3ab135e860 : RETRY_LIMIT in 2m 43s","accounts_in_message":[],"_revision_number":11},{"id":"171b215d90e23cccc477955e0fad1497892ae0e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-18 19:24:28.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"e20da0b8a4d8387e80b6e01663923c2ecf784376","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 19:34:50.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-overcloud-ubuntu https://zuul.opendev.org/t/openstack/build/ffc4b5419bd244d9bfaa61c1156d2986 : FAILURE in 6m 04s","accounts_in_message":[],"_revision_number":12},{"id":"e54996776d05f71a538d118a634f1675c180fbc8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-19 09:02:56.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"a4d8f7e47e858109d5abe3c6dd4982230c7ec51f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-19 09:10:14.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-overcloud-ubuntu https://zuul.opendev.org/t/openstack/build/62cfee2c565f4e6883ea47b42746314c : FAILURE in 6m 01s","accounts_in_message":[],"_revision_number":13},{"id":"99add7a9a475e46624d142c833a09fd0b1c45564","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-19 09:26:18.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"46d5affa6e693f5d37c2ba3c0afa60fbdcbd2c9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-19 09:34:47.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-overcloud-ubuntu https://zuul.opendev.org/t/openstack/build/ad5b2892cf754d85b3e34c773bbb7dbc : FAILURE in 6m 52s","accounts_in_message":[],"_revision_number":13},{"id":"0ec3eee7e6a7a697a3a11da738718e6bbc132dfb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-21 09:23:45.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"f27b8cee5867e20d0d0cebd0c96406f0bae98e2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-21 09:31:22.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-overcloud-ubuntu https://zuul.opendev.org/t/openstack/build/be9dc22fb5314773aab29522042a5427 : FAILURE in 5m 52s","accounts_in_message":[],"_revision_number":14},{"id":"8886c6af618489ddef5808cf72eccd2cda8ea1d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-21 09:32:49.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"79b5db338b460cd4dd95d73e7c3bb7eaca328770","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-21 09:39:04.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-overcloud-ubuntu https://zuul.opendev.org/t/openstack/build/6002a6feb88246719ce111632dc30fed : FAILURE in 5m 06s","accounts_in_message":[],"_revision_number":15},{"id":"262744a00be3ca9fa92686bd74497523115bdbb8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-21 09:41:54.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"77db0af2dc1c310b8ddbddec9240eb12c5a06b26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-21 09:53:19.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-overcloud-ubuntu https://zuul.opendev.org/t/openstack/build/bb307aa723c84fa6a112c957a02110c5 : FAILURE in 6m 12s","accounts_in_message":[],"_revision_number":16},{"id":"0c397bb8ee4a325a158c9c825b55c941e5493389","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-21 10:48:14.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"d80f3a5e5550d88dba6d34a8487df184f64b31f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-21 11:31:20.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-overcloud-ubuntu https://zuul.opendev.org/t/openstack/build/bb5a2486aa1c450f97a1bdbd8290ab30 : FAILURE in 37m 27s","accounts_in_message":[],"_revision_number":17},{"id":"a9b69950785195ac2bf18dbf4d62ab52b1214ad7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-21 12:09:48.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"dab9d8cbe5a9ef64b71de4e7140f8fdea249b820","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-21 12:55:21.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-overcloud-ubuntu https://zuul.opendev.org/t/openstack/build/1f0a37c1fab445c1a2169ecc90c54d0e : FAILURE in 38m 52s","accounts_in_message":[],"_revision_number":18},{"id":"0e63c3925a9b6d308e2301b3593102e0c3d39f72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-21 13:06:23.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"0e6008a976cf6ed5f2b51ae22498c2a249cdb2c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-21 14:01:56.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-overcloud-ubuntu https://zuul.opendev.org/t/openstack/build/3d408207c159496e99b18d6db08feaff : FAILURE in 52m 42s","accounts_in_message":[],"_revision_number":19},{"id":"4485b1f2f41245b893b3ed13f3ddd8d87f86b44e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-21 14:07:09.000000000","message":"Patch Set 20: Published edit on patch set 19.","accounts_in_message":[],"_revision_number":20},{"id":"6f2f2e0c26c184ab7b5e92e65b430b146270bf1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-21 14:45:54.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-overcloud-ubuntu https://zuul.opendev.org/t/openstack/build/159c9e040c5d4e6fb0afb809b18ba4fe : FAILURE in 34m 55s","accounts_in_message":[],"_revision_number":20},{"id":"9acd3212d1668042e990cf72d7249447fd1c00cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-21 14:59:56.000000000","message":"Patch Set 21: Published edit on patch set 20.","accounts_in_message":[],"_revision_number":21},{"id":"48594e85c527582d97c8ec260f422a362656e719","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-21 15:27:37.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"ce303dfe050a5b4bc330c8dc7b10efd5376d4de4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-21 15:28:42.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"284c78ce4953d4e3f05abe3a5e0883992861f06a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-21 16:33:02.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kayobe-overcloud-ubuntu https://zuul.opendev.org/t/openstack/build/176a0ce0c3de420491d5862560cd92cc : SUCCESS in 47m 12s","accounts_in_message":[],"_revision_number":23},{"id":"1d2b6184d49e6cecfdac7c6974a7a4f8ff79c66e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-21 16:48:17.000000000","message":"Patch Set 23:\n\nlovely!","accounts_in_message":[],"_revision_number":23},{"id":"1bb54715b631de8dd0f758889a6e508ed7deadb7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-22 13:57:56.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"638b59732c7c229ec279aea26e6b2706f04569ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 14:08:57.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-overcloud-ubuntu https://zuul.opendev.org/t/openstack/build/440886186c4b4b70bba6ce442dc8ec4b : FAILURE in 5m 34s","accounts_in_message":[],"_revision_number":24},{"id":"3c5aa83390a4bbbfa6e3920826f83b56cdae1bfe","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-22 14:29:33.000000000","message":"Patch Set 24:\n\nrecheck","accounts_in_message":[],"_revision_number":24},{"id":"85e1e3106c11f9f1402287ab0591ffd1b7d2e3b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 15:28:23.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kayobe-overcloud-ubuntu https://zuul.opendev.org/t/openstack/build/9509fdf41f214eb385ca9ea45bdd8e92 : SUCCESS in 53m 15s","accounts_in_message":[],"_revision_number":24},{"id":"4116c87aa1a13b8ec33c7ca117be82f50b5f2d54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-22 17:25:26.000000000","message":"Patch Set 25: Published edit on patch set 24.","accounts_in_message":[],"_revision_number":25},{"id":"b905e4cb32ad40afaec9b1243de8434e7e7b6a94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 17:32:57.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-overcloud-ubuntu https://zuul.opendev.org/t/openstack/build/b975475e31b34d15bfa63e860a3f941b : FAILURE in 5m 26s","accounts_in_message":[],"_revision_number":25},{"id":"9b749ac87b3634237a7b5d69f821e45372ca3d2f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-22 18:01:53.000000000","message":"Patch Set 25:\n\nrecheck","accounts_in_message":[],"_revision_number":25},{"id":"c6e9a8e4e3c552a99f03082a7d2f0975f600953b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 19:11:58.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kayobe-overcloud-ubuntu https://zuul.opendev.org/t/openstack/build/b060bda8757d480a8f48f88c88ecd0dc : SUCCESS in 1h 04m 14s","accounts_in_message":[],"_revision_number":25},{"id":"d6d3648133ea0f9f71ffe1124146019179923e55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-07 13:45:40.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"cc9a7b58f3887fc5e74b8d28e0216ebf33f269e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-07 13:54:51.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"1f93ab5be14c4e611992a87a67df0557a9ec8178","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 14:55:04.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/c3191dcc146c4e6cb7436cb1eb6aff03 : SUCCESS in 50m 23s\n- kayobe-overcloud-ubuntu https://zuul.opendev.org/t/openstack/build/8e5cbf19bd5c4104be4046bbc32b0a88 : SUCCESS in 48m 34s","accounts_in_message":[],"_revision_number":27},{"id":"a401b150a1e1b80c521a1b157fca430bc449e431","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-07 15:27:03.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"4288bd7c8af37d153398fb9d928a5aebb61db5c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 17:23:21.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ba3c2fe72a24b50815351699522866d : SUCCESS in 9m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3e3fbdee3dc40c39e32dc949feff3ac : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a9581f37d504c91890b069324bbf75a : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/56f40171529c431b89e31010722ff0be : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aefd28ae46a745599e9fd997376d1485 : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2150436c50c84f5c85ea0ec5fba2b12e : SUCCESS in 7m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7d4d308abbcc4e6b9ee16c03fa3b6df3 : SUCCESS in 5m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ffb883e616534262b51865417af624be : SUCCESS in 9m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e643004107144ed295c5c4a92493f00a : SUCCESS in 17m 48s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/5db6d6e1ea4747bfa31d3b43eb4e5683 : SUCCESS in 58m 43s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/81fc16f6c7eb47d6a0252468128c3384 : SUCCESS in 53m 27s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/be334ac509774294a114562b4f3db920 : SUCCESS in 1h 06m 18s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/ef723e43b9994750b6e0f0c27326a430 : SUCCESS in 19m 01s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e2825f4b0db64972ab1442e7e4561eb9 : SUCCESS in 1h 44m 26s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/0ec1c8365de64fb6b8a6f5012f91a166 : SUCCESS in 15m 09s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/8f64c68d598f4ff0b92597a710119616 : SUCCESS in 24m 49s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/2d75287f244c463c8986b2ab3a175961 : FAILURE in 16m 22s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":28},{"id":"3e1cb48cb900a13c6896809107c211ea8ceca99d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-07 17:27:24.000000000","message":"Patch Set 28:\n\nrecheck","accounts_in_message":[],"_revision_number":28},{"id":"3a48cdb00c55b57424958115a4841c1a8c4de52c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 19:06:52.000000000","message":"Patch Set 28:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c5c064d603e54b64aae6842bd12c0ebe : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df7c2cf259cc4a98bafaee33b191cac3 : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b901896ca0804401b6f0d9262f3bb24d : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e14d253a211b46f3a376f0bbd6768d60 : SUCCESS in 5m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/38694d4b59a04f29a394372f1cba009a : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a089ed159a4644d6b7ad791b13385e8f : SUCCESS in 10m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/323746b520ce41c982ed5739111e787b : SUCCESS in 6m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/44bf76bc75ea41bb8598ef6a6cbe70e8 : SUCCESS in 9m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/15eda32e46374a90841b3b583f31a5da : SUCCESS in 20m 22s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/f59f157843d34006bf05884fc97995e3 : SUCCESS in 59m 40s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cd12909366904ff09f7ec3c3bf52d256 : FAILURE in 53m 26s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/81e1e20f6d904e82ad51160fbbc2dd30 : SUCCESS in 1h 04m 25s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/2887faf7fdc14d59a61ba3441faba1af : SUCCESS in 17m 07s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f6d7b17f94474c1db511a94eea782404 : SUCCESS in 1h 30m 34s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/33c2f2c0e31a46768bc6b936f1920111 : SUCCESS in 17m 31s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/bbf9dfb74dc24fd9a8e99fd824acb4d7 : SUCCESS in 31m 18s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/9e93670d33264317a87be1a3f552b189 : FAILURE in 14m 07s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":28},{"id":"f982e2a4aa55e6cdf57faa06259d302c5d886282","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-08 09:22:48.000000000","message":"Patch Set 29: Published edit on patch set 28.","accounts_in_message":[],"_revision_number":29},{"id":"063697e4eceba895af894c5efe3415b3cac9e629","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 10:59:28.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c989b5fa20294196b76f32253e7a4b91 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6af950624f9642eaa92d6c989468e01c : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/159a872a63994f91bf07cfeded692b1e : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49ae888d5e0140b3a36e13e6734a78f5 : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/62ddff1c20314828b1ce7041580e48ee : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f09ad6146c2481789834417d0cc2840 : SUCCESS in 6m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/352f601e4a624e1ba306fb8137ef1ea0 : SUCCESS in 4m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4109095110df40d7a196d51deb7774d9 : SUCCESS in 9m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e4349c1e37d74d778f6e2d14a3340f1b : SUCCESS in 20m 28s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/544644011e3040d58273d8328667b569 : SUCCESS in 1h 04m 15s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d0a219bc3cd6498f89d6cbb950200c92 : FAILURE in 47m 44s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/376c3da71cba44bb89accb97d0e121a4 : SUCCESS in 1h 04m 22s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/1d43863caf9048e5b19f51cc16313c7c : SUCCESS in 15m 10s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/fd1cb9df5c4a4797985685ee75919a27 : SUCCESS in 1h 29m 18s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/09ec03ac4dc74d878c5f44d5008712c4 : SUCCESS in 21m 14s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a10805d6aac14833ad6659ed53c05afe : SUCCESS in 25m 11s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/4e1b783247fa45f681cc09d00c9f4278 : SUCCESS in 11m 37s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":29},{"id":"476311459aca7e7ffd2ef43dffab9a47628c7650","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-08 11:22:42.000000000","message":"Patch Set 29:\n\nrecheck","accounts_in_message":[],"_revision_number":29},{"id":"cee4ed016a02dec2754b422567b7bec6015921f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 13:15:44.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fad0fc0dadc447e9a95e53ccdcbbe69b : SUCCESS in 8m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9029714c09d9409785931afd45f8476d : SUCCESS in 9m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c945b0aed1f4cb3a0b5a004bd6ebd28 : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b2d557552db64019b757a6e03ce58627 : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5ff160cee476482ea77e53cdb96d9e99 : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d75872ef43b4b0c91a5fbde35656467 : SUCCESS in 8m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/388406f88c0c4ef4a771efa548061eb8 : SUCCESS in 7m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f8ad1010f9bd49889a744d747e725f5c : SUCCESS in 10m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4c270791e58a49d494e43af62ddad4d1 : SUCCESS in 19m 51s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/cdd4839e5c184d5c959c579939d6b549 : SUCCESS in 1h 10m 28s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/07460fa3ae554032a00d24567fe39275 : SUCCESS in 48m 43s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/d2b63cdfb14643fb89f268bd26c2b82e : SUCCESS in 1h 04m 13s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/78558ed33ff649219aa692ded51f9424 : SUCCESS in 18m 07s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a9181cb7693549158183dd523626d3b3 : SUCCESS in 1h 46m 12s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/c5ebf17731cd41ffa9e0f33dd1b99653 : SUCCESS in 15m 20s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/4c31af4a58ae40648a59461859a354d2 : SUCCESS in 27m 23s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/8536c2c052be4768b94d41d45a0a1177 : SUCCESS in 12m 28s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":29},{"id":"648f0073d0326aca14b5d40ef40a83e45efe97e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-11 11:03:21.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"f711acc7c12065a8d6d366a7328c67b5c60b3a37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 12:42:01.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ce9a80bc90e4560b1ddab7645805aef : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c87e367690924e6c8dccbd2b74b70cc4 : SUCCESS in 7m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c40a29b24ed04fdf82b3ef9379632700 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5580ea6b60b7454a9d758e7c22db38d4 : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/25367be0481846958fe9fe8172e800cc : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fd774b2b4f7400b93770d5d2a651bdc : SUCCESS in 6m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a439a28bc15349f38d227d270d12792e : SUCCESS in 5m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b7c7da3ff08e42f3b148c87e5e493ad1 : SUCCESS in 9m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b4e54cd11ece4bd8b924fdfb3a2f77a4 : SUCCESS in 18m 13s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/82373d8c08d54f7b95d67f76e1f27ed5 : SUCCESS in 1h 04m 33s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cb40411233a8472aa016eb95cd4e404c : SUCCESS in 48m 04s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/20ac05e75e1148a29777a56157d268e5 : SUCCESS in 1h 02m 25s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/1c8abb46b6b246d1b0506deb3702c794 : SUCCESS in 15m 37s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/4d980c7d064e49ab903cdebc5540882f : SUCCESS in 1h 34m 27s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/995aeb39d0d447e18a1923baa107dd3a : SUCCESS in 19m 47s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/ece2b791d3c546578ec21dc7bbc569d1 : SUCCESS in 24m 42s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/6f156d79d9f543609898fc64325c5ed7 : SUCCESS in 11m 21s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":30},{"id":"ac1bd4cdfb2dc152464ac3ba8d5344cc72ae68cc","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-28 15:10:43.000000000","message":"Patch Set 30:\n\n(1 comment)","accounts_in_message":[],"_revision_number":30},{"id":"9407b0860252056f3cfcef4e8b5566981f9151f4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-11 11:02:46.000000000","message":"Patch Set 30:\n\n(1 comment)","accounts_in_message":[],"_revision_number":30},{"id":"f9f26068852a6fc83d7d8a231019b80a37d2be75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-11 11:11:10.000000000","message":"Uploaded patch set 31.","accounts_in_message":[],"_revision_number":31},{"id":"496f29761df2b031ffcf155ed39783976f56fb4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 13:25:32.000000000","message":"Patch Set 31: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b682cb8ee42e4cfa99370a1a144b8a7a : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44a622339008484a82a2d6110b109e63 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/679953f0ecd64837a50e535fa3c12e20 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ade63c3d84e43fb913e0806ddb102d3 : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/58ee3a42ed0946168012570cd9a6db54 : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/589a819babfc4da5917c0be5a07829d2 : SUCCESS in 6m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f85fcc83fe2b4f3984b9ac31353bda0c : SUCCESS in 6m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/41b4cf346ae8417c90300b43bcb86bcb : SUCCESS in 7m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/46ef86c36b3440d091e70f3ec6359f51 : SUCCESS in 14m 02s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/2ad05fa7cffa4eb4aeb7aa876fe5b4e9 : FAILURE in 1h 43m 18s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2f0c7119f9854329b278a972059444c4 : SUCCESS in 38m 11s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/2bb4c6f54cca407eb4e3cd02e3050e96 : SUCCESS in 57m 08s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/dfcec91850dd4489a110b96f736c7e2d : SUCCESS in 17m 45s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/dde1df01b9f74143af7263137722491a : SUCCESS in 1h 43m 26s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/3a274a49f6504aa5a4dbd610ebd831f8 : SUCCESS in 19m 20s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/333dd49157314bfa96647843a74ae64c : SUCCESS in 23m 03s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/2e2251885df34d40b7d26b364f2bbc72 : SUCCESS in 9m 10s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":31},{"id":"13dbb2ab7472cc3a1572d876b3ea90fa8e93436c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-16 10:25:23.000000000","message":"Patch Set 31:\n\nrecheck","accounts_in_message":[],"_revision_number":31},{"id":"d935fef3b7c08ce8f86078a81785739b5bf12b12","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-16 11:16:17.000000000","message":"Patch Set 31:\n\nrecheck","accounts_in_message":[],"_revision_number":31},{"id":"581672079bf7fb209cb4c118d459864c43da8da9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 12:20:21.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c35ed26392a4471f89e4fa88f3a530cb : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35c70fc099244650a0c918e851314467 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f64e792b9b5a4df098a3b78d8a2b3b04 : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/62628276b48e47a5b7f1a851da19ab86 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0950def3a31f48369106c94a7a03f053 : SUCCESS in 4m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/472133241a0847e391c680a3f6f38aeb : SUCCESS in 6m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/28abc42314aa4d85a441c71ece3ee757 : SUCCESS in 5m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d25b7a89278446fe83c1983e9c1c9b44 : SUCCESS in 7m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/74cac74292694e4a8a993a2c3eb3e794 : SUCCESS in 18m 09s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/271c3c5992724a78b45400dee3aeb96e : SUCCESS in 1h 07m 38s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/167eccdd8d414676ac65f19927a3f111 : SUCCESS in 43m 04s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/4cdaf6cef8d64c2cb9f24d7125af02c0 : SUCCESS in 1h 14m 32s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/6dd61a0b9a434c759a958d9ad0e21c6f : SUCCESS in 14m 38s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/16087b87c0ea4768aec65c88d5fb8da3 : SUCCESS in 1h 35m 42s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/3015189821f5489c88ab93e9c17c36d0 : SUCCESS in 16m 03s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/996836de4626476b8025522c03dcab12 : SUCCESS in 23m 20s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/107a25f8a638466fa6134ecffeeac61a : SUCCESS in 9m 19s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":31},{"id":"ba9775cd55e75232e6b09aaf2783292cbcd8ac9a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-19 14:19:23.000000000","message":"Patch Set 31: Code-Review-1\n\n(2 comments)\n\nTiny issue to fix and a question.","accounts_in_message":[],"_revision_number":31},{"id":"000862550128929b8796e1a68b3db4dab19c5846","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-19 14:59:04.000000000","message":"Patch Set 31:\n\n(2 comments)","accounts_in_message":[],"_revision_number":31},{"id":"3ddad90282403ba0c3e094d3edf9a39ba7dab706","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-19 14:59:33.000000000","message":"Patch Set 32: Published edit on patch set 31.","accounts_in_message":[],"_revision_number":32},{"id":"d7133aa3ef9f742164ede00dacc096e925b396e7","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-19 17:08:51.000000000","message":"Patch Set 32: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":32},{"id":"5bb5551f8e46a278aad9eac5d93b399b235a3727","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 17:13:07.000000000","message":"Patch Set 32: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e597d095879469c8905f20f71391a3c : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f7ff252147c4d8db431e002d42f6e3c : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf5b652432e344629f554003343bac1d : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c63ee17888f4f6e8861c3c23187277f : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60f56c423a4a4802aa24b99f9c11cde4 : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2068d7a4a59d4daeac4c2f1c7d8b1151 : SUCCESS in 7m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4b0ad2c4d1e54de58bbb4d7afa5989c9 : SUCCESS in 5m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d79edce106d9438cac5464a9c41c395e : SUCCESS in 8m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/960554fb231840afbd8629673bafe34b : SUCCESS in 17m 16s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/aab0386c4f204e2e9a67779cc8f659d6 : SUCCESS in 58m 09s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/04a223ed92f3461ab0524a8a5a42e71a : SUCCESS in 41m 34s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/c29eb6eb62654132b04c20600d266c58 : TIMED_OUT in 2h 03m 05s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/152403e60dea405282ffdd59cd668a32 : SUCCESS in 15m 35s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/5b53061de95b4fd4a2a274008574a28a : SUCCESS in 1h 44m 21s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/ab1b635d11d34942ad44c1de10fc3a83 : SUCCESS in 24m 26s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/21653e0930174ba2a7f305cb1becb9eb : SUCCESS in 24m 35s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/02779175761142faad88d6359cbc7918 : SUCCESS in 14m 23s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":32},{"id":"e401407bb38ceb52b32de0e31c91225ea63a7bca","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-22 13:50:47.000000000","message":"Patch Set 32:\n\nrecheck","accounts_in_message":[],"_revision_number":32},{"id":"3828f21b1cf8ef4d2c78afd16f934e118a9fd7c9","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-22 13:52:45.000000000","message":"Topic set to story/2004960","accounts_in_message":[],"_revision_number":32},{"id":"ba2d2c2000a8c43b75f204eaee4a75dc84f5402f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 15:53:19.000000000","message":"Patch Set 32:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e0895b76d8264cafaf8b4b82904982c9 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5319890df49f43b3b14b5df5eefede74 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f0b70cad841347b18c31314028c79f3a : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e29c39049f3b42d2b04e3050d6e1d1e6 : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/110e68cdace3410e9eb32ba9bb12ffad : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba384b05d2644f99a7d2b08efd22d587 : SUCCESS in 7m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d333fcc479ae4f67934f55a7825f2644 : SUCCESS in 4m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6114d31afbe347dd814db2d069e14e01 : SUCCESS in 7m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/93ec9f98a96b4b6e8b697322f852f782 : SUCCESS in 14m 48s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/188bf9cbd775409e92c643b373e35474 : FAILURE in 1h 58m 37s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b819270c2dcd4cccbc802fab83d87841 : SUCCESS in 43m 10s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/1f2737db1b154effa3446a997909d4b6 : SUCCESS in 1h 05m 03s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/226c3334affa411aa7b7fd639130c7d9 : SUCCESS in 16m 40s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/62d9bc00aa0843ccb69bf5efdb9dc849 : SUCCESS in 1h 37m 05s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/775ca9c10e3e4ebc8195bd4d83888790 : SUCCESS in 27m 37s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/163cbcdeb3124b90ba2ce31c3087dd87 : SUCCESS in 23m 05s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/a1eac3e2d4244e09a8aa681a2d3a045e : SUCCESS in 11m 58s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":32},{"id":"74a460d789430cb545a28b41d106a19ed34a5796","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-23 10:01:43.000000000","message":"Patch Set 32:\n\nBare metal timed out waiting for callback.","accounts_in_message":[],"_revision_number":32},{"id":"5ef6057149d393e629a81616f59ff2f00ce89166","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-01 17:57:51.000000000","message":"Patch Set 33: Patch Set 32 was rebased","accounts_in_message":[],"_revision_number":33},{"id":"5f61029261ff7d20a756dcfe117a149fee92162a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 19:59:38.000000000","message":"Patch Set 33: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/55cc8bdfd47b4d6088ac34c776ae91c5 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6849716d1ec4eaba5fdf4786158505b : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c73c3fe5f2a4507851801c6c6e386cc : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee7e0c03ab584e63ae02ea0be464c3e6 : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f2a68a80ca9d400c82d58f7f92553f77 : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c305136f77f49fb86926a38f46c5f99 : SUCCESS in 6m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3111e36a71f44ec1b00125e546ed7c13 : SUCCESS in 4m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ce992cb25fba4a22adfd4b2002169cce : SUCCESS in 8m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/894b1a3b769b4f3c88a57a5fa7d5c709 : FAILURE in 6m 14s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/03ec2262aaf44d8892b720ab7eb10435 : SUCCESS in 55m 24s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d1684b0dfb6f479798a919f7fe6deea5 : SUCCESS in 47m 23s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/7b44dea818444ef9b9b07d302a053e11 : SUCCESS in 1h 01m 09s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/1521412cff22419893f8eae240c87cda : FAILURE in 14m 28s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/839bc5dbc65340deaedc18e398cc2c48 : SUCCESS in 1h 38m 08s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/c8f1e59c9bf14ba99e156e2253fed66d : SUCCESS in 24m 58s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/6e4df8b6d922480484924171be206e68 : SUCCESS in 26m 52s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/84fb1964c2314131a4862c2e995b0de7 : SUCCESS in 9m 35s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":33},{"id":"0b28a2d813e13b18c8f3432d7eb1e4d977473601","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-02 10:48:53.000000000","message":"Patch Set 33:\n\nrecheck","accounts_in_message":[],"_revision_number":33},{"id":"0ed09d21da8f3a55333f19d45e2b2d14e50175d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 13:12:49.000000000","message":"Patch Set 33: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c6307f8159640f1b1ae823a94f4645c : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1d124e7ffa241109b0d5364b4de200e : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bfbc0cb97f754448b8bb01a3fd921227 : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70d5016c01154e9d9035c08f6375a16a : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ce3adc8df4ac46f2a8e7109eb0bbfac2 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77d86371307243c28d1c2bec2c4d2d6f : SUCCESS in 8m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a854dc675b424a75b1bed5d78d879116 : SUCCESS in 5m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/12b53f56c87f4f3abd62e973c8547b17 : SUCCESS in 9m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/831dfc555971400db10cb0482b09605b : SUCCESS in 19m 58s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/6f1dfd25b4944bfd9ef9ec0c2ac392c8 : SUCCESS in 58m 42s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a949c9c3fc3407887b5a002b32414aa : SUCCESS in 47m 09s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/e71ca90ae24b4e34a8237f5b15d95626 : SUCCESS in 1h 03m 37s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/7bb44fee89e04e468816e491009f28bd : SUCCESS in 18m 31s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/9cfdcdd5bf0a41639b0c31a75c874575 : SUCCESS in 1h 41m 58s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/86c92db90b674de0bd827b042f691a79 : SUCCESS in 28m 27s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/dbf2a3c031fd470fb6725df4d1cbaa79 : SUCCESS in 23m 49s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/c6df1f21a99f4100970365ed4c669836 : SUCCESS in 11m 05s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":33},{"id":"97bf2a76119f7203a8b50edb8e1e735c66cdd889","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-03-04 11:12:11.000000000","message":"Patch Set 33: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":33},{"id":"c36f2842c5d4953125b4bdb69ce2a44c352d6798","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 11:12:39.000000000","message":"Patch Set 33: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":33},{"id":"1650c57cd40030605e52c8d16f7bca0ea30f004a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 14:22:28.000000000","message":"Patch Set 33: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24620956ff524e44b235902deea555ae : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8695806b05234696827b74e63ff3581a : SUCCESS in 6m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c0ca8397f3194bf89e4700b908c5d0f7 : SUCCESS in 7m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e42e07a464d4c60877ba80b6475a559 : SUCCESS in 7m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1dd34792beac426eaac557df80414089 : SUCCESS in 6m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/557bf497d8ce4b69b4821f8a660d1a80 : SUCCESS in 9m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6ab90e8cfdd049d5a53b13ecb06473b0 : SUCCESS in 18m 35s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/485e0d31bda74c489005adc0af793545 : SUCCESS in 1h 01m 05s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6126819a1c384befa9ce8a148de544df : SUCCESS in 45m 52s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/05b772a099c64cf996741d0de6073f85 : SUCCESS in 1h 02m 02s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/526c820f7d47453aba22b0fb5278f436 : SUCCESS in 15m 48s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/79c898a2253947d2b889b59a92ebbaea : SUCCESS in 1h 44m 31s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/0db18e6892474aff8cf3f8af3682eb8f : SUCCESS in 34m 25s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/169f9da4388f47f3a17b67e0e930a79d : SUCCESS in 26m 17s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/89e2d7cf371e49ba9a3678fcfd2ad3e0 : SUCCESS in 10m 11s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":33},{"id":"03c83ea641af7d8aa5ee5d84c740b4d8398b831c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 14:22:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":33},{"id":"bb8bb3753f56e3e3d097dda26ef42dc5c12fd7b9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 14:30:47.000000000","message":"Patch Set 33:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b77379f3ad8840fb972e8431b98f3207 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":33}],"current_revision_number":33,"current_revision":"9f41cefc15e240602d2c6ef9076e36135e9138e4","revisions":{"58a235ec8504b0a55a7efa857f87775b398e1744":{"kind":"REWORK","_number":1,"created":"2020-12-18 12:18:44.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/05/767705/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/05/767705/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/05/767705/1"}}},"commit":{"parents":[{"commit":"858c0f2a62dcd874fa3ac5601b38c70f6f46a535","subject":"Merge \"Remove retired Searchlight support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/858c0f2a62dcd874fa3ac5601b38c70f6f46a535"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-18 12:17:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-18 12:18:23.000000000","tz":0},"subject":"WIP: CI: add Ubuntu jobs","message":"WIP: CI: add Ubuntu jobs\n\nDepends-On: https://review.opendev.org/766984\nDepends-On: https://review.opendev.org/766958\n\nChange-Id: I1985efae7c18f55c3ff7c27c17d6242523904f3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/58a235ec8504b0a55a7efa857f87775b398e1744"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/58a235ec8504b0a55a7efa857f87775b398e1744"}]},"branch":"refs/heads/master"},"61f68d65f24488ef329c7063ace6d5425ad556e0":{"kind":"REWORK","_number":2,"created":"2020-12-18 12:22:44.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/05/767705/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/05/767705/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/05/767705/2"}}},"commit":{"parents":[{"commit":"858c0f2a62dcd874fa3ac5601b38c70f6f46a535","subject":"Merge \"Remove retired Searchlight support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/858c0f2a62dcd874fa3ac5601b38c70f6f46a535"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-18 12:17:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-18 12:22:35.000000000","tz":0},"subject":"WIP: CI: add Ubuntu jobs","message":"WIP: CI: add Ubuntu jobs\n\nDepends-On: https://review.opendev.org/766984\nDepends-On: https://review.opendev.org/766958\n\nChange-Id: I1985efae7c18f55c3ff7c27c17d6242523904f3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/61f68d65f24488ef329c7063ace6d5425ad556e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/61f68d65f24488ef329c7063ace6d5425ad556e0"}]},"branch":"refs/heads/master"},"fdc2bdf395f82c3e16992c013aa88badc52f97a3":{"kind":"REWORK","_number":3,"created":"2020-12-18 12:56:48.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/05/767705/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/05/767705/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/05/767705/3"}}},"commit":{"parents":[{"commit":"858c0f2a62dcd874fa3ac5601b38c70f6f46a535","subject":"Merge \"Remove retired Searchlight support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/858c0f2a62dcd874fa3ac5601b38c70f6f46a535"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-18 12:17:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-18 12:56:38.000000000","tz":0},"subject":"WIP: CI: add Ubuntu jobs","message":"WIP: CI: add Ubuntu jobs\n\nDepends-On: https://review.opendev.org/766984\nDepends-On: https://review.opendev.org/766958\n\nChange-Id: I1985efae7c18f55c3ff7c27c17d6242523904f3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fdc2bdf395f82c3e16992c013aa88badc52f97a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fdc2bdf395f82c3e16992c013aa88badc52f97a3"}]},"branch":"refs/heads/master"},"53fc1ff5fe9c5349edbca2e33ca23b860f2532f3":{"kind":"REWORK","_number":4,"created":"2020-12-18 13:58:38.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/05/767705/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/05/767705/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/05/767705/4"}}},"commit":{"parents":[{"commit":"858c0f2a62dcd874fa3ac5601b38c70f6f46a535","subject":"Merge \"Remove retired Searchlight support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/858c0f2a62dcd874fa3ac5601b38c70f6f46a535"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-18 12:17:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-18 13:58:27.000000000","tz":0},"subject":"WIP: CI: add Ubuntu jobs","message":"WIP: CI: add Ubuntu jobs\n\nDepends-On: https://review.opendev.org/766984\nDepends-On: https://review.opendev.org/766958\n\nChange-Id: I1985efae7c18f55c3ff7c27c17d6242523904f3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/53fc1ff5fe9c5349edbca2e33ca23b860f2532f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/53fc1ff5fe9c5349edbca2e33ca23b860f2532f3"}]},"branch":"refs/heads/master"},"e3d754e0a30c4127034006cfe914a61c0dc3d23b":{"kind":"REWORK","_number":5,"created":"2020-12-18 14:16:35.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/05/767705/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/05/767705/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/05/767705/5"}}},"commit":{"parents":[{"commit":"858c0f2a62dcd874fa3ac5601b38c70f6f46a535","subject":"Merge \"Remove retired Searchlight support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/858c0f2a62dcd874fa3ac5601b38c70f6f46a535"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-18 12:17:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-18 14:16:26.000000000","tz":0},"subject":"WIP: CI: add Ubuntu jobs","message":"WIP: CI: add Ubuntu jobs\n\nDepends-On: https://review.opendev.org/766984\nDepends-On: https://review.opendev.org/766958\n\nStory: 2004960\nTask: 29393\n\nChange-Id: I1985efae7c18f55c3ff7c27c17d6242523904f3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e3d754e0a30c4127034006cfe914a61c0dc3d23b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e3d754e0a30c4127034006cfe914a61c0dc3d23b"}]},"branch":"refs/heads/master"},"3f635cf99cd96ad8ad051d0d5223ccbae96515c1":{"kind":"REWORK","_number":6,"created":"2020-12-18 16:39:24.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/05/767705/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/05/767705/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/05/767705/6"}}},"commit":{"parents":[{"commit":"858c0f2a62dcd874fa3ac5601b38c70f6f46a535","subject":"Merge \"Remove retired Searchlight support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/858c0f2a62dcd874fa3ac5601b38c70f6f46a535"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-18 12:17:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-18 16:39:12.000000000","tz":0},"subject":"WIP: CI: add Ubuntu jobs","message":"WIP: CI: add Ubuntu jobs\n\nDepends-On: https://review.opendev.org/766984\nDepends-On: https://review.opendev.org/766958\n\nStory: 2004960\nTask: 29393\n\nChange-Id: I1985efae7c18f55c3ff7c27c17d6242523904f3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3f635cf99cd96ad8ad051d0d5223ccbae96515c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3f635cf99cd96ad8ad051d0d5223ccbae96515c1"}]},"branch":"refs/heads/master"},"222ebdf9f61ecc543267d57cf2e44c58a68e0962":{"kind":"REWORK","_number":7,"created":"2020-12-18 17:09:00.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/05/767705/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/05/767705/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/05/767705/7"}}},"commit":{"parents":[{"commit":"858c0f2a62dcd874fa3ac5601b38c70f6f46a535","subject":"Merge \"Remove retired Searchlight support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/858c0f2a62dcd874fa3ac5601b38c70f6f46a535"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-18 12:17:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-18 17:08:51.000000000","tz":0},"subject":"WIP: CI: add Ubuntu jobs","message":"WIP: CI: add Ubuntu jobs\n\nDepends-On: https://review.opendev.org/766984\nDepends-On: https://review.opendev.org/766958\n\nStory: 2004960\nTask: 29393\n\nChange-Id: I1985efae7c18f55c3ff7c27c17d6242523904f3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/222ebdf9f61ecc543267d57cf2e44c58a68e0962"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/222ebdf9f61ecc543267d57cf2e44c58a68e0962"}]},"branch":"refs/heads/master"},"0d5125ad3b77f575d32e5ca6e5846cbc9e5a4c93":{"kind":"REWORK","_number":8,"created":"2020-12-18 17:21:18.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/05/767705/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/05/767705/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/05/767705/8"}}},"commit":{"parents":[{"commit":"858c0f2a62dcd874fa3ac5601b38c70f6f46a535","subject":"Merge \"Remove retired Searchlight support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/858c0f2a62dcd874fa3ac5601b38c70f6f46a535"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-18 12:17:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-18 17:21:09.000000000","tz":0},"subject":"WIP: CI: add Ubuntu jobs","message":"WIP: CI: add Ubuntu jobs\n\nDepends-On: https://review.opendev.org/766984\nDepends-On: https://review.opendev.org/766958\n\nStory: 2004960\nTask: 29393\n\nChange-Id: I1985efae7c18f55c3ff7c27c17d6242523904f3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0d5125ad3b77f575d32e5ca6e5846cbc9e5a4c93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0d5125ad3b77f575d32e5ca6e5846cbc9e5a4c93"}]},"branch":"refs/heads/master"},"ea20580c8a297a1e6e4e855f4bff7f88a399d1dd":{"kind":"REWORK","_number":9,"created":"2020-12-18 17:41:01.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/05/767705/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/05/767705/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/05/767705/9"}}},"commit":{"parents":[{"commit":"858c0f2a62dcd874fa3ac5601b38c70f6f46a535","subject":"Merge \"Remove retired Searchlight support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/858c0f2a62dcd874fa3ac5601b38c70f6f46a535"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-18 12:17:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-18 17:40:53.000000000","tz":0},"subject":"WIP: CI: add Ubuntu jobs","message":"WIP: CI: add Ubuntu jobs\n\nDepends-On: https://review.opendev.org/766984\nDepends-On: https://review.opendev.org/766958\n\nStory: 2004960\nTask: 29393\n\nChange-Id: I1985efae7c18f55c3ff7c27c17d6242523904f3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ea20580c8a297a1e6e4e855f4bff7f88a399d1dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ea20580c8a297a1e6e4e855f4bff7f88a399d1dd"}]},"branch":"refs/heads/master"},"c624636ca542b1c55cc258e84dc5b082657aceb6":{"kind":"REWORK","_number":10,"created":"2020-12-18 18:06:04.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/05/767705/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/05/767705/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/05/767705/10"}}},"commit":{"parents":[{"commit":"858c0f2a62dcd874fa3ac5601b38c70f6f46a535","subject":"Merge \"Remove retired Searchlight support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/858c0f2a62dcd874fa3ac5601b38c70f6f46a535"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-18 12:17:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-18 18:05:53.000000000","tz":0},"subject":"WIP: CI: add Ubuntu jobs","message":"WIP: CI: add Ubuntu jobs\n\nDepends-On: https://review.opendev.org/766984\nDepends-On: https://review.opendev.org/766958\n\nStory: 2004960\nTask: 29393\n\nChange-Id: I1985efae7c18f55c3ff7c27c17d6242523904f3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c624636ca542b1c55cc258e84dc5b082657aceb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c624636ca542b1c55cc258e84dc5b082657aceb6"}]},"branch":"refs/heads/master"},"6fd234384c2500491ccac659ce59ea5c66520ff7":{"kind":"REWORK","_number":11,"created":"2020-12-18 18:36:27.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/05/767705/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/05/767705/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/05/767705/11"}}},"commit":{"parents":[{"commit":"858c0f2a62dcd874fa3ac5601b38c70f6f46a535","subject":"Merge \"Remove retired Searchlight support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/858c0f2a62dcd874fa3ac5601b38c70f6f46a535"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-18 12:17:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-18 18:36:19.000000000","tz":0},"subject":"WIP: CI: add Ubuntu jobs","message":"WIP: CI: add Ubuntu jobs\n\nDepends-On: https://review.opendev.org/766984\nDepends-On: https://review.opendev.org/766958\n\nStory: 2004960\nTask: 29393\n\nChange-Id: I1985efae7c18f55c3ff7c27c17d6242523904f3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6fd234384c2500491ccac659ce59ea5c66520ff7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6fd234384c2500491ccac659ce59ea5c66520ff7"}]},"branch":"refs/heads/master"},"0ce7cdca87b51294fa1190fb0d2d839c19a3cbeb":{"kind":"REWORK","_number":12,"created":"2020-12-18 19:24:28.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/05/767705/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/05/767705/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/05/767705/12"}}},"commit":{"parents":[{"commit":"858c0f2a62dcd874fa3ac5601b38c70f6f46a535","subject":"Merge \"Remove retired Searchlight support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/858c0f2a62dcd874fa3ac5601b38c70f6f46a535"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-18 12:17:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-18 19:24:17.000000000","tz":0},"subject":"WIP: CI: add Ubuntu jobs","message":"WIP: CI: add Ubuntu jobs\n\nDepends-On: https://review.opendev.org/766984\nDepends-On: https://review.opendev.org/766958\n\nStory: 2004960\nTask: 29393\n\nChange-Id: I1985efae7c18f55c3ff7c27c17d6242523904f3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0ce7cdca87b51294fa1190fb0d2d839c19a3cbeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0ce7cdca87b51294fa1190fb0d2d839c19a3cbeb"}]},"branch":"refs/heads/master"},"fc9e798b287c0de67b0769a8e9e75daa194351a5":{"kind":"REWORK","_number":13,"created":"2020-12-19 09:02:56.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/05/767705/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/05/767705/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/05/767705/13"}}},"commit":{"parents":[{"commit":"858c0f2a62dcd874fa3ac5601b38c70f6f46a535","subject":"Merge \"Remove retired Searchlight support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/858c0f2a62dcd874fa3ac5601b38c70f6f46a535"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-18 12:17:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-19 09:02:41.000000000","tz":0},"subject":"WIP: CI: add Ubuntu jobs","message":"WIP: CI: add Ubuntu jobs\n\nDepends-On: https://review.opendev.org/766984\nDepends-On: https://review.opendev.org/766958\n\nStory: 2004960\nTask: 29393\n\nChange-Id: I1985efae7c18f55c3ff7c27c17d6242523904f3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fc9e798b287c0de67b0769a8e9e75daa194351a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fc9e798b287c0de67b0769a8e9e75daa194351a5"}]},"branch":"refs/heads/master"},"9ebb9ba39e1a8e10c1dcb8008d35c334370dc6a0":{"kind":"REWORK","_number":14,"created":"2020-12-21 09:23:45.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/05/767705/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/05/767705/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/05/767705/14"}}},"commit":{"parents":[{"commit":"858c0f2a62dcd874fa3ac5601b38c70f6f46a535","subject":"Merge \"Remove retired Searchlight support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/858c0f2a62dcd874fa3ac5601b38c70f6f46a535"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-18 12:17:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-21 09:23:33.000000000","tz":0},"subject":"WIP: CI: add Ubuntu jobs","message":"WIP: CI: add Ubuntu jobs\n\n* Use source images\n* Need to specify bash for \u0026\u003e syntax\n\nIssues worked around:\n\n* Manually configuring bridge via ip commands makes ifup fail to bring\n  up the link\n\n* veth role had no Ubuntu support. Added manual commands\n\n* python-setuptools was required on a dev VM to install kolla-ansible.\n  Why is it using python2?\n\n* bridge not active after interfaces role bounce. Added a pause, similar\n  to https://github.com/michaelrigart/ansible-role-interfaces/pull/31\n\n* fails creating pip config for kolla user.\n  Failed to set permissions on the temporary files Ansible needs to\n  create when becoming an unprivileged user (rc: 1, err: chown: changing\n  ownership of\n  \u0027/var/tmp/ansible-tmp-1608370439.1820047-8575-75139340648971/\u0027:\n  Operation not permitted\\nchown: changing ownership of\n  \u0027/var/tmp/ansible-tmp-1608370439.1820047-8575-75139340648971/AnsiballZ_file.py\u0027:\n  Operation not permitted\\n}). For information on working around this,\n  see\n  https://docs.ansible.com/ansible/become.html#becoming-an-unprivileged-user\n\nDepends-On: https://review.opendev.org/766984\nDepends-On: https://review.opendev.org/766958\n\nStory: 2004960\nTask: 29393\n\nChange-Id: I1985efae7c18f55c3ff7c27c17d6242523904f3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9ebb9ba39e1a8e10c1dcb8008d35c334370dc6a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9ebb9ba39e1a8e10c1dcb8008d35c334370dc6a0"}]},"branch":"refs/heads/master"},"1e2bfeeb7753f4a548bfcf2120b6420ea49049d7":{"kind":"REWORK","_number":15,"created":"2020-12-21 09:32:49.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/05/767705/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/05/767705/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/05/767705/15"}}},"commit":{"parents":[{"commit":"858c0f2a62dcd874fa3ac5601b38c70f6f46a535","subject":"Merge \"Remove retired Searchlight support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/858c0f2a62dcd874fa3ac5601b38c70f6f46a535"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-18 12:17:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-21 09:32:41.000000000","tz":0},"subject":"WIP: CI: add Ubuntu jobs","message":"WIP: CI: add Ubuntu jobs\n\n* Use source images\n* Need to specify bash for \u0026\u003e syntax\n\nIssues worked around:\n\n* Manually configuring bridge via ip commands makes ifup fail to bring\n  up the link\n\n* veth role had no Ubuntu support. Added manual commands\n\n* python-setuptools was required on a dev VM to install kolla-ansible.\n  Why is it using python2?\n\n* bridge not active after interfaces role bounce. Added a pause, similar\n  to https://github.com/michaelrigart/ansible-role-interfaces/pull/31\n\n* fails creating pip config for kolla user.\n  Failed to set permissions on the temporary files Ansible needs to\n  create when becoming an unprivileged user (rc: 1, err: chown: changing\n  ownership of\n  \u0027/var/tmp/ansible-tmp-1608370439.1820047-8575-75139340648971/\u0027:\n  Operation not permitted\\nchown: changing ownership of\n  \u0027/var/tmp/ansible-tmp-1608370439.1820047-8575-75139340648971/AnsiballZ_file.py\u0027:\n  Operation not permitted\\n}). For information on working around this,\n  see\n  https://docs.ansible.com/ansible/become.html#becoming-an-unprivileged-user\n\nDepends-On: https://review.opendev.org/766984\nDepends-On: https://review.opendev.org/766958\n\nStory: 2004960\nTask: 29393\n\nChange-Id: I1985efae7c18f55c3ff7c27c17d6242523904f3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1e2bfeeb7753f4a548bfcf2120b6420ea49049d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1e2bfeeb7753f4a548bfcf2120b6420ea49049d7"}]},"branch":"refs/heads/master"},"d48b570800f3681720e3dcd08e2434d08d04c599":{"kind":"REWORK","_number":16,"created":"2020-12-21 09:41:54.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/05/767705/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/05/767705/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/05/767705/16"}}},"commit":{"parents":[{"commit":"858c0f2a62dcd874fa3ac5601b38c70f6f46a535","subject":"Merge \"Remove retired Searchlight support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/858c0f2a62dcd874fa3ac5601b38c70f6f46a535"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-18 12:17:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-21 09:41:45.000000000","tz":0},"subject":"WIP: CI: add Ubuntu jobs","message":"WIP: CI: add Ubuntu jobs\n\n* Use source images\n* Need to specify bash for \u0026\u003e syntax\n\nIssues worked around:\n\n* Manually configuring bridge via ip commands makes ifup fail to bring\n  up the link\n\n* veth role had no Ubuntu support. Added manual commands\n\n* python-setuptools was required on a dev VM to install kolla-ansible.\n  Why is it using python2?\n\n* bridge not active after interfaces role bounce. Added a pause, similar\n  to https://github.com/michaelrigart/ansible-role-interfaces/pull/31\n\n* fails creating pip config for kolla user.\n  Failed to set permissions on the temporary files Ansible needs to\n  create when becoming an unprivileged user (rc: 1, err: chown: changing\n  ownership of\n  \u0027/var/tmp/ansible-tmp-1608370439.1820047-8575-75139340648971/\u0027:\n  Operation not permitted\\nchown: changing ownership of\n  \u0027/var/tmp/ansible-tmp-1608370439.1820047-8575-75139340648971/AnsiballZ_file.py\u0027:\n  Operation not permitted\\n}). For information on working around this,\n  see\n  https://docs.ansible.com/ansible/become.html#becoming-an-unprivileged-user\n\nDepends-On: https://review.opendev.org/766984\nDepends-On: https://review.opendev.org/766958\n\nStory: 2004960\nTask: 29393\n\nChange-Id: I1985efae7c18f55c3ff7c27c17d6242523904f3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d48b570800f3681720e3dcd08e2434d08d04c599"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d48b570800f3681720e3dcd08e2434d08d04c599"}]},"branch":"refs/heads/master"},"7067f95f15f2fd5e33f7297d8fddcdcd2db9153e":{"kind":"REWORK","_number":17,"created":"2020-12-21 10:48:14.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/05/767705/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/05/767705/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/05/767705/17"}}},"commit":{"parents":[{"commit":"858c0f2a62dcd874fa3ac5601b38c70f6f46a535","subject":"Merge \"Remove retired Searchlight support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/858c0f2a62dcd874fa3ac5601b38c70f6f46a535"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-18 12:17:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-21 10:46:24.000000000","tz":0},"subject":"WIP: CI: add Ubuntu jobs","message":"WIP: CI: add Ubuntu jobs\n\n* Use source images\n* Need to specify bash for \u0026\u003e syntax\n\nIssues worked around:\n\n* Manually configuring bridge via ip commands makes ifup fail to bring\n  up the link\n\n* veth role had no Ubuntu support. Added manual commands\n\n* python-setuptools was required on a dev VM to install kolla-ansible.\n  Why is it using python2?\n\n* bridge not active after interfaces role bounce. Added a pause, similar\n  to https://github.com/michaelrigart/ansible-role-interfaces/pull/31\n\n* fails creating pip config for kolla user.\n  Failed to set permissions on the temporary files Ansible needs to\n  create when becoming an unprivileged user (rc: 1, err: chown: changing\n  ownership of\n  \u0027/var/tmp/ansible-tmp-1608370439.1820047-8575-75139340648971/\u0027:\n  Operation not permitted\\nchown: changing ownership of\n  \u0027/var/tmp/ansible-tmp-1608370439.1820047-8575-75139340648971/AnsiballZ_file.py\u0027:\n  Operation not permitted\\n}). For information on working around this,\n  see\n  https://docs.ansible.com/ansible/become.html#becoming-an-unprivileged-user\n\n* fails installing docker python module for kolla user. WARNING: The\n  repository located at mirror-int.ord.rax.opendev.org is not a trusted\n  or secure host and is being ignored ERROR: No matching distribution\n  found for docker\u003d\u003d\u003d4.4.0 Adding trusted host for PyPI mirror.\n\nDepends-On: https://review.opendev.org/766984\nDepends-On: https://review.opendev.org/766958\n\nStory: 2004960\nTask: 29393\n\nChange-Id: I1985efae7c18f55c3ff7c27c17d6242523904f3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7067f95f15f2fd5e33f7297d8fddcdcd2db9153e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7067f95f15f2fd5e33f7297d8fddcdcd2db9153e"}]},"branch":"refs/heads/master"},"43ca45ed2d284714d32d0ded87c4a94ebc79c7db":{"kind":"REWORK","_number":18,"created":"2020-12-21 12:09:48.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/05/767705/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/05/767705/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/05/767705/18"}}},"commit":{"parents":[{"commit":"858c0f2a62dcd874fa3ac5601b38c70f6f46a535","subject":"Merge \"Remove retired Searchlight support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/858c0f2a62dcd874fa3ac5601b38c70f6f46a535"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-18 12:17:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-21 12:08:56.000000000","tz":0},"subject":"WIP: CI: add Ubuntu jobs","message":"WIP: CI: add Ubuntu jobs\n\n* Use source images\n* Need to specify bash for \u0026\u003e syntax\n\nIssues worked around:\n\n* Manually configuring bridge via ip commands makes ifup fail to bring\n  up the link\n\n* veth role had no Ubuntu support. Added manual commands\n\n* python-setuptools was required on a dev VM to install kolla-ansible.\n  Why is it using python2?\n\n* bridge not active after interfaces role bounce. Added a pause, similar\n  to https://github.com/michaelrigart/ansible-role-interfaces/pull/31\n\n* fails creating pip config for kolla user.\n  Failed to set permissions on the temporary files Ansible needs to\n  create when becoming an unprivileged user (rc: 1, err: chown: changing\n  ownership of\n  \u0027/var/tmp/ansible-tmp-1608370439.1820047-8575-75139340648971/\u0027:\n  Operation not permitted\\nchown: changing ownership of\n  \u0027/var/tmp/ansible-tmp-1608370439.1820047-8575-75139340648971/AnsiballZ_file.py\u0027:\n  Operation not permitted\\n}). For information on working around this,\n  see\n  https://docs.ansible.com/ansible/become.html#becoming-an-unprivileged-user\n\n* fails installing docker python module for kolla user. WARNING: The\n  repository located at mirror-int.ord.rax.opendev.org is not a trusted\n  or secure host and is being ignored ERROR: No matching distribution\n  found for docker\u003d\u003d\u003d4.4.0 Adding trusted host for PyPI mirror.\n\n* switch from \u0027source\u0027 to \u0027.\u0027 in Ansible tasks, since Ubuntu uses sh\n  rather than bash for execution of Ansible tasks.\n\nDepends-On: https://review.opendev.org/766984\nDepends-On: https://review.opendev.org/766958\n\nStory: 2004960\nTask: 29393\n\nChange-Id: I1985efae7c18f55c3ff7c27c17d6242523904f3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/43ca45ed2d284714d32d0ded87c4a94ebc79c7db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/43ca45ed2d284714d32d0ded87c4a94ebc79c7db"}]},"branch":"refs/heads/master"},"04ebaad01b53e07178256523697d35a858c1bc9a":{"kind":"REWORK","_number":19,"created":"2020-12-21 13:06:23.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/05/767705/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/05/767705/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/05/767705/19"}}},"commit":{"parents":[{"commit":"858c0f2a62dcd874fa3ac5601b38c70f6f46a535","subject":"Merge \"Remove retired Searchlight support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/858c0f2a62dcd874fa3ac5601b38c70f6f46a535"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-18 12:17:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-21 13:05:25.000000000","tz":0},"subject":"WIP: CI: add Ubuntu jobs","message":"WIP: CI: add Ubuntu jobs\n\n* Use source images\n* Need to specify bash for \u0026\u003e syntax\n\nIssues worked around:\n\n* Manually configuring bridge via ip commands makes ifup fail to bring\n  up the link\n\n* veth role had no Ubuntu support. Added manual commands\n\n* python-setuptools was required on a dev VM to install kolla-ansible.\n  Why is it using python2?\n\n* bridge not active after interfaces role bounce. Added a pause, similar\n  to https://github.com/michaelrigart/ansible-role-interfaces/pull/31\n\n* fails creating pip config for kolla user.\n  Failed to set permissions on the temporary files Ansible needs to\n  create when becoming an unprivileged user (rc: 1, err: chown: changing\n  ownership of\n  \u0027/var/tmp/ansible-tmp-1608370439.1820047-8575-75139340648971/\u0027:\n  Operation not permitted\\nchown: changing ownership of\n  \u0027/var/tmp/ansible-tmp-1608370439.1820047-8575-75139340648971/AnsiballZ_file.py\u0027:\n  Operation not permitted\\n}). For information on working around this,\n  see\n  https://docs.ansible.com/ansible/become.html#becoming-an-unprivileged-user\n\n* fails installing docker python module for kolla user. WARNING: The\n  repository located at mirror-int.ord.rax.opendev.org is not a trusted\n  or secure host and is being ignored ERROR: No matching distribution\n  found for docker\u003d\u003d\u003d4.4.0 Adding trusted host for PyPI mirror.\n\n* switch from \u0027source\u0027 to \u0027.\u0027 in Ansible tasks, since Ubuntu uses sh\n  rather than bash for execution of Ansible tasks.\n\n* Tenks fails to create block devices - missing qemu-img (in qemu-utils)\n\nDepends-On: https://review.opendev.org/766984\nDepends-On: https://review.opendev.org/766958\n\nStory: 2004960\nTask: 29393\n\nChange-Id: I1985efae7c18f55c3ff7c27c17d6242523904f3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/04ebaad01b53e07178256523697d35a858c1bc9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/04ebaad01b53e07178256523697d35a858c1bc9a"}]},"branch":"refs/heads/master"},"90a4b973edf029beb4a55918c2c3c0b8a57bd73b":{"kind":"REWORK","_number":20,"created":"2020-12-21 14:07:09.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/05/767705/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/05/767705/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/05/767705/20"}}},"commit":{"parents":[{"commit":"858c0f2a62dcd874fa3ac5601b38c70f6f46a535","subject":"Merge \"Remove retired Searchlight support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/858c0f2a62dcd874fa3ac5601b38c70f6f46a535"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-18 12:17:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-21 14:07:04.000000000","tz":0},"subject":"WIP: CI: add Ubuntu jobs","message":"WIP: CI: add Ubuntu jobs\n\n* Use source images\n* Need to specify bash for \u0026\u003e syntax\n\nIssues worked around:\n\n* Manually configuring bridge via ip commands makes ifup fail to bring\n  up the link\n\n* veth role had no Ubuntu support. Added manual commands\n\n* python-setuptools was required on a dev VM to install kolla-ansible.\n  Why is it using python2?\n\n* bridge not active after interfaces role bounce. Added a pause, similar\n  to https://github.com/michaelrigart/ansible-role-interfaces/pull/31\n\n* fails creating pip config for kolla user.\n  Failed to set permissions on the temporary files Ansible needs to\n  create when becoming an unprivileged user (rc: 1, err: chown: changing\n  ownership of\n  \u0027/var/tmp/ansible-tmp-1608370439.1820047-8575-75139340648971/\u0027:\n  Operation not permitted\\nchown: changing ownership of\n  \u0027/var/tmp/ansible-tmp-1608370439.1820047-8575-75139340648971/AnsiballZ_file.py\u0027:\n  Operation not permitted\\n}). For information on working around this,\n  see\n  https://docs.ansible.com/ansible/become.html#becoming-an-unprivileged-user\n\n* fails installing docker python module for kolla user. WARNING: The\n  repository located at mirror-int.ord.rax.opendev.org is not a trusted\n  or secure host and is being ignored ERROR: No matching distribution\n  found for docker\u003d\u003d\u003d4.4.0 Adding trusted host for PyPI mirror.\n\n* switch from \u0027source\u0027 to \u0027.\u0027 in Ansible tasks, since Ubuntu uses sh\n  rather than bash for execution of Ansible tasks.\n\n* Tenks fails to create block devices - missing qemu-img (in qemu-utils)\n\nDepends-On: https://review.opendev.org/766984\nDepends-On: https://review.opendev.org/766958\n\nStory: 2004960\nTask: 29393\n\nChange-Id: I1985efae7c18f55c3ff7c27c17d6242523904f3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/90a4b973edf029beb4a55918c2c3c0b8a57bd73b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/90a4b973edf029beb4a55918c2c3c0b8a57bd73b"}]},"branch":"refs/heads/master"},"2906f5349511964b36ccd93c627c3c1ed0adc192":{"kind":"REWORK","_number":21,"created":"2020-12-21 14:59:56.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/05/767705/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/05/767705/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/05/767705/21"}}},"commit":{"parents":[{"commit":"858c0f2a62dcd874fa3ac5601b38c70f6f46a535","subject":"Merge \"Remove retired Searchlight support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/858c0f2a62dcd874fa3ac5601b38c70f6f46a535"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-18 12:17:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-21 14:59:51.000000000","tz":0},"subject":"WIP: CI: add Ubuntu jobs","message":"WIP: CI: add Ubuntu jobs\n\n* Use source images\n* Need to specify bash for \u0026\u003e syntax\n\nIssues worked around:\n\n* Manually configuring bridge via ip commands makes ifup fail to bring\n  up the link\n\n* veth role had no Ubuntu support. Added manual commands\n\n* python-setuptools was required on a dev VM to install kolla-ansible.\n  Why is it using python2?\n\n* bridge not active after interfaces role bounce. Added a pause, similar\n  to https://github.com/michaelrigart/ansible-role-interfaces/pull/31\n\n* fails creating pip config for kolla user.\n  Failed to set permissions on the temporary files Ansible needs to\n  create when becoming an unprivileged user (rc: 1, err: chown: changing\n  ownership of\n  \u0027/var/tmp/ansible-tmp-1608370439.1820047-8575-75139340648971/\u0027:\n  Operation not permitted\\nchown: changing ownership of\n  \u0027/var/tmp/ansible-tmp-1608370439.1820047-8575-75139340648971/AnsiballZ_file.py\u0027:\n  Operation not permitted\\n}). For information on working around this,\n  see\n  https://docs.ansible.com/ansible/become.html#becoming-an-unprivileged-user\n\n* fails installing docker python module for kolla user. WARNING: The\n  repository located at mirror-int.ord.rax.opendev.org is not a trusted\n  or secure host and is being ignored ERROR: No matching distribution\n  found for docker\u003d\u003d\u003d4.4.0 Adding trusted host for PyPI mirror.\n\n* switch from \u0027source\u0027 to \u0027.\u0027 in Ansible tasks, since Ubuntu uses sh\n  rather than bash for execution of Ansible tasks.\n\n* Tenks fails to create block devices - missing qemu-img (in qemu-utils)\n\nDepends-On: https://review.opendev.org/766984\nDepends-On: https://review.opendev.org/766958\n\nStory: 2004960\nTask: 29393\n\nChange-Id: I1985efae7c18f55c3ff7c27c17d6242523904f3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2906f5349511964b36ccd93c627c3c1ed0adc192"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2906f5349511964b36ccd93c627c3c1ed0adc192"}]},"branch":"refs/heads/master"},"ddab772ec57c48a8245ee051391e2a27f5cd65a5":{"kind":"REWORK","_number":22,"created":"2020-12-21 15:27:37.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/05/767705/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/05/767705/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/05/767705/22"}}},"commit":{"parents":[{"commit":"858c0f2a62dcd874fa3ac5601b38c70f6f46a535","subject":"Merge \"Remove retired Searchlight support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/858c0f2a62dcd874fa3ac5601b38c70f6f46a535"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-18 12:17:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-21 15:27:00.000000000","tz":0},"subject":"WIP: CI: add Ubuntu jobs","message":"WIP: CI: add Ubuntu jobs\n\n* Use source images\n* Need to specify bash for \u0026\u003e syntax\n\nIssues worked around:\n\n* Manually configuring bridge via ip commands makes ifup fail to bring\n  up the link\n\n* veth role had no Ubuntu support. Added manual commands\n\n* python-setuptools was required on a dev VM to install kolla-ansible.\n  Why is it using python2?\n\n* bridge not active after interfaces role bounce. Added a pause, similar\n  to https://github.com/michaelrigart/ansible-role-interfaces/pull/31\n\n* fails creating pip config for kolla user.\n  Failed to set permissions on the temporary files Ansible needs to\n  create when becoming an unprivileged user (rc: 1, err: chown: changing\n  ownership of\n  \u0027/var/tmp/ansible-tmp-1608370439.1820047-8575-75139340648971/\u0027:\n  Operation not permitted\\nchown: changing ownership of\n  \u0027/var/tmp/ansible-tmp-1608370439.1820047-8575-75139340648971/AnsiballZ_file.py\u0027:\n  Operation not permitted\\n}). For information on working around this,\n  see\n  https://docs.ansible.com/ansible/become.html#becoming-an-unprivileged-user\n\n* fails installing docker python module for kolla user. WARNING: The\n  repository located at mirror-int.ord.rax.opendev.org is not a trusted\n  or secure host and is being ignored ERROR: No matching distribution\n  found for docker\u003d\u003d\u003d4.4.0 Adding trusted host for PyPI mirror.\n\n* switch from \u0027source\u0027 to \u0027.\u0027 in Ansible tasks, since Ubuntu uses sh\n  rather than bash for execution of Ansible tasks.\n\n* Tenks fails to create block devices - missing qemu-img (in qemu-utils)\n\n* Tenks should use python3\n\n* Tenks qemu emulator is different on Ubuntu\n\nDepends-On: https://review.opendev.org/766984\nDepends-On: https://review.opendev.org/766958\n\nStory: 2004960\nTask: 29393\n\nChange-Id: I1985efae7c18f55c3ff7c27c17d6242523904f3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ddab772ec57c48a8245ee051391e2a27f5cd65a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ddab772ec57c48a8245ee051391e2a27f5cd65a5"}]},"branch":"refs/heads/master"},"1ab79bf195e79dffd5643361ce48ec0709e889e3":{"kind":"REWORK","_number":23,"created":"2020-12-21 15:28:42.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/05/767705/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/05/767705/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/05/767705/23"}}},"commit":{"parents":[{"commit":"858c0f2a62dcd874fa3ac5601b38c70f6f46a535","subject":"Merge \"Remove retired Searchlight support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/858c0f2a62dcd874fa3ac5601b38c70f6f46a535"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-18 12:17:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-21 15:28:29.000000000","tz":0},"subject":"WIP: CI: add Ubuntu jobs","message":"WIP: CI: add Ubuntu jobs\n\n* Use source images\n* Need to specify bash for \u0026\u003e syntax\n\nIssues worked around:\n\n* Manually configuring bridge via ip commands makes ifup fail to bring\n  up the link\n\n* veth role had no Ubuntu support. Added manual commands\n\n* python-setuptools was required on a dev VM to install kolla-ansible.\n  Why is it using python2?\n\n* bridge not active after interfaces role bounce. Added a pause, similar\n  to https://github.com/michaelrigart/ansible-role-interfaces/pull/31\n\n* fails creating pip config for kolla user.\n  Failed to set permissions on the temporary files Ansible needs to\n  create when becoming an unprivileged user (rc: 1, err: chown: changing\n  ownership of\n  \u0027/var/tmp/ansible-tmp-1608370439.1820047-8575-75139340648971/\u0027:\n  Operation not permitted\\nchown: changing ownership of\n  \u0027/var/tmp/ansible-tmp-1608370439.1820047-8575-75139340648971/AnsiballZ_file.py\u0027:\n  Operation not permitted\\n}). For information on working around this,\n  see\n  https://docs.ansible.com/ansible/become.html#becoming-an-unprivileged-user\n\n* fails installing docker python module for kolla user. WARNING: The\n  repository located at mirror-int.ord.rax.opendev.org is not a trusted\n  or secure host and is being ignored ERROR: No matching distribution\n  found for docker\u003d\u003d\u003d4.4.0 Adding trusted host for PyPI mirror.\n\n* switch from \u0027source\u0027 to \u0027.\u0027 in Ansible tasks, since Ubuntu uses sh\n  rather than bash for execution of Ansible tasks.\n\n* Tenks fails to create block devices - missing qemu-img (in qemu-utils)\n\n* Tenks qemu emulator is different on Ubuntu\n\nDepends-On: https://review.opendev.org/766984\nDepends-On: https://review.opendev.org/766958\n\nStory: 2004960\nTask: 29393\n\nChange-Id: I1985efae7c18f55c3ff7c27c17d6242523904f3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1ab79bf195e79dffd5643361ce48ec0709e889e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1ab79bf195e79dffd5643361ce48ec0709e889e3"}]},"branch":"refs/heads/master"},"01aab2080a798c232754796e4954b4e94381069d":{"kind":"REWORK","_number":24,"created":"2020-12-22 13:57:56.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/05/767705/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/05/767705/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/05/767705/24"}}},"commit":{"parents":[{"commit":"50dddc3ace4c4c40e3d1680ddcceb789a47ae3fb","subject":"Ubuntu: fix pip configuration for kolla user","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/50dddc3ace4c4c40e3d1680ddcceb789a47ae3fb"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-18 12:17:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-22 13:57:25.000000000","tz":0},"subject":"WIP: CI: add Ubuntu jobs","message":"WIP: CI: add Ubuntu jobs\n\n* Use source images\n* Need to specify bash for \u0026\u003e syntax\n\nIssues worked around:\n\n* Manually configuring bridge via ip commands makes ifup fail to bring\n  up the link\n\n* veth role had no Ubuntu support. Added manual commands\n\n* python-setuptools was required on a dev VM to install kolla-ansible.\n  Why is it using python2?\n\n* bridge not active after interfaces role bounce. Added a pause, similar\n  to https://github.com/michaelrigart/ansible-role-interfaces/pull/31\n\n* fails installing docker python module for kolla user. WARNING: The\n  repository located at mirror-int.ord.rax.opendev.org is not a trusted\n  or secure host and is being ignored ERROR: No matching distribution\n  found for docker\u003d\u003d\u003d4.4.0 Adding trusted host for PyPI mirror.\n\n* Tenks fails to create block devices - missing qemu-img (in qemu-utils)\n\n* Tenks qemu emulator is different on Ubuntu\n\nDepends-On: https://review.opendev.org/766984\nDepends-On: https://review.opendev.org/766958\n\nStory: 2004960\nTask: 29393\n\nChange-Id: I1985efae7c18f55c3ff7c27c17d6242523904f3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/01aab2080a798c232754796e4954b4e94381069d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/01aab2080a798c232754796e4954b4e94381069d"}]},"branch":"refs/heads/master"},"38967472361ded496b7ed3f6915990a1d5707b7b":{"kind":"REWORK","_number":25,"created":"2020-12-22 17:25:26.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/05/767705/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/05/767705/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/05/767705/25"}}},"commit":{"parents":[{"commit":"50dddc3ace4c4c40e3d1680ddcceb789a47ae3fb","subject":"Ubuntu: fix pip configuration for kolla user","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/50dddc3ace4c4c40e3d1680ddcceb789a47ae3fb"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-18 12:17:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-22 17:25:20.000000000","tz":0},"subject":"WIP: CI: add Ubuntu jobs","message":"WIP: CI: add Ubuntu jobs\n\n* Use source images\n* Need to specify bash for \u0026\u003e syntax\n\nIssues worked around:\n\n* Manually configuring bridge via ip commands makes ifup fail to bring\n  up the link\n\n* veth role had no Ubuntu support. Added manual commands\n\n* python-setuptools was required on a dev VM to install kolla-ansible.\n  Why is it using python2?\n\n* bridge not active after interfaces role bounce. Added a pause, similar\n  to https://github.com/michaelrigart/ansible-role-interfaces/pull/31\n\n* fails installing docker python module for kolla user. WARNING: The\n  repository located at mirror-int.ord.rax.opendev.org is not a trusted\n  or secure host and is being ignored ERROR: No matching distribution\n  found for docker\u003d\u003d\u003d4.4.0 Adding trusted host for PyPI mirror.\n\n* Tenks fails to create block devices - missing qemu-img (in qemu-utils)\n\n* Tenks qemu emulator is different on Ubuntu\n\nDepends-On: https://review.opendev.org/766984\nDepends-On: https://review.opendev.org/766958\n\nStory: 2004960\nTask: 29393\n\nChange-Id: I1985efae7c18f55c3ff7c27c17d6242523904f3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/38967472361ded496b7ed3f6915990a1d5707b7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/38967472361ded496b7ed3f6915990a1d5707b7b"}]},"branch":"refs/heads/master"},"544b7006fc5acd57e6eb51d0ffd8dbbfdcf1d280":{"kind":"REWORK","_number":26,"created":"2021-01-07 13:45:40.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/05/767705/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/05/767705/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/05/767705/26"}}},"commit":{"parents":[{"commit":"f9b02c8bf5d4cfe8ff29aac6181a38771f267037","subject":"Ubuntu: support non-persistent veth configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f9b02c8bf5d4cfe8ff29aac6181a38771f267037"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-18 12:17:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-01-07 13:42:49.000000000","tz":0},"subject":"WIP: CI: add Ubuntu jobs","message":"WIP: CI: add Ubuntu jobs\n\n* Use source images\n* Need to specify bash for \u0026\u003e syntax\n\nIssues worked around:\n\n* Manually configuring bridge via ip commands makes ifup fail to bring\n  up the link. Adds a kayobe-network-bootstrap Zuul CI role that adds\n  persistent configuration for the all-in-one network.\n\n* bridge not active after interfaces role bounce. Added a pause, similar\n  to https://github.com/michaelrigart/ansible-role-interfaces/pull/31\n\n* fails installing docker python module for kolla user. WARNING: The\n  repository located at mirror-int.ord.rax.opendev.org is not a trusted\n  or secure host and is being ignored ERROR: No matching distribution\n  found for docker\u003d\u003d\u003d4.4.0 Adding trusted host for PyPI mirror.\n\n* Tenks fails to create block devices - missing qemu-img (in qemu-utils)\n\n* Tenks qemu emulator is different on Ubuntu\n\nDepends-On: https://review.opendev.org/766984\nDepends-On: https://review.opendev.org/766958\n\nStory: 2004960\nTask: 29393\n\nChange-Id: I1985efae7c18f55c3ff7c27c17d6242523904f3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/544b7006fc5acd57e6eb51d0ffd8dbbfdcf1d280"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/544b7006fc5acd57e6eb51d0ffd8dbbfdcf1d280"}]},"branch":"refs/heads/master"},"4297624a7df41f48a4211a00a6ff85d9c02f7b90":{"kind":"REWORK","_number":27,"created":"2021-01-07 13:54:51.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/05/767705/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/05/767705/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/05/767705/27"}}},"commit":{"parents":[{"commit":"f9b02c8bf5d4cfe8ff29aac6181a38771f267037","subject":"Ubuntu: support non-persistent veth configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f9b02c8bf5d4cfe8ff29aac6181a38771f267037"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-18 12:17:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-01-07 13:54:41.000000000","tz":0},"subject":"WIP: CI: add Ubuntu overcloud deploy job","message":"WIP: CI: add Ubuntu overcloud deploy job\n\n* Use source images\n* Need to specify bash for \u0026\u003e syntax\n\nIssues worked around:\n\n* Manually configuring bridge via ip commands makes ifup fail to bring\n  up the link. Adds a kayobe-network-bootstrap Zuul CI role that adds\n  persistent configuration for the all-in-one network.\n\n* bridge not active after interfaces role bounce. Added a pause, similar\n  to https://github.com/michaelrigart/ansible-role-interfaces/pull/31\n\n* fails installing docker python module for kolla user. WARNING: The\n  repository located at mirror-int.ord.rax.opendev.org is not a trusted\n  or secure host and is being ignored ERROR: No matching distribution\n  found for docker\u003d\u003d\u003d4.4.0 Adding trusted host for PyPI mirror.\n\n* Tenks fails to create block devices - missing qemu-img (in qemu-utils)\n\n* Tenks qemu emulator is different on Ubuntu\n\nDepends-On: https://review.opendev.org/766984\nDepends-On: https://review.opendev.org/766958\n\nStory: 2004960\nTask: 29393\n\nChange-Id: I1985efae7c18f55c3ff7c27c17d6242523904f3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4297624a7df41f48a4211a00a6ff85d9c02f7b90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4297624a7df41f48a4211a00a6ff85d9c02f7b90"}]},"branch":"refs/heads/master"},"6b8eb899cef856825c85c3ff2f907794be7c1eb9":{"kind":"REWORK","_number":28,"created":"2021-01-07 15:27:03.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/05/767705/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/05/767705/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/05/767705/28"}}},"commit":{"parents":[{"commit":"f9b02c8bf5d4cfe8ff29aac6181a38771f267037","subject":"Ubuntu: support non-persistent veth configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f9b02c8bf5d4cfe8ff29aac6181a38771f267037"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-18 12:17:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-01-07 15:21:25.000000000","tz":0},"subject":"CI: add Ubuntu overcloud deploy job","message":"CI: add Ubuntu overcloud deploy job\n\n* Use source images\n* Need to specify bash for \u0026\u003e syntax\n\nIssues worked around:\n\n* Manually configuring bridge via ip commands makes ifup fail to bring\n  up the link. Adds a kayobe-network-bootstrap Zuul CI role that adds\n  persistent configuration for the all-in-one network.\n\n* bridge not active after interfaces role bounce. Added a pause, similar\n  to https://github.com/michaelrigart/ansible-role-interfaces/pull/31\n\n* fails installing docker python module for kolla user. WARNING: The\n  repository located at mirror-int.ord.rax.opendev.org is not a trusted\n  or secure host and is being ignored ERROR: No matching distribution\n  found for docker\u003d\u003d\u003d4.4.0 Adding trusted host for PyPI mirror.\n\n* Tenks fails to create block devices - missing qemu-img (in qemu-utils)\n\n* Tenks qemu emulator is different on Ubuntu\n\nDepends-On: https://review.opendev.org/766984\nDepends-On: https://review.opendev.org/766958\n\nStory: 2004960\nTask: 29393\n\nChange-Id: I1985efae7c18f55c3ff7c27c17d6242523904f3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6b8eb899cef856825c85c3ff2f907794be7c1eb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6b8eb899cef856825c85c3ff2f907794be7c1eb9"}]},"branch":"refs/heads/master"},"4fe1a1081f8d96fb6ca528023c78b33438cf6033":{"kind":"REWORK","_number":29,"created":"2021-01-08 09:22:48.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/05/767705/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/05/767705/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/05/767705/29"}}},"commit":{"parents":[{"commit":"f9b02c8bf5d4cfe8ff29aac6181a38771f267037","subject":"Ubuntu: support non-persistent veth configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f9b02c8bf5d4cfe8ff29aac6181a38771f267037"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-18 12:17:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-01-08 09:22:44.000000000","tz":0},"subject":"CI: add Ubuntu overcloud deploy job","message":"CI: add Ubuntu overcloud deploy job\n\n* Use source images\n* Need to specify bash for \u0026\u003e syntax\n\nIssues worked around:\n\n* Manually configuring bridge via ip commands makes ifup fail to bring\n  up the link. Adds a kayobe-network-bootstrap Zuul CI role that adds\n  persistent configuration for the all-in-one network.\n\n* bridge not active after interfaces role bounce. Added a pause, similar\n  to https://github.com/michaelrigart/ansible-role-interfaces/pull/31\n\n* fails installing docker python module for kolla user. WARNING: The\n  repository located at mirror-int.ord.rax.opendev.org is not a trusted\n  or secure host and is being ignored ERROR: No matching distribution\n  found for docker\u003d\u003d\u003d4.4.0 Adding trusted host for PyPI mirror.\n\n* Tenks fails to create block devices - missing qemu-img (in qemu-utils)\n\n* Tenks qemu emulator is different on Ubuntu\n\nDepends-On: https://review.opendev.org/766984\nDepends-On: https://review.opendev.org/766958\n\nStory: 2004960\nTask: 29393\n\nChange-Id: I1985efae7c18f55c3ff7c27c17d6242523904f3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4fe1a1081f8d96fb6ca528023c78b33438cf6033"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4fe1a1081f8d96fb6ca528023c78b33438cf6033"}]},"branch":"refs/heads/master"},"9daec627d2393b5b71d013e87258cddd181b54cd":{"kind":"REWORK","_number":30,"created":"2021-01-11 11:03:21.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/05/767705/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/05/767705/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/05/767705/30"}}},"commit":{"parents":[{"commit":"f9b02c8bf5d4cfe8ff29aac6181a38771f267037","subject":"Ubuntu: support non-persistent veth configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f9b02c8bf5d4cfe8ff29aac6181a38771f267037"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-18 12:17:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-01-11 11:01:59.000000000","tz":0},"subject":"CI: add Ubuntu overcloud deploy job","message":"CI: add Ubuntu overcloud deploy job\n\n* Use source images\n* Need to specify bash for \u0026\u003e syntax\n\nIssues worked around:\n\n* Manually configuring bridge via ip commands makes ifup fail to bring\n  up the link. Adds a kayobe-network-bootstrap Zuul CI role that adds\n  persistent configuration for the all-in-one network.\n\n* bridge not active after interfaces role bounce. Added a pause, similar\n  to https://github.com/michaelrigart/ansible-role-interfaces/pull/31\n\n* fails installing docker python module for kolla user. WARNING: The\n  repository located at mirror-int.ord.rax.opendev.org is not a trusted\n  or secure host and is being ignored ERROR: No matching distribution\n  found for docker\u003d\u003d\u003d4.4.0 Adding trusted host for PyPI mirror.\n\n* Tenks fails to create block devices - missing qemu-img (in qemu-utils)\n\n* Tenks qemu emulator is different on Ubuntu\n\nRemaining issues:\n\n* Bare metal testing is unreliable on Ubuntu - some jobs see IPMI\n  failures such as the following:\n\n    ipmitool chassis bootdev pxe\n\n    Error setting Chassis Boot Parameter 5\\nError setting Chassis Boot\n    Parameter 0\\n\n\n  Bare metal testing is disabled on Ubuntu for now.\n\nDepends-On: https://review.opendev.org/766984\nDepends-On: https://review.opendev.org/766958\n\nStory: 2004960\nTask: 29393\n\nChange-Id: I1985efae7c18f55c3ff7c27c17d6242523904f3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9daec627d2393b5b71d013e87258cddd181b54cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9daec627d2393b5b71d013e87258cddd181b54cd"}]},"branch":"refs/heads/master"},"8762af4ae520f6af7cb559915526d469cf27528b":{"kind":"REWORK","_number":31,"created":"2021-02-11 11:11:10.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/05/767705/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/05/767705/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/05/767705/31"}}},"commit":{"parents":[{"commit":"f9b02c8bf5d4cfe8ff29aac6181a38771f267037","subject":"Ubuntu: support non-persistent veth configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f9b02c8bf5d4cfe8ff29aac6181a38771f267037"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-18 12:17:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-11 11:04:35.000000000","tz":0},"subject":"CI: add Ubuntu overcloud deploy job","message":"CI: add Ubuntu overcloud deploy job\n\n* Use source images\n* Need to specify bash for \u0026\u003e syntax\n\nIssues worked around:\n\n* Manually configuring bridge via ip commands makes ifup fail to bring\n  up the link. Adds a kayobe-network-bootstrap Zuul CI role that adds\n  persistent configuration for the all-in-one network.\n\n* bridge not active after interfaces role bounce. Added a pause, similar\n  to https://github.com/michaelrigart/ansible-role-interfaces/pull/31\n\n* fails installing docker python module for kolla user. WARNING: The\n  repository located at mirror-int.ord.rax.opendev.org is not a trusted\n  or secure host and is being ignored ERROR: No matching distribution\n  found for docker\u003d\u003d\u003d4.4.0 Adding trusted host for PyPI mirror.\n\n* Tenks fails to create block devices - missing qemu-img (in qemu-utils)\n\n* Tenks qemu emulator is different on Ubuntu\n\nRemaining issues:\n\n* Bare metal testing is unreliable on Ubuntu - some jobs see IPMI\n  failures such as the following:\n\n    ipmitool chassis bootdev pxe\n\n    Error setting Chassis Boot Parameter 5\\nError setting Chassis Boot\n    Parameter 0\\n\n\n  Bare metal testing is disabled on Ubuntu for now.\n\nDepends-On: https://review.opendev.org/766984\nDepends-On: https://review.opendev.org/766958\n\nStory: 2004960\nTask: 29393\n\nChange-Id: I1985efae7c18f55c3ff7c27c17d6242523904f3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8762af4ae520f6af7cb559915526d469cf27528b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8762af4ae520f6af7cb559915526d469cf27528b"}]},"branch":"refs/heads/master"},"9b5fe8509fbe29f0ce67fef0035892e3f312fceb":{"kind":"REWORK","_number":32,"created":"2021-02-19 14:59:33.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/05/767705/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/05/767705/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/05/767705/32"}}},"commit":{"parents":[{"commit":"f9b02c8bf5d4cfe8ff29aac6181a38771f267037","subject":"Ubuntu: support non-persistent veth configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f9b02c8bf5d4cfe8ff29aac6181a38771f267037"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-18 12:17:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-19 14:59:26.000000000","tz":0},"subject":"CI: add Ubuntu overcloud deploy job","message":"CI: add Ubuntu overcloud deploy job\n\n* Use source images\n* Need to specify bash for \u0026\u003e syntax\n\nIssues worked around:\n\n* Manually configuring bridge via ip commands makes ifup fail to bring\n  up the link. Adds a kayobe-network-bootstrap Zuul CI role that adds\n  persistent configuration for the all-in-one network.\n\n* bridge not active after interfaces role bounce. Added a pause, similar\n  to https://github.com/michaelrigart/ansible-role-interfaces/pull/31\n\n* fails installing docker python module for kolla user. WARNING: The\n  repository located at mirror-int.ord.rax.opendev.org is not a trusted\n  or secure host and is being ignored ERROR: No matching distribution\n  found for docker\u003d\u003d\u003d4.4.0 Adding trusted host for PyPI mirror.\n\n* Tenks fails to create block devices - missing qemu-img (in qemu-utils)\n\n* Tenks qemu emulator is different on Ubuntu\n\nRemaining issues:\n\n* Bare metal testing is unreliable on Ubuntu - some jobs see IPMI\n  failures such as the following:\n\n    ipmitool chassis bootdev pxe\n\n    Error setting Chassis Boot Parameter 5\\nError setting Chassis Boot\n    Parameter 0\\n\n\n  Bare metal testing is disabled on Ubuntu for now.\n\nDepends-On: https://review.opendev.org/766984\nDepends-On: https://review.opendev.org/766958\n\nStory: 2004960\nTask: 29393\n\nChange-Id: I1985efae7c18f55c3ff7c27c17d6242523904f3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9b5fe8509fbe29f0ce67fef0035892e3f312fceb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9b5fe8509fbe29f0ce67fef0035892e3f312fceb"}]},"branch":"refs/heads/master"},"9f41cefc15e240602d2c6ef9076e36135e9138e4":{"kind":"TRIVIAL_REBASE","_number":33,"created":"2021-03-01 17:57:51.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/05/767705/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/05/767705/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/767705/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/05/767705/33"}}},"commit":{"parents":[{"commit":"a9bf637cc56fe960ed7edd4af4624490bfc60b69","subject":"Merge \"Ubuntu: support non-persistent veth configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a9bf637cc56fe960ed7edd4af4624490bfc60b69"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-18 12:17:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-01 17:57:51.000000000","tz":0},"subject":"CI: add Ubuntu overcloud deploy job","message":"CI: add Ubuntu overcloud deploy job\n\n* Use source images\n* Need to specify bash for \u0026\u003e syntax\n\nIssues worked around:\n\n* Manually configuring bridge via ip commands makes ifup fail to bring\n  up the link. Adds a kayobe-network-bootstrap Zuul CI role that adds\n  persistent configuration for the all-in-one network.\n\n* bridge not active after interfaces role bounce. Added a pause, similar\n  to https://github.com/michaelrigart/ansible-role-interfaces/pull/31\n\n* fails installing docker python module for kolla user. WARNING: The\n  repository located at mirror-int.ord.rax.opendev.org is not a trusted\n  or secure host and is being ignored ERROR: No matching distribution\n  found for docker\u003d\u003d\u003d4.4.0 Adding trusted host for PyPI mirror.\n\n* Tenks fails to create block devices - missing qemu-img (in qemu-utils)\n\n* Tenks qemu emulator is different on Ubuntu\n\nRemaining issues:\n\n* Bare metal testing is unreliable on Ubuntu - some jobs see IPMI\n  failures such as the following:\n\n    ipmitool chassis bootdev pxe\n\n    Error setting Chassis Boot Parameter 5\\nError setting Chassis Boot\n    Parameter 0\\n\n\n  Bare metal testing is disabled on Ubuntu for now.\n\nDepends-On: https://review.opendev.org/766984\nDepends-On: https://review.opendev.org/766958\n\nStory: 2004960\nTask: 29393\n\nChange-Id: I1985efae7c18f55c3ff7c27c17d6242523904f3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9f41cefc15e240602d2c6ef9076e36135e9138e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9f41cefc15e240602d2c6ef9076e36135e9138e4"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
