)]}'
{"id":"openstack%2Fkayobe~767919","triplet_id":"openstack%2Fkayobe~master~I976122f72bdca7f91cbfcd81fca431b81a6b6aef","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I976122f72bdca7f91cbfcd81fca431b81a6b6aef","subject":"CI: fix race condition in kolla-ansible-host-vars role tests","status":"MERGED","created":"2020-12-18 17:17:09.000000000","updated":"2020-12-22 14:09:52.000000000","submitted":"2020-12-22 14:08:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"767919","meta_rev_id":"c5d2dccff7a464180e753185a5753f3ac356372c","_number":767919,"virtual_id_number":767919,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-22 14:08:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-12-22 12:26:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-12-20 22:00:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-12-22 12:26:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-18 19:46:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-20 22:00:46.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"3c3d238497cd9dce7ad8afd3d4eef70613490703","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-18 17:17:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8d985ad23a01b8334589e650a8cc354c8bd78185","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 19:46:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a0b629000dfa461495314bba384f4174 : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c693918f21c841738dc54a7d7d14c542 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e15b521158774b3da7f631d7df276d95 : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/458a708590f245ddacd0394bd27a66ae : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/680f7ef8d751442ba52661539dfc2584 : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c67b61390524f56864e98e791dad8bd : SUCCESS in 6m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d0a66155d2ba4aa8aad5e263318db276 : SUCCESS in 4m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fedc277eeb2546a3ab6f3ced68e362f5 : SUCCESS in 7m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6470223bf67c4dccb34fa74b65436b7e : SUCCESS in 28m 06s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/106f4d17f18d4666b2702895d8701433 : SUCCESS in 1h 00m 30s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/40aa99c14b574ad1a0590d6e2234aad1 : SUCCESS in 1h 05m 57s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/507fbe2625904991a691c24cfd59900d : SUCCESS in 17m 53s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/0cabbe54f8534e8faaba1df5a7392c25 : SUCCESS in 2h 23m 07s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/e619a629a9204affa6756017fced873f : SUCCESS in 39m 12s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/ebe4ba7ed0594794ae924873749b88d4 : SUCCESS in 26m 03s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/12afeab2c3304bec9f760715478c2d6e : SUCCESS in 11m 13s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"14d452ebc369b46ea19972471fce3b1f938dc074","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-12-18 20:46:31.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"967fbb7b343f378c16e3aeb7ce6f28596ec26a43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 22:52:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/71d2f1f5f00e4e53a48b9d732b2c5fab : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d5eff45b2ee47ba888f8db2dce96122 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a7f076fda2b44f286e03a5fdc8bb594 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8eede87335364db68f7fcf8821a9e00f : SUCCESS in 4m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b958721c61f14274b99fddcb62dbf8ea : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5047db20de164c71a4be6c9f5db53c70 : SUCCESS in 7m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/24192daa9e0a454194aa6c491ae00bf4 : SUCCESS in 4m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e18637cdf05a4865b9d5ce556f7f04ac : SUCCESS in 7m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9e3acb3de0da4891b401676369a760ee : SUCCESS in 19m 25s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/4fe6c46bc3a74ec29387005672bd993b : TIMED_OUT in 2h 01m 40s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/966c03d76e0b4f45ae64dab65e1c3107 : SUCCESS in 57m 23s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/02a849cbaa3b48c1afe51d7af6fbcc2d : SUCCESS in 15m 42s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/037d1f7eb8e645f8b38e616adc6b64a0 : SUCCESS in 1h 30m 13s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/c16d0deda0314d7293dadeb531f60907 : SUCCESS in 17m 26s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/91c7e3847c924b9597702212bebd782f : FAILURE in 18m 45s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/0eb3e2c584224d018c307d85868d39a1 : SUCCESS in 9m 32s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":2},{"id":"a733754c4d3a2ccfc584b9ef1ec5133285f158e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-12-20 22:00:37.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"4866b87abdb8ce5e210c083ebf67e8b9aaf5e3f5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-12-20 22:00:46.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"9400e3db632387ab50170c0ccecbd3139412625a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-20 23:35:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6d9a5ab211cb45f49b2656a5b87c9712 : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ab7e8cc56204a9d96d2193ef080976a : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b42e8b5f48d04f28962be2c542fa30f0 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd694f789ba94411bdfdbcf6332c0f94 : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f1528e402bf347faab6cae282a880c4c : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de6d91c8b8b744e3ba168bf0f5fbb2c8 : SUCCESS in 6m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b5931c9dd6af4342824f933a82666c8e : SUCCESS in 4m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1b0297b4541b44cfbeb9287167049642 : SUCCESS in 8m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/617bf1a071c54543bc200026e1b9dd97 : SUCCESS in 17m 52s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/4d1673d99dad4367a1851db7b4b43d78 : SUCCESS in 54m 56s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/1a40478cdf504902aa0bbd03ae0ba778 : SUCCESS in 56m 22s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/26994f51894249b69822c59c2b59f4e4 : SUCCESS in 15m 00s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/c5681622d2754718b1f0fe948aa28dd6 : SUCCESS in 1h 32m 30s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/afe849f17c5347fb9131a48bd0affa27 : SUCCESS in 15m 16s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/ebf68ed94a824749b3c050859f2a35f7 : SUCCESS in 23m 32s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/9ed2520bb4aa4ab1a605cd627b997e05 : SUCCESS in 9m 52s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":3},{"id":"feeb3675d900090d5d111bda5331f838f71f9e6e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-22 12:26:13.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nHit this again - approving.","accounts_in_message":[],"_revision_number":3},{"id":"84cb29aa75c44151e558c331536cb1476af680c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 12:26:38.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0d33d6d69867e6f062a3818b8f1e7b6757850d55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 14:08:24.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d40985314b10411fb3444d2f5660acd8 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b894d8072bff4900ad1d6cca36387e06 : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f3db112457e040a2a584bd04e5422bbb : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/724686bd394b484cb28d78e024276fbf : SUCCESS in 6m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/da0e40c7d007492192a41fb43aa75102 : SUCCESS in 4m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cdf870fdac3147dcb416e5aeaf7e6971 : SUCCESS in 8m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ce6c29e02f4d4533a8b7951cbf9fe945 : SUCCESS in 19m 27s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/4ab8daba6fbe49a09889df53438b4d0d : SUCCESS in 1h 08m 18s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/a05439c7b8914f98b6d67b495002cd7b : SUCCESS in 1h 14m 41s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/2b31b8a5a8e2409b948b586c47983c77 : SUCCESS in 15m 08s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/739ba8ad67dc426eadd513cfaf06efd5 : SUCCESS in 1h 37m 12s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/16740e5df6934867bd4b86266fc0ad22 : SUCCESS in 14m 21s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/3729a48e889d44e19f23d31aa54a381d : SUCCESS in 28m 37s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/44eb658cad024000ae17e6eee24a495d : SUCCESS in 11m 56s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":3},{"id":"2f37bb0fe95bb8d5ca812acdf8cd22cc2d215fee","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 14:08:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"c5d2dccff7a464180e753185a5753f3ac356372c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 14:09:52.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21cd7516a6044e7bb6fdefba51f85bd7 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7a384b9ce4c8724b408132a71db9a2fffcf7839d","revisions":{"ef0788537e36c10d9dd2a2925783a3150e441a4b":{"kind":"REWORK","_number":1,"created":"2020-12-18 17:17:09.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/19/767919/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/19/767919/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/767919/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/767919/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/767919/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/19/767919/1"}}},"commit":{"parents":[{"commit":"858c0f2a62dcd874fa3ac5601b38c70f6f46a535","subject":"Merge \"Remove retired Searchlight support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/858c0f2a62dcd874fa3ac5601b38c70f6f46a535"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-18 17:16:01.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-18 17:16:17.000000000","tz":0},"subject":"CI: fix race condition in kolla-ansbile-host-vars role tests","message":"CI: fix race condition in kolla-ansbile-host-vars role tests\n\nA temporary file is removed by multiple test hosts, which all have\nconnection\u003dlocal. This leads to a race between multiple executions of\nthe file module on localhost.\n\nAdd run_once.\n\nChange-Id: I976122f72bdca7f91cbfcd81fca431b81a6b6aef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ef0788537e36c10d9dd2a2925783a3150e441a4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ef0788537e36c10d9dd2a2925783a3150e441a4b"}]},"branch":"refs/heads/master"},"51be3b921ee561dbdba662cd3f5b432006ecee77":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-12-18 20:46:31.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/19/767919/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/19/767919/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/767919/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/767919/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/767919/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/19/767919/2"}}},"commit":{"parents":[{"commit":"858c0f2a62dcd874fa3ac5601b38c70f6f46a535","subject":"Merge \"Remove retired Searchlight support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/858c0f2a62dcd874fa3ac5601b38c70f6f46a535"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-18 17:16:01.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-12-18 20:46:31.000000000","tz":0},"subject":"CI: fix race condition in kolla-ansible-host-vars role tests","message":"CI: fix race condition in kolla-ansible-host-vars role tests\n\nA temporary file is removed by multiple test hosts, which all have\nconnection\u003dlocal. This leads to a race between multiple executions of\nthe file module on localhost.\n\nAdd run_once.\n\nChange-Id: I976122f72bdca7f91cbfcd81fca431b81a6b6aef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/51be3b921ee561dbdba662cd3f5b432006ecee77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/51be3b921ee561dbdba662cd3f5b432006ecee77"}]},"branch":"refs/heads/master","description":"Edit commit message"},"7a384b9ce4c8724b408132a71db9a2fffcf7839d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-12-20 22:00:37.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/19/767919/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/19/767919/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/767919/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/767919/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/767919/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/19/767919/3"}}},"commit":{"parents":[{"commit":"e569036a05d06840148886cafd13a96b78e1df09","subject":"Merge \"Remove retired Qinling support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e569036a05d06840148886cafd13a96b78e1df09"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-18 17:16:01.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-12-20 22:00:37.000000000","tz":0},"subject":"CI: fix race condition in kolla-ansible-host-vars role tests","message":"CI: fix race condition in kolla-ansible-host-vars role tests\n\nA temporary file is removed by multiple test hosts, which all have\nconnection\u003dlocal. This leads to a race between multiple executions of\nthe file module on localhost.\n\nAdd run_once.\n\nChange-Id: I976122f72bdca7f91cbfcd81fca431b81a6b6aef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7a384b9ce4c8724b408132a71db9a2fffcf7839d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7a384b9ce4c8724b408132a71db9a2fffcf7839d"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
