)]}'
{"id":"openstack%2Fkayobe~768127","triplet_id":"openstack%2Fkayobe~stable%2Fvictoria~Ib869ae5146948abcb11eb20691ccc2311db24bb1","project":"openstack/kayobe","branch":"stable/victoria","topic":"story/2008458-stable/victoria","hashtags":[],"change_id":"Ib869ae5146948abcb11eb20691ccc2311db24bb1","subject":"Fix implementation of kayobe seed hypervisor host package update","status":"MERGED","created":"2020-12-21 20:56:00.000000000","updated":"2020-12-24 18:49:38.000000000","submitted":"2020-12-24 18:48:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"768127-story/2008458-stable/victoria","meta_rev_id":"22782f48472e8bb5565b6452520e01a5d3dbef53","_number":768127,"virtual_id_number":768127,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-24 18:48:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-12-22 09:17:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-12-24 16:58:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-12-24 16:58:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-21 22:54:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-22 09:17:52.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"20004b6f164b4377e0b0fbc69fe10f437d731d94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-12-21 20:56:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0c8ca6f7f6355d023d41871d0ae7900bdb3e1f5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-21 22:54:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6dd12941217a450b9bbf74b8ba1726ee : SUCCESS in 1m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aaaf829d1f19494eba1a191e39e05eb9 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82218d28d4d043e5a2e6d0d7a0cbc557 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/52a9b300f5164d3fb267490d13df89db : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6e46c0a89a4a40d7a80fc504e786ecdf : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/978a1c7e2f284cdb85b3e95a201d3307 : SUCCESS in 5m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4817c81a11544f42942ff5c69c0b993c : SUCCESS in 3m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/224fffc05a344bf2aadab2c48e237383 : SUCCESS in 4m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5177ffd6498e4692840f23c996f46288 : SUCCESS in 8m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a21b8189a75549a49db5e178b7fee2b6 : SUCCESS in 18m 27s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/5a88ef14b77e4c3ba1944a26341da269 : SUCCESS in 1h 06m 24s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/dce4003d92f8467aba395fccf134e1d2 : SUCCESS in 1h 17m 35s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/43d5cb3327cd456eb798c5490e8986d7 : SUCCESS in 15m 20s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/78e3a3b506e6466dbbf243ebb646ce20 : SUCCESS in 1h 53m 35s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/86c5510c7c964c33bd687d1db0ae52e0 : SUCCESS in 15m 33s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/5a9652ae0b7943909a7aa660141b9e82 : SUCCESS in 24m 09s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/8cc6c13247ee4529bf8ce761c1ed693a : SUCCESS in 10m 27s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"fc55d6ac090ce18c8c9a60d2cfaa64efdfa0729d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-22 09:17:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8206def587fc9de97239ca3b3219a9026adda771","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-12-24 16:58:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7e88fe798aa55f16867cccd2d8fc02c06b142c0c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-24 16:59:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"247d22b45eeeef1ca00e9d7ef9fdba45e823f7f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-24 18:48:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4155ce15d9e140d2944d3d958c13d442 : SUCCESS in 1m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ca096f4e0b845dcb736aace38d1e5cf : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17ab6ff2e8b84027bd956f6059474c7c : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e4b7b0771b69440c86c367d5184bd86d : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd1e03e35743433bbeb4ed6b969d0a50 : SUCCESS in 6m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39d8e2aeb2ec49fda45dcfaf0951eb11 : SUCCESS in 3m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/81eca5dc79da411a80aeb2f25d454584 : SUCCESS in 4m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9a4b60fae85c4d0496b72b16ce01b31a : SUCCESS in 7m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6ed349b3ce93481a8c30dc48e344e6f9 : SUCCESS in 18m 08s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/2d3c811c92114cccad38e6b052456d39 : SUCCESS in 1h 01m 37s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/36e1bff69d884679b202c50f8938726b : SUCCESS in 1h 05m 50s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/0178bae21ad7419690317f9d06f3dd4a : SUCCESS in 14m 56s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e62697a3f87347909b619d60ace0c689 : SUCCESS in 1h 44m 54s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/24674416044d44bfa202498c41427aac : SUCCESS in 17m 12s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/ef67bb1ff4634c24893e40ad41c993dd : SUCCESS in 31m 44s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/14437a2953c7455eb9c76384b6674e2f : SUCCESS in 10m 00s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"882c2a9a809f30f76e2331782a2f8324b8af2045","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-24 18:48:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"22782f48472e8bb5565b6452520e01a5d3dbef53","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-24 18:49:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/542a5174a3664fe1ab6f47fca841dd9e : SUCCESS in 1m 06s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79b932e849654fa3ad8f55341e8c25dc : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dc0b8b0787eaca9815650b36bbd77e646b2f0ad3","revisions":{"dc0b8b0787eaca9815650b36bbd77e646b2f0ad3":{"kind":"REWORK","_number":1,"created":"2020-12-21 20:56:00.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/27/768127/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/27/768127/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/27/768127/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/27/768127/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/27/768127/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/27/768127/1"}}},"commit":{"parents":[{"commit":"885dd064f48bbe5c72b921c972cc11ff85ac7858","subject":"Fixes for CentOS 8.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/885dd064f48bbe5c72b921c972cc11ff85ac7858"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-12-21 09:29:41.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-12-21 20:55:03.000000000","tz":60},"subject":"Fix implementation of kayobe seed hypervisor host package update","message":"Fix implementation of kayobe seed hypervisor host package update\n\nChange-Id: Ib869ae5146948abcb11eb20691ccc2311db24bb1\nStory: 2008458\nTask: 41485\n(cherry picked from commit eac1702b84c23755e98eb276ee9447ee1f02bf94)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dc0b8b0787eaca9815650b36bbd77e646b2f0ad3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dc0b8b0787eaca9815650b36bbd77e646b2f0ad3"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
