)]}'
{"id":"openstack%2Fkayobe~768128","triplet_id":"openstack%2Fkayobe~stable%2Fvictoria~I1264832398744c48a6f706efe91f084890a37772","project":"openstack/kayobe","branch":"stable/victoria","topic":"story/2008388-stable/victoria","hashtags":[],"change_id":"I1264832398744c48a6f706efe91f084890a37772","subject":"Fix overcloud provisioning","status":"MERGED","created":"2020-12-21 20:59:56.000000000","updated":"2020-12-24 18:39:12.000000000","submitted":"2020-12-24 18:38:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"768128-story/2008388-stable/victoria","cherry_pick_of_change":764362,"cherry_pick_of_patch_set":2,"meta_rev_id":"1140abd7c65c3214a9df790057ab6c3f1475f030","_number":768128,"virtual_id_number":768128,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-24 18:38:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-12-22 12:17:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-12-24 16:57:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-12-24 16:57:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-12-22 12:17:56.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-21 20:59:56.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-12-21 23:01:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"91e4ba9741c4204d186ee5b1d1d3946df6b2680d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-12-21 20:59:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"325016e15661f89c07deb1b057279a1b112695f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-21 23:01:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/87c56558b37648fe8e1875f5af00e216 : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1261b7ce98184742801398946dba6c5c : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6dd260971ff4df599a54b0122a522fd : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a750402df2d446f68e51c85bb7e87e7d : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c97acfb13ec45b1a0e4bdc2c1b81c90 : SUCCESS in 5m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e30c9520f4dd49d485ecaa4e9e81c727 : SUCCESS in 4m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f3e0b5ee1ff14d538871756b54f2c6a3 : SUCCESS in 7m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/50c5e65f389d46e5bfa407daca94b7cf : SUCCESS in 21m 27s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/721d5fdc17ec49d78f9998997c9023b3 : SUCCESS in 1h 06m 34s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/52964a84566241ba80d8d7ffc902f905 : FAILURE in 2h 00m 37s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/f6c6b97bad05458688c89c58f95dc60f : SUCCESS in 13m 05s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/42286ab34607423794f0094b388a7253 : SUCCESS in 1h 32m 21s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/4c686a53723c4923b3c41886d64317e7 : SUCCESS in 16m 31s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/938e205adfc14ab8af78aa119035e34a : SUCCESS in 21m 50s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/56bd8dd1716b4a2292178f680615f4eb : SUCCESS in 11m 22s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"9604d9451bd5affd08ac77a016215ba859c46f4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-22 09:16:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f0a760e85646439388304ef1c9570ff8c5a5b54c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 11:04:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79afeb6a08d7421d89f8760da4358374 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e677539228d548f9972ffdcd6378a9a9 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e53ab9c106c4173bb76e540682edfa1 : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/350ff7af832041cd8547768189b29ec6 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b2aebefb4174d2891c7c794eb4a7a22 : SUCCESS in 6m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d5275a7890c042159e3ceeec82156149 : SUCCESS in 5m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/129f7ec7d2c141c4ba07d92247e5341f : SUCCESS in 7m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2f7d37f0acb9412fb4f7b325443c59f3 : SUCCESS in 17m 24s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/8ae6a9215aff4f9996022ba6ee9910c0 : SUCCESS in 1h 00m 46s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/264a593b28894126a9fdcd24dce49b62 : SUCCESS in 1h 14m 06s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/e63d9686235b43198b225b5b8a8f9bf5 : SUCCESS in 15m 06s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/20cadb2108f04601a50de5be0cc1fcd1 : SUCCESS in 1h 44m 38s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/a93e913d18ac4e2495734dd918afadd7 : SUCCESS in 16m 11s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b1306c1277db45f58bb30ba63bb2e800 : SUCCESS in 25m 02s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/10f20fa677bf4e4e91a8e63e426508f9 : SUCCESS in 10m 56s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"cc9c88edbaea269ce96e78e5fb8f9f5714065efd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-22 12:17:38.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"584a2bf8ffb79749b60066381d29a3ab6bbfa802","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-22 12:17:56.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"33e9792af90ee8047442291ee92d2909ccc9badf","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-12-24 16:57:10.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"cc6695bf3d000dd36f95039cab1c40a0c8afa56f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-24 16:57:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"eb9d6d318bacca75dd458d3c99624215a0bd7463","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-24 18:38:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f765b8fd0fe548a988a4d6ddc10f3db3 : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b113cc0be534f9ab91ff823a743490a : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b642fad7dab14542a472b236b166fbce : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/212c86c7922849469e9e5b355af4fc4b : SUCCESS in 5m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/411e1355a5704c63b0ec8cb35f10169c : SUCCESS in 4m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1e4dfa689c374bb89b9418a19c5bd1cf : SUCCESS in 7m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c88d4255b3264a8a8f5eb3b19ee1e277 : SUCCESS in 16m 49s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/5a11e5bdfda448b9932388fa5ecc1c3e : SUCCESS in 1h 09m 11s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/32be52d2ed4f42b8be72c8ef265207c3 : SUCCESS in 57m 41s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/966528176fde4d23abd5eabc91d707b6 : SUCCESS in 15m 05s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/75d6bddc929148fa9258bbd80d972b97 : SUCCESS in 1h 37m 40s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/4f66611f9e0f41c7b3eb500f378903e3 : SUCCESS in 17m 20s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e7566b3f95044f138d1190668143ec31 : SUCCESS in 33m 08s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/b442b744782b48aba98bd9c9d195488d : SUCCESS in 9m 21s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":2},{"id":"2834e7e9f97c009d08ef15057b2996041b52e38f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-24 18:38:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"1140abd7c65c3214a9df790057ab6c3f1475f030","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-24 18:39:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1b43f63798a47619d39108665ce0dc7 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9eb982bbd9bde5b2aec1557886e47e949897336b","revisions":{"bfdbf3ef54d27b8b4b8238743d1aefbb8f755eed":{"kind":"REWORK","_number":1,"created":"2020-12-21 20:59:56.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/28/768128/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/28/768128/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/768128/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/768128/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/768128/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/28/768128/1"}}},"commit":{"parents":[{"commit":"885dd064f48bbe5c72b921c972cc11ff85ac7858","subject":"Fixes for CentOS 8.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/885dd064f48bbe5c72b921c972cc11ff85ac7858"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-26 15:41:37.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-12-21 20:59:46.000000000","tz":60},"subject":"Fix overcloud provisioning","message":"Fix overcloud provisioning\n\nAffects master \u0026 Victoria only.\n\nSince Victoria, Bifrost added a play to gather facts for localhost at\nthe beginning of deploy-dynamic.yaml. Kayobe executes the playbook with\na limit, which does not include localhost. This causes fact gathering to\nbe skipped for localhost:\n\n    PLAY [Gather facts on localhost] ***********************************\n      skipping: no hosts matched\n\nThis change fixes the issue by adding localhost to the limit.\n\nChange-Id: I1264832398744c48a6f706efe91f084890a37772\nStory: 2008388\nTask: 41311\n(cherry picked from commit 1ad073f0d0b181fc3eaf9b3523c29d96302a356a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bfdbf3ef54d27b8b4b8238743d1aefbb8f755eed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bfdbf3ef54d27b8b4b8238743d1aefbb8f755eed"}]},"branch":"refs/heads/stable/victoria"},"9eb982bbd9bde5b2aec1557886e47e949897336b":{"kind":"NO_CHANGE","_number":2,"created":"2020-12-22 09:16:12.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/28/768128/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/28/768128/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/768128/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/768128/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/768128/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/28/768128/2"}}},"commit":{"parents":[{"commit":"885dd064f48bbe5c72b921c972cc11ff85ac7858","subject":"Fixes for CentOS 8.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/885dd064f48bbe5c72b921c972cc11ff85ac7858"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-26 15:41:37.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-22 09:16:12.000000000","tz":0},"subject":"Fix overcloud provisioning","message":"Fix overcloud provisioning\n\nAffects master \u0026 Victoria only.\n\nSince Victoria, Bifrost added a play to gather facts for localhost at\nthe beginning of deploy-dynamic.yaml. Kayobe executes the playbook with\na limit, which does not include localhost. This causes fact gathering to\nbe skipped for localhost:\n\n    PLAY [Gather facts on localhost] ***********************************\n      skipping: no hosts matched\n\nThis change fixes the issue by adding localhost to the limit.\n\nChange-Id: I1264832398744c48a6f706efe91f084890a37772\nStory: 2008388\nTask: 41311\n(cherry picked from commit 1ad073f0d0b181fc3eaf9b3523c29d96302a356a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9eb982bbd9bde5b2aec1557886e47e949897336b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9eb982bbd9bde5b2aec1557886e47e949897336b"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
