)]}'
{"id":"openstack%2Fkayobe~768159","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~Ib869ae5146948abcb11eb20691ccc2311db24bb1","project":"openstack/kayobe","branch":"stable/train","topic":"story/2008458-stable/train","hashtags":[],"change_id":"Ib869ae5146948abcb11eb20691ccc2311db24bb1","subject":"Fix implementation of kayobe seed hypervisor host package update","status":"MERGED","created":"2020-12-22 09:19:38.000000000","updated":"2020-12-24 18:50:58.000000000","submitted":"2020-12-24 18:49:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"768159-story/2008458-stable/train","cherry_pick_of_change":768021,"cherry_pick_of_patch_set":3,"meta_rev_id":"33cfccee253aca9cfff916f4fd7e43d4632fdd27","_number":768159,"virtual_id_number":768159,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-24 18:49:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-12-22 17:20:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-12-24 16:58:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-12-24 16:58:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-22 09:19:38.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-12-22 10:32:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ea18e51c2284d305292226689215d1e851b2b32d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-22 09:19:38.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* setup.cfg","accounts_in_message":[],"_revision_number":1},{"id":"1b055fcfafa0b7be52ab4204e0bb4f490697f3e4","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-22 09:20:43.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":1},{"id":"0484bf613ab91eaf74331bfa9792e3794d1f6985","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-22 09:23:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9eb956f547907d752f3d358921d006d8a63506b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 10:32:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/72317cd8097645169a294c4ef807f2e7 : SUCCESS in 1m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/93f708596e474737884b421342aa385d : SUCCESS in 3m 59s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a06127b7f0e04b1c9348825c6bbd108c : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb58d04ec64945c8a8bdc30e6989878d : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e79725c91b842c79a0ebe130ce0eead : SUCCESS in 3m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ac573f82e40e4ae5b6d238f6a019b682 : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32d892894bd245e2a0fa344dd5a14dd6 : SUCCESS in 5m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/557c9728243e4f9184e016aeb0c2963f : SUCCESS in 3m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/68c39960c2894981b64a5d89cb7bfc4d : SUCCESS in 4m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/652693c30a424ed3941a3efba17f42b8 : SUCCESS in 9m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1bd35d14f1fb417b9e1594397120215c : SUCCESS in 16m 27s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/28487d9884b64fd7a829d2be26e4eddf : SUCCESS in 1h 03m 27s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/e82ddb6dc44646c7bd27dab6422768bd : SUCCESS in 58m 10s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/cd37a6af7bfa404abbf61c40c89e00a2 : SUCCESS in 13m 57s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/e3153d8709ca41b3b87c1c0294f5da99 : SUCCESS in 14m 37s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/deb36c1a7f0840bb93edf19e544181eb : FAILURE in 54m 06s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/54502ebc09e5463b998302fdc525b4f0 : SUCCESS in 24m 40s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/24253e0a9fed4c5ea4ef763fd402257a : SUCCESS in 19m 14s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/a2e5995043ac41a1b557a173c8a260d2 : FAILURE in 53m 27s","accounts_in_message":[],"_revision_number":2},{"id":"bcc143d4c93292cd58bcc27984ba860aabe9f72c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-22 12:16:27.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"1059ce8a1695937884db2b478fa021ade67beeb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 13:33:14.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d8134b61c3a1404aa9c4de9d4e9c1669 : SUCCESS in 3m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c4526f9ecb5047cea3f69b4d32f2578f : SUCCESS in 5m 19s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/31841dca76b2458386c58c35f57e0e14 : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cbd837bf63944635b72acb96a1d7c063 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da3984f4139548df98e7498fd983963f : SUCCESS in 4m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e4e74fcf50634c8d83c9dd5ffd81249f : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0df202674434a4888555130ff1b6782 : SUCCESS in 6m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d7e9a9c878be4c20bd412983d2065500 : SUCCESS in 3m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/210267f447984e7daf852ce79243e0c1 : SUCCESS in 4m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6ecef4fdf99e48ed9c1684e7d15c18be : SUCCESS in 8m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6e54713732bc47a380577c99d6c55445 : SUCCESS in 18m 00s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/348c9214cb2f4416a7eaaaba2b745f9c : SUCCESS in 1h 12m 46s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/9a9f4392dc304a258081679f821142dc : SUCCESS in 1h 16m 04s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/c082ce6f7a66472d8fb6dc6d7e80200c : SUCCESS in 14m 11s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/60efef32789143f697fea6e4b4bb5b12 : SUCCESS in 15m 48s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/312acc1689e04f55a59e4c32d9e4022c : FAILURE in 54m 33s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/cd33a66c272d4f52a68cbec736c98726 : SUCCESS in 24m 24s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/361f88e6fb304efa95bf3c497dd824a9 : SUCCESS in 17m 13s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/4fd2468b763a4662b53a43ceaa3373d4 : SUCCESS in 23m 10s","accounts_in_message":[],"_revision_number":2},{"id":"b9547d45e727f671b0d131aed8ee116f2b1a9299","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-22 13:40:34.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"61ffe34223655481d1b02840b9ae690389d3b475","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 15:51:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d5251c119cf14be587cb3f0582ceff3f : SUCCESS in 2m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef0affaf70cf4e3aa0a4c9a311d8d2ec : SUCCESS in 4m 34s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8d3d1aabcbbf4c19aeada7e350c4e5d2 : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf410d499e8249509dad1a284033f700 : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f7736f002fe436c9ea1a3d81232e285 : SUCCESS in 3m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/560996b9132c4ac6a2148de7420a7580 : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d50ab6294cf8499780d4abaffe860968 : SUCCESS in 6m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fcc76da5920740b6b8957119406a3ae2 : SUCCESS in 4m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d4509fb4b89e4424a2bcdafb6b4fbc1b : SUCCESS in 3m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/acd209eb334b488493a1ada01e12ecb4 : SUCCESS in 8m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a69f9772ab2c40738ca2cd2d34e0a72c : SUCCESS in 16m 33s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/c3ab669f60454ed6ba9863fc0ee27c1d : SUCCESS in 59m 31s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/4b8d828591fa4ed896ebe9a180f12b91 : SUCCESS in 58m 25s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/a2588f8531fc41a1bc0c272c49c86af3 : SUCCESS in 13m 19s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/ee6a7f02816a4247ada30e1e12738650 : SUCCESS in 15m 39s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/334f269de82746a8b9447d372eeab1f0 : SUCCESS in 2h 00m 31s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/ec6e2c67fade466b9ee30e170b038664 : SUCCESS in 21m 01s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/e302ba291de0483eae59fe96748df9ff : SUCCESS in 16m 07s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/99d893ece6d24e6d907e823f6a0c8ee2 : SUCCESS in 44m 45s","accounts_in_message":[],"_revision_number":2},{"id":"b2b8d8fe218d67dfe57cd264ce153a55ac91ff86","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-22 17:20:54.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"889d1993ebe9edc4089f2b9214ab11a32ada61b5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-12-24 16:58:55.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d80b5ad63af8f47538790b346f9bc73aaf52901e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-24 16:59:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1534161b2aefa290ac95c7d78e7ea1f257ea2e74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-24 18:49:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/54c63327b29e4188a55744ac6af98d5c : SUCCESS in 2m 00s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6cdc8ae8ae2c4d2798d2ca22d1fdbedf : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e07c91ffb684af1998de9c23350716b : SUCCESS in 3m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61c5be8f8fc84dc087bcaf17a32b3f89 : SUCCESS in 3m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a53b9ba47db74d5092621a25be8be47c : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72985fbe6fea4f9fa1ad164d44cf6a4a : SUCCESS in 5m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26636ecd419e495ea509dc54703aaa48 : SUCCESS in 3m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cfeddb4d2eb44225b1b2d85b39f21bdb : SUCCESS in 3m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/997958978c3b449f83e6658355792237 : SUCCESS in 7m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3dcf887edad5446e9528dd4593650c0f : SUCCESS in 16m 56s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/f34b499545d84b39802c0a30e3c5d4dc : SUCCESS in 1h 05m 41s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/50dec2a4c4c147629373c973375cf16e : SUCCESS in 1h 00m 05s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/7d84f3695d214a629f05a1d8d56341d3 : SUCCESS in 15m 35s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/495c4b250a13469ab5c3f2869d43b538 : SUCCESS in 14m 48s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/f89fee6057ac4b1dbfc778d11cc94874 : SUCCESS in 1h 44m 21s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/912e1e06368845a0982c5e2b8eb5ec0a : SUCCESS in 23m 47s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/57788e31856f4ba4a58d3a9458caa497 : SUCCESS in 15m 42s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/2db62b00691f4fcd81f731d3d23ca346 : SUCCESS in 39m 14s","accounts_in_message":[],"_revision_number":2},{"id":"73431eac07e7acdaf0bd06ab08030414fee346ac","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-24 18:49:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"33cfccee253aca9cfff916f4fd7e43d4632fdd27","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-24 18:50:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b279d0cf64be4221ae6d090f1f34891f : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e1386d642b14c348987a42b1a88d4f5 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"641772dbca48b428729ab5fd52056dca6656efb9","revisions":{"8f75815e94a253da99756e215de3edc8cc5f6e66":{"kind":"REWORK","_number":1,"created":"2020-12-22 09:19:38.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/59/768159/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/59/768159/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/768159/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/768159/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/768159/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/59/768159/1"}}},"commit":{"parents":[{"commit":"ac7adbb9b56cd9901c7a0bfb826f8b60d590a792","subject":"Fixes for CentOS 8.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ac7adbb9b56cd9901c7a0bfb826f8b60d590a792"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-12-21 09:29:41.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-22 09:19:38.000000000","tz":0},"subject":"Fix implementation of kayobe seed hypervisor host package update","message":"Fix implementation of kayobe seed hypervisor host package update\n\nChange-Id: Ib869ae5146948abcb11eb20691ccc2311db24bb1\nStory: 2008458\nTask: 41485\n(cherry picked from commit eac1702b84c23755e98eb276ee9447ee1f02bf94)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8f75815e94a253da99756e215de3edc8cc5f6e66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8f75815e94a253da99756e215de3edc8cc5f6e66"}]},"branch":"refs/heads/stable/train"},"641772dbca48b428729ab5fd52056dca6656efb9":{"kind":"REWORK","_number":2,"created":"2020-12-22 09:23:52.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/59/768159/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/59/768159/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/768159/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/768159/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/768159/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/59/768159/2"}}},"commit":{"parents":[{"commit":"ac7adbb9b56cd9901c7a0bfb826f8b60d590a792","subject":"Fixes for CentOS 8.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ac7adbb9b56cd9901c7a0bfb826f8b60d590a792"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-12-21 09:29:41.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-22 09:23:36.000000000","tz":0},"subject":"Fix implementation of kayobe seed hypervisor host package update","message":"Fix implementation of kayobe seed hypervisor host package update\n\nChange-Id: Ib869ae5146948abcb11eb20691ccc2311db24bb1\nStory: 2008458\nTask: 41485\n(cherry picked from commit eac1702b84c23755e98eb276ee9447ee1f02bf94)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/641772dbca48b428729ab5fd52056dca6656efb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/641772dbca48b428729ab5fd52056dca6656efb9"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
