)]}'
{"id":"openstack%2Fkayobe~768165","triplet_id":"openstack%2Fkayobe~stable%2Fussuri~I976122f72bdca7f91cbfcd81fca431b81a6b6aef","project":"openstack/kayobe","branch":"stable/ussuri","hashtags":[],"change_id":"I976122f72bdca7f91cbfcd81fca431b81a6b6aef","subject":"CI: fix race condition in kolla-ansible-host-vars role tests","status":"MERGED","created":"2020-12-22 14:30:07.000000000","updated":"2020-12-23 13:45:53.000000000","submitted":"2020-12-23 13:44:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"768165","cherry_pick_of_change":767919,"cherry_pick_of_patch_set":3,"meta_rev_id":"73bc5637a774fd1e8fafa3688754a2a6532ca239","_number":768165,"virtual_id_number":768165,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-23 13:44:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-12-22 17:19:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-12-23 10:15:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-22 16:18:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dd220b7e14c686ab29133fb28d523a314ecb9297","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-22 14:30:07.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"72d69e7aac66c363cea3c06bac7c23f89d64578a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 16:18:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0245a826d9be429abbd3bfd42239b021 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/722022e109f147a7a8f7ba2f1a9eda36 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a6d16c990d1405f8d01a361f63c4909 : SUCCESS in 5m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/177dbb2eb78e40aaa32d0023bd6dde04 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f811beae5fdc4b488bcb59c3e8ba3ec5 : SUCCESS in 6m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0855c85d20da45b0b173f2ca3f7a8b7c : SUCCESS in 5m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ade18eacc61b4c88890f2c05f4776209 : SUCCESS in 9m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2c196bef8f694add9214c63999705780 : SUCCESS in 19m 21s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/409bcfe09bbd41c094336225688b51b4 : SUCCESS in 54m 59s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/18d0a6dce543419f847925a35de8518e : SUCCESS in 17m 11s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/0ccb015a5e7b40a6a2a40811281c6b05 : SUCCESS in 1h 45m 55s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/64d7289bb558430781de5dc6ef303017 : SUCCESS in 19m 28s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/6d35ffb3b2554b74a11c57aabe39d125 : SUCCESS in 26m 30s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/69ebf3ce076e4543b871a483defd0c66 : SUCCESS in 12m 10s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"d7a5a97a1075407c672b333fd26e5ebcc280a731","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-22 17:19:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3d571430ca45d8cd427f3db30dd2364446bd5817","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-23 10:15:15.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"490962e5ffce700d931e86185d531e0fb1b2aa56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 10:15:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d1bae81542864d52569bbbf801200621e2b40cf4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 13:44:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d99f4c530b6049f2b727a141d7da3ee5 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/09945287bec0404696da379bad9677a0 : SUCCESS in 7m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/add48560b6df47aba01be119f97407bf : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90177486e2374817874626aa99becba2 : SUCCESS in 6m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c0dbb7e1701f4c968b99ffa4eace0eb9 : SUCCESS in 5m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8955db1c259b4be38cdf344bac9231cb : SUCCESS in 8m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8f73e594a83743e58a9ced93ed169458 : SUCCESS in 13m 46s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/4ae5dafb5b4e4b77a0391d14a221dd81 : SUCCESS in 1h 00m 14s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/806aac235b304a7eaf95a9653ca5ac21 : SUCCESS in 17m 26s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/78ee4ecf8a7941599048e47f7743640d : SUCCESS in 1h 43m 18s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/32a5ca572e5b46f79b5ffdde7716825b : SUCCESS in 16m 43s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/c3c36c43dcd14662a3ec318122c6ed6a : SUCCESS in 25m 46s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/cfafc5a03ffb46398d2864de3c939bdb : SUCCESS in 11m 47s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"67d58b44bbfdb45cfb2c201136295c752a512b95","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 13:44:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"73bc5637a774fd1e8fafa3688754a2a6532ca239","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 13:45:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22fb2ca3156c477f9c76b6d50db9497d : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5871473c42228273f5b06726b0460e364471eb0a","revisions":{"5871473c42228273f5b06726b0460e364471eb0a":{"kind":"REWORK","_number":1,"created":"2020-12-22 14:30:07.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/65/768165/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/65/768165/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/768165/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/768165/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/768165/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/65/768165/1"}}},"commit":{"parents":[{"commit":"5cfd2ba5aef07e31efc6695a8f987a1b7391ba66","subject":"Fixes for CentOS 8.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5cfd2ba5aef07e31efc6695a8f987a1b7391ba66"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-18 17:16:01.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-22 14:30:07.000000000","tz":0},"subject":"CI: fix race condition in kolla-ansible-host-vars role tests","message":"CI: fix race condition in kolla-ansible-host-vars role tests\n\nA temporary file is removed by multiple test hosts, which all have\nconnection\u003dlocal. This leads to a race between multiple executions of\nthe file module on localhost.\n\nAdd run_once.\n\nChange-Id: I976122f72bdca7f91cbfcd81fca431b81a6b6aef\n(cherry picked from commit 7a384b9ce4c8724b408132a71db9a2fffcf7839d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5871473c42228273f5b06726b0460e364471eb0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5871473c42228273f5b06726b0460e364471eb0a"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
