)]}'
{"id":"openstack%2Fkayobe~768205","triplet_id":"openstack%2Fkayobe~master~Ic36531c0f26fc7dee361bebb83a1d3089166c1eb","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"Ic36531c0f26fc7dee361bebb83a1d3089166c1eb","subject":"Ubuntu: Use . to source shell scripts","status":"MERGED","created":"2020-12-22 13:57:56.000000000","updated":"2021-01-14 15:10:29.000000000","submitted":"2021-01-14 15:08:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"768205","meta_rev_id":"ac44dfa739d8ba01c3da9caa85945248b04306b6","_number":768205,"virtual_id_number":768205,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-14 15:08:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-01-14 13:27:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-13 11:58:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-01-14 13:27:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-22 15:35:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-13 11:58:36.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-01-14 13:27:20.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"dc0a2d4b146295f38317ef9cdc660e117473f0eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-22 13:57:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"339b1e0f3ee10b94f4b2e40b1d56b42f6b767c15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 15:35:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb2c45fa9d6844309976945844a20fc5 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c57c503d81e34a1a8c965fff5451654b : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/826744b6be0340dd9fbf45624f21bf75 : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d76d1b8966ad43659d6f6bb9e306ac1c : SUCCESS in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ae96526fe1f142978b48b3fc7678454c : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d58598ece60846a18a5b6923c28f2999 : SUCCESS in 5m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/92f1010001a8471da7474330d5d771d4 : SUCCESS in 4m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/53a83e78a729480d9b52bd4f0799e26d : SUCCESS in 8m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/adcabda6e2a34928a875e724b16150b9 : SUCCESS in 17m 12s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/362129dc903d42c29a808294095d8f30 : SUCCESS in 56m 17s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/01d02fe5cef44db2a90d6b56c32e7142 : SUCCESS in 59m 35s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/ecf361e906264a72acab2da2501a4872 : SUCCESS in 14m 21s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f9a21ede39c04faf8841e170e3b11ef5 : SUCCESS in 1h 33m 53s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/5c622d5904cd42238519d6d9d8e1382f : SUCCESS in 15m 08s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/99a0dad42da74af88d1a4db9acd71b7a : SUCCESS in 21m 54s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/4708dc0084db43ac8bce3ea429f974ec : SUCCESS in 10m 38s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"ebbf694c342fe87499becf176dbafff43f1b056b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-13 11:58:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"09eecc6fa2735ce9755a9f45dcb995bcf8d4b151","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-01-14 13:27:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b1fac6af369ce888546c19aba7026d0189b9f3a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 13:27:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"52795562b6726813fb29c7d4fa06f27fe22126fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 15:08:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d00b829122d7485e9ad4a3ff64ec9683 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6608b8703e2b4a6fa29348ac6159e844 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d980f56ded641a191c2696c527a6c09 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7eb44dc3d003473db4a4f9560fbc1fa9 : SUCCESS in 11m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8f0c2c3da83b47c299e992849924e8ee : SUCCESS in 4m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ce2406cf96034740921717ee726d9a1d : SUCCESS in 7m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fde087b91bb9418f9e38a758c45cf651 : SUCCESS in 16m 20s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/8d08b4f84e0f4678908a89e004ac717d : SUCCESS in 1h 07m 12s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/16796d119c544563b89dfc14f5c7984a : SUCCESS in 56m 25s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/57bca9d638a148c5beb73e8138c60333 : SUCCESS in 13m 48s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/34bed929aa5e4eba9d592020446e9847 : SUCCESS in 1h 38m 10s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/70dd990699a1490ca081f325bb556690 : SUCCESS in 16m 46s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/d07de497355048b8b38b58c27da555a6 : SUCCESS in 25m 09s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/2ae04c26ad8b429eb4d7a38676092d0c : SUCCESS in 12m 31s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"52d6e65a5dfc675536d4b5764d014f56b4860af3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 15:08:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ac44dfa739d8ba01c3da9caa85945248b04306b6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 15:10:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a4773e4a63e456a86d5bbb1d71d7284 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e360f7cd62e89f8bceb1519f13f676652790252a","revisions":{"e360f7cd62e89f8bceb1519f13f676652790252a":{"kind":"REWORK","_number":1,"created":"2020-12-22 13:57:56.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/05/768205/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/05/768205/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/768205/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/768205/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/768205/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/05/768205/1"}}},"commit":{"parents":[{"commit":"05867bfa1899b7dab854816b2f29252950d7e436","subject":"Merge \"Fix overcloud provisioning\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/05867bfa1899b7dab854816b2f29252950d7e436"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-22 13:20:28.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-22 13:22:30.000000000","tz":0},"subject":"Ubuntu: Use . to source shell scripts","message":"Ubuntu: Use . to source shell scripts\n\nUbuntu uses sh rather than bash by default, which does not support the\nsource keyword.\n\nChange-Id: Ic36531c0f26fc7dee361bebb83a1d3089166c1eb\nStory: 2004960\nTask: 41490\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e360f7cd62e89f8bceb1519f13f676652790252a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e360f7cd62e89f8bceb1519f13f676652790252a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
