)]}'
{"id":"openstack%2Fkayobe~769820","triplet_id":"openstack%2Fkayobe~master~I87260fd428b428f32e1b544e65ed09eba200c210","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I87260fd428b428f32e1b544e65ed09eba200c210","subject":"DNM: try increasing seed VM SSH timeout","status":"ABANDONED","created":"2021-01-07 19:34:24.000000000","updated":"2021-01-08 09:56:09.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9e8e2d3608c99e3483642c93c9a739e2d371bcfd","_number":769820,"virtual_id_number":769820,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-01-08 09:27:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-07 19:57:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"babd8bd18fa68702069e009776701ff13f80481d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-07 19:34:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"57fca6ad39ecabca7ae24494c591dce65c5e6d90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 19:57:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/72da414265bd46308ea84f71649a4982 : FAILURE in 18m 52s","accounts_in_message":[],"_revision_number":1},{"id":"07a54829f65bfd9ce5d5b7c4de5d9478f4dfaf50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-08 09:00:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c6a6a559afa3aa0411bf86ebc5e3980043c98b80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 09:27:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/c22c10d4764d468688ab5e275faab588 : FAILURE in 18m 42s","accounts_in_message":[],"_revision_number":2},{"id":"9e8e2d3608c99e3483642c93c9a739e2d371bcfd","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-08 09:56:09.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a4daac6b9d999f4894847127a6df813df917467d","revisions":{"50ac68d09f29a32925510a03072cb484f523a27f":{"kind":"REWORK","_number":1,"created":"2021-01-07 19:34:24.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/20/769820/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/20/769820/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/769820/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/769820/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/769820/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/20/769820/1"}}},"commit":{"parents":[{"commit":"6b8eb899cef856825c85c3ff2f907794be7c1eb9","subject":"CI: add Ubuntu overcloud deploy job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6b8eb899cef856825c85c3ff2f907794be7c1eb9"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-01-07 19:33:49.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-01-07 19:33:49.000000000","tz":0},"subject":"DNM: try increasing seed VM SSH timeout","message":"DNM: try increasing seed VM SSH timeout\n\nChange-Id: I87260fd428b428f32e1b544e65ed09eba200c210\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/50ac68d09f29a32925510a03072cb484f523a27f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/50ac68d09f29a32925510a03072cb484f523a27f"}]},"branch":"refs/heads/master"},"a4daac6b9d999f4894847127a6df813df917467d":{"kind":"REWORK","_number":2,"created":"2021-01-08 09:00:21.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/20/769820/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/20/769820/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/769820/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/769820/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/769820/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/20/769820/2"}}},"commit":{"parents":[{"commit":"337a0026ce18a227d601b9513f575b6fae03cdc3","subject":"Merge \"Remove Retired Karbor Support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/337a0026ce18a227d601b9513f575b6fae03cdc3"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-01-07 19:33:49.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-01-08 09:00:10.000000000","tz":0},"subject":"DNM: try increasing seed VM SSH timeout","message":"DNM: try increasing seed VM SSH timeout\n\nChange-Id: I87260fd428b428f32e1b544e65ed09eba200c210\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a4daac6b9d999f4894847127a6df813df917467d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a4daac6b9d999f4894847127a6df813df917467d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
