)]}'
{"id":"openstack%2Fkayobe~770561","triplet_id":"openstack%2Fkayobe~stable%2Fussuri~I8b04a3dd0403f98850185436ddf8a632b46baa3a","project":"openstack/kayobe","branch":"stable/ussuri","hashtags":[],"change_id":"I8b04a3dd0403f98850185436ddf8a632b46baa3a","subject":"DNM: test seed jobs (ussuri)","status":"ABANDONED","created":"2021-01-13 10:01:49.000000000","updated":"2021-02-03 12:07:34.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"16ce0060f1e0f94f927fc197c5e94b33ba730485","_number":770561,"virtual_id_number":770561,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-01-13 11:29:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-13 11:29:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5611127ec94db2f609aafcdfa186f9748b6992b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-13 10:01:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c2f20643688dbaba62a33397060f1456d15134ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 11:29:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/9dd71aa91ccc4d25a6e46f42acf7b6d0 : SUCCESS in 17m 32s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/67861cd4e0b4469fa38a6b4f73baf386 : RETRY_LIMIT in 2m 00s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/165ffbbf74074059af2008ab3e473be1 : SUCCESS in 15m 24s","accounts_in_message":[],"_revision_number":1},{"id":"16ce0060f1e0f94f927fc197c5e94b33ba730485","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-03 12:07:34.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ea554524c0fcd97d66776c4b8076c3f004b9ae30","revisions":{"ea554524c0fcd97d66776c4b8076c3f004b9ae30":{"kind":"REWORK","_number":1,"created":"2021-01-13 10:01:49.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/61/770561/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/61/770561/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/61/770561/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/61/770561/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/61/770561/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/61/770561/1"}}},"commit":{"parents":[{"commit":"ea073b8cb4960356cd3318f89f0ea7476b21aa03","subject":"Merge \"Fix implementation of kayobe seed hypervisor host package update\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ea073b8cb4960356cd3318f89f0ea7476b21aa03"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-01-13 10:01:18.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-01-13 10:01:38.000000000","tz":0},"subject":"DNM: test seed jobs (ussuri)","message":"DNM: test seed jobs (ussuri)\n\nChange-Id: I8b04a3dd0403f98850185436ddf8a632b46baa3a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ea554524c0fcd97d66776c4b8076c3f004b9ae30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ea554524c0fcd97d66776c4b8076c3f004b9ae30"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
