)]}'
{"id":"openstack%2Fkayobe~770908","triplet_id":"openstack%2Fkayobe~master~I54cb2e723ec3cb0f01cfebcef4ec385c166fb5e4","project":"openstack/kayobe","branch":"master","topic":"kolla-build-customizations-hyphens","hashtags":[],"change_id":"I54cb2e723ec3cb0f01cfebcef4ec385c166fb5e4","subject":"Make it clearer that hyphen replacement is not automatic","status":"MERGED","created":"2021-01-15 09:13:11.000000000","updated":"2021-02-15 20:15:50.000000000","submitted":"2021-02-15 20:13:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"770908-kolla-build-customizations-hyphens","meta_rev_id":"f523c0e9d23e7012de0c7a3d509f104554d83038","_number":770908,"virtual_id_number":770908,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-15 20:13:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"recommended":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"all":[{"value":1,"date":"2021-02-11 10:36:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":2,"date":"2021-02-15 14:26:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2021-01-15 09:16:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":1,"date":"2021-02-15 14:26:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-15 09:16:19.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-01-15 11:08:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-04 09:39:34.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-02-04 09:39:34.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2021-02-04 09:39:34.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2021-02-11 10:36:41.000000000","updated_by":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"reviewer":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"state":"REVIEWER"}],"messages":[{"id":"bb0d8b7d23705513a5f8108dd634ba227c089dc5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-15 09:13:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"629ef3f5378e775aadd88b1898be95aeaad5db7c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-15 09:16:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a9ec1bdd8a982f0a2537b6d4df74e446e754cf92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 11:08:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ccc7eb04c27040af8f8fc132f63f4a6c : ERROR Failed to update project None in 1s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a68bbfec6dff4766a4c5f66dd0353bbf : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/95f6bbfdd8bd4fc7bad9d9ba58c46fae : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4688d1b3e18243c8a89b4be0943707b0 : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/667fbb47b46342d6813ed5482cbde94f : SUCCESS in 4m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e2feb71c3e349baa7c626b37c37b984 : SUCCESS in 8m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7a2b905d62294b3fb95d3b0559d71d7b : SUCCESS in 4m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8a0fc506a77249b0a631fefbfe4673ce : SUCCESS in 9m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/640ea8921c5c435c9dcf52bf6e9d2ff1 : SUCCESS in 16m 06s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/304fc9b7eb4d4be8950f41bdc63f7947 : SUCCESS in 53m 45s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/565d60d97aeb4add86ecbd49a42a7b2a : SUCCESS in 1h 03m 32s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/d6ae169c11e440ccaeea9b65874393a3 : SUCCESS in 15m 01s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/82a6a13ac8fb41479894c009e3bf26b9 : SUCCESS in 1h 43m 46s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/67eb816b590141c1a9b76a37510ebe5b : SUCCESS in 17m 08s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/74092448d8bb47699a5f25b453d172b6 : SUCCESS in 24m 08s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/a8c20cb7adfc4c2ca02a7cb346903b17 : FAILURE in 13m 24s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"b0dda6bab4ffeeb86a6edc97e3a793ef10a79f66","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-15 11:52:57.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c4857ce37a887b3fa1dcf8f98f4b0851cfcba093","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 13:51:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79e7bb49ebe64d79ada48b3becc70ded : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8061c68d6fb49fc9f62791c2452d56a : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c033c09fe51433d895f8ac942c16d9b : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/844eb237de024780adb0e8c2194bad10 : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d18bec9d07247cb9bd0d945c1f0f902 : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4541c215acd46d5b03a9c02466bfa28 : SUCCESS in 6m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/459beb4261ed468b9f7bb0e44e8b8b65 : SUCCESS in 4m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c6ee1fc6cdec4ad58d17f5a6b72da70c : SUCCESS in 8m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3c0ba87f5e074cb798f7dedcc7fc8eff : SUCCESS in 15m 05s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/b27b44a7da624bafa6fcc89a544888eb : SUCCESS in 1h 02m 40s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/cc957de2024f46b3a0c147a236006bf4 : SUCCESS in 1h 04m 44s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/501092bf5ca2475ea47697eb5012c670 : SUCCESS in 16m 28s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/0158eec835674c469ce385139e2ee664 : SUCCESS in 1h 50m 16s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/34065754416547319dbd60dc4817b52a : SUCCESS in 16m 18s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/9b585c94bfa545d283466b36e13e0397 : SUCCESS in 22m 31s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/f473c47c467947e8bab59afab16fabf3 : SUCCESS in 13m 37s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"69f8045bb9c3e30b902f7ce42bea2afa0f3e845b","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2021-02-11 10:36:41.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c52f17da4198cd13d5428d490a76f4b28c731fff","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2021-02-15 14:26:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n#RFC-2219 :)\n\nhttps://tools.ietf.org/html/rfc2119","accounts_in_message":[],"_revision_number":1},{"id":"3fb89b420166f9c390320ee018ccad7463622b88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 14:33:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1e2f1106ad8e8f04cc2b02f4cb0c47446ec744a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 16:30:52.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1931e9dc01ac4b6d82eb9c2272d19f16 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c2f4c5713ca496baaa8eec5cd295697 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e7a74f0a66dd42f4821a74b4909ab182 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbe3649f248e40898f33c042141347b1 : SUCCESS in 7m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d0c40884b9aa4a49a0c254f1971a249a : SUCCESS in 5m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/23a55de446ea470d86695300ebcfe303 : SUCCESS in 6m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2fef82c711144a5faa16a41951d9341d : SUCCESS in 17m 17s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/7d2d06050e934002948f6de83fe81d09 : SUCCESS in 55m 47s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/c64bc208ab2c4015acae57dbb0419e97 : SUCCESS in 1h 13m 31s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/f6bbaf16a52f4c9c80fd6e083d97f694 : SUCCESS in 13m 44s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a106bb60ff034dc68fd859cd9e78b8e4 : FAILURE in 1h 53m 31s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/533dbe9e5ba14c98b3e14a1d5b4b2a00 : SUCCESS in 20m 58s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/6924843c80d5404fad5e9abf9b2b155a : SUCCESS in 24m 27s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/f58948494eab4528a09cf2ff8c0d4ad8 : SUCCESS in 10m 32s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"155c6b22b6d6e4c6a67895ad523e59a4b1374852","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-15 17:05:10.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"25471323ae43b1b0e5dd38664e5473851c38190f","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2021-02-15 17:19:26.000000000","message":"Patch Set 1:\n\nAny idea what the SSH to VM fail was?","accounts_in_message":[],"_revision_number":1},{"id":"8b8bd4fe5bf3979e24ccf982ef488db41233b308","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-15 17:21:08.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e Any idea what the SSH to VM fail was?\n\nNo, let\u0027s see if it happens again…","accounts_in_message":[],"_revision_number":1},{"id":"7ef0da034c585975bc4007e5db8c89eef0c3ffd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 18:41:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d98a541125184ed0abaf2581a471a072 : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a3f34c62b124fb09039db2ff1045c8e : SUCCESS in 7m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a7e9f762deaa4dbeae7f1e3dc55bfddc : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a2fbc87187844289b3b9c56802d6719 : SUCCESS in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7c2094ca471c45c188237b09ad725935 : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75899f25bfd641aa90244cc628f1cf4d : SUCCESS in 6m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8b0efd1756a44b5fa7a21da73df2eee0 : SUCCESS in 4m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a6ecedc4b27747fbbe060e8be1842423 : SUCCESS in 11m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cd867d403e8c41ae8c59bb89e5b68ea9 : SUCCESS in 17m 11s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/13adf22e97f848dca43958533771272d : SUCCESS in 56m 43s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/bfc62d399d47436ea0da1613652d7d90 : SUCCESS in 58m 30s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/15a9a0382ce742ec80a02d3ff6d12319 : SUCCESS in 14m 52s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/05eab6c072f746629df37598fe0b3caf : SUCCESS in 1h 31m 15s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/253eda3cea7547d0abd5532aeac404a4 : SUCCESS in 15m 53s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/93f964f3b46f4b38a9fb966a3d8894ec : SUCCESS in 21m 47s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/f682a766ad1147558ef49acf10308601 : SUCCESS in 11m 02s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"83e142e0c524aa6b9707812448e54e8a94cb7d7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 18:42:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2ad4f55472a5547f7eb0709791bf3bcc4810bccb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 20:13:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5d75a8894b043eea0f7a83ecc5999b6 : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81954d88f6e547a2bbee185af4337ff6 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7957ab75b0b74fc39ab8758476ff055d : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2143131204e747b694f063607ebbc159 : SUCCESS in 7m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7b218acf4b51436bb5369057188e8f49 : SUCCESS in 5m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/afd9a5d46c82419ba61154b611bbc8cc : SUCCESS in 10m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b5b8289a91e543c4ac633bf30e577642 : SUCCESS in 19m 50s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/9262040aa3c8463fb6ba9130d3bec935 : SUCCESS in 1h 03m 40s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/e5d15b7ff28b4bcebb25c01499dd8246 : SUCCESS in 57m 19s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/fbc115f6bad84dffb1861f2e4148f01e : SUCCESS in 16m 11s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/ad68242a5ded45d5901d422a151cb682 : SUCCESS in 1h 28m 47s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/ff524373bbf04a5a991db3fc72d14aa8 : SUCCESS in 17m 03s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/74501e8df33a4a89a53d7cfe93367690 : SUCCESS in 21m 40s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/18cadfd9435746cc982683b9928b5e46 : SUCCESS in 10m 20s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"3b2562ca1b354400e7a2240e2d6e850d48057964","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 20:13:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f523c0e9d23e7012de0c7a3d509f104554d83038","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 20:15:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a08e506b97449c6868fbb53010177fb : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"648afb721a666cf80d993c5cc682d44b7791739a","revisions":{"648afb721a666cf80d993c5cc682d44b7791739a":{"kind":"REWORK","_number":1,"created":"2021-01-15 09:13:11.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/08/770908/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/08/770908/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/08/770908/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/08/770908/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/08/770908/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/08/770908/1"}}},"commit":{"parents":[{"commit":"f248be81ecb2f8f9b54516c619795425c37e0fac","subject":"Merge \"Fix --limit with commas\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f248be81ecb2f8f9b54516c619795425c37e0fac"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-01-15 09:11:59.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-01-15 09:11:59.000000000","tz":60},"subject":"Make it clearer that hyphen replacement is not automatic","message":"Make it clearer that hyphen replacement is not automatic\n\nChange-Id: I54cb2e723ec3cb0f01cfebcef4ec385c166fb5e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/648afb721a666cf80d993c5cc682d44b7791739a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/648afb721a666cf80d993c5cc682d44b7791739a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
