)]}'
{"id":"openstack%2Fkayobe~770959","triplet_id":"openstack%2Fkayobe~stable%2Fussuri~Ibe42fa372c6fa0c539d2c2b0e238601286dc213d","project":"openstack/kayobe","branch":"stable/ussuri","hashtags":[],"change_id":"Ibe42fa372c6fa0c539d2c2b0e238601286dc213d","subject":"Fix --limit with commas","status":"MERGED","created":"2021-01-15 16:59:38.000000000","updated":"2021-02-07 17:23:08.000000000","submitted":"2021-02-07 17:21:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"770959","cherry_pick_of_change":758459,"cherry_pick_of_patch_set":2,"meta_rev_id":"f1e31f4fba5492c20fd944596435a7168ac0a36c","_number":770959,"virtual_id_number":770959,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-07 17:21:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-02-03 15:56:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-18 08:45:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2021-02-07 13:49:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2021-02-03 15:56:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-15 19:28:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-18 08:45:47.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-02-07 13:49:41.000000000","updated_by":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"reviewer":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"state":"REVIEWER"}],"messages":[{"id":"1ba740b51da08b5085af29853ed07c97d85cb16c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-15 16:59:38.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"387b24ee27118cc042a853b22884419b90299621","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 19:28:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/315dd7a355e0444098ae0b43a7be73f1 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/484e907e74344548a3d659580851d557 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/29ac9018b3274aafa8185046fc5d4c52 : SUCCESS in 4m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5316762e03e34c7da81527cff9caffd4 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/568340233419459faefeef0943a4a7f8 : SUCCESS in 7m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71164be20eb24fbba2c46ea33530a1fd : SUCCESS in 3m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ed7bff9033e2482695d6030ae3181db8 : SUCCESS in 4m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9237eb12d90a48d480e83ded822206bb : SUCCESS in 8m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/481b157c62e743089cea3b0f1f7c0bad : SUCCESS in 17m 18s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/08657363fc0948bba8d904f420dcd71d : SUCCESS in 56m 57s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/be70bdf377fe4a2abf319c34c2254692 : SUCCESS in 15m 10s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/0f42e4b80d0347d38e5361fad02f5aa6 : SUCCESS in 2h 20m 26s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/f166aa74267c4b129a26962b2487e91b : SUCCESS in 16m 41s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/96b9214221434324a5cf410b90c5d328 : SUCCESS in 26m 39s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/53e4ab1e15c44923b387e3422fd9dfff : SUCCESS in 11m 05s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"01cfb7f110213dae933ac1b5db05de982a6c6c21","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-18 08:45:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e8f21ad7b72638cc6dff2e92766319cd94a29032","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-03 15:56:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a8a3d45be373c602865f1b975a89663237736f8a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 15:57:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"28f31c72992ce5b515252c685c656f121f8aaf4a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 17:10:54.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a123965692df45c8b30c22b2fd133c8c : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b8d337a46cf4b42aa732f1a6d3bb4ee : SUCCESS in 4m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/86e08d13c9f24e1a9dbb4b8b2a5ae978 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ce224cf507a426f9623049f5e5d29e6 : SUCCESS in 6m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/08991c3f0ecd4ce58a30113fbff5aefc : SUCCESS in 3m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/35ed4b10dcd74454ab80ca794aabadff : SUCCESS in 5m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8d164575a34b4eb199225d4c73017f1f : SUCCESS in 9m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/10784fcde7074ebb91d88f0ab3148494 : SUCCESS in 16m 58s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/2a9bc6be541a47fbb50b18076471ab8c : SUCCESS in 56m 36s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/e420c776f73142ada9dd4df1f1eb3bb5 : SUCCESS in 14m 36s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/db608bcd7a724ef79107daf3ff62fd80 : FAILURE in 14m 02s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/dd34f7dda0b64c4fb867057ac3d78791 : SUCCESS in 13m 04s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b0fb4d7117884d6f97702f1da6fe31a6 : SUCCESS in 19m 35s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/e5cebb4eaf024d5ca5e458f3856c53c9 : FAILURE in 14m 46s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"81c5d6e6f25444c87ef34ea641fac20dc7565ab2","author":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"date":"2021-02-07 13:49:41.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c6f4ecbfb61318f664376dc321b638c83d2bff10","author":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"date":"2021-02-07 13:49:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5a709a827ce8590367402a1977ba85b91e53d290","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-07 15:31:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6f845bdb95404d6ba4a3e481c48fe3c9 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e83c9dc904948a1acc6a0511ad9b19d : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eff08b71d2944cac9f76adc48612e48d : SUCCESS in 4m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8feb8029d0c74c11b7ffc1aec29e9408 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f53c4dd7dfbb403ba0c1b77a638cd7de : SUCCESS in 6m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e31a10d317442859a69277314114a35 : SUCCESS in 3m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ffe6b1057c974c9abbd8ec91d98c5201 : SUCCESS in 4m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8534f9444122477087056199ceff1c1d : SUCCESS in 8m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3989aa84148f4a1d9ef03f7484e27c65 : SUCCESS in 17m 21s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/f9be51ebb1c74558b39d1aa6cd078110 : SUCCESS in 59m 27s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/480f8391dae644afaeaa16adffa182a5 : SUCCESS in 14m 45s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/4600761ce082489b80f3f15054779b43 : SUCCESS in 1h 39m 46s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/75db66551d6642769739533ab3ab40d2 : SUCCESS in 15m 00s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/8a55328f1d224cc5ad84ac64a4ae9c4b : SUCCESS in 23m 09s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/4465487edc4546ddb10c188c4a64b870 : SUCCESS in 10m 03s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"3ad84a8e09fbf5e7e7e0e4e595ae3bd7070ee942","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-07 15:32:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2762fe2a32f4eea4c80be25d3fece0b4b2e618ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-07 17:21:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e28a13437fdf4756af780d1ac80347e5 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a3a5004e5a274aceb161b65feacab7ec : SUCCESS in 3m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/76fc8bc1c9264223bcdb5074f399a919 : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/581f990b88d047eea7b8170319f38544 : SUCCESS in 7m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/535f43fe352e482eb21331d27ef39496 : SUCCESS in 3m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6b395dca1557454e90d5ce0eb612f101 : SUCCESS in 5m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3e07f6d98de24c85a3df38463713e408 : SUCCESS in 8m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2a72f4284b8a457eaa85560930ddde8d : SUCCESS in 17m 54s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/6cc34bb4d18e4a13b13bf60699d91c95 : SUCCESS in 1h 07m 03s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/4a8f0e5df53848bd90667d58610b8b2b : SUCCESS in 15m 12s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/bfaea8f9263242b79d119b0be1d45d4c : SUCCESS in 1h 43m 31s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/4e0f699fec7542c9a1b33ec0d79a6797 : SUCCESS in 16m 19s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/fcfbb06dce4242df998e4efb84e040a0 : SUCCESS in 30m 18s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/25813f1ddc1841779a9d2f8e1d521d17 : SUCCESS in 10m 14s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"89c02d05aef862c4c39f619275cef4828402f1a9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-07 17:21:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f1e31f4fba5492c20fd944596435a7168ac0a36c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-07 17:23:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa98772683cc4395a9aaadb28d2bb026 : SUCCESS in 1m 03s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/689087f363cc471f8b5be32dd05a8656 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"783d0f2e3626bab169d6f2268b0bc1e97514dbed","revisions":{"783d0f2e3626bab169d6f2268b0bc1e97514dbed":{"kind":"REWORK","_number":1,"created":"2021-01-15 16:59:38.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/59/770959/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/59/770959/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/770959/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/770959/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/770959/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/59/770959/1"}}},"commit":{"parents":[{"commit":"ea073b8cb4960356cd3318f89f0ea7476b21aa03","subject":"Merge \"Fix implementation of kayobe seed hypervisor host package update\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ea073b8cb4960356cd3318f89f0ea7476b21aa03"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-10-15 15:35:51.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-01-15 16:59:38.000000000","tz":0},"subject":"Fix --limit with commas","message":"Fix --limit with commas\n\nKayobe allows specifying a --limit argument, which is passed through to\nAnsible. In some cases we wish to add an intersection with a group. This\nallows us to reuse playbooks for the seed, overcloud etc.\n\nFor example, the lvm.yml playbook specifies a host list of\nseed-hypervisor:seed:overcloud. When executed as part of a kayobe\novercloud host configure command, Kayobe passes a limit of overcloud. If\nthe user specifies a --limit argument, this gets intersected with the\novercloud limit: host1:\u0026overcloud.\n\nThe problem happens if the user specifies multiple parts to the host\npattern in their limit using a comma, e.g. host1,host2. This results in\nhost1,host2:\u0026overcloud. Ansible ignores the colon, and treats this as\nhost1 or host2:\u0026overcloud.\n\nThe solution is to use a comma to join the patterns if the user has used\na comma: host1,host2,\u0026overcloud\n\nChange-Id: Ibe42fa372c6fa0c539d2c2b0e238601286dc213d\nStory: 2008255\nTask: 41111\n(cherry picked from commit 017b092df70706177c542c3a3cab3f9d0a75e5c2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/783d0f2e3626bab169d6f2268b0bc1e97514dbed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/783d0f2e3626bab169d6f2268b0bc1e97514dbed"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
