)]}'
{"id":"openstack%2Fkayobe~770960","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~Ibe42fa372c6fa0c539d2c2b0e238601286dc213d","project":"openstack/kayobe","branch":"stable/train","hashtags":[],"change_id":"Ibe42fa372c6fa0c539d2c2b0e238601286dc213d","subject":"Fix --limit with commas","status":"MERGED","created":"2021-01-15 16:59:59.000000000","updated":"2021-02-08 20:39:47.000000000","submitted":"2021-02-08 20:38:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"770960","cherry_pick_of_change":758459,"cherry_pick_of_patch_set":2,"meta_rev_id":"3af769f84a0abca02c25cb8c8b5b25d123ba10bf","_number":770960,"virtual_id_number":770960,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-08 20:37:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-02-03 15:56:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-18 08:45:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2021-02-03 15:56:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-15 18:55:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-18 08:45:59.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"c8b936f1d3652b1493e599092d4bc7345df2e9f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-15 16:59:59.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b0a78a512a67269682e2003e3cc970ffe77843ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 18:55:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f4bfa3354a64c1aaa3ec67f0e0633df : SUCCESS in 3m 52s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9c226c75ec8d4e64a0d10eb11ae82759 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8f921afe2f54561a59641bf50a13708 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2511132a71f4dd3a993d1a5377949b7 : SUCCESS in 3m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/93e717675fb2425db8cefce8a3fd0a04 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ab5fd1370a340dbb3d4dcb99a6783a1 : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a198ca1703bd491fb997130800a53ec7 : SUCCESS in 3m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fbddbc9b2d0d4b27a7b9bc1581f20ada : SUCCESS in 4m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/706be710cba84bf4879d8b42631dc722 : SUCCESS in 10m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/120de284c9194c27bbd03adc548487cc : SUCCESS in 17m 07s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/fb3f81ffded14e988a20d4b2592eebd6 : SUCCESS in 57m 50s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/2ef4ba5d43e147dcba3c28f4a1bc4ae5 : SUCCESS in 1h 00m 32s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/8736c8ec135f46d2a73481716dde25b4 : SUCCESS in 15m 02s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/681024590ca24c5facbaf96f628851c0 : SUCCESS in 17m 06s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/6b08649c8ce1470bb24e25caaba6aba2 : SUCCESS in 1h 45m 43s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/cfb17e008cf14733bbd7a67cb3fef120 : SUCCESS in 19m 46s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/bca08b4d770e4cf48a3318cff8f6d46f : SUCCESS in 16m 26s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/8458e4ae605e46dd84f9052eef97087d : SUCCESS in 34m 18s","accounts_in_message":[],"_revision_number":1},{"id":"e598f4486de65d0cfd84d2ea8f283485ae1b2012","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-18 08:45:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6ba576097afb415365639a02b3f8b62209011d0a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-03 15:56:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2fa5d1eff205eb22fe10007a043ce64f08dd3c38","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 15:57:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"89a3298687c09663a693417e229f997938d7c9b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 16:28:12.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a2fb8a4b1b374d79838c0054ba8b6341 : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ea6ec66e2fa46e2a7a6010f01fd1be9 : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83454aacbb3d441599b3bb8fe5610715 : SUCCESS in 4m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e82df18c6765400fafa77a55ff9a9656 : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caa5f807e06848b19b48dd1f5a20bb60 : SUCCESS in 7m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f8042d71737643ae80d99f6df10502dd : SUCCESS in 4m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ca3dbf178523420e97a2fc5841ffa45e : SUCCESS in 4m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1c60ca5a8a974c88bca408f17df33ca5 : SUCCESS in 8m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cbe475d539de49a5b1cd261da065197a : FAILURE in 6m 03s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/9b70d3c7be9a4cbf977fb203aba7d138 : FAILURE in 10m 29s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/2ee84ae660154a2fa74dcfa93702a78a : FAILURE in 14m 09s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/3f8b569701714d55bf670154dac24a25 : FAILURE in 13m 18s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/75db922a59334a278d29ebb7aef8b17d : FAILURE in 13m 31s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/47ab2737dbe24171aa72c883dbd6eccb : FAILURE in 10m 53s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/381a87d0c9504541ac6c46b82f8fbd97 : FAILURE in 10m 01s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/964e48f0f3374f34ace6617f505e352e : FAILURE in 11m 27s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/e14eff5f01fa4618a7dc5faef49f7007 : SUCCESS in 23m 37s","accounts_in_message":[],"_revision_number":1},{"id":"384b69511f993576b4b01066195048c2c764fbdc","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-08 09:03:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"33321bfdea1ef6d0b45a9b180b51094d2448cb26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 10:44:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e3d4c4873274cea9abe016d4e127156 : SUCCESS in 4m 56s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f0b77989d45c4945a6d803cd68ea44e8 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a6a61db451814fa99410f54adb737024 : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10f9f805b9f94617ba26c8ebfd0e3026 : SUCCESS in 4m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9898a33f56ce49d6aeeb2db08223ba59 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed5c84e7449f4f56b702fcf54d38811c : SUCCESS in 6m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/356cbda7e2564febb7f74fa5a1d651dc : SUCCESS in 3m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5eb46de13b614d36a7efc0a56cace6a2 : SUCCESS in 3m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8a306e32728c4ffb9240eb43c50d644d : SUCCESS in 8m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/62db121bb4a74ddaa385745056808f4b : SUCCESS in 16m 35s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/9a21a923058f4370a4f8e45ba75091fc : SUCCESS in 1h 11m 48s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/cc23db2b6347461595a1825a1dc82640 : SUCCESS in 1h 02m 51s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/4019d7b601da4cdfa2d9ab9107d2b2c7 : SUCCESS in 14m 55s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/cfe37b09fa294386bfcc6be99d39821d : SUCCESS in 14m 10s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/f562fe38c12443e39de3fddab9fa7440 : SUCCESS in 1h 35m 10s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/b137a8ecabb440dd95e18bf4fe0f51dd : SUCCESS in 18m 35s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/995df7b8c7784847afccbc823c2f904f : SUCCESS in 17m 37s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/6b1c0381bae349338f1fd4b4314afaf1 : SUCCESS in 33m 19s","accounts_in_message":[],"_revision_number":1},{"id":"6f4a80c48eea938a659ba19763fe86874a0c9800","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 10:44:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5ca2eeefaad6d036e835f60e00a12c3199d78cce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 12:14:20.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a4b69f0a727047659fd2fd29370f34d5 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d33a8ac8cb0425b96367ef51a67a40c : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9e7f659990047e6b9b9ab41d175093e : SUCCESS in 3m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/49fb535c3fc34202ba6bfd155497300a : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ace1c2fe75c487cade71cffece98d85 : SUCCESS in 6m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1619a47614f64d2bbfce97910745c51c : SUCCESS in 4m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/978a55bb60a14e0eaf20f8bd7f015a8d : SUCCESS in 3m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b1e0f5be8d3842a2aec15a282da6d337 : SUCCESS in 9m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4c7fdbaf9abb456fb7712eccded78a7f : SUCCESS in 13m 54s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/5f4d1f0870b244b4aef96d028ff66bfa : FAILURE in 51m 36s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/6fc4ee763ccc4a6b9e46c069adb5dba6 : FAILURE in 49m 13s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/ef923ed713aa412ca96011a255512a3b : SUCCESS in 13m 29s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/44958af630b840f2a270ac3cfd1051a7 : SUCCESS in 13m 21s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/a07e86d50a1c4137aabedce93dd39d74 : SUCCESS in 1h 27m 14s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/1146025e9afe47de90b7ab0c2e59ce46 : SUCCESS in 19m 56s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/89609f9da9ef4dfa9d8305a96ce806fc : SUCCESS in 16m 42s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/1b2cca25d6724aaebc99dfbecc7b7ffd : SUCCESS in 33m 12s","accounts_in_message":[],"_revision_number":1},{"id":"835687560aa3d0fbc75211ad3e9744f4722efb2a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-08 17:34:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"083bbaf4585f8c55e7567595badb520e34129387","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 19:10:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a78eb98e761450a94c8219c86daf3e5 : SUCCESS in 4m 32s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d8aa10dcbee94fc293be2606b8905a69 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf02583cec934507a92b2a7d28d8af32 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28072a48e0444f99a1fa43572b13609e : SUCCESS in 5m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e5c25c6c47e54340b702c2c592679523 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ac7960fa1454b71851ff2b956be44d9 : SUCCESS in 6m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d33d3471a3f4353aaf7260c044f83a5 : SUCCESS in 3m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/100565bc45d940c4a2468d43d5bfaa6c : SUCCESS in 4m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d43a50517e2b4d80a21941c80e03a72e : SUCCESS in 7m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/89ca29ff1fc94bda9aaec4d001706ccb : SUCCESS in 16m 51s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/7833dbbe720049a38d496dbca869d730 : SUCCESS in 1h 08m 24s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/4f0257bac0184ca6898cb913a7e58269 : SUCCESS in 1h 07m 47s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/618ce4446cf747c9a70836075a3ddb57 : SUCCESS in 14m 58s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/6a11c54488de4839854cc26c340dc07b : SUCCESS in 15m 32s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/081c27032e674f50be305f83c37aae27 : SUCCESS in 1h 32m 23s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/dba71efd30ff438d84f3fb3b91a6b266 : SUCCESS in 20m 42s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/dd7a3be9aaed44fda6f7096dddbffb0d : SUCCESS in 15m 47s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/45d6a670f8404951ac77ff3ccf0b2d8e : SUCCESS in 30m 53s","accounts_in_message":[],"_revision_number":1},{"id":"5fc9c6f407b15b2ced4bdb3a5ead31b6eaa2649c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 19:10:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f46949c558b7dd1e32c7c27c148b517854b908ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 20:37:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b6f9129ddad24148a186a698a45d73b2 : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1c35758665842a8b5b7e1c4afa3003d : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2701ce099c29433ab67927cfa814c385 : SUCCESS in 4m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c39efdd1e83b48e0ba041a24f4a00479 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b2b749ebb6e473f9bc3d58a94f81136 : SUCCESS in 5m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2279b44b306a4cc2bb44753ae9cd90d3 : SUCCESS in 3m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b434bed71beb41ec8398038b6a2a1b65 : SUCCESS in 3m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2e289edddfe64917934a2fbb9e6f3d0b : SUCCESS in 8m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4b26f0222adc4b8a9dbd677285c72110 : SUCCESS in 17m 24s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/c4372626060e48178aab7bf3253e894e : SUCCESS in 1h 02m 49s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/08dbb0fd1dfb4f4b99e25b323cc3074a : SUCCESS in 57m 06s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/a69ae41c48634f68a87f6ce5e29c1fef : SUCCESS in 16m 16s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/88db547e7a2c447b8cca9be7b130b9b1 : SUCCESS in 18m 30s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/3038bd3236bc4331b15f0427ac1e08ff : SUCCESS in 1h 25m 06s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/81c125f289ba4993bfae0ff3adab71a3 : SUCCESS in 25m 35s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/9c3814bf3f5f47518a7a9adb61344599 : SUCCESS in 14m 51s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/10a7031c165a42ab8173ed232d6fab6f : SUCCESS in 34m 37s","accounts_in_message":[],"_revision_number":1},{"id":"5c8cea25b2405548ab175adc224644b804dcac94","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 20:38:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3af769f84a0abca02c25cb8c8b5b25d123ba10bf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 20:39:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/039d5641c7d5411dbd336e77851e3e64 : SUCCESS in 1m 02s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5722cb45382a44e09d6249e183efeb16 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b3dd061250745f81bfcde292f13464eb8203eba2","revisions":{"b3dd061250745f81bfcde292f13464eb8203eba2":{"kind":"REWORK","_number":1,"created":"2021-01-15 16:59:59.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/60/770960/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/60/770960/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/770960/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/770960/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/770960/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/60/770960/1"}}},"commit":{"parents":[{"commit":"cf0146097249fc211131732f858399c8ea1e30de","subject":"Merge \"Fix implementation of kayobe seed hypervisor host package update\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cf0146097249fc211131732f858399c8ea1e30de"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-10-15 15:35:51.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-01-15 16:59:59.000000000","tz":0},"subject":"Fix --limit with commas","message":"Fix --limit with commas\n\nKayobe allows specifying a --limit argument, which is passed through to\nAnsible. In some cases we wish to add an intersection with a group. This\nallows us to reuse playbooks for the seed, overcloud etc.\n\nFor example, the lvm.yml playbook specifies a host list of\nseed-hypervisor:seed:overcloud. When executed as part of a kayobe\novercloud host configure command, Kayobe passes a limit of overcloud. If\nthe user specifies a --limit argument, this gets intersected with the\novercloud limit: host1:\u0026overcloud.\n\nThe problem happens if the user specifies multiple parts to the host\npattern in their limit using a comma, e.g. host1,host2. This results in\nhost1,host2:\u0026overcloud. Ansible ignores the colon, and treats this as\nhost1 or host2:\u0026overcloud.\n\nThe solution is to use a comma to join the patterns if the user has used\na comma: host1,host2,\u0026overcloud\n\nChange-Id: Ibe42fa372c6fa0c539d2c2b0e238601286dc213d\nStory: 2008255\nTask: 41111\n(cherry picked from commit 017b092df70706177c542c3a3cab3f9d0a75e5c2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b3dd061250745f81bfcde292f13464eb8203eba2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b3dd061250745f81bfcde292f13464eb8203eba2"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
