)]}'
{"id":"openstack%2Fkayobe~771244","triplet_id":"openstack%2Fkayobe~stable%2Fvictoria~I82313fcd258e58969e0b68ae91a78943df537a7f","project":"openstack/kayobe","branch":"stable/victoria","hashtags":[],"change_id":"I82313fcd258e58969e0b68ae91a78943df537a7f","subject":"Fix overcloud host image resolv workaround on CentOS 8.3","status":"MERGED","created":"2021-01-18 16:07:17.000000000","updated":"2021-01-20 18:33:47.000000000","submitted":"2021-01-20 18:31:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"771244","meta_rev_id":"6db7918505d9636d880eae67e9f96a003d39c96c","_number":771244,"virtual_id_number":771244,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-20 18:31:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-01-19 09:30:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-20 16:00:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-20 16:00:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-18 16:07:17.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-01-18 18:29:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"62c9336cc4d2fda9db9481a6358da3f827d9a883","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-18 16:07:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e28cc9c84fcd6ba58415c3a49ed9c85540dfbae9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 18:29:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b0ad104e22f4bd29b5176a51df09bc6 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9d62394f5ea417eaa6555d0012b881c : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/18cca2b673ac406ca094dbae9fc2a599 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e16e44134f384bc1b12a96e36eacd3ea : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2bc394cec584b459f5f67fd03f30d08 : SUCCESS in 6m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3bfc54bf191a464d929d65f29ff20166 : SUCCESS in 3m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/40af1e9bec00435681e249b6dc23f404 : SUCCESS in 5m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7a0c00a05b7544e79405a8ab2837d60c : SUCCESS in 8m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/90afe886a915445ca7598aafc5f600a8 : SUCCESS in 20m 42s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/0efb9e954b09487285eeef3855e088e0 : SUCCESS in 1h 14m 56s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/c5d0d62f50eb4fd49660ea548b2cd92e : SUCCESS in 1h 14m 55s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/afe38774f8d3496496e8fe4fb0742b17 : SUCCESS in 17m 28s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/22bfc4ca7f7d43668c4475ea65f89045 : SUCCESS in 2h 01m 55s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/414e5494e24a49f6a74fcc023c80b70a : SUCCESS in 21m 33s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/7cdb1654d123446fb9145b5b03ef0445 : SUCCESS in 22m 43s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/8324fad2260e40269c133c684a53f456 : FAILURE in 13m 43s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"e3f1034d8586b97fe11a681381d0fbef69898931","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-19 08:51:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"08ac5106e2362f7428f9500526d9ccfe7294b6cf","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-19 09:30:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ad719501c8e2c4dab35381403e53384d8da94a10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 11:03:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cb7566d1877e4ccda952162a433cd3a7 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96602fda80d7495f9b9aa81068f6a8f8 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/483cda816c1c4ccaa9e3c7d0fcadb9b5 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/30363d3445044198b036e3d148d5c0df : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a45c9493c154229a63494272b6a72d5 : SUCCESS in 7m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/040779e6c8d34950af8e7816206cbaf7 : SUCCESS in 3m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6d004a42803b411281df96625bfd32a7 : SUCCESS in 4m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/827fbcec21b7418080edfe189a483262 : SUCCESS in 7m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/db01e76d1b0847c8ab06d74f99546b5f : SUCCESS in 18m 58s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/2d16864e75134beb84f36f07b825272d : SUCCESS in 1h 02m 36s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/0a6c600ce28040bfa510c18e577aeb0e : SUCCESS in 1h 03m 28s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/bb5dac2b49c643c5bc61cfd75e5aa635 : SUCCESS in 16m 15s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/90a433b06da84dc48436590ae2179598 : SUCCESS in 2h 04m 28s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/8fe35457f614447c844549886925404c : SUCCESS in 15m 00s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/984a9906645c414293c51d612d74cd6b : SUCCESS in 22m 54s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/87a106c8bb34476e8329861177602ab6 : SUCCESS in 10m 35s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"e8a0f738f554926ba7cdc07e76e77bd1aab4a134","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-20 16:00:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8b88793d2f44a00eb19e99d0d7fcb68029257aaa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 16:00:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dd589bfacf34d42861aa715cd4fbd7cf06e83c2f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 18:31:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/807adaf5a01c4cccb6dfe30f616ce413 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f68c4c0a8a4b48a3b22981745ef186b1 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f992baf47ad451890a42bf615313e05 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b985b601b44d409d9164910b11aa0178 : SUCCESS in 6m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe9394c36ab244cab1081d2757ac8ffd : SUCCESS in 3m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a4444a31cca84e0fa714d97c5bae0d99 : SUCCESS in 5m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a3c9f55e14564a97b439ff9c3ca0a2de : SUCCESS in 8m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/40b086886715427a8c1b633c9881b2c4 : SUCCESS in 19m 35s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/7b8d80614d1a463b8bf53c70a38315e4 : SUCCESS in 1h 16m 27s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/30c0bf8de4c34363adce16400f478770 : SUCCESS in 1h 17m 20s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/36b54adf259b4b15a93d16a690770e05 : SUCCESS in 17m 19s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f26517e2123a42459dc7512a1a86861d : SUCCESS in 2h 24m 16s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/8b759f35a131488a906a58c618441e49 : SUCCESS in 16m 03s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/51cf8fafee0d48d09662596e1e68795a : SUCCESS in 32m 40s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/36892748988f4529b71d1bf3fb2f891b : SUCCESS in 9m 39s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"71c6c1d700ab9dc5092d7604aa9102b4b79a5e9d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 18:31:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6db7918505d9636d880eae67e9f96a003d39c96c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 18:33:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88289f8bf02b43419a92d698dde77555 : SUCCESS in 1m 37s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/18a2463e19d04b4fb7705b9949f970df : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e660db0088d2553d43e8859573c88c94da061ebb","revisions":{"e660db0088d2553d43e8859573c88c94da061ebb":{"kind":"REWORK","_number":1,"created":"2021-01-18 16:07:17.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/44/771244/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/44/771244/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/771244/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/771244/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/771244/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/44/771244/1"}}},"commit":{"parents":[{"commit":"f26cd52325c9a4ae259f4c1cdacb9d24517aa413","subject":"Merge \"Fix implementation of kayobe seed hypervisor host package update\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f26cd52325c9a4ae259f4c1cdacb9d24517aa413"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-15 14:55:13.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-01-18 16:07:04.000000000","tz":60},"subject":"Fix overcloud host image resolv workaround on CentOS 8.3","message":"Fix overcloud host image resolv workaround on CentOS 8.3\n\nIf a seed\u0027s bifrost_deploy container is deployed using a CentOS 8.2 base\nimage, prior to the CentOS 8.3 release, then a user runs \u0027kayobe seed\nservice deploy\u0027, the following task may fail:\n\n    TASK [Ensure the overcloud host image has bogus name server entries removed]\n\nThis seems to be an issue with libgcrypt not being upgraded with\nqemu-kvm, and becoming incompatible. The workaround is to ensure\nlibgcrypt is upgraded also.\n\nStory: 2008430\nTask: 41379\n\nChange-Id: I82313fcd258e58969e0b68ae91a78943df537a7f\n(cherry picked from commit bd5c5c018783028a4748de28d1287f29ef714c8e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e660db0088d2553d43e8859573c88c94da061ebb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e660db0088d2553d43e8859573c88c94da061ebb"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
