)]}'
{"id":"openstack%2Fkayobe~771245","triplet_id":"openstack%2Fkayobe~stable%2Fussuri~I82313fcd258e58969e0b68ae91a78943df537a7f","project":"openstack/kayobe","branch":"stable/ussuri","hashtags":[],"change_id":"I82313fcd258e58969e0b68ae91a78943df537a7f","subject":"Fix overcloud host image resolv workaround on CentOS 8.3","status":"MERGED","created":"2021-01-18 16:07:50.000000000","updated":"2021-01-21 01:23:13.000000000","submitted":"2021-01-21 01:21:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"771245","meta_rev_id":"964b77d478c32d9d27e99a5fa1036f1f7a21f1e1","_number":771245,"virtual_id_number":771245,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-21 01:21:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-01-19 09:33:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-20 16:00:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-20 16:00:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-18 16:07:50.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-01-18 18:21:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"04bb8e88a30f4c0221db0a3a5933b6171a595d51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-18 16:07:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"87ad1aae05d2df433396297573354c1e7af7c164","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 18:21:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db22eb18b31e4342b7d05d2592304b92 : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9eee6e09a9b43dcb02c8e386830b60f : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/80c11f43b2044e4ab968b2ba7c6bd35b : SUCCESS in 5m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2a8c4c7af2ad4f188627d46d6ac5b9c5 : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e1c735956c845a5802b024584cfe966 : SUCCESS in 7m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb547402471d4ff7b52758438cbab5b4 : SUCCESS in 4m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/db5c588f3dd247b4a19fc6200fbccf5d : SUCCESS in 4m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6d287b8eb03d420384bfe7b0dd13c797 : SUCCESS in 7m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/35aca327448e455f9f5ac43efa685ce9 : SUCCESS in 18m 05s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/d747f6275fec461290671a00145fdd49 : SUCCESS in 1h 13m 36s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/f900b31d3c9c418aa7ef18195dd83005 : SUCCESS in 14m 44s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/c4f80e08fa284fe384711feaae9a96f5 : SUCCESS in 1h 49m 10s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/f37338a89de54dfb80e2857a7bb26ee0 : SUCCESS in 16m 31s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/83122b99100c49829dd52d2cb08a0bab : SUCCESS in 25m 43s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/6d4309bc68c9416bba13a2f8a201c9fe : SUCCESS in 11m 55s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"6b36bf90574f9e620bf332d7e54d437416b19939","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-19 09:33:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f5da98c2f094fcd568bb321923f100d119488a69","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-20 16:00:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3b2786db9cc84c0f0f07babb44b17e889615d434","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 16:00:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"32f4c811a90cc3128b4213916ea9d8988b3a7d4f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 18:31:35.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d31eb482493445e81edf12be690636c : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bdb65997e56840d19498ef42b13f208d : SUCCESS in 5m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/62956418879c42afb575eb201389d6b0 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a72d5b0648b40279e9d1a0cc46e620c : SUCCESS in 6m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/580aa56bb3bf4303954f70004f1c99b1 : SUCCESS in 4m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/855baa6fb27b415f93379b58019c1b15 : SUCCESS in 4m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e2a28f4cca704e12a3dff051adff61c2 : SUCCESS in 9m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/898cfd39c9874d12b0ace0954a5956ab : SUCCESS in 18m 45s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/6f76a6c84c444055b0d303b5735615a2 : FAILURE in 1h 51m 44s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/27ae42aac9be492ea08b581c0466bb92 : SUCCESS in 15m 48s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/2cebe027263e47d1bc0af3b540d313a3 : SUCCESS in 1h 51m 45s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/aded5ac5d17c426287a0b43c0378ec00 : SUCCESS in 16m 19s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/495df6083cb347229b2b0b1cfe90a1ee : SUCCESS in 36m 29s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/f9942a897c0c426d847fbb4173e717a8 : SUCCESS in 9m 03s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"a4a276052db0f0d715fa8daeff647b262afe2497","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-20 21:16:18.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2a3881c6ae44d91c3cef8c044077a5210811b47c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 23:22:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d096f4f0fad4785a271e795e33c4804 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/967b41ea72844c86ae32052f07472371 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e7909240ed844cb879591733c968bc7 : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/30b16df90f3a442eaa344dbf68c376c0 : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c049d374f54b46fc91d5da6a11f7ef0c : SUCCESS in 6m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7b89f2b8adf449aa75a95c02af29bd1 : SUCCESS in 3m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9eabc452209147059ec147a375fbd31d : SUCCESS in 4m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/de9887df1cd14bc9b14ce9a71131c14a : SUCCESS in 9m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7f1f384d7c36434697c3f6587a3d9048 : SUCCESS in 17m 11s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/73bae3cb9ca7477a9763f107ed636935 : SUCCESS in 1h 03m 15s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/3a92b82f2d8644d8bf4307921df9384c : SUCCESS in 15m 39s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/bf6bf12cfa09455688ee5483e2b216a1 : SUCCESS in 1h 49m 34s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/6bf1fa24593a47848f236ef359fe9981 : SUCCESS in 15m 57s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e4e76812f6d144f5bd02c68f23eaad68 : SUCCESS in 23m 00s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/0e24da1840e8482da7ce411883a70804 : SUCCESS in 12m 08s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"ea0e8b30c88a9864160fe8faeb5d35fa65f3bc5b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 23:22:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d7f9bf54efa7db34c9154cfebf75bb19a7003691","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 01:21:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3c899b58b57469aab7d0f95d146a242 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f29aca70ff724151848e2201522818da : SUCCESS in 4m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7a25c659679d43c8b038b055ad89b3ce : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/668a3338cd0a44a6a8007db447a3cfd6 : SUCCESS in 7m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/176e8efd0d9e4cadb79857ada3f309b1 : SUCCESS in 3m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/90f3ea6c186e439f8f3ac22731e9a436 : SUCCESS in 4m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/77ae5e52f6914c95bbef344818d29fb5 : SUCCESS in 9m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/89f2a92e5ad04928873efbe9cc071d01 : SUCCESS in 16m 23s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/e5990841264748ac9fff032e596f200e : SUCCESS in 56m 41s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/fcbc77c6142f49a3965a97a42918bdce : SUCCESS in 15m 59s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/66d8cc1edb964a9f8676c6f474eead47 : SUCCESS in 1h 52m 12s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/5bca5994ff064a8d8cbc3be3dd26e78e : SUCCESS in 14m 38s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/af57fa4a0eaf4e5180bd2e18dc4accb5 : SUCCESS in 27m 16s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/bce52e1ef2074ffeb5eaecf30404a90d : SUCCESS in 10m 00s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"fdffac70e0c77f6dd1d37251725cd6df2d84d4fe","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 01:21:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"964b77d478c32d9d27e99a5fa1036f1f7a21f1e1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 01:23:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/046f5c2c8be04bcda785a601c354d71a : SUCCESS in 1m 07s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e7815da2ffa946dc9abdcdd2c20daa00 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"304a869fe1658e461af777974c97df6e5966610f","revisions":{"304a869fe1658e461af777974c97df6e5966610f":{"kind":"REWORK","_number":1,"created":"2021-01-18 16:07:50.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/45/771245/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/45/771245/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/771245/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/771245/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/771245/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/45/771245/1"}}},"commit":{"parents":[{"commit":"6ee7333e7e6ebe6862e84df7fb69ddde431b0d4a","subject":"Fix Python setup when venv is not used","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6ee7333e7e6ebe6862e84df7fb69ddde431b0d4a"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-15 14:55:13.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-01-18 16:07:42.000000000","tz":60},"subject":"Fix overcloud host image resolv workaround on CentOS 8.3","message":"Fix overcloud host image resolv workaround on CentOS 8.3\n\nIf a seed\u0027s bifrost_deploy container is deployed using a CentOS 8.2 base\nimage, prior to the CentOS 8.3 release, then a user runs \u0027kayobe seed\nservice deploy\u0027, the following task may fail:\n\n    TASK [Ensure the overcloud host image has bogus name server entries removed]\n\nThis seems to be an issue with libgcrypt not being upgraded with\nqemu-kvm, and becoming incompatible. The workaround is to ensure\nlibgcrypt is upgraded also.\n\nStory: 2008430\nTask: 41379\n\nChange-Id: I82313fcd258e58969e0b68ae91a78943df537a7f\n(cherry picked from commit bd5c5c018783028a4748de28d1287f29ef714c8e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/304a869fe1658e461af777974c97df6e5966610f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/304a869fe1658e461af777974c97df6e5966610f"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
