)]}'
{"id":"openstack%2Fkayobe~771620","triplet_id":"openstack%2Fkayobe~master~I2482c86efebd66a5d22a761df82763cecd764386","project":"openstack/kayobe","branch":"master","topic":"fix-macos-tests","hashtags":[],"change_id":"I2482c86efebd66a5d22a761df82763cecd764386","subject":"Fix unit tests on macOS","status":"MERGED","created":"2021-01-20 11:58:17.000000000","updated":"2021-01-21 11:04:15.000000000","submitted":"2021-01-21 11:02:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"771620-fix-macos-tests","meta_rev_id":"8de3703e92e70a3368821a06f224d17f1ef44648","_number":771620,"virtual_id_number":771620,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-21 11:02:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-01-20 15:28:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2021-01-21 09:15:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-20 13:47:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-20 15:28:07.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"79d5ef543ac4bc752d8f91b7d43082eb8eb193cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-20 11:58:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dbf2c50b3f47a2eb2cf40b6d17d0eeed3ee37d85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 13:47:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1dff81fe77bb4bc9a210b795005e09ee : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49e8ad5641ed4b2fac0745d6f1d4004c : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d2c0a8dc9974699a120f22cfc3750f1 : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b9af5459aece443f8ca59e0fac00bead : SUCCESS in 5m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/546c0cc3482c4635980eef9c04c699ab : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/888b8883d9504222bafa1c3c9c77295e : SUCCESS in 7m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7557fedb3501443c835db3a1a59cbdd2 : SUCCESS in 4m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9daad0ddb5614ff28610343862599119 : SUCCESS in 8m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6f38c11529eb452a9212cf149bc7f46d : SUCCESS in 18m 24s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/729c9efc91bb4383b44fa479b5f57f4b : SUCCESS in 1h 31m 16s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/7c6ae586956d493eb88b36dec45b22f0 : SUCCESS in 58m 55s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/62a2d153e0b747bbb3ef7a4bf3eb81b6 : SUCCESS in 16m 53s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/2acf2784a1e34c90b34e9ac9823b5e41 : FAILURE in 1h 47m 20s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/633228a469a34ca5bb73a01e597b7482 : SUCCESS in 17m 47s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/704f06bd61824a46ae71ae60bc257497 : SUCCESS in 19m 02s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/60e959d31cc0497dad0316b34987c62a : SUCCESS in 10m 53s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"dffee8690cb620b2560e0e1cb140a33f45b5dbed","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-20 14:06:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c4e5fae2e187ea7d4d3ae303c1be802bb0e488b0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-20 15:28:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1cecf0f29124e2ef7d637c2750c7d7a31437c723","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 15:50:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f839b4ea7dbd4ed2af182152b1f4c1d7 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afb43a1857dd4326a79d3dd559b8a6ce : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f9f19135f40e47e6a1a13b7d0ce89b40 : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/467b8c25bf204684be4132cb081e7a4f : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bbe4f3f3602749ee93f712ec6032872f : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/508ffd4a107a48a79549e2b41b526fdf : SUCCESS in 5m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ee2a152fbb9044e59fa15350e33d35e6 : SUCCESS in 5m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e144eba9b3ee46d1a28d070069c53d4f : SUCCESS in 8m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d30086253ca14597a6d8a1c71b4a347d : SUCCESS in 17m 57s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/c6928102699b4a2ba3bd5c77573ad03d : SUCCESS in 57m 57s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/9a5920c5bd17440b9a96c0c938a07450 : SUCCESS in 1h 30m 00s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/0b6b8523b4fc4d79b6e587d4658e4363 : SUCCESS in 17m 05s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e15e784676464623881b502692a9c4c9 : SUCCESS in 1h 37m 23s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/11016b67445a4631ba4ef35e06bf07e6 : SUCCESS in 15m 03s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/fd3fa2ae15a14370bfdfce69730e6074 : SUCCESS in 28m 24s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/01e11414ab294733b504301a0bb07695 : SUCCESS in 9m 52s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"d4ab950fcace92295f9c448b0c2c3aba12fb78b2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-21 09:15:31.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e7c8a8ec640e8eb5c5d8e1bcbbeffc890c679e0c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 09:15:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"155a1860fdf144d3635dd78544b0eae231ef00ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 11:02:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1cf8473c7c4473bab8d85a6ba082120 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de169e584cb54d2e9aff8bb486bafe95 : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/daf22d6602104314bac8416e246b37cf : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6810aed45ec649b798cda9fbdce7f1d4 : SUCCESS in 6m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/04bd1f945e914f2d8d4e45933820443f : SUCCESS in 5m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e28d1e8f99734b219b4b99eaf31390be : SUCCESS in 8m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e6e2b9bb6d6a42d69cb914b27a40b517 : SUCCESS in 17m 32s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/ea9184268bf9476e95c7cdf2c0d5d54a : SUCCESS in 1h 00m 44s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/7aee63654a084860aefb4e679ef9c766 : SUCCESS in 58m 25s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/6cb04a08ebdd4422aec5726df9b02492 : SUCCESS in 15m 25s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/0df57fcd756a4deea85414b8b3076efa : SUCCESS in 1h 42m 07s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/688ba094502a4d46863d6f0f5dcb3f7d : SUCCESS in 14m 15s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/262ce087edfd43139c146e2f0a586cea : SUCCESS in 23m 43s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/2608dbf32d5341a297ccb46fa7c68a20 : SUCCESS in 11m 13s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"5dfa281d17759bc8b4bc9a2ed4ae7c1748a038ad","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 11:02:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8de3703e92e70a3368821a06f224d17f1ef44648","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 11:04:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6d5971cdf854f078d96172e73c25fab : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e0120b43ed9b4e243fd9efaba3a9b1c33f94a958","revisions":{"e0120b43ed9b4e243fd9efaba3a9b1c33f94a958":{"kind":"REWORK","_number":1,"created":"2021-01-20 11:58:17.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/20/771620/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/20/771620/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/771620/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/771620/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/771620/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/20/771620/1"}}},"commit":{"parents":[{"commit":"e96bd3309ab0f121a9b496f31f626314b5786115","subject":"Merge \"Fix overcloud host image resolv workaround on CentOS 8.3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e96bd3309ab0f121a9b496f31f626314b5786115"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-01-20 11:56:24.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-01-20 11:56:24.000000000","tz":60},"subject":"Fix unit tests on macOS","message":"Fix unit tests on macOS\n\nThis unit test was failing because /tmp is a symlink to /private/tmp on\nmacOS. _detect_install_prefix() returns a different path because it\ncalls os.path.realpath().\n\nChange-Id: I2482c86efebd66a5d22a761df82763cecd764386\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e0120b43ed9b4e243fd9efaba3a9b1c33f94a958"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e0120b43ed9b4e243fd9efaba3a9b1c33f94a958"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
