)]}'
{"id":"openstack%2Fkayobe~771840","triplet_id":"openstack%2Fkayobe~master~Ia885a226de30d72610dca2eb2f194cef10193095","project":"openstack/kayobe","branch":"master","topic":"story/2008546","hashtags":[],"change_id":"Ia885a226de30d72610dca2eb2f194cef10193095","subject":"Quote kolla limit when calling kolla-ansible","status":"MERGED","created":"2021-01-21 15:54:58.000000000","updated":"2021-03-01 19:00:48.000000000","submitted":"2021-03-01 18:58:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"771840-story/2008546","meta_rev_id":"84dee255005cc551aa49cf41b966909a7342a33a","_number":771840,"virtual_id_number":771840,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-01 18:58:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-02-22 14:19:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-01-25 10:51:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-02-22 14:19:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2021-01-25 10:51:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-21 18:05:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-25 10:51:09.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-02-22 14:19:40.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"6118e8d6ff517049d6b70d80980d617ef9eaf4d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-21 15:54:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8701ffce7bff2a61a60db54dc0ebdcdf6996e6b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-21 16:12:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"046569468722ff615a75b64bf1bccb96e317eacd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 18:05:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac5b0f5dd9cc4d949b1d5978c01e3f44 : SUCCESS in 6m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eaf6d8a203d441da8ec8c36c8d92d944 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0ec4def4398420b965283040f5a2194 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb89f00946984e459bdd66862db952a2 : SUCCESS in 5m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1f9a77826f824ac38a61968dd4c189fe : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9dcd049ecdd488589248aa8ae8ecb7b : SUCCESS in 7m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/611e8f51a2574d6d9178ac116def84e6 : SUCCESS in 4m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7c78deaa71994181a13fe14615284f43 : SUCCESS in 5m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/14547e6d13324a9e8dec751a23c6d8b1 : SUCCESS in 8m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ccb94748090a4af08547a17deda3b57d : SUCCESS in 19m 07s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/c25bf4abf6ff409287ca241c09b25731 : SUCCESS in 53m 12s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/763b9e71a34948a3b17d401b2785a48a : SUCCESS in 1h 00m 12s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/a041c8a817b34337980e13ff95b297dd : SUCCESS in 15m 27s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/6b014684ed1e441ab8f061db0ff3179b : SUCCESS in 1h 42m 37s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/5a6a53c3e8a74b83894c0630ccfdab8b : SUCCESS in 15m 31s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/7860829ff46740428bce84bf87281a92 : SUCCESS in 27m 51s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/f577df4954754fb9bea4eeab894a5f94 : SUCCESS in 13m 12s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":2},{"id":"f7595bd185a56154741f7685b49fbcd798f50805","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-25 10:51:09.000000000","message":"Patch Set 2: Backport-Candidate+2 Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3dc1939fc4d5e92e82d4e070337b03153f888a8c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-02-22 14:19:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a62aeaa58f6e1d4078cb302dacf0340574d3a6e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 14:24:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e96b3cb6695b8b1384363c0576a848f1027589ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 18:06:01.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf63f5d9cf264c609e3206f1fce5e364 : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7388148533154dcfbb698e9f123032ac : SUCCESS in 7m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b081fbbea7164d5e85fa811cf26807cb : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1961210873f40798387d74511af92b3 : FAILURE in 13m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2de5d6454bc64175ba7326d44d240ffc : SUCCESS in 3m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c4f437d8b8ae4daea2084a302a712a79 : SUCCESS in 5m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d9c0d0769e9d4d5ebc46380552bf8cce : SUCCESS in 9m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4a7d1a72e9c3403fbd8f578018063c29 : SUCCESS in 18m 37s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/65d06d240214482dbc74b10c3e6888f6 : FAILURE in 1h 24m 45s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/118d200a03b5450187dde7fc8933fb76 : FAILURE in 1h 48m 21s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/0a6e392526ed4a08a6657172fb9060c9 : FAILURE in 6m 59s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/030290695b04484587b1aac2de3fc9ac : SUCCESS in 1h 52m 29s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/ee3758a15db040da9f76ee4e1f9800c5 : SUCCESS in 27m 03s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/2fcf49d3999e48748402ce214215fda8 : SUCCESS in 24m 40s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/9bebafd7a7044a52ac22bbab80caebf0 : SUCCESS in 12m 10s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"73b40935a230f365300ba7dc6d2bcb8b2443f1d0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-22 19:33:44.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"040864bf72f1e52f69c87d988d355203bd08d421","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 21:41:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd1a17607be04cbb8207ad8384560b79 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89203e64f5b84393a95b1d607bcf5a38 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e066a66df6ae4f47b320f3580f5210b0 : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90b6095a7a8541108a10da9f08edd08c : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a4e7b780390c4878be08f55afa2d94ce : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b2ca5ca730e4fd2aca9e22a5bbfa0fd : SUCCESS in 7m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e1d196c217a4b72a6c42bf349bd732f : SUCCESS in 4m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5a95719618f0449cabde7096d034e0bd : SUCCESS in 5m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/853cc0c9c80a45e9b93f42d9c7e41689 : SUCCESS in 8m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dd465c1ab0f4485796f17264208c8577 : SUCCESS in 19m 02s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/07cad60e782c4cadb985b26ee5a610d5 : SUCCESS in 56m 52s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/927a0145368d4d08af2ffd41a8965245 : TIMED_OUT in 2h 01m 41s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/2c3b28fc46e945ca9f4d8242cb3d044b : SUCCESS in 16m 57s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/ba145a3ba72440e3a6137343fd909a64 : SUCCESS in 1h 36m 46s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/cdb4b4871a93459980f2ab6b0fa758b5 : SUCCESS in 28m 45s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/d3bef3953bd243b1a40fd57626c5a557 : SUCCESS in 22m 13s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/83db18bb10954e99bbc60541572c350b : SUCCESS in 9m 53s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"820b67811d69b1f7d0b7f4ad39b4c1c1edc8d16c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-22 22:02:14.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"794713f9863469d6ab576f80a12834dd0ea1e88e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 00:11:50.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b1eb133449aa418297ff04595f3622b9 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfa4e1d7978a4596bbcc8e912847a343 : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ae94e1e3f2c4a5bb3b373766b2b4680 : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac46d91439fb41b18d2b52fb6272907e : SUCCESS in 6m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df5f52c4325440ccbfe8c8dcefea1893 : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd90af31eac7490b9ba696a4b5b5bfb6 : SUCCESS in 7m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d37f0d2cdda44dfe90e3b1b6fefb84d6 : SUCCESS in 4m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/102e9d9b6bc04c1cbd0afc950fc49ab0 : SUCCESS in 5m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a1953cefa0e74b6f97e2058d6fc5e62c : SUCCESS in 10m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/93a45c52c28948e1826c214de2088b1c : SUCCESS in 21m 05s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/fff337c4bd8a4fd593a76c7521b48c11 : SUCCESS in 1h 00m 16s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/0dd92adca5064ec1808ed2f41acedc25 : TIMED_OUT in 2h 01m 56s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/4e8feb0405e64845839135f684c51471 : SUCCESS in 13m 56s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a2bdf447b31b484fbded9d6f486a0ded : SUCCESS in 1h 33m 06s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/25cb5868ba4c432aad1bb1ce36ccaeea : SUCCESS in 26m 38s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/af9723089fab4f3f887bd84b0c06250c : SUCCESS in 20m 45s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/ef49d702fe1c42769407fa00cfd863bc : SUCCESS in 11m 15s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"b5e970d7ccbf921c98d468ddb718034267641e25","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-23 08:43:08.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7ff0bfb1ccb779b77c469035804c059d7c4a03da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 10:53:44.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a7e249e332e64da4918ce7cf544c7247 : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82a0da73631d4f04b8345ff4d49e7d43 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5ecafaf86b14e5da0c52079e868a3f9 : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fcecd149c3c04074a411be74a22da551 : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d7aa8e0637c7490eabdb3f6a800c520d : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/652cee707b254b34add36becc788534b : SUCCESS in 7m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c2d73741c1184099b5f305a2f85aeb94 : SUCCESS in 4m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b5265392b519484895dde9b780e22921 : SUCCESS in 4m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fcdc2fde9ee34dcf9c4ad5185ad046a1 : SUCCESS in 13m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/173707533e4e4d82b8de6c68269d81d3 : SUCCESS in 18m 36s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/460db52bb6524ae984080b5057947b96 : SUCCESS in 1h 14m 36s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/0ce06b438f474b9783be8366c927bc1d : TIMED_OUT in 2h 01m 49s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/5ad2b78d45df400f976f8f4fc9afe12a : SUCCESS in 15m 37s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/983522e7ef1e435285205e086dd360ff : SUCCESS in 1h 34m 09s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/e157679bca2644d9a335138176a80cda : SUCCESS in 27m 09s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/af813614f67b4685a7bd2b601132d6d6 : SUCCESS in 30m 42s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/a2acbd5bd3154f0ab41ba3a6b5627801 : SUCCESS in 13m 29s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":2},{"id":"87b05fb9fa72c5f19fcf637251267988e1348278","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-23 15:52:46.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d90d40292b381d00cd14feb18b9050e32e74fb36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 18:13:12.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a28caa51267644f5942419172876a7b8 : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3845d861b7cc49f297e04fedd3619d3a : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/60bdea5297ec4058bea48e1340e864dc : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/515154a4780d4bf29eb0ca8d00c681cd : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7d3b14f6b79b4879beb03b9c1747eb79 : SUCCESS in 7m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f840a581ad484a0eae248e6c81495b1a : SUCCESS in 6m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7590593b94774dbe836021590f43887b : SUCCESS in 4m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3042447966b046259f99a615e5835b8e : SUCCESS in 6m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6fe5bb3ad80245dfab97aa6c76eb64b9 : SUCCESS in 8m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6326cb4e240647a9872d5d16c23676d5 : SUCCESS in 16m 00s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/3590a6b6178f4f5ab2d3b26f82470476 : FAILURE in 1h 50m 56s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/4d4805964d8e41abb1de80b865f9d67e : TIMED_OUT in 2h 02m 29s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/a96223df648742069ad65ebbe92a086d : SUCCESS in 15m 52s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/7d1183d84dcb4dc2b27ff7a2586b4114 : SUCCESS in 1h 29m 17s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/96fc0f0555184a81b53ad161fadcd22c : SUCCESS in 26m 10s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/55c683ccb4fb4dd69d9335ea05be5c21 : SUCCESS in 32m 47s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/fd661b9c5cfa42978f749bff0590e2a7 : SUCCESS in 10m 09s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":2},{"id":"4b6204e47fc740f24d7990de179cc37b4c326a17","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-01 15:10:02.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f6a362ed92b8d523a50a1298535788e4af7b6368","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 17:09:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0dac0fc09d794dbba88bbd2be333b87d : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e4baf5204274fe99dd8c11e859482de : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c8df417d8c94564907599c68370ed42 : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/17114cb834e446a88da16e1e762c200c : SUCCESS in 6m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8c18ce0261dd45ae9c787728cacc4629 : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b774dd1496004d929b4b502698840c42 : SUCCESS in 6m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/458eda6fb1a04fc9947e26d3982961b8 : SUCCESS in 5m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/76847aea0997464ab21827909136d9bb : SUCCESS in 5m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e3a5995c2628476d92c6b3afb60efe95 : SUCCESS in 7m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/65d14ad60dfd43b4b1d319eb8b2dffba : SUCCESS in 20m 51s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/8ba12e7067cb4119a73bd6ff97f25e40 : SUCCESS in 59m 59s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/c4cd6a34d5414c2cbef4379705eac3c0 : SUCCESS in 1h 02m 29s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/47854fb571d84f078eca96d0f503085c : SUCCESS in 15m 32s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/cfb0988a9bb842b5960bbe34ab52c46b : SUCCESS in 1h 47m 01s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/37530da47565461b8eb6ee52c2fa3049 : SUCCESS in 28m 46s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/1f865155c64041e182757298e792c040 : SUCCESS in 24m 39s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/5d9c36fd604a45598a9bc5e46f322d29 : SUCCESS in 10m 10s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":2},{"id":"6e35c852533c02280e596317c5f98a4d0a39244b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 17:09:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"946c28184735e8a4ce8afa1b865f4f986349a755","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 18:58:36.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19d35abbe3d44f8c959bd48c3d8876c3 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dee555f0a88c4b1fbdb9d18508e9d273 : SUCCESS in 6m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c3fe0ae613294289a821201a2e16eb91 : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eebb176ab6304b7abd072253dfbbf917 : SUCCESS in 6m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e9dd7439440142e7a0fb57db83fa4fed : SUCCESS in 4m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/799a5bcc1f67425db651e845c9dab17a : SUCCESS in 5m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2ffa45753cb14615a0c09a2ff2b422a9 : SUCCESS in 11m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/aaa85a86dd8c4c5da0c5ca1742e15617 : SUCCESS in 19m 19s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/360925c0fe8a4477b75199c594d97fff : SUCCESS in 57m 41s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/05603cd1e33744f8918d538afe32361d : SUCCESS in 1h 08m 26s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/faf06d42aa25421783ea5706ebfb2bee : SUCCESS in 15m 28s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/76b25bb67147417ca0756925eae65dde : SUCCESS in 1h 30m 22s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/38ac406800e04e28b445ff6660718142 : SUCCESS in 25m 38s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/27c73718e9064ef98422a43c28e5b99f : SUCCESS in 23m 31s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/e3d8d36c68c0446b878ee5e15064b2f8 : SUCCESS in 9m 37s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":2},{"id":"760ae5bbcf6a07ee2b78bf197bbebcc41acac9c4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 18:58:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"84dee255005cc551aa49cf41b966909a7342a33a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 19:00:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/970d8d922be84634952111dd22910527 : SUCCESS in 1m 25s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca54f25d9e5042ddbd0759efeb3c8ebf : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a1c9a8152508785d5babe4ed98eb6a743932a229","revisions":{"e055d445f63bbb3cd6e0bec3ea25c1b20bc2b182":{"kind":"REWORK","_number":1,"created":"2021-01-21 15:54:58.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/40/771840/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/40/771840/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/771840/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/771840/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/771840/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/40/771840/1"}}},"commit":{"parents":[{"commit":"4d965689123163b9f5bc340437fa2c318e30283a","subject":"Merge \"Fix /tmp/swift-rings/backups/ files deletion\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4d965689123163b9f5bc340437fa2c318e30283a"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-01-21 15:54:18.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-01-21 15:54:18.000000000","tz":60},"subject":"Quote kolla limit when calling kolla-ansible","message":"Quote kolla limit when calling kolla-ansible\n\nThis is necessary for correctly passing limits containing special\ncharacters such as ~ or \u0026.\n\nChange-Id: Ia885a226de30d72610dca2eb2f194cef10193095\nStory: 2008546\nTask: 41651\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e055d445f63bbb3cd6e0bec3ea25c1b20bc2b182"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e055d445f63bbb3cd6e0bec3ea25c1b20bc2b182"}]},"branch":"refs/heads/master"},"a1c9a8152508785d5babe4ed98eb6a743932a229":{"kind":"REWORK","_number":2,"created":"2021-01-21 16:12:15.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/40/771840/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/40/771840/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/771840/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/771840/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/771840/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/40/771840/2"}}},"commit":{"parents":[{"commit":"4d965689123163b9f5bc340437fa2c318e30283a","subject":"Merge \"Fix /tmp/swift-rings/backups/ files deletion\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4d965689123163b9f5bc340437fa2c318e30283a"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-01-21 16:12:07.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-01-21 16:12:07.000000000","tz":60},"subject":"Quote kolla limit when calling kolla-ansible","message":"Quote kolla limit when calling kolla-ansible\n\nThis is necessary for correctly passing limits containing special\ncharacters such as ~ or \u0026.\n\nChange-Id: Ia885a226de30d72610dca2eb2f194cef10193095\nStory: 2008546\nTask: 41651\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a1c9a8152508785d5babe4ed98eb6a743932a229"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a1c9a8152508785d5babe4ed98eb6a743932a229"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
