)]}'
{"id":"openstack%2Fkayobe~771928","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~I7fa88a203274c3030b3d8e5208f94b2230d43c68","project":"openstack/kayobe","branch":"stable/train","hashtags":[],"change_id":"I7fa88a203274c3030b3d8e5208f94b2230d43c68","subject":"Fix /tmp/swift-rings/backups/ files deletion","status":"MERGED","created":"2021-01-22 17:19:34.000000000","updated":"2021-02-08 23:52:24.000000000","submitted":"2021-02-08 23:50:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"771928","cherry_pick_of_change":764600,"cherry_pick_of_patch_set":3,"meta_rev_id":"4d68a1e954f82fb38ce3588158a1f750047e71b1","_number":771928,"virtual_id_number":771928,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-08 23:50:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-02-08 23:52:24.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-29 20:52:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2021-02-08 23:52:24.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-29 20:52:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-22 17:19:34.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"state":"REVIEWER"},{"updated":"2021-01-22 20:28:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-29 20:52:27.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"0499fb1783a4501c2bf4521876abe2cac248170d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-22 17:19:34.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9cda3a20c35b2084f9ae8b409d837ac07b970710","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 20:28:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a1e79d3100a043f0bf3371b4440f1283 : SUCCESS in 4m 05s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/76d2ee81069d4fcd8325fbb134795fc9 : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a2edec52fdb64a509fd370a9221e7293 : SUCCESS in 9m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4fd5108323a4ac187c9c9c243e93961 : SUCCESS in 4m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b74e9c0433e04c5b95a3d074d131d7c7 : SUCCESS in 10m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b926f72c2a814e6793b53c600f2805ab : SUCCESS in 7m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b009120d2882452ca789e8c96af2c6eb : SUCCESS in 5m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a2a4df212dc34544ae56c8063c41a53c : SUCCESS in 4m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/626e4671a7364c3580b62b448cff48f6 : SUCCESS in 7m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2fb002efc0464b859ea30a2b6cd80796 : SUCCESS in 19m 56s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/b730de2010ba4bcea91743720d5bf590 : SUCCESS in 1h 04m 39s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/6c44d4151c7742e3a8d0ccc1585b0f00 : SUCCESS in 1h 11m 13s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/130f3c8aa589433d9c2d4cdba3a27521 : RETRY_LIMIT in 16m 57s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/4eccfc640cba497ca996ed4c44cc4bad : RETRY_LIMIT in 11m 46s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/c4f14230424d47a0a493a7686fe3f316 : SUCCESS in 1h 36m 12s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/0bb90eb2606b44e195b360013a4b5dee : SUCCESS in 23m 30s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/4ebfe62389314d3eba0725d70860962d : SUCCESS in 17m 18s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/ea35584e0bce4a3e94400642e7d3b211 : FAILURE in 53m 03s","accounts_in_message":[],"_revision_number":1},{"id":"e057b3f90c9e2679f6a9d4d80f6a3566df6c6842","author":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"date":"2021-01-23 10:39:17.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1dd77c56249c2e4716de2f4ec9c89b9e2f234138","author":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"date":"2021-01-25 04:34:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fbe338062ddfc9c2e84dc36ee7f3123bd6cd7d93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 06:12:18.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba9c0cbd122740f0a6651b9f96c213a3 : SUCCESS in 4m 29s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9fa61cedb756446fb31bb8b01b49f594 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5457b2aa748347bc82906e73130b4bdb : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a504a84542a645bbadfc627f9e4c884f : SUCCESS in 5m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/09c8bee67e6645129809fa20cfa017c3 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/081e0f8e4f464c37975328f34efb8986 : SUCCESS in 5m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/62d10e75542c4994bf86de31832deb03 : SUCCESS in 3m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/61d3c3c2ab5443f38298f2e11bdeaf6c : SUCCESS in 3m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4fdccf4bc41748a298dba9404761126d : SUCCESS in 8m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/03890f560b9a421e937d5031a1c8456b : SUCCESS in 18m 30s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/c6bfe98c84f340c6ba0b0561decfef9c : SUCCESS in 58m 21s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/4247300f381448d2ad098f5b148e42bd : SUCCESS in 1h 02m 02s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/daa078c9d6ca484a98b862f61ba9be0f : SUCCESS in 14m 33s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/018d03d814074cba9133f44fcf2c1c51 : SUCCESS in 20m 52s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/9cb2a63a10264ac5b369909342b6c3da : SUCCESS in 1h 34m 39s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/cc7123ab65834e10a363f90ee774897d : SUCCESS in 18m 46s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/dc097b98c55e422caf0f0223e9143cee : SUCCESS in 17m 18s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/ba70ef2784054ddfad7d898a3ae6bf51 : RETRY_LIMIT in 18m 00s","accounts_in_message":[],"_revision_number":1},{"id":"fd91727ecd0661e656bcc67f1fa53a63969a5334","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-25 10:35:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b0d7c04afcfd62ffa29643ba5b103c093518251e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-27 09:21:01.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"82f9bbb3dc305a672daac5f1054e19aa81760355","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 11:13:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/376899580c0c4889a065e68c6eb1922f : SUCCESS in 5m 11s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/509d438775b449a69a2637223e9526f1 : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f2b68f5112b4218b76e5619d733d0a2 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b3da3389e784aec85b1e6c1ea7b987a : SUCCESS in 4m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/55bea4315a09450ba48b75c04964a425 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e14ff4eab9f4d8c84b4f145a823f0ac : SUCCESS in 5m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6c29c1e8442b48cf9ee3ff8f6f099385 : SUCCESS in 3m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2472f2a1ed2a46c99e4614314b72a013 : SUCCESS in 4m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0c7b0ea29dab4052af12cdaf96580a2a : SUCCESS in 9m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/de8a737ae6cf45ed8f259da1b7bb4137 : SUCCESS in 17m 50s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/93bda17a4f5d41bd88c0508a6a589cfe : SUCCESS in 59m 41s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/9f8d8870d3224d11a47f5b286f26f147 : SUCCESS in 1h 06m 48s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/3325aade79154587a89a28c68f5d934c : SUCCESS in 14m 16s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/6c4d9707d20d4daaa40e8cbd9a8af270 : SUCCESS in 14m 40s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/a7f540b8169645888c5d1fab8fc8c7fc : SUCCESS in 1h 47m 59s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/9464f90ad2c44542b17bc500e090441a : SUCCESS in 18m 19s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/9efd3256e8514685b7f268fc7bf32ce3 : SUCCESS in 19m 26s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/2b70e4406a854ab6bb8abc5d0aa23115 : SUCCESS in 29m 42s","accounts_in_message":[],"_revision_number":2},{"id":"9db240fc02e7fcc388c8a1ab3f31e7d2d56f2fb9","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-29 20:52:27.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b0423dedb489d294cf5a8411a94111c5631077e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 20:52:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"018c8b4aca58d061b340a890eb0a5ca2243974f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-30 02:03:31.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d1471cecb3d54e36b1d3a4c99ab4d3e8 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/666b60933ea1441eb443b8a7403ff818 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59d1c74ad2d14d4b94b79f6b8f211f50 : SUCCESS in 4m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7ab7ec4af5bd43f098fa1aa1387af6fb : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/699d8f2107b34f4b86f09b3f10a43dc7 : SUCCESS in 6m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a2b4b978c74a4f1c9f0f57730da70f57 : SUCCESS in 4m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bafa1cb4e31e4dfcbd7f223a8769acdb : SUCCESS in 5m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cec37ecfe58b43cc9e95dd6922141de2 : SUCCESS in 8m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6f53667b6d7c470ba0533b0932f0b3f5 : SUCCESS in 17m 45s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/5f4c493fa98a425eb09493d5f729892e : SUCCESS in 57m 57s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/00debc19b6c04dc68f120f7634c1a453 : SUCCESS in 1h 01m 39s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/5a8d3aaf1e8e4054a29d6b9996e19b7b : SUCCESS in 12m 34s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/9bcfc7ffcd8b40c095b0df01b302fa31 : SUCCESS in 15m 33s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/e5137bab7bec468ab1e2484a253a4605 : SUCCESS in 1h 40m 44s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/1cc196d6846145acb148aa386925fc02 : SUCCESS in 22m 50s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/f8aa322df60d4594a23754921bcc4ed2 : FAILURE in 13m 31s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/9b7a7de8abdc4a589feffffdf0ec481c : SUCCESS in 32m 04s","accounts_in_message":[],"_revision_number":2},{"id":"2f6e6fb0d0a4c5ab46c9000cd4ba8ba982c817f3","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-02 06:28:31.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f17ed10244b945da385c4ea92736dff19213c7e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 07:48:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f20d55767d1a4d7fa3511382d8a6aaaa : SUCCESS in 5m 11s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d43c960dbf074947b9dfd4733d53d3c5 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b7c53e152ab4bf58be0542506134e1b : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9c6d3af2d524e7f849b603731e04d97 : SUCCESS in 4m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4b579bff80e04abba41d1893838c639b : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40c47eb112c141c9a4b65b41284932df : SUCCESS in 6m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/396461c25dd7479b8d0ffcbd310e51cd : SUCCESS in 3m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/74a4b0fd07474243ac41adb8c52b8387 : SUCCESS in 5m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/10701a7254d14e399eb467629a2bd110 : SUCCESS in 9m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8c8469f49f05419bb1234320cc0798e6 : SUCCESS in 18m 30s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/30932ec866a9477b86e50af05f66e6a2 : SUCCESS in 1h 05m 22s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/f8540fa55374441482bdcbe922aca3e8 : SUCCESS in 54m 50s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/68460eb79fd044148e0531cf59e5718c : SUCCESS in 12m 41s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/ac89a92b6be349918c04f48397c58ca7 : SUCCESS in 11m 52s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/324766b82a5f490eb2960072bb7c21ce : FAILURE in 1h 15m 45s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/e86aeb2469be4c02ba43a91d2d0430c3 : SUCCESS in 21m 07s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/bf354a35b2c94c1f808e21418a18d161 : SUCCESS in 18m 52s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/1c9baf6377e94fe1932ff6c2f26f6380 : SUCCESS in 34m 50s","accounts_in_message":[],"_revision_number":2},{"id":"5713f9e4fef6d5d5b9e3bb64882ca6ba2b779406","author":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"date":"2021-02-02 08:19:01.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"64dbb8886f68089b9800d5fa9dd2d87c99aefad0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 09:48:42.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39c9b7cf4a1447d581c3d6b4240be40e : SUCCESS in 4m 12s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8f05ab27d6ee425b8835f27eb88f0a95 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c77b9e936ec94840a03c6bb022765de2 : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8de3334a607747658e148efe4d73dca8 : SUCCESS in 3m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b881b4e7455749a5893d890dde37e312 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bfcf46e222747acac83b39e415de7e3 : SUCCESS in 6m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5fb15aa60af64f88b8bb33bddc93e0ff : SUCCESS in 3m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1f68fd637eb84f1fbe3ce1b679e5f23c : SUCCESS in 4m 00s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/00c9343ced354a488f0ce5a7c368d114 : SUCCESS in 8m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/46658908d11e40828cb8c6028ada7517 : SUCCESS in 16m 10s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/d1ba462638f54d378fa50affba275053 : SUCCESS in 1h 05m 55s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/467079745a5b479589456cc22f92b87c : SUCCESS in 53m 07s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/8b85b9cdc1cf4a23a4c0dbdd3bd46ed1 : SUCCESS in 13m 42s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/782a89b0955d4b7c856f4cc58c324bfe : SUCCESS in 12m 10s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/a472f5d915054695828a7fe1bc5c46c7 : FAILURE in 1h 19m 07s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/14124f72b97e43028ed22f20158ad366 : SUCCESS in 17m 44s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/b5e7e0ba3eb042c0b7f6cd0132fd75f7 : SUCCESS in 13m 38s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/7f9cdb04b42f4b83a4f8b216a4a1306f : SUCCESS in 24m 40s","accounts_in_message":[],"_revision_number":2},{"id":"48d390e924e5e08e27b7b6f1e707f5a6ca027df9","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-02 10:35:44.000000000","message":"Patch Set 2:\n\nrecheck\n\nfailed to register layer: too many links","accounts_in_message":[],"_revision_number":2},{"id":"28df99780c005d7b42d289a60f4b4690aa928c10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 12:38:57.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4d5bc45ff0d54088b1a9d270ecd6df91 : SUCCESS in 4m 22s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f38ddcd81d5a49ee837349b2b2c65acc : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/29e4dfa02c1a416dab63dbea0758a8cf : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1bfada1a757408fbdd7b405811112f7 : SUCCESS in 3m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c939f1858f24458cb89338ace3ebc001 : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb4e3f57aaa64539bf4a5ba207779fe9 : SUCCESS in 5m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e44176bdd3c84e9783d3e0374767cc96 : SUCCESS in 3m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1b65209f27c248ad994cbdd0813a6883 : SUCCESS in 5m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7e248f9ebfc04314adaf0dec106a007b : SUCCESS in 8m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e2f6e5f664aa479cbbb40abd959ec073 : SUCCESS in 15m 55s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/1c8a93b2604f45e0a3d73e508906812f : SUCCESS in 1h 10m 09s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/507832a7df0040e2b23219db08fb9348 : FAILURE in 1h 55m 10s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/52ed720c5d004554875b0c9b65a61cf2 : SUCCESS in 13m 09s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/2faaa46cb10645d3a28a9146c09b8197 : SUCCESS in 15m 06s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/50fa71d789da4d9e8010515b19c1e6a9 : FAILURE in 53m 29s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/67fb09fc93a542359de57e7a96d86973 : SUCCESS in 18m 41s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/fafb1b615b604e04a47028973762a7e0 : SUCCESS in 14m 51s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/f05a10df1a3a4874a95d6058fac32e36 : SUCCESS in 32m 15s","accounts_in_message":[],"_revision_number":2},{"id":"9e8d811c942a762e545d122ba94d6eea9f482423","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-08 20:41:30.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"54de6f890fe7d6a466069337170c99593458be24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 22:02:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e4c6f4d72c8a48da83cce9ed105ee416 : SUCCESS in 3m 59s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c538343680ec4e37a2245d2df06090c4 : SUCCESS in 7m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d471ce7e25f24eecac8e1a7e327c2ebb : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbf9aa0459e34fe1872734ac6a98c841 : SUCCESS in 3m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8cc78172f4104ed7b4e856f22a6362f4 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dcf1495620642fba4b727cab8ef553d : SUCCESS in 6m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6a639ad06484ad7929907eb547649d1 : SUCCESS in 2m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1b7dfce5885d4064921e1f4e30f23ce3 : SUCCESS in 4m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/da2b0f34b0cf47e49fa8005910c34b5b : SUCCESS in 8m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c16da303511c41959507516b87b566f8 : SUCCESS in 16m 33s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/44d5ede788a54d2db63a11da25d2c61d : SUCCESS in 58m 53s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/4852c3317a6e403594acce9fe8e1769b : SUCCESS in 53m 58s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/435b33a4d4644f319304e01c49fe8151 : SUCCESS in 13m 16s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/ac90012b836c4d7e9db7684ed7b0604f : SUCCESS in 14m 09s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/38dac5b551e74e2db0af431fad590266 : SUCCESS in 1h 18m 54s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/764f8a4c2c904660b35b709578cae3e9 : SUCCESS in 18m 54s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/0d20d2967f9946d7994dc4aa28cfd165 : SUCCESS in 16m 12s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/a2e4623ca5ef42128c0f5b9f2172ad62 : SUCCESS in 30m 24s","accounts_in_message":[],"_revision_number":2},{"id":"d70a0aef35d7ea2dda89dce3901e3feef27c42d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 22:02:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9293cc9495f5889f18f20f341d060df8f636a670","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 23:50:50.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ee5ace631b784aa0addb5335d77f5f78 : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/55571158303842e1a7f3e525e090d81a : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18fbc5deccb54a7ab00db9ac9d0d83ee : SUCCESS in 3m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/63abd58c5f2d41769bfaa873b6e0a6b1 : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f37647ed8e2b491e897fe41f76fa6ba8 : SUCCESS in 6m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14f550166df449599c7f96becfb6108f : SUCCESS in 3m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c820f397e7234cba82f19b71c16659c7 : SUCCESS in 3m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9c6d4e4c901a43c0834a206192b756b5 : SUCCESS in 8m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7ce2fb925bf8448e8338de4e34a149ed : SUCCESS in 16m 43s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/7d59d602059c47c99672c37c5702650d : SUCCESS in 1h 05m 32s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/fb2a694420024c50a83283374162bfb0 : SUCCESS in 1h 00m 00s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/1240ba6287324097807cb39f244af2f9 : SUCCESS in 12m 42s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/d41a155840a44057b19b630be5b0257c : SUCCESS in 14m 56s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/59c98901d37043eea137ceeed32af0da : SUCCESS in 1h 43m 04s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/86601512f92a4c0391b36464c63ffb62 : SUCCESS in 23m 19s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/81965e6ba1a944d2bfa65345f989feb8 : SUCCESS in 14m 32s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/c6d9e008903141a4ae398ab82322a3bb : SUCCESS in 35m 10s","accounts_in_message":[],"_revision_number":2},{"id":"57da6a3a8fd679206ca21b81f357740a77f637b6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 23:50:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"702ce41fe28f89855f8e652912a4d0e75a15d4eb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 23:52:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39ec1cf85d1f4178a6fe6f6fd8fdffa0 : SUCCESS in 56s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/efa7cfd3c5494d0fa9543e68ca49fcf8 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f4a8db8f9d94142979bbf6183062f30e42a3cae8","revisions":{"6b38a9032a6a54a5f45ba0ccb0d3fc7216fa3941":{"kind":"REWORK","_number":1,"created":"2021-01-22 17:19:34.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/28/771928/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/28/771928/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/771928/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/771928/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/771928/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/28/771928/1"}}},"commit":{"parents":[{"commit":"cf0146097249fc211131732f858399c8ea1e30de","subject":"Merge \"Fix implementation of kayobe seed hypervisor host package update\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cf0146097249fc211131732f858399c8ea1e30de"}]}],"author":{"name":"Margarita Shakhova","email":"shakhova.margarita@gmail.com","date":"2020-11-30 04:53:46.000000000","tz":-300},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-01-22 17:19:34.000000000","tz":0},"subject":"Fix /tmp/swift-rings/backups/ files deletion","message":"Fix /tmp/swift-rings/backups/ files deletion\n\nAdd permission escalation to the swift ring generation playbook\n\nTask: 41256\nStory: 2008354\nChange-Id: I7fa88a203274c3030b3d8e5208f94b2230d43c68\n(cherry picked from commit f811d3148ac542f32b476146ffc791aa4ea72693)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6b38a9032a6a54a5f45ba0ccb0d3fc7216fa3941"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6b38a9032a6a54a5f45ba0ccb0d3fc7216fa3941"}]},"branch":"refs/heads/stable/train"},"f4a8db8f9d94142979bbf6183062f30e42a3cae8":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-01-27 09:21:01.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/28/771928/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/28/771928/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/771928/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/771928/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/771928/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/28/771928/2"}}},"commit":{"parents":[{"commit":"1073fcfb3eb06c0b581a16aca9ea2a198ad1ddc7","subject":"Fix overcloud host image resolv workaround on CentOS 8.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1073fcfb3eb06c0b581a16aca9ea2a198ad1ddc7"}]}],"author":{"name":"Margarita Shakhova","email":"shakhova.margarita@gmail.com","date":"2020-11-30 04:53:46.000000000","tz":-300},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-01-27 09:21:01.000000000","tz":0},"subject":"Fix /tmp/swift-rings/backups/ files deletion","message":"Fix /tmp/swift-rings/backups/ files deletion\n\nAdd permission escalation to the swift ring generation playbook\n\nTask: 41256\nStory: 2008354\nChange-Id: I7fa88a203274c3030b3d8e5208f94b2230d43c68\n(cherry picked from commit f811d3148ac542f32b476146ffc791aa4ea72693)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f4a8db8f9d94142979bbf6183062f30e42a3cae8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f4a8db8f9d94142979bbf6183062f30e42a3cae8"}]},"branch":"refs/heads/stable/train","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
