)]}'
{"id":"openstack%2Fkayobe~772607","triplet_id":"openstack%2Fkayobe~master~I25078e69acdf41a4ef9957f99fe5047de54b778d","project":"openstack/kayobe","branch":"master","topic":"story/2008558","hashtags":[],"change_id":"I25078e69acdf41a4ef9957f99fe5047de54b778d","subject":"Adapt to changes from the Bifrost Victoria release","status":"MERGED","created":"2021-01-26 22:20:08.000000000","updated":"2021-02-11 11:11:23.000000000","submitted":"2021-02-11 10:17:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"772607-story/2008558","meta_rev_id":"4e155b68c5c2664ec08ced5fd60c8c3ba8832f2e","_number":772607,"virtual_id_number":772607,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-11 10:17:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-02-11 08:14:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-02-03 09:40:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-02-11 08:14:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-11 11:11:23.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-11 08:09:07.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-27 00:05:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-27 09:53:03.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2021-02-03 09:40:37.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-02-11 08:14:55.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"e503440e1a1d02029f5547199adfc10f098ebb0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-26 22:20:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6ce9c24c54a40af0ae95e9a736f39f6b08b7280a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-26 22:29:29.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"38ed7313a0b642f681d1c6614d9483de6a5c4eb0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-26 22:30:57.000000000","message":"Patch Set 2: Backport-Candidate+2\n\nBackport to Victoria","accounts_in_message":[],"_revision_number":2},{"id":"f6b3d1d51f52e552874db20db0b93aabb92f1531","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-26 22:36:21.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"56a0c07617ccc707947a5efb6ef91d89d136abba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 00:05:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ea66bf668c764b0d9534eb095362d81a : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecd012d0efe84562869a419064454ce5 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e14dc7d3d6464fa8ab1a0fb5b2cd501e : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b135262a2654476a3c67c1f7e05dcfa : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fb2163fdfa44456885c45e9b88ff6808 : SUCCESS in 4m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9528a3d16ca48e29b5f41d76281cca5 : SUCCESS in 6m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d1b80cb5f0a47f2881f7961988907b1 : SUCCESS in 3m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2c0d3a04ab9d4207b3b02e7ba22dc44f : SUCCESS in 4m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b8e96a9ec8094d1d9637a662e66ae671 : SUCCESS in 7m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cc7cd4111b8140ea821f687ce580076d : SUCCESS in 18m 12s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/769602c68d53464f9fcae663a3ad3921 : SUCCESS in 51m 24s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/f0cbd5e9df1a484493d4d249afdef9f9 : SUCCESS in 55m 10s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/935e2f2d2350402fa3c5d7ab86947904 : SUCCESS in 15m 05s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/55c424a89571434bbc696ed998d221e4 : SUCCESS in 1h 27m 20s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/8543597e5f74428399c35aa056a5b86d : SUCCESS in 14m 46s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/c3168726191f4fd68cd45318c0cca272 : SUCCESS in 21m 37s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/54184f0301fb4f8fade739765e871181 : SUCCESS in 10m 16s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":3},{"id":"47d3462e62236cf164722ac8bc9281cd48b643ca","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-27 09:53:03.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1c296ffed47eb5a2ddc8807baca845be1350c707","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-27 10:00:47.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c70cfe1cd8c4a3ebac862ba2546b0c2c9538410e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-27 10:02:52.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1159b797aaa72f86489196d6702cb8b4e2f49d0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 11:35:59.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b2a9b7f5d5ee420d907aff2e0baab8ca : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71ec7bf94d164dc2bebee9c2875e9acc : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/617f4ab1545849068cfb89bedd7eb1b7 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/154d8d93dba7444d99ef38652a5eb324 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/774e760e75384c3bb1c8f2fe77d82713 : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19fdb49f10a145d382efa45af12c0c80 : SUCCESS in 6m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a73e7594d154f11abcc4a5e7f26dc92 : SUCCESS in 3m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e71a5c1164df4e66bd9f671332f1cb53 : SUCCESS in 4m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8ed7f154c16e4f4aa00417f1b116f69d : SUCCESS in 7m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d1f1a501ddb04f21b17e01b1cff753c1 : SUCCESS in 17m 12s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/18ca10fecf7f4ad8ac1374662d98b060 : SUCCESS in 1h 01m 31s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/0efddbf5b50c49e79e31a6b020f4195e : SUCCESS in 1h 22m 07s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/3a1f61d428d74ecda4a88310091fdd71 : SUCCESS in 14m 22s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/afdcdfe859cf43368b1715016361ce41 : SUCCESS in 1h 34m 06s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/08e704b647614ad094fc9743bfe11804 : SUCCESS in 17m 07s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/ae733d49b39745c0b2bbbc3ee82f53cf : SUCCESS in 20m 32s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/1fa33580841344de82138e534b983063 : SUCCESS in 9m 54s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":4},{"id":"dc96b82b53948fd034c30b60fb45c4235b46dde7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-03 09:40:37.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"48800b207bf34ff36f3588f4c8acc8d986ba08f2","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-11 08:09:07.000000000","message":"Patch Set 4: Review-Priority+1","accounts_in_message":[],"_revision_number":4},{"id":"22e3131195771c582450f531ab10c389486298f5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-02-11 08:14:55.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"250b9813c9ff42561375a176611c611a491696a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 08:15:27.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"3e5cf1e9a972d9319204ad3e67ee95576c71a310","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 10:17:25.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a06312456a844f5c89eca4390bad4c36 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6367c4f6be1d4137bb29634139a28c89 : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75be167605c04bdaa0aa8a9744c983de : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea076335897b4017b6f9ab39ba510414 : SUCCESS in 6m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e7e7dfd41fb4dcab9815141b2571ffc : SUCCESS in 4m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fa8323bd07e24841a98786b221837b29 : SUCCESS in 4m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/dbd8286a6e9240d8bc05ec35077195c0 : SUCCESS in 8m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e08395ba140941b8938ab0b4ab0a95e7 : SUCCESS in 17m 05s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/a96d0cd7732f47ff93efddf3cff44c08 : SUCCESS in 57m 44s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/3386e65ddd734a49af33dd60d857518b : SUCCESS in 1h 02m 39s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/d510af9f948b4e758a265da46dceccbd : SUCCESS in 14m 16s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/4b54a4027edd44ed8bf64c7011e527f2 : SUCCESS in 1h 33m 26s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/de780cb576724e0d9985b4ed20d25d12 : SUCCESS in 14m 09s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/3b92aa82cf2f4a3199a19fc2601fc5f4 : SUCCESS in 22m 25s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/abeade9322e84405a2f333bad8fa155d : SUCCESS in 9m 29s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":4},{"id":"514a642b655dc677b186d3780ec45e48853a6bb2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 10:17:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"e7e2dcab6ed9ae0b3b3e1b1ad594d2595455bb22","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 11:11:23.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/248720ee08184a50803dc81585961070 : SUCCESS in 1m 35s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a9acff89bd8468088193e073772cd2b : SUCCESS in 1m 32s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f3c0526c092760d6c08c244679a0fc47f7e6b821","revisions":{"a4c2cf454d7f27ffcba6a3f461ca3f20c2f4f166":{"kind":"REWORK","_number":1,"created":"2021-01-26 22:20:08.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/07/772607/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/07/772607/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/772607/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/772607/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/772607/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/07/772607/1"}}},"commit":{"parents":[{"commit":"ff7c1741d86c580dcd2a68cc3c49bb047d61455b","subject":"Merge \"Add variables for API VIP address and FQDN\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ff7c1741d86c580dcd2a68cc3c49bb047d61455b"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-01-26 22:15:34.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-01-26 22:15:34.000000000","tz":60},"subject":"Adapt to changes from the Bifrost Victoria release","message":"Adapt to changes from the Bifrost Victoria release\n\nThe critical part of this commit is adapting code that was still\nsourcing env-vars. This file was removed from Bifrost in the Victoria\nrelease, breaking the `kayobe seed deployment image build` command.\n\nThe other changes are not yet breaking Kayobe:\n\n1) Release notes claim that OpenStackClient is no longer installed when\n   keystone is not enabled, but it appears to still be available. Use\n   the ironic native baremetal command instead except in playbooks\n   related to baremetal compute nodes (i.e. overcloud ironic).\n\n2) The use of OS_CLOUD\u003dbifrost-inspector is deprecated and should be\n   replaced by OS_CLOUD\u003dbifrost.\n\nChange-Id: I25078e69acdf41a4ef9957f99fe5047de54b778d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a4c2cf454d7f27ffcba6a3f461ca3f20c2f4f166"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a4c2cf454d7f27ffcba6a3f461ca3f20c2f4f166"}]},"branch":"refs/heads/master"},"64489f967c286afc458b4da073ea6e1a19d7fad8":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-01-26 22:29:29.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/07/772607/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/07/772607/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/772607/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/772607/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/772607/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/07/772607/2"}}},"commit":{"parents":[{"commit":"ff7c1741d86c580dcd2a68cc3c49bb047d61455b","subject":"Merge \"Add variables for API VIP address and FQDN\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ff7c1741d86c580dcd2a68cc3c49bb047d61455b"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-01-26 22:29:11.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-01-26 22:29:11.000000000","tz":60},"subject":"Adapt to changes from the Bifrost Victoria release","message":"Adapt to changes from the Bifrost Victoria release\n\nThe critical part of this commit is adapting code that was still\nsourcing env-vars. This file was removed from Bifrost in the Victoria\nrelease, breaking the `kayobe seed deployment image build` command.\n\nThe other changes are not yet breaking Kayobe:\n\n1) Release notes claim that OpenStackClient is no longer installed when\n   keystone is not enabled, but it appears to still be available. Use\n   the ironic native baremetal command instead except in playbooks\n   related to baremetal compute nodes (i.e. overcloud ironic).\n\n2) The use of OS_CLOUD\u003dbifrost-inspector is deprecated and should be\n   replaced by OS_CLOUD\u003dbifrost.\n\nChange-Id: I25078e69acdf41a4ef9957f99fe5047de54b778d\nStory: 2008558\nTask: 41696\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/64489f967c286afc458b4da073ea6e1a19d7fad8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/64489f967c286afc458b4da073ea6e1a19d7fad8"}]},"branch":"refs/heads/master"},"2580292a53e74300e7ada1ffa8d019805697e489":{"kind":"REWORK","_number":3,"created":"2021-01-26 22:36:21.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/07/772607/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/07/772607/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/772607/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/772607/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/772607/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/07/772607/3"}}},"commit":{"parents":[{"commit":"ff7c1741d86c580dcd2a68cc3c49bb047d61455b","subject":"Merge \"Add variables for API VIP address and FQDN\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ff7c1741d86c580dcd2a68cc3c49bb047d61455b"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-01-26 22:36:13.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-01-26 22:36:13.000000000","tz":60},"subject":"Adapt to changes from the Bifrost Victoria release","message":"Adapt to changes from the Bifrost Victoria release\n\nThe critical part of this commit is adapting code that was still\nsourcing env-vars. This file was removed from Bifrost in the Victoria\nrelease, breaking the `kayobe seed deployment image build` command.\n\nThe other changes are not yet breaking Kayobe:\n\n1) Release notes claim that OpenStackClient is no longer installed when\n   keystone is not enabled, but it appears to still be available. Use\n   the ironic native baremetal command instead except in playbooks\n   related to baremetal compute nodes (i.e. overcloud ironic).\n\n2) The use of OS_CLOUD\u003dbifrost-inspector is deprecated and should be\n   replaced by OS_CLOUD\u003dbifrost.\n\nChange-Id: I25078e69acdf41a4ef9957f99fe5047de54b778d\nStory: 2008558\nTask: 41696\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2580292a53e74300e7ada1ffa8d019805697e489"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2580292a53e74300e7ada1ffa8d019805697e489"}]},"branch":"refs/heads/master"},"f3c0526c092760d6c08c244679a0fc47f7e6b821":{"kind":"REWORK","_number":4,"created":"2021-01-27 10:00:47.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/07/772607/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/07/772607/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/772607/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/772607/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/772607/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/07/772607/4"}}},"commit":{"parents":[{"commit":"ff7c1741d86c580dcd2a68cc3c49bb047d61455b","subject":"Merge \"Add variables for API VIP address and FQDN\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ff7c1741d86c580dcd2a68cc3c49bb047d61455b"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-01-27 10:00:36.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-01-27 10:00:36.000000000","tz":60},"subject":"Adapt to changes from the Bifrost Victoria release","message":"Adapt to changes from the Bifrost Victoria release\n\nThe critical part of this commit is adapting code that was still\nsourcing env-vars. This file was removed from Bifrost in the Victoria\nrelease, breaking the `kayobe seed deployment image build` command.\n\nThe other changes are not yet breaking Kayobe:\n\n1) Release notes claim that OpenStackClient is no longer installed when\n   keystone is not enabled, but it appears to still be available. Use\n   the ironic native baremetal command instead except in playbooks\n   related to baremetal compute nodes (i.e. overcloud ironic).\n\n2) The use of OS_CLOUD\u003dbifrost-inspector is deprecated and should be\n   replaced by OS_CLOUD\u003dbifrost.\n\nChange-Id: I25078e69acdf41a4ef9957f99fe5047de54b778d\nStory: 2008558\nTask: 41696\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f3c0526c092760d6c08c244679a0fc47f7e6b821"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f3c0526c092760d6c08c244679a0fc47f7e6b821"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
