)]}'
{"id":"openstack%2Fkayobe~772609","triplet_id":"openstack%2Fkayobe~master~I93d242889e225b4e60254f6b9cc5eeb457294ac8","project":"openstack/kayobe","branch":"master","topic":"story/2002098","attention_set":{},"removed_from_attention_set":{"32657":{"account":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"last_update":"2021-11-26 13:28:25.000000000","reason":"Piotr Parczewski replied on the change"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2021-11-26 10:29:32.000000000","reason":"Michal Nasiadka replied on the change"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2021-11-09 10:25:58.000000000","reason":"Mark Goddard replied on the change"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2021-11-29 10:51:43.000000000","reason":"Pierre Riteau replied on the change"}},"hashtags":[],"change_id":"I93d242889e225b4e60254f6b9cc5eeb457294ac8","subject":"Build overcloud host image directly with DIB","status":"MERGED","created":"2021-01-26 22:48:19.000000000","updated":"2021-11-29 11:51:12.000000000","submitted":"2021-11-26 22:57:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":39,"unresolved_comment_count":8,"has_review_started":true,"submission_id":"772609-story/2002098","meta_rev_id":"67b68aee2e9f3c719067b972dc0e79387595c342","_number":772609,"virtual_id_number":772609,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-26 22:57:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":2,"date":"2021-11-26 09:23:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2021-11-26 10:29:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2021-11-26 10:29:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"}],"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-27 00:26:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-19 07:41:23.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2021-10-05 16:03:50.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-10-06 10:19:10.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-11-19 14:59:14.000000000","updated_by":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"reviewer":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"state":"REVIEWER"},{"updated":"2021-11-24 15:58:13.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"}],"messages":[{"id":"2c1a3ac11fef9276af0f18c1a6003d4262d2c08a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-26 22:48:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4976f06a16dd0fae96bee997803c1e3c816d2e43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 00:26:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/94b56e81fea3482d8b55646f7003a3f2 : SUCCESS in 2m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1ae1316149d44f29da518df72091d1b : FAILURE in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd29dc0339c547daa847078bbf54329a : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c98232dd0fa04fa5b1dccd3385c299e7 : FAILURE in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a1e7f639fae41bf82c626561110710b : FAILURE in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0610429a3aa54313acbf0628431816c4 : FAILURE in 4m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80b83ab72c9247e1b0ac54c6be994763 : SUCCESS in 5m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8abb0e2b963641048c754b5c57804dfa : SUCCESS in 5m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a2f69da1f4634b7bbf1d0a321b0f1bf5 : SUCCESS in 10m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8e6340975d0f444eb46c69a8216b66bb : SUCCESS in 18m 37s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/99c1b53d7e534e38b2532820484b076c : SUCCESS in 56m 49s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/3d046678ca2b45029cf5f60e85d9694b : SUCCESS in 1h 02m 39s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/a42d9f78b8c54fbba6d226fb0699c26e : SUCCESS in 16m 14s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/0e45b649f68943aaae8675f8d452f889 : SUCCESS in 1h 34m 02s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/cfa1149c9c114d08ba86feeb767e3287 : SUCCESS in 13m 05s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/09838149833b455a98d201f291847c43 : SUCCESS in 21m 26s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/aeea5d29cc814492affe3b368bead405 : SUCCESS in 9m 35s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"ec9deebe0e2b34dc4b016d23ed08a3288ace7b7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-11 21:20:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e0efcd253d5773d0af43d488541fbf1305fbc4cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 23:07:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b2c4aac8427f4a2c9c25795ede85e8fc : SUCCESS in 2m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/725d85c0e8434c6fa40887cea8384035 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e91dccd19ae941dfadebed4ec27323de : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e4681fa6f3f41f3870a8d3e2ed24e95 : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b426e797c86e47af94e1a9e1ee59e1aa : SUCCESS in 6m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea3fef8b07df4af88d898c265459ff47 : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/062f17659d4a4f8b8c73ba9a38e97f3d : SUCCESS in 6m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/561e45bfa9ca4aeda70720254b33212b : SUCCESS in 4m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4ae60606c99b4314bccfa40a40b93ae7 : SUCCESS in 8m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f1136bac5307423c98036b3d9a090020 : SUCCESS in 16m 51s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/9db46a8e466140779c227c43341df629 : SUCCESS in 58m 00s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/f6f1e5aa29464d26a51fa2dfa0d3ec00 : SUCCESS in 1h 12m 20s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/2aebae3ae2544354851a2cc5a83d22dc : SUCCESS in 15m 57s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/8bbf7a0ef1334acd9037fb9c23ce7a1a : SUCCESS in 1h 41m 56s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/400a2f915b134f05a104b42a6b9a3ee5 : SUCCESS in 19m 56s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/4d02390ecffa4f899a6bf0f823ad1249 : SUCCESS in 20m 40s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/64de5852a4054894964e5bfd970a56b2 : SUCCESS in 9m 58s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"fdc2089bb6f3abd81ed3b3d4f2fa4dc11373dcc5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-07-30 09:42:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ae9483982d6a5b6f82e1c68fa9d147d440baf668","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-30 11:32:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/270a63a0765c494f833ec364344dfe63 : SUCCESS in 2m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f63b70ce954340e1b1465abf9da69481 : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d8cbde092004f429fba4c72f6f9233f : SUCCESS in 6m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/23e8499831594ab88a1f480b085cd30a : SUCCESS in 7m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab075a11065740e1a03005746e4c0a2b : SUCCESS in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b8215e46742b46fc8e3c3d9d55f16109 : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fa0c0eebda547c5b3802aa5dbd259b4 : SUCCESS in 9m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e48534cc499d4d59913e458389520b3b : SUCCESS in 6m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/feff3918a02c4be6aaaf48c45df00127 : SUCCESS in 10m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1bf706d1de064e4b8f883c0f672205d5 : SUCCESS in 32m 29s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/6306c063de5240a095ab5fa1026f90f1 : SUCCESS in 58m 15s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a1e2742335a04a7d824b404fa851e099 : SUCCESS in 44m 16s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/4cba8c6316314914a7f39144e1408a64 : SUCCESS in 1h 08m 17s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/e8ca517c6dc64e4d9aaa45fad7636859 : SUCCESS in 17m 38s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e990bb3d152043d8b40126ff4a309a15 : SUCCESS in 14m 24s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e8d0cd8161b64bc7b322cfa9dc3821bf : SUCCESS in 1h 49m 13s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/7eb6031f250a46a8adc11bd46ab99654 : SUCCESS in 38m 44s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/308488d961734790904423733c70be84 : SUCCESS in 25m 22s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6bc8ad5ad4bd42ca9d7c6dc0610a2032 : SUCCESS in 26m 22s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/349e4a2aa4bc497caa56df4b70614169 : SUCCESS in 14m 42s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/156766da3017413ab545525d6adac2f7 : SUCCESS in 10m 40s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":3},{"id":"ffd7d7acc83c7fc40fb8fddbfa0b33b9ffc15de6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-08-30 15:14:22.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"56e048a2d2e2c9f0423e9e5c102b026d031e338f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 17:20:22.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/efc86caa2f8e456ba68a8d244e21b269 : SUCCESS in 3m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/94ea2999d3ac45da84380638664e1a5b : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cfd64940bc847c288849f2e0ae08e46 : SUCCESS in 8m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/74f1cb5b3b7144ec8f231a2b7bdeb232 : SUCCESS in 7m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79436ba130984bb79b7e7216b7e0759c : SUCCESS in 7m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d8552abab3a459584aea3aacd913302 : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f243108b977f4685821f0382ff33c245 : SUCCESS in 9m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bc126806c3ba45d3b18a1b7185980fcb : SUCCESS in 7m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/85f75fba00974d97ac3f6a286763d0ee : SUCCESS in 11m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/769f2ea1a51f4aedba35d35ceb6b1323 : SUCCESS in 29m 21s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/7717a07280744d18aa4e1385bc5b8163 : SUCCESS in 1h 04m 22s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cbc607d7ef684bc8b4d6b4321d3493ee : FAILURE in 37m 46s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/7b65030f5c844911b9e4a3c181167c80 : SUCCESS in 1h 12m 20s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/d8359cd075404fae92a269374e950cd0 : SUCCESS in 20m 28s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6f2f4a95f606419093f15ca12fe11397 : SUCCESS in 13m 25s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2b7e64f88fe64d88be7db7ffdfb597fb : SUCCESS in 1h 58m 55s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/db1d0fac12d94144aed325108aecdfc5 : SUCCESS in 34m 54s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b0baa9cf6114a8ba120e80b47701635 : SUCCESS in 26m 26s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3b503dce561a4eca8c20b002429c8f53 : SUCCESS in 28m 55s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/8d6805a90b5d4f239f4e49792d87e192 : SUCCESS in 16m 21s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a2a2b70fbe0e47ca9388fd34420c0abd : SUCCESS in 12m 22s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh","accounts_in_message":[],"_revision_number":4},{"id":"e7567833626b0e1f4b2f2f3ed252f8c831292289","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-02 15:35:02.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"277e4e70031c68728c44a989148de8caef8c77fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 16:58:14.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a61efb91f2914ae7abf7dda6b667622f : SUCCESS in 2m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d98ceb5e2f3144938354cfa0698a93af : TIMED_OUT in 52m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b2f667431904674a1fd5afdcc1122ec : TIMED_OUT in 32m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/249d72cc98b9410e8971724ec60bd0f9 : TIMED_OUT in 41m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90761508230f4c2784b3b3d6b2d0b80a : TIMED_OUT in 41m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0c734ff53581447caebf3059bfcc9b8b : TIMED_OUT in 41m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e3ecefbea124b1f8e0d6069857016b5 : SUCCESS in 8m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d053eef4a1ee41dca340dbcfb7ec9556 : TIMED_OUT in 31m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/130df97bb4fc4a3abec306ee6ce6341b : TIMED_OUT in 31m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/14bf1a530710403ea6af727402e57a29 : TIMED_OUT in 1h 01m 19s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/1b598b419d1e452c908ebc9a9529c09f : FAILURE in 31m 56s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c7bb7985cc474091a1ad817cd6a745f0 : FAILURE in 20m 42s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/7f9181098de742b2aa11346987e5b8cb : FAILURE in 31m 35s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/f9230a39efe44431b6937b18716657b3 : SUCCESS in 20m 12s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/28081b68fe7643ff9881e1de1e7a4c61 : SUCCESS in 15m 54s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2426588fb0ab4a9c818ebacec30861df : FAILURE in 1h 17m 29s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/91f4e430a42848b993d8e4a1423550eb : SUCCESS in 38m 39s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5be3a805d96b4d1c876a78e25b47230a : SUCCESS in 28m 01s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1ae79ab874374f13b75c87186dbbf554 : SUCCESS in 28m 57s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/885fff60669344bfac0938290e1a7c36 : SUCCESS in 13m 33s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c94d36ea050d465ab294a5cc8e8358a8 : SUCCESS in 9m 47s","accounts_in_message":[],"_revision_number":5},{"id":"5a22319148a9b2086f166a036b776f48d200cfa7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-05 20:43:34.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"be80c863a734cd94d9d2a70d4e2a89127250c277","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-05 22:17:37.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2cf62a359e6f4dcabe81bb2adafedd89 : SUCCESS in 2m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/803a1017776244e49a3f5b75c2a9396a : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c6a1d69365548d599006a7b89379f50 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e0bb1a443df4b0b9d49ec5bd3cee370 : SUCCESS in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/360b51c48ecf426fa8e56b00b68d83b4 : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b8053e0affa24c4193fd04f553847c80 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/577d3c42f17a40c0b1436dd3857b3eb2 : SUCCESS in 8m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4685766ef6b140aabe20fc7da3880811 : SUCCESS in 5m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8d8e396f7df8480bb61023a08f62b7fa : SUCCESS in 9m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c42013120b2a476b8ab3baebf9f07151 : SUCCESS in 31m 08s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/1bd8f19c93514a31bebb4acf87f248cd : SUCCESS in 51m 56s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/247bebbf34cf4adabb0baac8c3518a5b : SUCCESS in 43m 22s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/27322873988040c6aa385e3df89dbf31 : SUCCESS in 1h 01m 58s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/cbfd5143904a4a6fb435e347713343fc : SUCCESS in 18m 08s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b1c2ac9d4c2f41989439419c2cdbef71 : SUCCESS in 11m 40s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/57302577ec94449ba8d58f5804e79c31 : SUCCESS in 1h 28m 33s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/a41468f1731d4c07962f6eb8f8d6e4e8 : SUCCESS in 30m 16s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/309921d8ab48437abed8d8ab528cc213 : SUCCESS in 31m 13s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e00738a5af4b4a1786d4a1b9b4361429 : SUCCESS in 32m 57s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/abb18536146247caa54fc9548c951549 : SUCCESS in 10m 49s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/29cfb96176214934aae78f34060ccf84 : SUCCESS in 9m 06s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":6},{"id":"eae587852e56777e99671e4b4f78fe93c751e404","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-05 09:35:58.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"451bf052f996bd810ab4cb140703a50e814be770","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-05 09:44:19.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"501d3f2fe01b9ad24f0fb71a845fe3e9dddfcd47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 11:58:07.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0e3d25c36725469b9ef57f65de985866 : SUCCESS in 4m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7584764320ee459bb78786c9f18c6956 : SUCCESS in 7m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c78ff7496e0a4b4db9a2f6c7677234d5 : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c6b5f26b2854148b3dc68b3fa4a05a8 : SUCCESS in 7m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/03a38c9b25064818baced7767f479984 : SUCCESS in 7m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/393ee24f74a649f3a23ce22347dedec9 : SUCCESS in 11m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff06074bfd20464f88e85a779af70f60 : SUCCESS in 11m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/49c0c27cdf9f4957b69354495984b146 : SUCCESS in 8m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e8d32bbc30fb4263a4daefeab8535178 : SUCCESS in 13m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bbc75d668c3e487baae336843ac79ae8 : SUCCESS in 35m 41s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/a9dceef110a54559bea8152bbeda837c : SUCCESS in 1h 11m 23s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/270665d1e0c842c09a08470354f471c4 : SUCCESS in 52m 01s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/85747121d8dd4601bc5a3561134db894 : SUCCESS in 1h 21m 29s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/eb74bd239c524ff4bb36d00bb4737df0 : SUCCESS in 23m 16s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e13fc7cf7533493f8ea7915ee8d2da8c : SUCCESS in 18m 14s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ff096eefaff24c47b465e007b70fc688 : SUCCESS in 2h 06m 07s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c079600000a74140a26c95975f5ccb0e : SUCCESS in 1h 30m 03s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d3f3dd6d32a541578bde9f5bd0729107 : SUCCESS in 40m 21s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8b436c1d15b48079f2a94f4293f70a0 : SUCCESS in 32m 33s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2c28e1aa30e94f4395671c809b4c3779 : SUCCESS in 35m 45s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94174de4c2fe49c095411c5837a6f80c : SUCCESS in 28m 15s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/36e3dad80a2b4fdaad83140d49592ca5 : SUCCESS in 17m 01s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0eba8d7aaefd475d88c77c49bd9875ab : SUCCESS in 14m 17s","accounts_in_message":[],"_revision_number":8},{"id":"b1661af503ba7da4e4f129c942b5d6aa351a9679","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-05 12:17:16.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"6ac74800f3fcc0f197efa5e891580a975691f869","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-05 12:18:51.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":9},{"id":"cca3bd7ca3245b8dcdcb196f0cafcb08dbfe1adb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 14:11:16.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2ebc6912ad2e4c5584f4eb3648c568f6 : SUCCESS in 2m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f401050db5c84531971052175ef18827 : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a002c13cb77a4956a9e1468c6769ddd6 : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0699dc65fa7148d5ac598386a5a88d49 : SUCCESS in 6m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f4876c2a3699433990ac64b7dcb517fb : SUCCESS in 6m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a7c3e811268346429983344397201473 : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d65571ba6bc546b1a838e9f1624e2383 : SUCCESS in 9m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/63cdf7d445cd4ccfa7ff5a9c50f3878e : SUCCESS in 6m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5a6aebf997b449f389255dc49e613274 : SUCCESS in 11m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a2bb909d16424303b3377a717c3f928b : SUCCESS in 29m 43s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/7a2cee2fa9954b5297b6c84a89877913 : SUCCESS in 1h 10m 35s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/894ed0a7598b410ba2f8facfeda311c4 : SUCCESS in 45m 56s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/fa8960d690d84c9e8b99827095c1a832 : SUCCESS in 1h 07m 13s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/a0d1d1c92f08457eadfef928ce3f2973 : SUCCESS in 24m 20s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3463130f13c7448cbc126fe825eafd98 : SUCCESS in 15m 19s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/54e5693be0a7422680d5ae4b074f8b04 : SUCCESS in 1h 49m 26s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e8cf4539d70488581f70d4ba154ff05 : SUCCESS in 1h 24m 34s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/0993caca49cc4990847e0edf60c1cc9f : SUCCESS in 43m 28s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2b5ea7f270274bfb8a6b44a5bf195f26 : SUCCESS in 38m 37s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ec69e0eeb80d42f380f9c0756993dfc2 : SUCCESS in 37m 28s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d9c2dda69a4f47bd8a2e23290feb0654 : SUCCESS in 25m 27s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/5f2dd5b6dae34366a465974a491efafb : SUCCESS in 17m 36s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/084e59e2812f46e28162f07e216769cf : SUCCESS in 14m 47s","accounts_in_message":[],"_revision_number":9},{"id":"7bbd73e4681e06eb7e5e0a7cb19de4993c348d20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-05 15:14:23.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"ff4adb1786c3c6b294cd50f0c6b4437bfe40b010","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-05 16:03:50.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"0b094c69d8bca78d1a301795c03bf93a13df9a71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 17:11:55.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/21c5a4294d6245c7a534062c51452e8e : SUCCESS in 3m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/87a673ad156f46329279dbb1a5f3c393 : SUCCESS in 7m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52e6a456e9314dcf86e0bc138a1269fc : SUCCESS in 7m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5dc5b509f0a14c68bde0707ffc071b79 : SUCCESS in 7m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae55799408504ba485c1abe8ff6acf35 : SUCCESS in 7m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f385f6ba5f144202adfd9b9e2a419920 : SUCCESS in 8m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb5fb270e37f4c94aff45fbd62d920a7 : SUCCESS in 10m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7d06aebac79a423881410df903cd9107 : SUCCESS in 9m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/78fb84d3e68a44838bfe24737490c691 : SUCCESS in 11m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a4f9e12cf1184ec5989522de8036be75 : SUCCESS in 33m 27s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/7a1524a9eebc4431b4a178c9832a0556 : SUCCESS in 1h 06m 53s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cb92c3196ae34ae0af77e5bc3f085d09 : SUCCESS in 45m 36s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/19f499671bcf4ae3ab470198326f7e03 : SUCCESS in 1h 10m 56s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/749d2a00e85449fe8c4240123779a5b6 : SUCCESS in 21m 51s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2bec30b9e30d45489093ea73df14153e : SUCCESS in 19m 02s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5e218a6ddfba40fc963b87a4e9656a0c : SUCCESS in 1h 55m 26s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/50c9c8c041cf49389ee4062c29bd5a91 : SUCCESS in 1h 23m 19s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/c7198af60bfb45398de83d2b3b10cec1 : SUCCESS in 37m 47s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/24db91b6c67046f794002d45412747d1 : SUCCESS in 30m 55s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/610b3200cf3d4ec9accfb4c5787653a6 : SUCCESS in 31m 26s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a75d84c630634f17b299bb2b0e0aae84 : SUCCESS in 25m 24s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/7319bbb001c249da8912b36b0e52d263 : SUCCESS in 16m 14s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b5405067e207456daa1cc5fecbc26420 : SUCCESS in 14m 30s","accounts_in_message":[],"_revision_number":10},{"id":"38854c53da2f0414c440e31c7d82cdeddfa24591","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-06 10:19:10.000000000","message":"Patch Set 10: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":10},{"id":"0d5dab46bebeccc6ef81ce1aae206602babfa3d7","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-06 10:42:16.000000000","message":"Patch Set 10:\n\n(6 comments)","accounts_in_message":[],"_revision_number":10},{"id":"7d53b7af78e86b137a2853b2ed9e446e3e295961","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-06 10:49:39.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"a5acd044199107b414c1d9e2df616d1688dd5630","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-06 11:27:55.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"124db7d74f4aa8771aafe441e7468d7f88fa6769","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 13:27:54.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a21433d29d72473888cbf649c092f24f : SUCCESS in 3m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9a7945c79ed49939ddc6eaef07dddb0 : SUCCESS in 7m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a992314e86304ba09b8fb524ce8a085a : SUCCESS in 7m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d043c2e8a0549bca432a07b89871293 : SUCCESS in 7m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/036b1096fa8a46f486d9255b120f2c31 : SUCCESS in 6m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/55ffb682d73b4c52b96b8fc575beda7d : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1155e1c6b10410db00e13c6b457e60a : SUCCESS in 11m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4111a55d73a94d29bfba86efe86b9b05 : SUCCESS in 6m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/041df791294643f482d45a3142a60cd2 : SUCCESS in 12m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ca71ed2aa221445da4dfed4bf8189e78 : SUCCESS in 31m 23s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/dfcc6bd385bb4773b93dc47539860441 : SUCCESS in 1h 21m 46s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c55c3592834e434e9fc6753d2b9df414 : SUCCESS in 1h 00m 46s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/ab0845addde34d36a5b2cd97c2a141db : SUCCESS in 1h 18m 14s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/2e8c737b3d3b4670a54dbe0f2a541ecf : SUCCESS in 37m 28s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f296efb2b4094752b8ee4d33c3782711 : SUCCESS in 18m 24s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/70e947ff67fa4e20880a45a56e25b899 : TIMED_OUT in 2h 33m 43s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a406a0e443e45d18b79b1ca9b2109aa : SUCCESS in 1h 19m 51s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/5cec0cd2dded4e039b056a17ef818611 : SUCCESS in 48m 27s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/69efa0ad626042a78d443830631ebf9a : SUCCESS in 43m 24s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/da3fa2e622eb458597c5854b7bd8eb2d : SUCCESS in 40m 37s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/617761f114d24bbabef96eae5d73e32d : SUCCESS in 37m 46s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/fa5d9e545b94418995630c402651f227 : SUCCESS in 33m 24s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2766eec41a064877ad16eee2a1abe193 : SUCCESS in 14m 45s","accounts_in_message":[],"_revision_number":11},{"id":"080840e3ab9e4382ab0d39bafb69e285b0129f56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-26 08:50:09.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"b5f5dba44beb67a45adbb0385db8c4f8a0396ad6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 08:54:49.000000000","message":"Patch Set 12: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":12},{"id":"501fb944bc1dc001407ba48bcfa40e1910371801","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-26 09:26:32.000000000","message":"Patch Set 12: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":12},{"id":"473c4f5e06e3dfc2fc8cc2adef3f90aabed8c566","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-26 10:31:42.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"8bfa99c654a832f08eff7f5b6d50016d32894a65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-26 10:35:16.000000000","message":"Patch Set 14: Commit message was updated.","accounts_in_message":[],"_revision_number":14},{"id":"f21c2d8749e36364d668eaebd3030ad03c519d6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 13:00:40.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3916cdbef5554e29afebfc96ea3159f6 : SUCCESS in 3m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f7745c97ed0249bd9deac2e5cc96c585 : SUCCESS in 7m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d1de9a4b8d740a18f44dde0d8d14525 : FAILURE in 7m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd62658efc7b4c27bade21e625c271e5 : SUCCESS in 7m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed55d43f4e8547ac800de5ed8d2131b9 : SUCCESS in 7m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/51356928c6644db2b5182cd903ebfe9a : SUCCESS in 7m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/877d9d454a974f60a510e66c16cdcf67 : SUCCESS in 10m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3ee781c2d2f148f695c85491db161141 : SUCCESS in 7m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/849c97ebfc5949588587abbd6a2b112f : SUCCESS in 13m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5fb172f4a8834edf824efb18ef8cff6b : SUCCESS in 30m 17s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/cfac9f1760ab4e90a4a5f07625e164fb : SUCCESS in 1h 14m 40s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa789e7c634c43d89808285b076d7080 : SUCCESS in 42m 11s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/ece54bfd6f2c47c0944a0dfbda2ce564 : SUCCESS in 1h 20m 29s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/a5170a04afc8452483602e5e297c486a : SUCCESS in 24m 40s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c8a61f1fe1f446dc929c58be3dbf898e : SUCCESS in 15m 32s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fb9008a670214f7d9b579dfb9ee3d002 : SUCCESS in 2h 10m 21s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f403ac83b3d4771b278ffe3110e32cc : SUCCESS in 1h 26m 55s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/ba992f6a722c46c4bcce297ae97b09fd : SUCCESS in 39m 43s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/40f1689245db4377a3b9c7d28d1db610 : SUCCESS in 28m 03s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c270b14be0704c029431d12270b0320a : SUCCESS in 31m 02s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e639177a708241db977a27381dafbd9f : SUCCESS in 24m 42s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/2676fe2bf9824a1b9cccc05aceb55611 : SUCCESS in 22m 50s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52255509abd340afa16ea7a424283358 : SUCCESS in 16m 51s","accounts_in_message":[],"_revision_number":14},{"id":"cfcd99cf0786940aad79e847a6075ae98bb3cebe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-27 13:49:26.000000000","message":"Patch Set 15: Published edit on patch set 14.","accounts_in_message":[],"_revision_number":15},{"id":"4728f954d361f4eac6e2d5a04600e6e2cad4fbd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 15:40:38.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3a15943a71374a62ac6fce73b6f1064c : SUCCESS in 3m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f92fc608a1c24106a3e0493efcc2fb75 : SUCCESS in 7m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0e975c448f74de2a160e1ba77d1114a : FAILURE in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7509ab0d645a43c0b5c62f7b7a68a220 : SUCCESS in 6m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d7a9ac669ce49cbb4bfd5941645a513 : SUCCESS in 7m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/701b12ab999c4cfb9618a9552f62d8c5 : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e716210b3435436b899927248e181d84 : SUCCESS in 8m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3ce8f0cad4c5495d8767ad71e22820e8 : SUCCESS in 7m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cc961a80cbca4495bae43bed6dac7a1f : SUCCESS in 14m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e596b4e0481b4c95a8f3fee9971c1600 : SUCCESS in 30m 38s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/f9408aa011dd4476860b7c4633f9480e : SUCCESS in 1h 03m 45s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d6a12c52a00a40fb96ce699494ca4534 : SUCCESS in 50m 36s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/635ded0d6f7041cf954f8d6a7841bd78 : SUCCESS in 1h 06m 06s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/e4160434125c45c58a74cf3f90f28c43 : SUCCESS in 24m 12s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4fa24009b73247a48e2ac92b4db6f0b4 : SUCCESS in 18m 52s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6658d5d0412b462eac04e1ee683975c1 : SUCCESS in 1h 46m 13s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e18f3b5350f4a018772b589921ca689 : SUCCESS in 1h 09m 27s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/47a8b94665a741879d8a39e8c8eb872a : SUCCESS in 36m 45s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f8d89ec718c64b2caebbec89ccf7b913 : SUCCESS in 33m 18s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5c0fecb8672245549d94b31b389ce7c3 : SUCCESS in 33m 30s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5265c0965e6f4d5db88cfe29623a1019 : SUCCESS in 27m 25s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ae388ed315f94ab894fcd2e4cdb0f8cf : SUCCESS in 17m 25s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/017e58ec06a844c8bbc61ae16dcc66e2 : SUCCESS in 13m 48s","accounts_in_message":[],"_revision_number":15},{"id":"c458f9380b5a74c49de1ca8688ecbadaba323550","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-28 11:59:13.000000000","message":"Patch Set 16: Published edit on patch set 15.","accounts_in_message":[],"_revision_number":16},{"id":"780f7f9bd9cef4510c8751bf05b2637685d04b31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-28 13:32:09.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8ad25af8c46c4397b66e8b152883f5ff : SUCCESS in 3m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/350dbd7e181d4e7786e4a88242e5b261 : SUCCESS in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffd655875cff48df9e1612d4f59e5420 : SUCCESS in 7m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0f2852bbebe451bb5d8563bb57c8f7b : SUCCESS in 7m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/673e52f6b9ef4ebbbb25af1b98715188 : SUCCESS in 8m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/208b4337262c431bbb421cb7c06d1a8d : SUCCESS in 7m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c441458514a74d63a773f018e3147b45 : SUCCESS in 10m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b5f56d181ddb4113a4729c9b5cb6e490 : SUCCESS in 6m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a1305e75f90344c3a01c377c10afff93 : SUCCESS in 14m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f681ca61158849ca86fa8d9acd5bcc77 : SUCCESS in 34m 30s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/57152353d0d44c5d8e28c44a41197761 : SUCCESS in 1h 02m 48s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ed45e19142b4475b9db849d1e456fc6 : SUCCESS in 52m 51s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/afefe65bf15446799b37c222685f09a4 : SUCCESS in 1h 15m 51s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/efdf937cbf0b44ecb2ae7f68a03baf47 : SUCCESS in 25m 40s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6823bb71625b41a687497407ebd6410f : SUCCESS in 17m 43s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e5556132f5fc45ad80a4d2e787d5235a : SUCCESS in 1h 29m 15s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/323c064aae9749b9bdb0db1a6dae750e : SUCCESS in 1h 13m 51s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/8bb975ff50c249f2ab729fd4b9ca8cff : SUCCESS in 41m 12s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6ce47815f46340b3881c1c0ae7858f75 : SUCCESS in 36m 03s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b404d7038364452dab8db8fe48f07376 : SUCCESS in 30m 09s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b281f10a4d0a4b50a60051eff36a74ff : SUCCESS in 29m 19s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/cce24d4a733f402590a423d1470c9b8b : SUCCESS in 19m 27s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4381acaf3994421892815ce6c52c01c8 : SUCCESS in 14m 23s","accounts_in_message":[],"_revision_number":16},{"id":"1d32a76bb45ded172113b4f92a2c1ac94e1b231a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2021-10-29 09:56:03.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"10762ba4dffd602f340246835a93261f0c3fbf1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-29 11:56:49.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0d305e8498144e188591c2c08e028ea7 : SUCCESS in 3m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9edaa2644aa2464283d19ea16ce49017 : SUCCESS in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11e5f7f35b5c4afbb74ab7efd81e2de2 : SUCCESS in 7m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e0e5a02547174b849c93a9a227e3decd : SUCCESS in 6m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ebec813200384cb38e6be1d2fe42c604 : SUCCESS in 6m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c424c9f0374f4f9581018f00cf768340 : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5f807237a0a4b5bad716c92a7c51264 : SUCCESS in 10m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/916c3fbe1215494481418e3930264023 : SUCCESS in 6m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d35203203c12442694a7a20c07b2fbc2 : SUCCESS in 12m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b35086a70d544507bf15b440ea9d4271 : SUCCESS in 26m 52s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/bd453686f3804779bedfdfe27251f012 : SUCCESS in 1h 18m 07s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c9f2ff091b174dbea0c19e2155b1ff3c : SUCCESS in 48m 58s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/596b38f17ce14b6a89eb9983360333dd : SUCCESS in 1h 14m 20s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/6e3050e2474145e6b9c3a80b8698a14d : SUCCESS in 23m 56s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd33133c10c241a0bc49833c84a57b58 : SUCCESS in 17m 48s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/91ed39e8bfb949dc93e3f7b7e27ce665 : SUCCESS in 1h 51m 27s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c080c23c3111492385565db942b239f0 : SUCCESS in 1h 21m 00s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/a50ff1763c104bb8919647c64211b58b : SUCCESS in 38m 30s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/82734953b35b4e15936ff55b601bf03a : SUCCESS in 26m 28s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/82d6a9053fcf4ee9b6071547d335246d : SUCCESS in 33m 51s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b21e08b053184334940c97a60344cbfb : SUCCESS in 21m 54s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/5bed054b0a3b4e36b4db4efd8168c6fc : SUCCESS in 15m 56s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ce3c76393b44ce58a9e6b3bf971d04d : SUCCESS in 12m 59s","accounts_in_message":[],"_revision_number":17},{"id":"21ef1183d959bb2e58dcd3c1e74f4df114c87a83","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-11-01 10:06:05.000000000","message":"Patch Set 17: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":17},{"id":"b523b8608c14f56239fb4e15ea9571767bd9ce1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-11-04 09:59:44.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"67af855cdcac0bda8edca48b6b51686f7e35e8a6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-11-04 10:00:35.000000000","message":"Patch Set 18:\n\n(4 comments)","accounts_in_message":[],"_revision_number":18},{"id":"a120249b087318ea40ad8a72c2bb16e6f0618ae4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 11:57:22.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f6dbfbe00db34f53a00546dcc35bd94d : SUCCESS in 3m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd12a7931f164e83ac9f8aa504f3bbcf : SUCCESS in 7m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e5e9b50761b4eecbaf425b2b0c83025 : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e794269b16c4037ba3f2db643b79e26 : SUCCESS in 7m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bbf74b8081dc49c7b44f111453ca32db : SUCCESS in 7m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b1aeda23e434ea4b0e738db8052bcb0 : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce6acdbc9ec64df291cbc49f6f0826c1 : SUCCESS in 10m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/244957602e0a4d6ea11698f31b22d9ba : SUCCESS in 7m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5ee929bfd9db4aeb8ba0ba482d59f9fb : SUCCESS in 13m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cd38c84af66c46368d4a021bbbd73cf4 : SUCCESS in 31m 22s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c3dd0b9083fe4bcc9eb42c61ad6c01ce : SUCCESS in 1h 17m 46s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/361a31f5c54d4d6186cb6f7f929e8a33 : SUCCESS in 46m 05s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/6d69b141c5904e4f945da1980396c6a1 : SUCCESS in 1h 11m 54s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/1b10203addf54dad97ffc0859a983d65 : SUCCESS in 26m 06s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a7e730123c9249f4aa430a8f6dfb9cb9 : SUCCESS in 16m 53s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3d883469b7a24493997974f5187bc6ed : SUCCESS in 1h 52m 58s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/75076789c88c49c383d3657a908932a5 : SUCCESS in 1h 27m 28s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/e1da50b2151d4a4ca23a4a6aec837057 : SUCCESS in 44m 23s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fe08e9d6debe4eaeb4aa63f243f3f4e7 : SUCCESS in 34m 25s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/693b590893be4b54933c5d1bd38d84ad : SUCCESS in 32m 51s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7295106f5ca6486e80b818afc5b23d3b : SUCCESS in 24m 33s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/d8979d6b88d74912a9012faa8b5f3d8b : SUCCESS in 16m 48s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2bff740f87484d0bb1e4af05f987b820 : SUCCESS in 13m 38s","accounts_in_message":[],"_revision_number":18},{"id":"a59e9f2745200e57bf0588518c95d428b07d4e68","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-11-09 10:25:58.000000000","message":"Patch Set 18: Code-Review+2","accounts_in_message":[],"_revision_number":18},{"id":"e80868a5b54d0774b654e867ab763cc4de341f33","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2021-11-19 14:59:14.000000000","message":"Patch Set 18: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":18},{"id":"8cdf76066a0f9d5455958d82ad77bd45bb518322","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2021-11-24 15:58:13.000000000","message":"Patch Set 18: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"333f936f667f01a331a823476862bd223b56510e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-11-24 16:08:15.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"29169156de0806b9226ee3aa2159cf09bbdb7db9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-11-24 16:21:08.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"6129e6df24f22f0dc4c13339ed0c412696026620","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-11-24 17:23:43.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"3deea134aa795575ca07939ca74d3ab5e7e6b622","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 19:38:06.000000000","message":"Patch Set 21: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9545087f5edc4e6d8da379fccfd8cf66 : SUCCESS in 2m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae190755ccec4ce98fbebf4fb0818aa7 : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4715c68764254d3b8fd6ef23f28999de : FAILURE in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fecbbd46e0694a90936f4ec2ee1857d4 : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/793049e92bda478fbe773057ea0d5198 : SUCCESS in 6m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9d8bd8f2dc874b588da49349468f7de2 : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d2c5eaa08894c888afdb657df7e12f1 : SUCCESS in 9m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8784194602f84da3811a66c198072f7c : SUCCESS in 6m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fc0325439ef34bd5ace8894bf8f1c78b : SUCCESS in 12m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4244d9278c9047629d8891326b23054b : SUCCESS in 32m 49s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/7e79e57633a249be940cbbf77455aeeb : FAILURE in 19m 58s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/03694789f6db48e6bf72b85e6a8c80e4 : FAILURE in 18m 00s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/b3f5d8635cf648eba0db4e2fa7b5b607 : SUCCESS in 1h 15m 35s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/7332c8a4e80d424f93f58899ffc6fc12 : SUCCESS in 22m 21s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d465583c11f14dab869137c90d72fe02 : SUCCESS in 15m 46s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f695185d5cef4e1486777ab44a403c21 : SUCCESS in 2h 02m 29s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94dae2d284cf4b5b9df1d0ec4c71d210 : SUCCESS in 1h 25m 53s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/a1612cd19020486894e194e1b7640111 : SUCCESS in 48m 10s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/733fd2416dc74d36af1b25535c52a69e : SUCCESS in 35m 12s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/abea4a806eea4b9cbf52b21f12597ed7 : FAILURE in 17m 53s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b2ecef74e61446b8523aec78d720a8f : SUCCESS in 26m 20s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/930af413e9a44a5d957a9ee163f0169a : SUCCESS in 15m 03s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3dd08cd8b80b407ca63c36192cb72b86 : SUCCESS in 10m 26s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/f1e91f97f2044eae97c804b333239f90 : SUCCESS in 17m 59s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db02ffadd3e24c878b684a035e5203f3 : SUCCESS in 14m 03s","accounts_in_message":[],"_revision_number":21},{"id":"e94ad88f1718f4b5a298351298fe5931459dc9f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-11-25 11:59:37.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"5fcfae385d7f50586a8c7248b460ee81ca4bb3dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 13:42:43.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a05aafd405ef49d4a1d554bb73e7460d : SUCCESS in 2m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f5aaf714f16493c80647941e3d98a7c : SUCCESS in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8f371825f3a40efb85246ebb22f4c38 : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f7c60fd2868440a9cd6c8e61767ac37 : SUCCESS in 7m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ca235ce7cfbd463ab2d5848491d3b182 : SUCCESS in 6m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa85be5dd12d4193819a79f738c31eb2 : SUCCESS in 10m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1aa8afaae79d4f3eb5c1a927785bf44a : SUCCESS in 6m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3aefe28c3ae742b2ac75b707e48920f0 : SUCCESS in 11m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/90fc5d9971f24a1da16d2c80977e8ee6 : SUCCESS in 32m 28s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/cbb28f096491455aadeb7148070344cb : SUCCESS in 56m 56s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5924c2d90c1949f6b58ae0b241dc1064 : SUCCESS in 46m 25s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/25cda2a73f35448f84b076c02dcefb20 : SUCCESS in 1h 14m 02s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/e6374914ce1b4287aa45462ae1e1fed8 : SUCCESS in 22m 22s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c1f4873f04d4f9d966c181f776f37db : SUCCESS in 17m 35s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e703a021fd544fbcb3d0640672876381 : SUCCESS in 1h 40m 31s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/630725b02f3840c38af5ce8f2bd44988 : FAILURE in 28m 23s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/e87f6f63aedf4ea18473b2d254902357 : SUCCESS in 46m 07s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dfc6076f026f45c4894829f42e16cff4 : SUCCESS in 40m 20s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/65cb2dac4e3d4b46a8e022728395d498 : SUCCESS in 29m 46s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cebc06bdddaf4bb3a47a0377dc4df7c6 : SUCCESS in 19m 02s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/abb6ca7c372f4815ab0c6e5788d7e8c0 : SUCCESS in 16m 21s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b51320753558494c960704989595fc0b : SUCCESS in 12m 28s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/11109b5fcc164cf6a171d647097d832d : SUCCESS in 20m 00s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dfa9823935a846f6adf73f7e29f660ab : SUCCESS in 13m 23s","accounts_in_message":[],"_revision_number":22},{"id":"5504a6a224535c2121a08b9ac493500d07e0dcae","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-11-25 13:57:47.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"8fd12300659034d019b533f1eba749e563053c80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 15:55:47.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e8b826a9844e4647b6f9789b95a92c2e : SUCCESS in 2m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/397fbd00b24041fab62ff09a57a74022 : SUCCESS in 7m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1918d5f97b044f4abb24769dce03131f : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d609c1077e344b5ebd37572c6c03472d : SUCCESS in 6m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/603f89b8767c48d5adcdd0af93e5acad : SUCCESS in 6m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c2816ddfd994b72b5e420ec34f1dc76 : SUCCESS in 9m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/61fa6bc7f22f448db661471ef58c4fd3 : SUCCESS in 7m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/675ddc2bfb6b42568f2615f5fae8dbfe : SUCCESS in 12m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dba875d91e224c539017f75018f70560 : SUCCESS in 28m 35s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/3429dae61a8f4ac49b31d0f016ad6cb9 : SUCCESS in 1h 12m 05s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ec22deb20494be185e9b08ee21c755f : SUCCESS in 39m 54s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/acbda5cc96814563bccf7a3971710c47 : SUCCESS in 1h 10m 31s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/8d852239237a4fff96d5b47de1b75bf2 : SUCCESS in 23m 58s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e72398fb73c42d084c954e154f96d18 : SUCCESS in 22m 03s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b047446c793443e1a6abf15d574e782f : SUCCESS in 1h 50m 38s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e327e5c286bb48bbaad3b768e0027959 : SUCCESS in 1h 08m 23s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/3993594562ce4db8a1c36bf404422835 : SUCCESS in 41m 46s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a9ae9ba3c0084c54aab01a86be829e8a : SUCCESS in 31m 37s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b31f0b2c4df444e6adda860fb0cbd4d4 : SUCCESS in 28m 35s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/33a6e1e517114b2595a934fc200a6c42 : SUCCESS in 24m 40s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/0cafbf00913d41d6b0219664ef833d56 : SUCCESS in 14m 12s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ce244f529824dd29d30f3ef40e7dae0 : SUCCESS in 11m 58s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/506d291e4102433598d750c95c2dac3b : SUCCESS in 16m 24s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ecc520927f4e4037bcac6ba05cb4ec0c : SUCCESS in 11m 31s","accounts_in_message":[],"_revision_number":22},{"id":"ca6df01a929b9e6a95344566b44bdb5ea1b763f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-11-26 08:28:10.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"bab77fcddb7040fbcc8051077e3235809e5db4b2","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-11-26 08:28:34.000000000","message":"Patch Set 23:\n\n(2 comments)","accounts_in_message":[],"_revision_number":23},{"id":"de83b5275d157b15a7fe7b943060d91dd76b9fc9","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2021-11-26 09:23:11.000000000","message":"Patch Set 23: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"fee9cb62d0fe954d56ae29d2372b1e3d086dab82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-26 10:22:24.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/dc5fbf631df3400cafe6ebdecc57e29a : SUCCESS in 2m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/987156e4078846d681ce4992dbc27c21 : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbf436368a054740a5f4788ae8957b8a : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b9c95e1a21d94805b9e55211c78e9a34 : SUCCESS in 6m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e746929352d43eba411bb1b33992a27 : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c458cb7af4a04b3ebc38bf082122fdba : SUCCESS in 8m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/495219a37cc54933a5a24bada49880f2 : SUCCESS in 6m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0ba47aac660c4bceb9e12037a114a97e : SUCCESS in 12m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d1caa2cff0224e0d8581248c49131b32 : SUCCESS in 28m 40s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/483df1551791417794f164e12fef6dec : SUCCESS in 1h 19m 14s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/248ce85431bc465dbbe8d647b0ddb621 : SUCCESS in 47m 08s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/5eb82d468e07486abee25ff0271a1e41 : SUCCESS in 1h 09m 27s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/3b36cd2cda534f70a52e9e38b3578742 : SUCCESS in 24m 48s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f44a0ca5994442d084f1715fb4f99afc : SUCCESS in 15m 17s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/44dfae68154b41d48b1fbc490a100881 : SUCCESS in 1h 48m 24s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9891a74b666e44979ecc5c5f6bc9391c : SUCCESS in 1h 29m 04s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/854a32809b7d4722ae41c9536bd59ffd : SUCCESS in 47m 22s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4632f83a73184113b87af37af940c15e : SUCCESS in 36m 55s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/76f5a13216cd4cd2976f7bcf5cd3d6a0 : SUCCESS in 30m 22s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3398532f6219495d814cae285db6786c : SUCCESS in 21m 54s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/9311a467185e4c72b1f784fc8135bc3d : SUCCESS in 15m 58s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/939cb5643dac48178f83d758b8bfac73 : SUCCESS in 11m 42s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/04d5b69c62a84cee80cf591007f8cdb4 : SUCCESS in 18m 12s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/875ac1e30b9b4f188a99ad2ec064e2c8 : SUCCESS in 14m 16s","accounts_in_message":[],"_revision_number":23},{"id":"4d792598f6d62624608e2d16f1f3210dae630ec2","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-11-26 10:25:20.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"7c2af75f6e9f573344c503ad2dc2a55f9b6e161b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-11-26 10:29:32.000000000","message":"Patch Set 23: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":23},{"id":"fbbc516919cd5c0b48a6be3ce472b7f16b784567","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-26 10:30:14.000000000","message":"Patch Set 23: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":23},{"id":"a7036792c5281c04159ccb5066bb7e458ed4ba31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-26 12:20:24.000000000","message":"Patch Set 23: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2f81d67e2eb1445697993626504c2973 : SUCCESS in 2m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d154c38305de45d0927cfce603da3ee1 : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a9e4f68f984427f9afa3c9a3d7201b2 : SUCCESS in 6m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/72eb74caf0d8461b9731a08e9d74e9d6 : SUCCESS in 6m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fb31e58375e4447a0e11806965bf2cd : SUCCESS in 9m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/582f2a633cf44ea9ad6d2ee12b3de534 : SUCCESS in 5m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ca010c413fd04b2f94db03c8a8b2630e : SUCCESS in 10m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b485c0cfb9724186991f35783b820bfa : SUCCESS in 29m 07s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/f9ee7855405e42659060bf3dea01f465 : SUCCESS in 1h 08m 00s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a20b921c45b9405284cf614ed9fd2bc2 : FAILURE in 31m 43s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/b704e45dcdfe41148cd9f7225fa5e0a4 : SUCCESS in 1h 11m 07s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/71bfb687373c4cd187d79db8bf7987c6 : SUCCESS in 18m 52s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7fd63c651c6649509b8af0b328d68ee8 : SUCCESS in 16m 25s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0f343989779b4914b0722677b174de05 : SUCCESS in 1h 49m 24s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b3f7f554d3b4aff8e1e1ae58547e297 : SUCCESS in 1h 07m 27s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/592fa901c16b49f98a0ea359391e3953 : SUCCESS in 45m 38s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3d62ea8405d84e499ef5dc0475b6de6e : SUCCESS in 28m 55s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/05a6ba812bfa495b853dd122b762d5c9 : SUCCESS in 27m 43s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6db026afb4da4ed48b7c88cbbb51aa9c : SUCCESS in 22m 39s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/c043b4e9b7574789a5d4da1d19c4577e : SUCCESS in 17m 45s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c6c0d23d92cc47d7a62f45c9e482cf24 : SUCCESS in 12m 25s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/65780a8e0d384e97b039942572372cae : SUCCESS in 15m 45s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5487b1736ff24b13a7c7e57d19754fbd : SUCCESS in 12m 20s","accounts_in_message":[],"_revision_number":23},{"id":"5f312a224ad8333027a1dea732c713a8e1a4deb0","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2021-11-26 13:28:25.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"06feb1c87621c75e4c026c5992c0e01ef15ab9d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-26 15:37:34.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d33d5de5b8584fbe99f2a81a4ea5f000 : SUCCESS in 4m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b35859cec020401fa4727f176db0864b : SUCCESS in 7m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95d203ad9e3d4b97bf84d1b97bf3e340 : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8043294d81bc4519a07ceb0653f19450 : SUCCESS in 6m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8242b4dd333849e38334b281e4ebcac5 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cbdb7b6d69d4e09aea76fcafbdc420c : SUCCESS in 9m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6c29d3aeace74d3e9e2dec7a1bc87175 : SUCCESS in 6m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a63e7f41370c4b74b5dbaad020d27fc1 : SUCCESS in 13m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0ca399420faf465eb4d2b2c01865b084 : SUCCESS in 30m 45s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/99d35da5769e4056a279a8498ff4988a : FAILURE in 1h 50m 09s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/88528d092ca34bbb863f9f9011888827 : SUCCESS in 52m 15s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c18ccff57aa54fc39cbe2efe1ed4a80a : SUCCESS in 1h 07m 19s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/499c55ab99c0400b99814df9580522b9 : SUCCESS in 18m 54s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/44c24eb5e7074ad3b427007fb7cb063f : SUCCESS in 15m 54s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c9136dfbd1e245868be7a737ad74a224 : SUCCESS in 1h 51m 31s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce1fefb89aa54c6f95ee82b4dab86f18 : SUCCESS in 1h 22m 50s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/ab66765de38a4183992099605f21ace1 : SUCCESS in 49m 43s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7770e1d37b0949ecaf6c6085989cff9c : SUCCESS in 34m 38s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0809f8bc217c485e9b2c3a55bc6c45eb : SUCCESS in 29m 34s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6647f9ac687346eb81452867f93a13c6 : SUCCESS in 24m 20s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/e1754b9525a44688b9bbfec6ba7d8c46 : SUCCESS in 15m 40s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7d18857f995a4ea1a4da8eb584923f65 : SUCCESS in 13m 42s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/5abaa39fbce94eada99c181e7dd74007 : SUCCESS in 17m 01s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52f8cbf56cb14c02943d7af9c8e40e81 : SUCCESS in 14m 27s","accounts_in_message":[],"_revision_number":23},{"id":"8132409a8cd765e24bcf505b8fe82765259c72f7","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2021-11-26 19:58:36.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"ada509fde1b26dc9812b8896f5c133e366e4286a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-26 21:29:26.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c109b275b72b45298151bf72aaa3a92a : SUCCESS in 2m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b3cbfedde17546179cfe20d8edd0482c : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91d2dd8692334baeb37c6580dde739e9 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b07fdd9ca774aafb94f73d3e3e7b661 : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6c0cde9137b1499eaa5c9ca114582fb0 : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/001d4f534e72437caaf65eec3ed2aee6 : SUCCESS in 7m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/77cfb73c61004cbbb358cb05b59ee4ec : SUCCESS in 6m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/72fd0fe3093b424d99acf99506b11635 : SUCCESS in 9m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d5cb5372b47047659415e2594ff26a94 : SUCCESS in 30m 04s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/e00b4396eeab48068762a05d864216ac : SUCCESS in 54m 41s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6337a6cd13fa43b38d87320128fb3e05 : SUCCESS in 49m 07s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/0037880097f340cca0bf3f32f0a687fe : SUCCESS in 51m 15s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/56c1b2c58e584b85baad8492f72f3db6 : SUCCESS in 18m 01s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c450a269323d44ad8f11db4428f88058 : SUCCESS in 16m 43s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9a554559be654cd3ade938699316239a : SUCCESS in 1h 29m 24s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/45c034cd04b44d648a9af16f74a7ab90 : SUCCESS in 1h 15m 12s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/21c71d62a86f441187e21806193bf4e8 : SUCCESS in 49m 29s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a6a6710aafe947faa3305958f2fd2a58 : SUCCESS in 36m 51s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7efdce6a34e440e1a34e45e8d850d87e : SUCCESS in 33m 14s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/81da5540d9ed460da854bd684748e6bd : SUCCESS in 26m 06s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/b0609fd1b35b4bd4aebd8abcae989769 : SUCCESS in 15m 35s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3e3882122d39414aaaf94eee1478721d : SUCCESS in 10m 03s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/4c5727367a134acb95cabed549720b83 : SUCCESS in 14m 06s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/93f6eabe889746808703a1d5aebcbd12 : SUCCESS in 10m 47s","accounts_in_message":[],"_revision_number":23},{"id":"7075f5f955e963731e9efec056a2ea5389db2380","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-26 21:30:01.000000000","message":"Patch Set 23: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":23},{"id":"e4a73e7033f52197684e37a26619c219937cf78f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-26 22:57:09.000000000","message":"Patch Set 23: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e6d3d0370c3f4bcab1b60906fca2435f : SUCCESS in 2m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/802bd76ad8cf48018a91a638b2e2bfbe : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e1c87f3310f94f318162606c1003f59e : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/145b4f5d1ec74b77b97fa17ab9cfd9d1 : SUCCESS in 6m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58d97e1286bf49e9ba666b1b6b6024dc : SUCCESS in 7m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/11a46bebd1de427f974da8e91a236d11 : SUCCESS in 5m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6cf73198fa134bb4ae57f36c541b859b : SUCCESS in 12m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/23231017188049468dc0960b6d76ca70 : SUCCESS in 25m 32s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c9371d5274114fe9aedc08a73ea98ccd : SUCCESS in 1h 02m 53s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e2b9a073ecf546fb8c7f52201e5216ef : SUCCESS in 37m 07s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/5ad4bfb492a34a969e1cc4765ab94960 : SUCCESS in 56m 05s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/4154764ab3c94f3b8184b29b9bc4e699 : SUCCESS in 19m 27s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/31edbc65dc8c4f19ace758248686977a : SUCCESS in 13m 16s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b3de056d51764255a4a26d39b6cb9fcd : SUCCESS in 1h 25m 27s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/939f38289f6c465c812e7be0fd742a45 : SUCCESS in 59m 09s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/f2bc6692a3fa4c999c247624a49f3619 : SUCCESS in 46m 01s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/135c58b34c344191bd8dd12779c07dda : SUCCESS in 44m 47s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9d81f2f36adf46bf81028c94938f968d : SUCCESS in 24m 37s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/259355cfaece4f799893d5bc4d9d5e07 : SUCCESS in 24m 05s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/47a79177d8714fa5bf7af3f986bb3660 : SUCCESS in 11m 58s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0588a4354ef54a2f923ffa8149fd2228 : SUCCESS in 10m 46s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/8398c68f9d024f868b09e62860ca3554 : SUCCESS in 14m 28s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/71769805d56c479390492265fc27b324 : SUCCESS in 11m 11s","accounts_in_message":[],"_revision_number":23},{"id":"e32299c7db5f03fed60fab7d4ee1090d0348d3ee","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-26 22:57:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":23},{"id":"fc629c1c1c48fcc1423aaaaa837d75fcca2981d3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-26 22:58:32.000000000","message":"Patch Set 23:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0ab4b8bdea04a9cab9d8a663f86ec99 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":23},{"id":"cd12ebba4a573dfca0a09642aa34d869cdbc176e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-11-29 09:08:16.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"5afeef1ee2a9ff1da8482a992285788e140afcfa","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-11-29 10:51:43.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"67b68aee2e9f3c719067b972dc0e79387595c342","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-11-29 11:51:12.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23}],"current_revision_number":23,"current_revision":"c35f112a14d5e5bd49bd49429179b49b5ae0da40","revisions":{"4e541c07ff13b3d9ab85247d186294b8583e3730":{"kind":"REWORK","_number":1,"created":"2021-01-26 22:48:19.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/09/772609/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/772609/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/772609/1"}}},"commit":{"parents":[{"commit":"ff7c1741d86c580dcd2a68cc3c49bb047d61455b","subject":"Merge \"Add variables for API VIP address and FQDN\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ff7c1741d86c580dcd2a68cc3c49bb047d61455b"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-01-26 22:47:52.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-01-26 22:47:52.000000000","tz":60},"subject":"[DNM] Work in progress of multiple host image support","message":"[DNM] Work in progress of multiple host image support\n\nChange-Id: I93d242889e225b4e60254f6b9cc5eeb457294ac8\nStory: 2002098\nTask: 41693\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4e541c07ff13b3d9ab85247d186294b8583e3730"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4e541c07ff13b3d9ab85247d186294b8583e3730"}]},"branch":"refs/heads/master"},"d641b1a3565d8947e494edd436a59734ceb3cffb":{"kind":"REWORK","_number":2,"created":"2021-02-11 21:20:10.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/09/772609/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/772609/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/772609/2"}}},"commit":{"parents":[{"commit":"38409a7c1ff06a5b335d8ea1fa03ec2c1ca19eea","subject":"Merge \"Remove trailing whitespace\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/38409a7c1ff06a5b335d8ea1fa03ec2c1ca19eea"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-11 21:15:19.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-11 21:15:19.000000000","tz":60},"subject":"[DNM] Work in progress of multiple host image support","message":"[DNM] Work in progress of multiple host image support\n\nChange-Id: I93d242889e225b4e60254f6b9cc5eeb457294ac8\nStory: 2002098\nTask: 41693\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d641b1a3565d8947e494edd436a59734ceb3cffb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d641b1a3565d8947e494edd436a59734ceb3cffb"}]},"branch":"refs/heads/master"},"d3a099106fff415766395013aaa2fbb973521d72":{"kind":"REWORK","_number":3,"created":"2021-07-30 09:42:14.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/09/772609/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/772609/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/772609/3"}}},"commit":{"parents":[{"commit":"da304da58eb184546f10083990b08005bafcd392","subject":"Merge \"Fix capitalisation of Ansible task name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/da304da58eb184546f10083990b08005bafcd392"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-07-30 09:42:08.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-07-30 09:42:08.000000000","tz":120},"subject":"[WIP] Build overcloud host image directly with DIB","message":"[WIP] Build overcloud host image directly with DIB\n\nAs a first step towards supporting multiple overcloud disk images, this\nchange introduces a new command to build a disk image directly with DIB:\n`kayobe overcloud host image build`.\n\nIt also disables building a root disk image during Bifrost bootstrap.\n\nChange-Id: I93d242889e225b4e60254f6b9cc5eeb457294ac8\nStory: 2002098\nTask: 41693\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d3a099106fff415766395013aaa2fbb973521d72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d3a099106fff415766395013aaa2fbb973521d72"}]},"branch":"refs/heads/master"},"83206e72e446afe354fdb6ba4d817372e84308b0":{"kind":"REWORK","_number":4,"created":"2021-08-30 15:14:22.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/09/772609/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/772609/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/772609/4"}}},"commit":{"parents":[{"commit":"37524ff2a67c2619eaea3934a0635f09232326d6","subject":"Merge \"Fix documentation links for routed control plane networks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/37524ff2a67c2619eaea3934a0635f09232326d6"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-08-30 15:14:07.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-08-30 15:14:07.000000000","tz":120},"subject":"[WIP] Build overcloud host image directly with DIB","message":"[WIP] Build overcloud host image directly with DIB\n\nAs a first step towards supporting multiple overcloud disk images, this\nchange introduces a new command to build a disk image directly with DIB:\n`kayobe overcloud host image build`.\n\nIt also disables building a root disk image during Bifrost bootstrap.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/806560\nChange-Id: I93d242889e225b4e60254f6b9cc5eeb457294ac8\nStory: 2002098\nTask: 41693\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/83206e72e446afe354fdb6ba4d817372e84308b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/83206e72e446afe354fdb6ba4d817372e84308b0"}]},"branch":"refs/heads/master"},"0d92c35c6d872e0a1124f6bbd898c7517180639c":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":5,"created":"2021-09-02 15:35:02.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/09/772609/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/772609/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/772609/5"}}},"commit":{"parents":[{"commit":"a09742b1b9873397b38f3f720a6ad8568803d57e","subject":"Allow to configure name of root disk image to provision","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a09742b1b9873397b38f3f720a6ad8568803d57e"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-02 15:34:50.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-02 15:34:50.000000000","tz":120},"subject":"[WIP] Build overcloud host image directly with DIB","message":"[WIP] Build overcloud host image directly with DIB\n\nAs a first step towards supporting multiple overcloud disk images, this\nchange introduces a new command to build a disk image directly with DIB:\n`kayobe overcloud host image build`.\n\nIt also disables building a root disk image during Bifrost bootstrap.\n\nChange-Id: I93d242889e225b4e60254f6b9cc5eeb457294ac8\nStory: 2002098\nTask: 41693\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0d92c35c6d872e0a1124f6bbd898c7517180639c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0d92c35c6d872e0a1124f6bbd898c7517180639c"}]},"branch":"refs/heads/master"},"25335b4f7391b881d3ab1c46dbd5d5753b865953":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2021-09-05 20:43:34.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/09/772609/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/772609/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/772609/6"}}},"commit":{"parents":[{"commit":"a8f3e13dd063292869b9b15df036428c2fa1b1a5","subject":"Allow to configure name of root disk image to provision","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a8f3e13dd063292869b9b15df036428c2fa1b1a5"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-02 15:34:50.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-05 20:43:22.000000000","tz":120},"subject":"[WIP] Build overcloud host image directly with DIB","message":"[WIP] Build overcloud host image directly with DIB\n\nAs a first step towards supporting multiple overcloud disk images, this\nchange introduces a new command to build a disk image directly with DIB:\n`kayobe overcloud host image build`.\n\nIt also disables building a root disk image during Bifrost bootstrap.\n\nChange-Id: I93d242889e225b4e60254f6b9cc5eeb457294ac8\nStory: 2002098\nTask: 41693\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/25335b4f7391b881d3ab1c46dbd5d5753b865953"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/25335b4f7391b881d3ab1c46dbd5d5753b865953"}]},"branch":"refs/heads/master"},"d5af8fd391595d61910c2e9749bf455574fa7622":{"kind":"REWORK","_number":7,"created":"2021-10-05 09:35:58.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/09/772609/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/772609/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/772609/7"}}},"commit":{"parents":[{"commit":"c583922c27eefc2fda6158f3bc46d4715bbcdf4d","subject":"Add support for infrastructure VMs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c583922c27eefc2fda6158f3bc46d4715bbcdf4d"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-05 09:35:01.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-05 09:35:01.000000000","tz":120},"subject":"Build overcloud host image directly with DIB","message":"Build overcloud host image directly with DIB\n\nAs a first step towards supporting multiple overcloud disk images, this\nchange introduces a new command to build a disk image directly with DIB:\n`kayobe overcloud host image build`.\n\nIt also disables building a root disk image during Bifrost bootstrap if\ndib_build_host_images is set to true.\n\nChange-Id: I93d242889e225b4e60254f6b9cc5eeb457294ac8\nStory: 2002098\nTask: 41693\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d5af8fd391595d61910c2e9749bf455574fa7622"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d5af8fd391595d61910c2e9749bf455574fa7622"}]},"branch":"refs/heads/master"},"d6835c82c5ce6520dd751f10ab7e0f72c1f4a021":{"kind":"REWORK","_number":8,"created":"2021-10-05 09:44:19.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/09/772609/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/772609/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/772609/8"}}},"commit":{"parents":[{"commit":"c583922c27eefc2fda6158f3bc46d4715bbcdf4d","subject":"Add support for infrastructure VMs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c583922c27eefc2fda6158f3bc46d4715bbcdf4d"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-05 09:44:10.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-05 09:44:10.000000000","tz":120},"subject":"Build overcloud host image directly with DIB","message":"Build overcloud host image directly with DIB\n\nAs a first step towards supporting multiple overcloud disk images, this\nchange introduces a new command to build a disk image directly with DIB:\n`kayobe overcloud host image build`.\n\nIt also disables building a root disk image during Bifrost bootstrap if\ndib_build_host_images is set to true.\n\nChange-Id: I93d242889e225b4e60254f6b9cc5eeb457294ac8\nStory: 2002098\nTask: 41693\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d6835c82c5ce6520dd751f10ab7e0f72c1f4a021"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d6835c82c5ce6520dd751f10ab7e0f72c1f4a021"}]},"branch":"refs/heads/master"},"e55357fb3742d14190792e0ef1328f959c0f2307":{"kind":"REWORK","_number":9,"created":"2021-10-05 12:17:16.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/09/772609/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/772609/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/772609/9"}}},"commit":{"parents":[{"commit":"c583922c27eefc2fda6158f3bc46d4715bbcdf4d","subject":"Add support for infrastructure VMs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c583922c27eefc2fda6158f3bc46d4715bbcdf4d"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-05 12:16:56.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-05 12:16:56.000000000","tz":120},"subject":"Build overcloud host image directly with DIB","message":"Build overcloud host image directly with DIB\n\nAs a first step towards supporting multiple overcloud disk images, this\nchange introduces a new command to build a disk image directly with DIB:\n`kayobe overcloud host image build`.\n\nIt also disables building a root disk image during Bifrost bootstrap if\ndib_build_host_images is set to true.\n\nChange-Id: I93d242889e225b4e60254f6b9cc5eeb457294ac8\nStory: 2002098\nTask: 41693\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e55357fb3742d14190792e0ef1328f959c0f2307"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e55357fb3742d14190792e0ef1328f959c0f2307"}]},"branch":"refs/heads/master"},"f879b0767ed5f834c31bf7c04e52fc27adaa45a1":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2021-10-05 15:14:23.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/09/772609/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/772609/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/772609/10"}}},"commit":{"parents":[{"commit":"79be93baf510940497f88f8b73c86022a811b6b8","subject":"Enable consistent network device naming","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/79be93baf510940497f88f8b73c86022a811b6b8"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-05 12:16:56.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-05 14:44:03.000000000","tz":120},"subject":"Build overcloud host image directly with DIB","message":"Build overcloud host image directly with DIB\n\nAs a first step towards supporting multiple overcloud disk images, this\nchange introduces a new command to build a disk image directly with DIB:\n`kayobe overcloud host image build`.\n\nIt also disables building a root disk image during Bifrost bootstrap if\ndib_build_host_images is set to true.\n\nChange-Id: I93d242889e225b4e60254f6b9cc5eeb457294ac8\nStory: 2002098\nTask: 41693\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f879b0767ed5f834c31bf7c04e52fc27adaa45a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f879b0767ed5f834c31bf7c04e52fc27adaa45a1"}]},"branch":"refs/heads/master"},"4ec5812b67b423597d531fba1ce9df4e9df5c0f2":{"kind":"REWORK","_number":11,"created":"2021-10-06 10:49:39.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/09/772609/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/772609/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/772609/11"}}},"commit":{"parents":[{"commit":"f11c73ff03ede87e4a3b8945f836edfe6bd0539d","subject":"Merge \"Support configuration of firewalld\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f11c73ff03ede87e4a3b8945f836edfe6bd0539d"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-06 10:49:23.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-06 10:49:23.000000000","tz":120},"subject":"Build overcloud host image directly with DIB","message":"Build overcloud host image directly with DIB\n\nAs a first step towards supporting multiple overcloud disk images, this\nchange introduces a new command to build a disk image directly with DIB:\n`kayobe overcloud host image build`.\n\nIt also disables building a root disk image during Bifrost bootstrap if\ndib_build_host_images is set to true.\n\nChange-Id: I93d242889e225b4e60254f6b9cc5eeb457294ac8\nStory: 2002098\nTask: 41693\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4ec5812b67b423597d531fba1ce9df4e9df5c0f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4ec5812b67b423597d531fba1ce9df4e9df5c0f2"}]},"branch":"refs/heads/master"},"bd5bdcc01513c7107a36b8fc9cf62d2d5f55cc93":{"kind":"REWORK","_number":12,"created":"2021-10-26 08:50:09.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/09/772609/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/772609/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/772609/12"}}},"commit":{"parents":[{"commit":"f11c73ff03ede87e4a3b8945f836edfe6bd0539d","subject":"Merge \"Support configuration of firewalld\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f11c73ff03ede87e4a3b8945f836edfe6bd0539d"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-06 10:49:23.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-10-26 08:50:06.000000000","tz":120},"subject":"Build overcloud host image directly with DIB","message":"Build overcloud host image directly with DIB\n\nAs a first step towards supporting multiple overcloud disk images, this\nchange introduces a new command to build a disk image directly with DIB:\n`kayobe overcloud host image build`.\n\nIt also disables building a root disk image during Bifrost bootstrap if\ndib_build_host_images is set to true.\n\nChange-Id: I93d242889e225b4e60254f6b9cc5eeb457294ac8\nStory: 2002098\nTask: 41693\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bd5bdcc01513c7107a36b8fc9cf62d2d5f55cc93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bd5bdcc01513c7107a36b8fc9cf62d2d5f55cc93"}]},"branch":"refs/heads/master"},"28166fa2cee788f0e76b1a04d13d88fa53b9a598":{"kind":"REWORK","_number":13,"created":"2021-10-26 10:31:42.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/09/772609/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/772609/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/772609/13"}}},"commit":{"parents":[{"commit":"bf440c5dd206699558ba009937d3bd846d04f783","subject":"Merge \"Support Ansible diff mode\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bf440c5dd206699558ba009937d3bd846d04f783"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-06 10:49:23.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-10-26 10:31:40.000000000","tz":120},"subject":"Build overcloud host image directly with DIB","message":"Build overcloud host image directly with DIB\n\nAs a first step towards supporting multiple overcloud disk images, this\nchange introduces a new command to build a disk image directly with DIB:\n`kayobe overcloud host image build`.\n\nIt also disables building a root disk image during Bifrost bootstrap if\ndib_build_host_images is set to true.\n\nChange-Id: I93d242889e225b4e60254f6b9cc5eeb457294ac8\nStory: 2002098\nTask: 41693\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/28166fa2cee788f0e76b1a04d13d88fa53b9a598"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/28166fa2cee788f0e76b1a04d13d88fa53b9a598"}]},"branch":"refs/heads/master"},"c1d7f0dfdc2ab502a6bddab8e0dfe1f68a20f705":{"kind":"NO_CODE_CHANGE","_number":14,"created":"2021-10-26 10:35:16.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/09/772609/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/772609/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/772609/14"}}},"commit":{"parents":[{"commit":"bf440c5dd206699558ba009937d3bd846d04f783","subject":"Merge \"Support Ansible diff mode\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bf440c5dd206699558ba009937d3bd846d04f783"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-06 10:49:23.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-10-26 10:35:16.000000000","tz":0},"subject":"Build overcloud host image directly with DIB","message":"Build overcloud host image directly with DIB\n\nAs a first step towards supporting multiple overcloud disk images, this\nchange introduces a new command to build a disk image directly with DIB:\n`kayobe overcloud host image build`.\n\nIt also disables building a root disk image during Bifrost bootstrap if\novercloud_dib_build_host_images is set to true.\n\nChange-Id: I93d242889e225b4e60254f6b9cc5eeb457294ac8\nStory: 2002098\nTask: 41693\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c1d7f0dfdc2ab502a6bddab8e0dfe1f68a20f705"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c1d7f0dfdc2ab502a6bddab8e0dfe1f68a20f705"}]},"branch":"refs/heads/master","description":"Edit commit message"},"b911bad745caee11115a4a9e2b9a6e70f1b112ef":{"kind":"REWORK","_number":15,"created":"2021-10-27 13:49:26.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/09/772609/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/772609/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/772609/15"}}},"commit":{"parents":[{"commit":"bf440c5dd206699558ba009937d3bd846d04f783","subject":"Merge \"Support Ansible diff mode\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bf440c5dd206699558ba009937d3bd846d04f783"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-06 10:49:23.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-10-27 13:49:26.000000000","tz":0},"subject":"Build overcloud host image directly with DIB","message":"Build overcloud host image directly with DIB\n\nAs a first step towards supporting multiple overcloud disk images, this\nchange introduces a new command to build a disk image directly with DIB:\n`kayobe overcloud host image build`.\n\nIt also disables building a root disk image during Bifrost bootstrap if\novercloud_dib_build_host_images is set to true.\n\nChange-Id: I93d242889e225b4e60254f6b9cc5eeb457294ac8\nStory: 2002098\nTask: 41693\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b911bad745caee11115a4a9e2b9a6e70f1b112ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b911bad745caee11115a4a9e2b9a6e70f1b112ef"}]},"branch":"refs/heads/master"},"a2e4e9be498baf5e10d4f66ea29ec4563116f6e7":{"kind":"REWORK","_number":16,"created":"2021-10-28 11:59:13.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/09/772609/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/772609/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/772609/16"}}},"commit":{"parents":[{"commit":"bf440c5dd206699558ba009937d3bd846d04f783","subject":"Merge \"Support Ansible diff mode\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bf440c5dd206699558ba009937d3bd846d04f783"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-06 10:49:23.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-10-28 11:59:12.000000000","tz":0},"subject":"Build overcloud host image directly with DIB","message":"Build overcloud host image directly with DIB\n\nAs a first step towards supporting multiple overcloud disk images, this\nchange introduces a new command to build a disk image directly with DIB:\n`kayobe overcloud host image build`.\n\nIt also disables building a root disk image during Bifrost bootstrap if\novercloud_dib_build_host_images is set to true.\n\nChange-Id: I93d242889e225b4e60254f6b9cc5eeb457294ac8\nStory: 2002098\nTask: 41693\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a2e4e9be498baf5e10d4f66ea29ec4563116f6e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a2e4e9be498baf5e10d4f66ea29ec4563116f6e7"}]},"branch":"refs/heads/master"},"41a960aad8b4ec2bd941e2e80a36812cc25f1b1e":{"kind":"REWORK","_number":17,"created":"2021-10-29 09:56:03.000000000","uploader":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"ref":"refs/changes/09/772609/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/772609/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/772609/17"}}},"commit":{"parents":[{"commit":"bf440c5dd206699558ba009937d3bd846d04f783","subject":"Merge \"Support Ansible diff mode\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bf440c5dd206699558ba009937d3bd846d04f783"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-06 10:49:23.000000000","tz":120},"committer":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2021-10-29 09:54:59.000000000","tz":120},"subject":"Build overcloud host image directly with DIB","message":"Build overcloud host image directly with DIB\n\nAs a first step towards supporting multiple overcloud disk images, this\nchange introduces a new command to build a disk image directly with DIB:\n`kayobe overcloud host image build`.\n\nIt also disables building a root disk image during Bifrost bootstrap if\novercloud_dib_build_host_images is set to true.\n\nChange-Id: I93d242889e225b4e60254f6b9cc5eeb457294ac8\nStory: 2002098\nTask: 41693\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/41a960aad8b4ec2bd941e2e80a36812cc25f1b1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/41a960aad8b4ec2bd941e2e80a36812cc25f1b1e"}]},"branch":"refs/heads/master"},"739a5e4d73b1b5fa880cffc9db192ae43084ff00":{"kind":"REWORK","_number":18,"created":"2021-11-04 09:59:44.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/09/772609/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/772609/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/772609/18"}}},"commit":{"parents":[{"commit":"97a7dc6b229bbaad3e693cf4c2ffcc311058e031","subject":"Merge \"Fix --check argument for overcloud host configure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/97a7dc6b229bbaad3e693cf4c2ffcc311058e031"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-11-04 09:59:08.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-11-04 09:59:08.000000000","tz":60},"subject":"Build overcloud host image directly with DIB","message":"Build overcloud host image directly with DIB\n\nAs a first step towards supporting multiple overcloud disk images, this\nchange introduces a new command to build a disk image directly with DIB:\n`kayobe overcloud host image build`.\n\nIt also disables building a root disk image during Bifrost bootstrap if\novercloud_dib_build_host_images is set to true.\n\nChange-Id: I93d242889e225b4e60254f6b9cc5eeb457294ac8\nStory: 2002098\nTask: 41693\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/739a5e4d73b1b5fa880cffc9db192ae43084ff00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/739a5e4d73b1b5fa880cffc9db192ae43084ff00"}]},"branch":"refs/heads/master"},"bb702aa0b0bf67dfc0da8161937d81a54bfb7853":{"kind":"REWORK","_number":19,"created":"2021-11-24 16:08:15.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/09/772609/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/772609/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/772609/19"}}},"commit":{"parents":[{"commit":"52d7523c0c3e6a6f7f321459b9d7fc055282057f","subject":"Merge \"CI: always return host configure test results\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/52d7523c0c3e6a6f7f321459b9d7fc055282057f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-11-24 16:08:06.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-11-24 16:08:06.000000000","tz":60},"subject":"Build overcloud host image directly with DIB","message":"Build overcloud host image directly with DIB\n\nAs a first step towards supporting multiple overcloud disk images, this\nchange introduces a new command to build a disk image directly with DIB:\n`kayobe overcloud host image build`.\n\nIt also disables building a root disk image during Bifrost bootstrap if\novercloud_dib_build_host_images is set to true.\n\nChange-Id: I93d242889e225b4e60254f6b9cc5eeb457294ac8\nStory: 2002098\nTask: 41693\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bb702aa0b0bf67dfc0da8161937d81a54bfb7853"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bb702aa0b0bf67dfc0da8161937d81a54bfb7853"}]},"branch":"refs/heads/master"},"8fa338c3e2577c62715fd22f25c0af0da9a16d75":{"kind":"REWORK","_number":20,"created":"2021-11-24 16:21:08.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/09/772609/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/772609/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/772609/20"}}},"commit":{"parents":[{"commit":"52d7523c0c3e6a6f7f321459b9d7fc055282057f","subject":"Merge \"CI: always return host configure test results\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/52d7523c0c3e6a6f7f321459b9d7fc055282057f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-11-24 16:21:01.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-11-24 16:21:01.000000000","tz":60},"subject":"Build overcloud host image directly with DIB","message":"Build overcloud host image directly with DIB\n\nAs a first step towards supporting multiple overcloud disk images, this\nchange introduces a new command to build a disk image directly with DIB:\n`kayobe overcloud host image build`.\n\nIt also disables building a root disk image during Bifrost bootstrap if\novercloud_dib_build_host_images is set to true.\n\nChange-Id: I93d242889e225b4e60254f6b9cc5eeb457294ac8\nStory: 2002098\nTask: 41693\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8fa338c3e2577c62715fd22f25c0af0da9a16d75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8fa338c3e2577c62715fd22f25c0af0da9a16d75"}]},"branch":"refs/heads/master"},"e8178b068692f4b7a4cbe212887c1482b9a63f07":{"kind":"REWORK","_number":21,"created":"2021-11-24 17:23:43.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/09/772609/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/772609/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/772609/21"}}},"commit":{"parents":[{"commit":"52d7523c0c3e6a6f7f321459b9d7fc055282057f","subject":"Merge \"CI: always return host configure test results\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/52d7523c0c3e6a6f7f321459b9d7fc055282057f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-11-24 17:23:35.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-11-24 17:23:35.000000000","tz":60},"subject":"Build overcloud host image directly with DIB","message":"Build overcloud host image directly with DIB\n\nAs a first step towards supporting multiple overcloud disk images, this\nchange introduces a new command to build a disk image directly with DIB:\n`kayobe overcloud host image build`.\n\nIt also disables building a root disk image during Bifrost bootstrap if\novercloud_dib_build_host_images is set to true.\n\nChange-Id: I93d242889e225b4e60254f6b9cc5eeb457294ac8\nStory: 2002098\nTask: 41693\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e8178b068692f4b7a4cbe212887c1482b9a63f07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e8178b068692f4b7a4cbe212887c1482b9a63f07"}]},"branch":"refs/heads/master"},"bc91239aedb36fa4a7401db0a765ab00a97614c3":{"kind":"REWORK","_number":22,"created":"2021-11-25 11:59:37.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/09/772609/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/772609/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/772609/22"}}},"commit":{"parents":[{"commit":"52d7523c0c3e6a6f7f321459b9d7fc055282057f","subject":"Merge \"CI: always return host configure test results\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/52d7523c0c3e6a6f7f321459b9d7fc055282057f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-11-25 11:59:31.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-11-25 11:59:31.000000000","tz":60},"subject":"Build overcloud host image directly with DIB","message":"Build overcloud host image directly with DIB\n\nAs a first step towards supporting multiple overcloud disk images, this\nchange introduces a new command to build a disk image directly with DIB:\n`kayobe overcloud host image build`.\n\nIt also disables building a root disk image during Bifrost bootstrap if\novercloud_dib_build_host_images is set to true.\n\nChange-Id: I93d242889e225b4e60254f6b9cc5eeb457294ac8\nStory: 2002098\nTask: 41693\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bc91239aedb36fa4a7401db0a765ab00a97614c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bc91239aedb36fa4a7401db0a765ab00a97614c3"}]},"branch":"refs/heads/master"},"c35f112a14d5e5bd49bd49429179b49b5ae0da40":{"kind":"REWORK","_number":23,"created":"2021-11-26 08:28:10.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/09/772609/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/772609/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/772609/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/772609/23"}}},"commit":{"parents":[{"commit":"52d7523c0c3e6a6f7f321459b9d7fc055282057f","subject":"Merge \"CI: always return host configure test results\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/52d7523c0c3e6a6f7f321459b9d7fc055282057f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-11-26 08:28:00.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-11-26 08:28:00.000000000","tz":60},"subject":"Build overcloud host image directly with DIB","message":"Build overcloud host image directly with DIB\n\nAs a first step towards supporting multiple overcloud disk images, this\nchange introduces a new command to build a disk image directly with DIB:\n`kayobe overcloud host image build`.\n\nIt also disables building a root disk image during Bifrost bootstrap if\novercloud_dib_build_host_images is set to true.\n\nChange-Id: I93d242889e225b4e60254f6b9cc5eeb457294ac8\nStory: 2002098\nTask: 41693\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c35f112a14d5e5bd49bd49429179b49b5ae0da40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c35f112a14d5e5bd49bd49429179b49b5ae0da40"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
