)]}'
{"id":"openstack%2Fkayobe~772777","triplet_id":"openstack%2Fkayobe~stable%2Fvictoria~I5efdbd52556721914fe69d7c6ba454b2c721b643","project":"openstack/kayobe","branch":"stable/victoria","topic":"story/2008550-stable/victoria","hashtags":[],"change_id":"I5efdbd52556721914fe69d7c6ba454b2c721b643","subject":"Fix deployment image build in Victoria","status":"MERGED","created":"2021-01-27 22:00:06.000000000","updated":"2021-02-01 01:15:13.000000000","submitted":"2021-02-01 01:14:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"772777-story/2008550-stable/victoria","meta_rev_id":"0af835d77362ebd5f0a132eddcd08421b0203ca3","_number":772777,"virtual_id_number":772777,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-01 01:14:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2021-01-28 15:55:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2021-01-28 09:31:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2021-01-28 15:55:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-28 03:00:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-28 09:31:39.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-01-28 15:54:37.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"bf65411d219a2390be493fc2a365651dde7f3823","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-27 22:00:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b3c8cf4acf91e4ed2360b7d7be56dfd387120670","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 03:00:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f6f7188a548749c3ad9e690da3e7aa95 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72aa27a341784b0088e635f6bfe743f9 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e998fea6a9549b7b83fca7affab8436 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/deffc812d5894b8fa4dac63d9fc5dbd5 : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf505cd23529458bb497235db63a75af : SUCCESS in 5m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/89d60f3d35d14f7286b6171c62122ffb : SUCCESS in 4m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f6555846edf245808635b456e85bdc16 : SUCCESS in 8m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f117e5ea88194c03a40c5000d0026bb9 : SUCCESS in 17m 36s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/5e29867e12b6497780f44bf70ae64aee : SUCCESS in 1h 02m 53s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/14f64da0f47c423f934980f6862de6f2 : SUCCESS in 1h 05m 18s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/e1711f49a18d4599b6a8e4961a977a54 : SUCCESS in 26m 34s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/865b210c79254876b8a5b8710f2a862c : SUCCESS in 1h 40m 00s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/1dbd1b399c1240eeabd55e0f56b5e310 : SUCCESS in 17m 29s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/d3fbcbaa01da45728f26b8ab5da85e43 : SUCCESS in 26m 27s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/4cdd5760101e42249235c5272f6b142b : SUCCESS in 12m 51s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"0cdac8ca4bf0488f78f83471b6e392a5e30763a6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-28 09:31:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6e289fa57a0c4c22b9628d7e948bcecc5598d4f5","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-01-28 15:55:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"acc7c23f2d6feff329bfbab18e4d8cf10c90862f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 15:56:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"10a32ef9db8d6d249875eeee122765e4a9829b3e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 20:33:43.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a7a0b39b02e41b08592c344970e8c63 : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5bc2935e760e46ef97b5f06f6c58b973 : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b386cc279ec4fada83c0e24cdb24337 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9f8f598caf54901a932a7173869a020 : SUCCESS in 5m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e72369adeba84e10b70331bb9f3f30f4 : SUCCESS in 4m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8533d7f02e024e608a808ed96ab49ea5 : SUCCESS in 7m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f58ae4725cee408ea05dbb3957a3b9e7 : SUCCESS in 16m 36s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/ec3b78b3777e4a2e81edfbf085d3398e : TIMED_OUT in 2h 01m 46s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/0c2375793d684b198c1982d290ba9a4f : SUCCESS in 57m 26s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/e4f3156019cc4582a449e124c325a757 : SUCCESS in 14m 08s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/15bc62cd642b409cb24104ed4cdfeb4c : RETRY_LIMIT in 1h 21m 18s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/cba6fd85ef224455be54f041325831d3 : SUCCESS in 16m 10s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a1a0365899464a55ae1a71ef1509397c : FAILURE in 10m 06s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/e4b2aa46f3954d629f728becaf25c63b : SUCCESS in 9m 39s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"f4e0dfb29deba5c6d2a1f1a8d102ff7910048e92","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-28 21:44:22.000000000","message":"Patch Set 1:\n\nFailed because of Docker Hub limits","accounts_in_message":[],"_revision_number":1},{"id":"ac21a5a4bd7f578b14b94d559f27bbe364fa0148","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-29 09:23:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"341abe7e95ebe4d2450fccd5ee8f2c0e5f7eea8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 12:26:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a9364b766d4d42c5815dc35b9a5f40d4 : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/163a15b74c5841849090b0e3c256ee06 : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/854f35fb5604482c9ed4faee8888c75e : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1bead13b4f7e41db985f77ace4290f60 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b156f3192a25450b954f62a81c0c7836 : SUCCESS in 13m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/233d3132e6c1404ab262c2cdd46af9ee : SUCCESS in 5m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/859d24a80ff549d99c8a5d53d690c0f7 : SUCCESS in 9m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/152b57a610ba4ca89207174e6f1b368f : SUCCESS in 18m 52s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/8a1826989f324bada89c4b42d9082cfc : SUCCESS in 58m 08s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/49112bc76ce24b8c9d3404631a99c2de : SUCCESS in 1h 00m 40s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/5fd6a8a455b8406caf08ff3fe1606791 : SUCCESS in 14m 24s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/bf59017576d74bbf912d68f2dad7a6e0 : SUCCESS in 1h 43m 13s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/5d25a77b11b54d64b58a013a057036ed : SUCCESS in 14m 41s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e84f35ef136b4d71bbda924d1ed511ad : SUCCESS in 23m 17s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/2da3ec292f8340b8b73554fc76580d1a : SUCCESS in 9m 26s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"333e5b3a21ca3f22c85f62f6764e87fd0a2e94ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 12:26:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1ad5e8d250d6ef911e737fa3b013c227e6bdfc69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 14:20:39.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5c7f258e41844268dce78971a00986c : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2cc903dc24364297a5f1786b1115083a : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bdc25d6ef913490ead110fed7c53593b : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/587bfa37ef5d47bca1c2922873b6dd9d : SUCCESS in 6m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0c06362769ec41ecb8ece48e8bd04bf1 : SUCCESS in 4m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1efe55275cd94553b4fd0bb3f99bccef : SUCCESS in 9m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/52e548938afd40b68dc89ab8c62d53b6 : SUCCESS in 18m 32s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/f2fa3e72c01e4928bd4c791dc411e6ac : SUCCESS in 1h 00m 01s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/93deeeb18306458d822189b222bc6158 : SUCCESS in 1h 02m 33s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/4b09451a6702486f91509782f428e479 : SUCCESS in 16m 41s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/33c52bf7a0ac4e6e8b991e1dc1e9305d : SUCCESS in 1h 52m 13s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/80942f4fd30d49e6946029915cfcbd79 : SUCCESS in 13m 29s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/fa4e34f472634be7ba58f0b28d21b165 : FAILURE in 25m 23s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/fd9b5588e13643d6be7c188f12be5207 : SUCCESS in 9m 57s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"b36ffbc924595783a7cf593cfd188a9a02e06718","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-29 16:02:58.000000000","message":"Patch Set 1:\n\nrecheck\n\nCirros image download failed","accounts_in_message":[],"_revision_number":1},{"id":"ae3df77adcd23653c4283e56b13bad0500bfbea9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 17:29:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3343083b3c074320b5e0fe2dc91a7ffa : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39f8f958bbf64c699fbf371a73c50179 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f83516966a344aadaaedce4d1354d893 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/efbd3f821d8344259aaaae73510b2c92 : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22bdfaa00b6b4a1a9f2320f6ebf0684b : SUCCESS in 5m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b24ed86487b34cc3a3122b2cf4d5db6b : SUCCESS in 4m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/72abbe7cdcc4422385773b8e148aaef7 : SUCCESS in 8m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e4463c03e0d14b11bf4a939eb9520472 : SUCCESS in 18m 56s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/8ff30b6274a34a58a4d30a3f544f7163 : SUCCESS in 57m 52s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/2e1c18c7228340339a4b7955221d52b4 : SUCCESS in 1h 07m 27s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/a55e98272415481988024985331dd153 : SUCCESS in 18m 08s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/574a71b6731e46bdae20b0f9bb37f811 : POST_FAILURE in 1h 19m 29s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/7d23344e222c4c5c802eeb48a5812ba6 : SUCCESS in 14m 24s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/784f52c01d9049168297a7c2691a1826 : SUCCESS in 27m 50s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/35d055df881a4da58d2261ef17608776 : SUCCESS in 9m 59s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"613f58039e8d6e9cb0f7dd92bb36d43a2bc02464","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-29 18:30:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3a218dfa516d4236fb70b43eafc6fa5e8e659abf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 21:36:35.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa83d8a1f78d42f0a415ee139da09433 : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f95083b919649718ea3161279dded3f : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/38530df5470f4723b60d08974f589087 : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a6b91f6d839f4c18a067aaef9961b1fb : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67588a11c3314e53a99ec5491859fcc6 : SUCCESS in 6m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/95f95dc43a124c1091783d575f60f948 : SUCCESS in 4m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2439d280f53749bf83b95def6f27eb99 : SUCCESS in 8m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/90b49f2217264b1091a6822c31c888eb : SUCCESS in 19m 36s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/62b29cc8b50549658cff7df8ed8b5268 : SUCCESS in 1h 00m 32s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/0c168f57e1344efcb61f76d303fc1a6c : SUCCESS in 56m 50s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/d75af4a4256e4eaf81e8243f0e79343b : SUCCESS in 13m 57s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/2802773649af4183bd5b631cb05d447a : FAILURE in 1h 44m 28s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/4048a01c858944139070e72265e75a1c : SUCCESS in 13m 34s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/ebd611e093ee4c2e951f5d0718651932 : SUCCESS in 22m 21s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/72ee99f4d5244acb8beb85ae9f582a71 : SUCCESS in 9m 29s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"306dfcace4ba81799c62655a87530ec5cb7682a8","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-29 22:45:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b0ad92daec0ffb48edd904311f1949046894abf3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-30 00:33:50.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/029f749c703c4d0a8441bf7012de6605 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d59a82e9cb7647419d9e7b46e80b6faf : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7fb6e1d8e146482899bde8ce87f2e71a : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77fe6e9445b14f7f82aab2e6b8b09dcd : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa7743557ff54e4988bc020a0171657c : SUCCESS in 7m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b5d69423a64f48fd986e4ea80c3d7638 : SUCCESS in 4m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3b81a7157b6a424dba54fa6a797924e6 : SUCCESS in 8m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2f2dfa7c3b864a4dbba75066d8bea9d8 : SUCCESS in 18m 58s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/6fbc5d98c38a4bcb978cfb812c480d24 : SUCCESS in 1h 01m 34s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/f59a17b01afc4dd4ac0f739d5f3cb2cc : SUCCESS in 1h 03m 07s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/24f96fbc2b9044e183dfa1a3d843a429 : SUCCESS in 17m 54s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/c2938ab1d9b24e33939bb06a6233b718 : SUCCESS in 1h 46m 05s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/e88dd698040444aeaee267f56b7da96b : SUCCESS in 17m 32s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/8b3a5a5592ff4ce0bf9ceeaa6a1d5a7c : SUCCESS in 26m 11s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/f6207be1a4c94877bcd5899196d24a64 : FAILURE in 16m 32s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"617332d84b4d5b2f45c54a3deb170d2ad5db54ee","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-31 21:49:06.000000000","message":"Patch Set 1:\n\nrechec","accounts_in_message":[],"_revision_number":1},{"id":"490803813d72e548c4f96a29a5c7b3c95846cbcf","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-01-31 21:49:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1448afcec9434d0e6f56c10d1fff379d81042086","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-31 23:36:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f77cb6894dc146a7b121681789288248 : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08d16a632a2345498e9918ddac472610 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e01e265ea36d434ab8dab43cdbc733dc : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d7bfb12a75654fff938784eb5a4c0e5f : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0cae3ef1e41432092b997d7d52036b3 : SUCCESS in 5m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/40df5086f3dd454e88865a769418d472 : SUCCESS in 3m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a262c1f0a554447b85c2012035a9e657 : SUCCESS in 8m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/45f32c7ea42d4e1ebc322621ff5916ca : SUCCESS in 15m 48s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/9b81213c343e4b2ea7af7e465396a7e4 : SUCCESS in 52m 21s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/0484c37344b347ef9ea020d0dbc5f39f : SUCCESS in 53m 33s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/6329d25f13524bf1aab725aae1a8ad8e : SUCCESS in 12m 35s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/291280a0cd5047d89208b39a3b9699d8 : SUCCESS in 1h 44m 53s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/d55bf5e3169c470eb746b8a8de504751 : SUCCESS in 16m 03s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/d0ace60a4a384d01a011382b3109d04c : SUCCESS in 20m 54s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/2d78150244df4909919960e78e68874a : SUCCESS in 9m 38s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"b9fd67bdde94a6ca7ebc98209f9d7433467ba5b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-31 23:36:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aab2beaa695d61aadab2a0cb7397e29d98fc80a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 01:14:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bbc6382d04b470098f0ce57b6be52b0 : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7973ab66ec3b48168e0dfe2b811754c9 : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d83e45f0f06743b097552e92baee5f0d : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73aecc8c741f46a6ae49a253820a87f5 : SUCCESS in 7m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4bf90236dd794b6e8ce7a739e2ef075b : SUCCESS in 4m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4ddbeae95f9a46baa3c059763c25b9d8 : SUCCESS in 8m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c83c34c54f3d49669511e8794608b057 : SUCCESS in 17m 21s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/9eaa6ec4c96b48278db7eff1442df11a : SUCCESS in 1h 09m 30s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/74a35ba4d3c04afb8088a5d897eab777 : SUCCESS in 53m 56s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/fd25457465354c3d95e78d624f3c0168 : SUCCESS in 16m 02s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/5e6295eeb6a5443aaf009e562ac6e67f : SUCCESS in 1h 35m 39s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/fde13e7fd4c442f2b85326a0d0c2020d : SUCCESS in 15m 53s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/0a9ac4252fd34f569bde2c152561a35d : SUCCESS in 21m 19s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/bfb52ad70aad4731977bc4f593c6768d : SUCCESS in 9m 53s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"86a82f306033ae14f063b3315cda33d4198a82f9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 01:14:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0af835d77362ebd5f0a132eddcd08421b0203ca3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 01:15:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb0226914d6648d6abbaf533e4bf5074 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ffce488618dfe8a6961f81a26f2dfeb6afbbf6d8","revisions":{"ffce488618dfe8a6961f81a26f2dfeb6afbbf6d8":{"kind":"REWORK","_number":1,"created":"2021-01-27 22:00:06.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/77/772777/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/772777/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/772777/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/772777/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/772777/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/772777/1"}}},"commit":{"parents":[{"commit":"e660db0088d2553d43e8859573c88c94da061ebb","subject":"Fix overcloud host image resolv workaround on CentOS 8.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e660db0088d2553d43e8859573c88c94da061ebb"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-01-27 20:54:17.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-01-27 21:59:58.000000000","tz":60},"subject":"Fix deployment image build in Victoria","message":"Fix deployment image build in Victoria\n\nIn the Victoria release, deployment images fail to build with the\nfollowing error:\n\n/opt/kayobe/venvs/ipa-build-dib/lib/python3.6/site-packages/diskimage_builder/lib/common-functions: line 343: element-info: command not found\n\nThis is caused by the virtualenv bin directory not being included in\nPATH while executing code from common-functions, which results in the\nelement-info script not being found.\n\nThis doesn\u0027t happen in Ussuri because older versions of DIB were\nactivating the virtualenv automatically [1].\n\nThis doesn\u0027t happen in master because its upper constraints allow a\nnewer DIB which doesn\u0027t use element-info anymore.\n\nThe fix is to activate the DIB virtualenv, which is done since\nstackhpc.os-images v1.10.1 [2]. Bump in master first then backport to\nstable/victoria.\n\n[1] https://review.opendev.org/c/openstack/diskimage-builder/+/734414\n[2] https://github.com/stackhpc/ansible-role-os-images/releases/tag/v1.10.1\n\nChange-Id: I5efdbd52556721914fe69d7c6ba454b2c721b643\nStory: 2008550\nTask: 41673\n(cherry picked from commit 6563a02f85dc4833a11608d0e360a5478919fac5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ffce488618dfe8a6961f81a26f2dfeb6afbbf6d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ffce488618dfe8a6961f81a26f2dfeb6afbbf6d8"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
