)]}'
{"id":"openstack%2Fkayobe~773411","triplet_id":"openstack%2Fkayobe~master~Id446d038f2a47b5ad9d8adeb223f3d132c7a56b3","project":"openstack/kayobe","branch":"master","topic":"multiple-environments","hashtags":[],"change_id":"Id446d038f2a47b5ad9d8adeb223f3d132c7a56b3","subject":"[DNM] Add jobs testing support for multiple environments","status":"NEW","created":"2021-02-01 15:05:04.000000000","updated":"2021-05-07 18:36:01.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":2,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"6623022c32a81f95af7b6192362b113a295e799c","_number":773411,"virtual_id_number":773411,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-05-07 18:36:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-01 16:28:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-10 14:07:49.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2021-03-30 11:12:48.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"}],"messages":[{"id":"68df91ce1645bf390e051f11176937deeda122de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-01 15:05:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9685ceff17ec53de48a6645e83b2dc7d7e109e77","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-01 15:29:02.000000000","message":"Topic multiple-environments-4 removed","accounts_in_message":[],"_revision_number":1},{"id":"acbc24c9481b5cb3ce3ad5117f35796314f9be68","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-01 15:29:12.000000000","message":"Topic set to multiple-environments","accounts_in_message":[],"_revision_number":1},{"id":"c0c4639c9c492fbdf5e93774d69c3f65eb402b3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 16:28:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba371df31e364861a57d5bace022ceca : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25540a82de5442839a3cd36bec55fd95 : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ccfde3fdbe07418a875fd939c8541e39 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/198766c7a2a0468f8f9c95705c08dc32 : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c8bbb5f579fa48038df31157439bcd0f : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6e49c3aeda34a1c9a83448bb3dd8fb0 : SUCCESS in 6m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/47235ca7869e4a00a0d66190b69f6cb6 : SUCCESS in 4m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c819742222474693a96ce475def53c87 : SUCCESS in 7m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/47e294dafedc456da5f3f1f28b363163 : SUCCESS in 17m 12s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/1de10614c9d740ca9cd165b485f65e1a : SUCCESS in 1h 00m 26s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/2d1e76ef03d848f297a46157727d36f1 : RETRY_LIMIT in 2m 48s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/95867481a0134186be5a538200935e5a : FAILURE in 16m 42s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/d306ca8459a24bbbaf2bff266862cbb6 : FAILURE in 41m 03s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/1d059d2d88d0430e9809a5d0f29e5760 : SUCCESS in 21m 23s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/44d4ef1545b141d7b53850fff33821b8 : FAILURE in 20m 33s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/fa629f2ee8cf45d99256f239e49dc011 : RETRY_LIMIT in 3m 03s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"11fc328a35d9b92db5f5578343cd604565f022ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-02 16:40:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a76e0f97d536c1b978e8b3112a78612a4d793792","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 17:54:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a5f0f833d483451bb1e247c609528358 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9ec661bc8574308a1350091913b3985 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15d74cb3871a40f99c9ed04266f82ac1 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a2428404aef497a8e22ea90b5259698 : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/85fb0e7a46a44107aae99972d16c37ff : SUCCESS in 4m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d8ea940918c4ac885ec069aabe1893d : SUCCESS in 5m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4a44a783344d4c42b889afa974bba507 : SUCCESS in 4m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/175035c6f61441dfb240dd9876f25480 : SUCCESS in 12m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fb7179576c074f3f93775689b2a16259 : SUCCESS in 17m 04s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/4f7d046b66c84f58abae1e3d82d558aa : SUCCESS in 1h 01m 02s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/49e75128d00344d2849b1a779fe2711e : RETRY_LIMIT in 3m 03s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/a462a4b2c8084284962475a0160f7bed : FAILURE in 15m 12s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/94474d2242bd4eb6ae680bac4b0404d9 : FAILURE in 42m 35s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/bc62f53ffbc2440d85b98f21ee9edab3 : SUCCESS in 20m 46s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/d766450a6c4345b3822daad21a4ebb87 : FAILURE in 21m 45s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/cd706610206e4bdb98b86648bc1f2710 : RETRY_LIMIT in 3m 18s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":2},{"id":"7dec7713c244336716b0d3348bd2ddba0e65a77f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-10 14:07:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2eb636c7b76929dca4d48e0638e9f8a26097c5aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-15 14:18:44.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"cbfd9092b5bd8cf1eaa098ae621e3d080524e454","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 15:48:23.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e6ba92313d57435eb5e22041a94b2ac6 : SUCCESS in 9m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c52185bd3264618be6556aec1ec3bd2 : SUCCESS in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d51c045254a847ba99004113954c07d5 : SUCCESS in 7m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d5354645961c46378a55639f02d2517e : SUCCESS in 7m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8353a52139264e609a4d98b95b04a256 : SUCCESS in 8m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0141a7a23d98476c9c77ec94d5ce0cef : SUCCESS in 11m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8c0a96f2847b475da10fa301e6b75c46 : SUCCESS in 6m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/293396044f2c46e6a4f07aeed3c8946c : SUCCESS in 11m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4d405110a81443f9a1f1815db2096076 : SUCCESS in 21m 24s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/98a8be5601c14eec9030ce4671fe29c7 : RETRY_LIMIT in 4m 47s\n- kayobe-overcloud-centos8-env https://zuul.opendev.org/t/openstack/build/fd1659d1d0f547009d41a8d450a48f82 : RETRY_LIMIT in 3m 09s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":3},{"id":"cda6ef4ece846f988fd966cda20cee44591927f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-17 15:12:03.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5d25b99fea13f6e94ed2364c6fb473500b0f8f73","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-17 15:36:12.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"1ff43562314fe1806fa51749949f241bf33316ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 16:52:30.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/197e3a6711fd466680eebdb5e3c7cb2d : SUCCESS in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffd06423178645fbaf93101a74b3ce05 : SUCCESS in 8m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6aa8ff47523b4334bdcdfc5724f79c82 : SUCCESS in 7m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97984e38e3df429b9bcc05ae587c6dd9 : SUCCESS in 6m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e6e3b3decc034259a1e103318e4f674c : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af429ba357214951a88402d7734ff1e6 : SUCCESS in 8m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4f86ab79c5d341728374a69b4d203122 : SUCCESS in 5m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a822c456b8934c4da47d8a071011703a : SUCCESS in 12m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b6bfc8de70ff432fa079c5c81433483e : SUCCESS in 19m 50s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/9f81e6ffc4c64b8093013742165a95ce : SUCCESS in 1h 10m 46s\n- kayobe-overcloud-centos8-env https://zuul.opendev.org/t/openstack/build/846f19dbd92c4dcbbd6a832801d99ca0 : SUCCESS in 53m 49s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":5},{"id":"c7f71d25fb6aa026b55c859eef7caccb1101eda3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-17 20:52:19.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"90576c95f1dc6405c746b7f7ee142c414fd753ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 23:09:49.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9957f6c55f2047aead408c5474dec58e : SUCCESS in 6m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63f32966ab9b4dec922ca86395936fdf : SUCCESS in 8m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ad9b98e6d1840f0b35f779bb33e11ea : SUCCESS in 6m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c48fc49d14054816b293718d018ff985 : SUCCESS in 5m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21906cbf5b4442b0b4c76a0f9f537c7e : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fea905de28449f8ac49f37a82ffed61 : SUCCESS in 9m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cb7ff22f96764400879950919709f5a0 : SUCCESS in 6m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/56856feb66aa40949cfa339bfac702b0 : SUCCESS in 8m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6250f277406a4b609d9936f17b25a8d4 : SUCCESS in 20m 05s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/c5a9f8a9b5464b38af8d330c84a8ef76 : SUCCESS in 57m 20s\n- kayobe-overcloud-centos8-env https://zuul.opendev.org/t/openstack/build/b3088303f674425a8d8c3f01f9a35196 : SUCCESS in 55m 10s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f86675dc9bde4fa7a18e8297690def0f : SUCCESS in 54m 21s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/6dd6ad7bd3464e979e4ccf0087a07cd9 : RETRY_LIMIT in 2m 59s\n- kayobe-overcloud-tls-centos8-env https://zuul.opendev.org/t/openstack/build/1782759a52854d698a4afac1d813c7ae : FAILURE in 18m 54s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/2ea3dfd0fcad4cd38b706c5a68cc4ca1 : FAILURE in 19m 40s\n- kayobe-overcloud-host-configure-centos8-env https://zuul.opendev.org/t/openstack/build/7d35882cac6c4384917f03fb8d5d0f95 : FAILURE in 16m 48s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d9f568fac3664ff1b9eacf9f42ea84b5 : FAILURE in 11m 46s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/f2f64e7b07774d52bce262b4d5b37b32 : SUCCESS in 44m 20s\n- kayobe-seed-centos8-env https://zuul.opendev.org/t/openstack/build/0042dbd9b9d847f9b121671ef9a24377 : SUCCESS in 21m 50s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b7ce09e7f2d4cb28730c10c056ed03b : SUCCESS in 33m 18s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/12e5e38316914c4281d84c6143f249ab : RETRY_LIMIT in 3m 07s\n- kayobe-seed-vm-centos8-env https://zuul.opendev.org/t/openstack/build/c0eb3f0af3aa475ba23173e75b5455e8 : SUCCESS in 13m 19s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e0ec34caffd54d928b321f25dd20f7e3 : RETRY_LIMIT in 2m 39s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":6},{"id":"a4f12915ea64a65c67aa40a431176a5207e15dd0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-18 08:54:02.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"230b7d8f64905f53cfb4ca29b4be93c92c0d83e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 10:28:12.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac60f2440d2a4850bff5e9d9850f6920 : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/205308f551d34f00bd4bc85da5ee3355 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/98c44c9310b744a4809979e3a7b73391 : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1cea210c3c924c2dadc9b3a5cb947c46 : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/464c2b368e964be19ab049eae9717960 : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9bcea209e7f4c39b5d5a5418c5bd4af : SUCCESS in 7m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/78f28451245149e4a0193909895f7e5a : SUCCESS in 5m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c1d2373208ea4d4daecbf60bcb4d77d4 : SUCCESS in 9m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ad3441bebb4f452298e8baa55c6a4401 : SUCCESS in 22m 25s\n- kayobe-overcloud-centos8-env https://zuul.opendev.org/t/openstack/build/de0090224caa4ce2a0092c087b05e909 : SUCCESS in 1h 04m 28s\n- kayobe-overcloud-tls-centos8-env https://zuul.opendev.org/t/openstack/build/3bc34e9379d343a48a37e0ab0065342f : SUCCESS in 1h 00m 10s\n- kayobe-overcloud-host-configure-centos8-env https://zuul.opendev.org/t/openstack/build/d524a01cf32e49898d1bc474cec9bbc2 : SUCCESS in 16m 23s\n- kayobe-seed-centos8-env https://zuul.opendev.org/t/openstack/build/cacda5ff070b409785886beb10530b74 : SUCCESS in 27m 06s\n- kayobe-seed-vm-centos8-env https://zuul.opendev.org/t/openstack/build/306d77041d1b4a198cd853a3bb0a114d : SUCCESS in 12m 47s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":7},{"id":"8a546f1c49c49dea096492d7d15700b99353b7bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-24 15:16:09.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"ed6157c6b02af13d496154259ad96120f5a02800","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-24 15:53:33.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"c9d8fef318769861f476a1b7ed2d15f823c5b584","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 16:29:59.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/25a8883483f240b4bd153740ed27ed9f : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9effc66b60d4afdbed22aed7ebd090b : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f865ef1b677471abe7ff396160b4a09 : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6708b10c8c4646f58aaf593e595b5378 : SUCCESS in 6m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3afdbdb7e0df47aa89719936fe6abd03 : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfdfb9faaa494ee5a51f11af2f2be4c5 : SUCCESS in 8m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2f16b7a8b68140fbb7315d0a3dff9f06 : SUCCESS in 5m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e0ca3050b1b149258b5fa597f473a472 : SUCCESS in 10m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/82d1b72f7074485eacb921c36f3ed1bb : SUCCESS in 20m 16s\n- kayobe-overcloud-centos8-env https://zuul.opendev.org/t/openstack/build/07ad86a1ee0d499f99cd4782fac58442 : FAILURE in 7m 33s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":9},{"id":"0ab8fddf38755131f3558d030f9754601ce80abb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-24 16:37:10.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"c20c3adfa676a32209b87187a1bf287408d3c273","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-24 16:56:41.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"ae1890dc7a12518d05f84ffe24c2eb044643618c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 17:38:42.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4dbbaac4a5146359f8bdac418e62e54 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f21e7f56d9547f9ab27860654e5bbf7 : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f62286ab66847f49996c95624675715 : SUCCESS in 7m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0b99acc10d3746f4a5b21985392e9ecc : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39738f2722e44c6486695ec2da0c0297 : SUCCESS in 8m 14s\n- kayobe-overcloud-centos8-env https://zuul.opendev.org/t/openstack/build/4b603d63c5dd49ff8cd92e064caaf160 : RETRY_LIMIT in 4m 55s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":11},{"id":"0f7b9d481e9b8d6834a3517b0600f2f0d1de06f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-24 17:42:18.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"212118680a703c3bf2ea7bc01785f0304d5e7e5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 17:59:01.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08f00d82626043d984c2112399aef948 : SUCCESS in 6m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0fd9f04e0a4b4419ae005f9c6e5b7c7a : SUCCESS in 7m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/339cf74f1b8b4ab9a198b4b8a8e2559b : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df7f28194d134cd88cd60b8ee0dd0ba5 : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b66cb1c67354a43983f5791140e78b0 : SUCCESS in 7m 02s\n- kayobe-overcloud-centos8-env https://zuul.opendev.org/t/openstack/build/2812540486fc4fe2ad0c5c066c17b7cb : RETRY_LIMIT in 5m 54s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":12},{"id":"d1876700ee7af55eab49f6873ab90ddc016e5c61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-24 18:01:19.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"0ca77e55bc0cd3b32ac1ba7ef72f795b779af5ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 19:06:30.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84dcf1f6c09548579697e32fef9eaa8e : SUCCESS in 7m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cae22babe0df49f0aaa1e2dda99449a8 : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/30e1188b7adf461d89babdb2ba01c6b5 : SUCCESS in 8m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/01629ec19c5e40e9a0012f8313b30079 : SUCCESS in 8m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a99b20de70ae40199eda5c663ebfc0ff : SUCCESS in 8m 28s\n- kayobe-overcloud-centos8-env https://zuul.opendev.org/t/openstack/build/de58be0386ec4774a734e5ffaea7d4bd : SUCCESS in 1h 01m 06s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":13},{"id":"fa4ad8572a8f9672f020c02a9a729d2eb6900bf8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-25 13:31:54.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"68648435eb1e6a509bd063715fd04518ffaea7a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 14:33:45.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8043db416cda449e9b3892cac73124f3 : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd1922467883499ca4f45f5cf96d43f0 : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c09bdb0f83714cfb9647a52e272b7fc0 : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/28688e3eb49a4fa1bab8133a72138e6e : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/80ab8b8ed375486798f7f9c002f5c30d : SUCCESS in 7m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5290d642a7644298a63c841f5ac3a41 : SUCCESS in 8m 54s\n- kayobe-overcloud-centos8-env https://zuul.opendev.org/t/openstack/build/47f714dda3bf4e7e96e67c87df2b68de : SUCCESS in 57m 54s\n- kayobe-overcloud-ubuntu-focal-env https://zuul.opendev.org/t/openstack/build/433aeefd4a91440b81c241aa995c8017 : SUCCESS in 41m 46s\n- kayobe-overcloud-tls-centos8-env https://zuul.opendev.org/t/openstack/build/c82178c43f694c0694f9d7f6f749e645 : SUCCESS in 57m 32s\n- kayobe-overcloud-host-configure-centos8-env https://zuul.opendev.org/t/openstack/build/0a7a96faf9264417a0859baa75f39ba2 : SUCCESS in 16m 58s\n- kayobe-overcloud-host-configure-ubuntu-focal-env https://zuul.opendev.org/t/openstack/build/dddb011ab3ce4d519c80433b4da6e0bd : SUCCESS in 13m 45s\n- kayobe-overcloud-upgrade-centos8-env https://zuul.opendev.org/t/openstack/build/c00e98b6bb5b4fae984ba986c7bc0dbc : FAILURE in 43m 33s\n- kayobe-seed-centos8-env https://zuul.opendev.org/t/openstack/build/bc9b9bedd779451ea2a42f0cdae7baff : SUCCESS in 26m 03s\n- kayobe-seed-ubuntu-focal-env https://zuul.opendev.org/t/openstack/build/338c3feaae814f4aac4941da865ee072 : SUCCESS in 29m 59s\n- kayobe-seed-upgrade-centos8-env https://zuul.opendev.org/t/openstack/build/b6c9774c8f374678a278dce6cdb09e72 : SUCCESS in 24m 02s\n- kayobe-seed-vm-centos8-env https://zuul.opendev.org/t/openstack/build/91a94d0ac67746f2a2fed02f3f719cab : SUCCESS in 11m 38s\n- kayobe-seed-vm-ubuntu-focal-env https://zuul.opendev.org/t/openstack/build/f7294663c7674b9fbaa5c502735e885f : SUCCESS in 9m 22s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":14},{"id":"2e690f1222d99acf38815fb48fdeaf88733da3bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-25 15:01:49.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"ff21ea02a3a785c0d9e5ba15bbf04a0497ba1e96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 16:01:38.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54be3d5957f34b668f8f6025065863db : SUCCESS in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a48fcd05b204eeca8b9535958bc306c : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba30883281b64a278dae77dd46e59a21 : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/74d4c62072cc4757ad356c6973cdefae : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/defeabdedb3647d49e19d54c4039f2d9 : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c36c7662cd964874aacf59e29516cd15 : SUCCESS in 8m 06s\n- kayobe-overcloud-upgrade-centos8-env https://zuul.opendev.org/t/openstack/build/a9bc90a3963b4c35ac93c3d17de77cf5 : FAILURE in 54m 30s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":15},{"id":"1f552e4068b46196a707853f7c6199323ca695ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-25 16:53:01.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"2b3a69bbff97c924a7e6f0e6c424386f746ab3f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 18:55:09.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a8269ddc01c94e0988713a0b3a9d42cf : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a22738e03229418a93b76a06d90fbbd8 : SUCCESS in 7m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5044954d62274dbebdde2416fb2c3431 : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a74d1bb0230d434ab6e13d17d3b7aa7c : SUCCESS in 6m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/43a70146287d452590bb5958add70177 : SUCCESS in 7m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71b831f69bae47d1ae069519184fedb2 : SUCCESS in 8m 58s\n- kayobe-overcloud-upgrade-centos8-env https://zuul.opendev.org/t/openstack/build/292a2affa7794ac1a6e8686bc2cb6c55 : SUCCESS in 1h 55m 16s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":16},{"id":"065d3f67dd533130c128755334b32685402ddc3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-25 21:42:42.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"10e2812fa10c01f7f8b28842a70ad6d20e538ce4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 23:40:30.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/539b5f7ed5d64bdbb51e047b7126ca06 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3d3a16fb83b4e7c9f27c54ca260f8a7 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/46877d2feed84380a7099857ad190113 : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a150803512f442578c8faf4ca0ccd734 : SUCCESS in 6m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d54316fb9fa1471894e84369ec83c48d : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/616b3bb3fc434df88c831a071f2c77ba : SUCCESS in 8m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ab3e753a208a453991c2993928ebaba6 : SUCCESS in 6m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/152bf1d9637f4afdaa42bffd38e6b06b : FAILURE in 11m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0606a30e18aa47fca993095f117f0caa : SUCCESS in 21m 30s\n- kayobe-overcloud-centos8-env https://zuul.opendev.org/t/openstack/build/d9e8427cb0a249f79f9e0f0a0aaa3e67 : SUCCESS in 1h 06m 27s\n- kayobe-overcloud-ubuntu-focal-env https://zuul.opendev.org/t/openstack/build/3d2d32d7734a4f8e8916afd60f585b44 : SUCCESS in 52m 13s\n- kayobe-overcloud-tls-centos8-env https://zuul.opendev.org/t/openstack/build/e9b1b852842d4cc8871376b351cb6d79 : SUCCESS in 1h 12m 12s\n- kayobe-overcloud-host-configure-centos8-env https://zuul.opendev.org/t/openstack/build/426d4fc434764392b28084005d1190b7 : SUCCESS in 17m 56s\n- kayobe-overcloud-host-configure-ubuntu-focal-env https://zuul.opendev.org/t/openstack/build/64d32f52edb0407cba97530709178d51 : SUCCESS in 13m 40s\n- kayobe-overcloud-upgrade-centos8-env https://zuul.opendev.org/t/openstack/build/273c1ee295974e97bea081dd1a069c50 : SUCCESS in 1h 50m 49s\n- kayobe-seed-centos8-env https://zuul.opendev.org/t/openstack/build/c5f716a900ea43f4a6cb379f96bf9aac : SUCCESS in 27m 41s\n- kayobe-seed-ubuntu-focal-env https://zuul.opendev.org/t/openstack/build/5d590e3480294b8f86ce20b1cb8fa8d0 : SUCCESS in 25m 57s\n- kayobe-seed-upgrade-centos8-env https://zuul.opendev.org/t/openstack/build/16f225cc4c7e40dea5b5740deb58bacd : SUCCESS in 29m 28s\n- kayobe-seed-vm-centos8-env https://zuul.opendev.org/t/openstack/build/565a3243e59c4e69924883406de6156b : SUCCESS in 13m 02s\n- kayobe-seed-vm-ubuntu-focal-env https://zuul.opendev.org/t/openstack/build/a12cc26a6c9648508758ce46ef367a82 : SUCCESS in 11m 39s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":17},{"id":"63ec4190f48bc1da7942a2ac82852978163318b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-26 08:34:09.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.","accounts_in_message":[],"_revision_number":18},{"id":"f3404c91186cb0c9475683be99e85692219b97d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 11:03:23.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e139004063cc41deb6ea096e1612d22e : SUCCESS in 9m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/024e286f60d04813a95c05fca6d31159 : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6878c19a17f9443aa2aa0ea563bb7ada : SUCCESS in 6m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/058e6ac22ca748ef899fcbbb9118e0f4 : SUCCESS in 6m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4940ccc1289f45eea01ce6aa30fb0381 : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2616eca9d47246c2a590db90f789bdf0 : SUCCESS in 12m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/71517ecfc99d4b76b24b7c681e537696 : SUCCESS in 6m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/724d08b19cf1450e8b86b91537450550 : SUCCESS in 11m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/61c3ceaf8660481d8939949ca693fa9d : SUCCESS in 21m 19s\n- kayobe-overcloud-centos8-env https://zuul.opendev.org/t/openstack/build/7de096905f664b609cf3fec50f64817c : SUCCESS in 1h 01m 50s\n- kayobe-overcloud-ubuntu-focal-env https://zuul.opendev.org/t/openstack/build/3ec9167708424e13811a8d7f9194d148 : SUCCESS in 38m 32s\n- kayobe-overcloud-tls-centos8-env https://zuul.opendev.org/t/openstack/build/c13b5b53cf374c7191283162b948b191 : SUCCESS in 1h 06m 45s\n- kayobe-overcloud-host-configure-centos8-env https://zuul.opendev.org/t/openstack/build/516b11d6c9b74848a227ee7c39ce8aef : SUCCESS in 17m 07s\n- kayobe-overcloud-host-configure-ubuntu-focal-env https://zuul.opendev.org/t/openstack/build/630083c7c9eb4c91b576b9f58224d7b2 : SUCCESS in 13m 31s\n- kayobe-overcloud-upgrade-centos8-env https://zuul.opendev.org/t/openstack/build/8ad8eab2dcf94bdc90c147b59e638ce6 : SUCCESS in 1h 59m 54s\n- kayobe-seed-centos8-env https://zuul.opendev.org/t/openstack/build/307f5b7686dd40d6b9f0d92c0f09a6d9 : SUCCESS in 23m 04s\n- kayobe-seed-ubuntu-focal-env https://zuul.opendev.org/t/openstack/build/dd429b7ac86947178f4441961a8773c1 : SUCCESS in 29m 59s\n- kayobe-seed-upgrade-centos8-env https://zuul.opendev.org/t/openstack/build/74c9da7033fb449590cfdb46974ddc89 : SUCCESS in 24m 21s\n- kayobe-seed-vm-centos8-env https://zuul.opendev.org/t/openstack/build/aa7562ed044241d9ba7d1b10ca19197d : SUCCESS in 12m 45s\n- kayobe-seed-vm-ubuntu-focal-env https://zuul.opendev.org/t/openstack/build/fe1b524348ce494983d6a84241e68a08 : SUCCESS in 10m 05s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":18},{"id":"e7945fea259864e1afd1dd4d84a65c2dbc278c8e","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-03-30 11:12:48.000000000","message":"Patch Set 18:\n\nThis one looks pretty good to me. Aside from the TODO, what else you thinking of changing?","accounts_in_message":[],"_revision_number":18},{"id":"fac009b1f9c8f7b43431739f07f1bcf808e8e997","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-30 12:49:09.000000000","message":"Patch Set 18:\n\n\u003e Patch Set 18:\n\u003e \n\u003e This one looks pretty good to me. Aside from the TODO, what else you thinking of changing?\n\nMark suggested that we should alternate environment and non-environment jobs, to keep good coverage while not doubling the number of jobs.","accounts_in_message":[],"_revision_number":18},{"id":"c867988dcf1f03a0be4667e6d299a9b77fe21478","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-31 09:11:06.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"5b76e7f1a17243fc00f80786650936a08397338c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-07 16:45:15.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"6623022c32a81f95af7b6192362b113a295e799c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-07 18:36:01.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b14a18494e61469f969fc04d59c25860 : SUCCESS in 6m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4ecf320f64b46c78b0d623d2ac31cff : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81fa8626967b409aac3d6feb87ae9bac : SUCCESS in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3a078f40c6b54011b47739b7ee468026 : SUCCESS in 6m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/db5271d005a342e0adadcd4988699e7b : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3651d4c75cd4912840a4f0b5f23d3e7 : SUCCESS in 8m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/67d50875bb56431083c42cd10edef639 : SUCCESS in 6m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/252beb2e411c48d5945d11084362a100 : SUCCESS in 10m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/eb82c84d30904e5f96f7f50dc7523cde : SUCCESS in 18m 47s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/baf23ce7eac84f4e9ceaa8495b3d17f8 : SUCCESS in 1h 02m 57s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3935aff5cacc4bd9b2794d62e3bd00e5 : FAILURE in 40m 43s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/dc68ab0a854a4f33be2a3c16f61d9258 : SUCCESS in 1h 08m 46s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/e65b4974b02d4e60a4788ee83c2d9ddc : SUCCESS in 17m 55s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/98aed21453664b41a194671abfb49b0d : SUCCESS in 14m 15s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/58ad5cbf3ef24549bbc5df0c6664019e : SUCCESS in 1h 45m 34s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d8df0a98b8ee4f058cda14a027432732 : SUCCESS in 33m 59s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a1421f152bea4c748b4bcced79bfebf8 : RETRY_LIMIT in 3m 23s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d1debd5776e442daaed7d9432c602520 : SUCCESS in 22m 55s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/e1e613fccdd64970926aea663a8af37b : SUCCESS in 13m 19s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94961fa38fa842b185a4ae87e88c326f : SUCCESS in 10m 25s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":19}],"current_revision_number":19,"current_revision":"f636fd17d30d8202c11e5cb3a0dac0587d8dd569","revisions":{"cad48a3dd98005544ef3f764f800a138bd14a526":{"kind":"REWORK","_number":1,"created":"2021-02-01 15:05:04.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/11/773411/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/11/773411/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/11/773411/1"}}},"commit":{"parents":[{"commit":"0c8ddee6681c4aefa33f10b3950d43fea220ae5f","subject":"[docs] Update main documentation page","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0c8ddee6681c4aefa33f10b3950d43fea220ae5f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-01 15:04:13.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-01 15:04:53.000000000","tz":60},"subject":"[DNM] Set environment for testing","message":"[DNM] Set environment for testing\n\nChange-Id: Id446d038f2a47b5ad9d8adeb223f3d132c7a56b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cad48a3dd98005544ef3f764f800a138bd14a526"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cad48a3dd98005544ef3f764f800a138bd14a526"}]},"branch":"refs/heads/master"},"00d584f58b6d6e9ae0762edf14192564179607dd":{"kind":"REWORK","_number":2,"created":"2021-02-02 16:40:18.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/11/773411/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/11/773411/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/11/773411/2"}}},"commit":{"parents":[{"commit":"586dccc236d1466e114f371573b39e5a2129a995","subject":"Merge \"Remove extra space characters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/586dccc236d1466e114f371573b39e5a2129a995"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-02 16:40:07.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-02 16:40:07.000000000","tz":60},"subject":"[DNM] Set environment for testing","message":"[DNM] Set environment for testing\n\nChange-Id: Id446d038f2a47b5ad9d8adeb223f3d132c7a56b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/00d584f58b6d6e9ae0762edf14192564179607dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/00d584f58b6d6e9ae0762edf14192564179607dd"}]},"branch":"refs/heads/master"},"9ef4695e4acef765f43b3833deebfa9fbd92ce42":{"kind":"REWORK","_number":3,"created":"2021-03-15 14:18:44.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/11/773411/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/11/773411/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/11/773411/3"}}},"commit":{"parents":[{"commit":"f8d9861c55532a265c24b1ecba3ec5267af60e41","subject":"Add documentation for multiple environments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f8d9861c55532a265c24b1ecba3ec5267af60e41"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-15 14:16:35.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-15 14:17:58.000000000","tz":60},"subject":"[DNM] Set environment for testing","message":"[DNM] Set environment for testing\n\nChange-Id: Id446d038f2a47b5ad9d8adeb223f3d132c7a56b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9ef4695e4acef765f43b3833deebfa9fbd92ce42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9ef4695e4acef765f43b3833deebfa9fbd92ce42"}]},"branch":"refs/heads/master"},"c494c9c4f762cce7b657f6b0329bdf791f1c4512":{"kind":"REWORK","_number":4,"created":"2021-03-17 15:12:03.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/11/773411/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/11/773411/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/11/773411/4"}}},"commit":{"parents":[{"commit":"f8d9861c55532a265c24b1ecba3ec5267af60e41","subject":"Add documentation for multiple environments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f8d9861c55532a265c24b1ecba3ec5267af60e41"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-17 15:11:01.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-17 15:11:01.000000000","tz":60},"subject":"[DNM] Set environment for testing","message":"[DNM] Set environment for testing\n\nChange-Id: Id446d038f2a47b5ad9d8adeb223f3d132c7a56b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c494c9c4f762cce7b657f6b0329bdf791f1c4512"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c494c9c4f762cce7b657f6b0329bdf791f1c4512"}]},"branch":"refs/heads/master"},"cda6db519228654463551cc35d4407e9d5812c12":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2021-03-17 15:36:12.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/11/773411/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/11/773411/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/11/773411/5"}}},"commit":{"parents":[{"commit":"f8d9861c55532a265c24b1ecba3ec5267af60e41","subject":"Add documentation for multiple environments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f8d9861c55532a265c24b1ecba3ec5267af60e41"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-17 15:35:39.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-17 15:35:39.000000000","tz":60},"subject":"[DNM] Set environment for testing","message":"[DNM] Set environment for testing\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/752469/\nChange-Id: Id446d038f2a47b5ad9d8adeb223f3d132c7a56b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cda6db519228654463551cc35d4407e9d5812c12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cda6db519228654463551cc35d4407e9d5812c12"}]},"branch":"refs/heads/master"},"6a46d25333eb89f38c8c15e5fcce776927ce0994":{"kind":"REWORK","_number":6,"created":"2021-03-17 20:52:19.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/11/773411/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/11/773411/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/11/773411/6"}}},"commit":{"parents":[{"commit":"f8d9861c55532a265c24b1ecba3ec5267af60e41","subject":"Add documentation for multiple environments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f8d9861c55532a265c24b1ecba3ec5267af60e41"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-17 20:51:22.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-17 20:51:22.000000000","tz":60},"subject":"[DNM] Set environment for testing","message":"[DNM] Set environment for testing\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/752469/\nChange-Id: Id446d038f2a47b5ad9d8adeb223f3d132c7a56b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6a46d25333eb89f38c8c15e5fcce776927ce0994"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6a46d25333eb89f38c8c15e5fcce776927ce0994"}]},"branch":"refs/heads/master"},"0c43f90c221098286105134d99213e4ff44da722":{"kind":"REWORK","_number":7,"created":"2021-03-18 08:54:02.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/11/773411/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/11/773411/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/11/773411/7"}}},"commit":{"parents":[{"commit":"f8d9861c55532a265c24b1ecba3ec5267af60e41","subject":"Add documentation for multiple environments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f8d9861c55532a265c24b1ecba3ec5267af60e41"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-18 08:52:55.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-18 08:52:55.000000000","tz":60},"subject":"[DNM] Set environment for testing","message":"[DNM] Set environment for testing\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/752469/\nChange-Id: Id446d038f2a47b5ad9d8adeb223f3d132c7a56b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0c43f90c221098286105134d99213e4ff44da722"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0c43f90c221098286105134d99213e4ff44da722"}]},"branch":"refs/heads/master"},"79751c87d14b43f6d14ef35881965b840a60d28e":{"kind":"REWORK","_number":8,"created":"2021-03-24 15:16:09.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/11/773411/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/11/773411/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/11/773411/8"}}},"commit":{"parents":[{"commit":"f8d9861c55532a265c24b1ecba3ec5267af60e41","subject":"Add documentation for multiple environments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f8d9861c55532a265c24b1ecba3ec5267af60e41"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-24 15:15:34.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-24 15:15:34.000000000","tz":60},"subject":"[DNM] Set environment for testing","message":"[DNM] Set environment for testing\n\nChange-Id: Id446d038f2a47b5ad9d8adeb223f3d132c7a56b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/79751c87d14b43f6d14ef35881965b840a60d28e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/79751c87d14b43f6d14ef35881965b840a60d28e"}]},"branch":"refs/heads/master"},"36c1c2f6f008c8b0cfcd6eeb59f6f94361a09b25":{"kind":"REWORK","_number":9,"created":"2021-03-24 15:53:33.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/11/773411/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/11/773411/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/11/773411/9"}}},"commit":{"parents":[{"commit":"f8d9861c55532a265c24b1ecba3ec5267af60e41","subject":"Add documentation for multiple environments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f8d9861c55532a265c24b1ecba3ec5267af60e41"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-24 15:53:13.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-24 15:53:13.000000000","tz":60},"subject":"[DNM] Set environment for testing","message":"[DNM] Set environment for testing\n\nChange-Id: Id446d038f2a47b5ad9d8adeb223f3d132c7a56b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/36c1c2f6f008c8b0cfcd6eeb59f6f94361a09b25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/36c1c2f6f008c8b0cfcd6eeb59f6f94361a09b25"}]},"branch":"refs/heads/master"},"4ae606a6f3e452590f976959de8da7c13fd7fc80":{"kind":"REWORK","_number":10,"created":"2021-03-24 16:37:10.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/11/773411/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/11/773411/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/11/773411/10"}}},"commit":{"parents":[{"commit":"f8d9861c55532a265c24b1ecba3ec5267af60e41","subject":"Add documentation for multiple environments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f8d9861c55532a265c24b1ecba3ec5267af60e41"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-24 16:36:35.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-24 16:36:35.000000000","tz":60},"subject":"[DNM] Set environment for testing","message":"[DNM] Set environment for testing\n\nChange-Id: Id446d038f2a47b5ad9d8adeb223f3d132c7a56b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4ae606a6f3e452590f976959de8da7c13fd7fc80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4ae606a6f3e452590f976959de8da7c13fd7fc80"}]},"branch":"refs/heads/master"},"09898c9d57fdeeeb3e548947416d58ad92425635":{"kind":"REWORK","_number":11,"created":"2021-03-24 16:56:41.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/11/773411/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/11/773411/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/11/773411/11"}}},"commit":{"parents":[{"commit":"f8d9861c55532a265c24b1ecba3ec5267af60e41","subject":"Add documentation for multiple environments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f8d9861c55532a265c24b1ecba3ec5267af60e41"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-24 16:56:21.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-24 16:56:21.000000000","tz":60},"subject":"[DNM] Set environment for testing","message":"[DNM] Set environment for testing\n\nChange-Id: Id446d038f2a47b5ad9d8adeb223f3d132c7a56b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/09898c9d57fdeeeb3e548947416d58ad92425635"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/09898c9d57fdeeeb3e548947416d58ad92425635"}]},"branch":"refs/heads/master"},"5afc41335cb496bbf121726eb73376c87357b855":{"kind":"REWORK","_number":12,"created":"2021-03-24 17:42:18.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/11/773411/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/11/773411/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/11/773411/12"}}},"commit":{"parents":[{"commit":"f8d9861c55532a265c24b1ecba3ec5267af60e41","subject":"Add documentation for multiple environments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f8d9861c55532a265c24b1ecba3ec5267af60e41"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-24 17:41:38.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-24 17:41:38.000000000","tz":60},"subject":"[DNM] Set environment for testing","message":"[DNM] Set environment for testing\n\nChange-Id: Id446d038f2a47b5ad9d8adeb223f3d132c7a56b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5afc41335cb496bbf121726eb73376c87357b855"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5afc41335cb496bbf121726eb73376c87357b855"}]},"branch":"refs/heads/master"},"d86ccc38283aad20245bb885ff2ad509df8497f3":{"kind":"REWORK","_number":13,"created":"2021-03-24 18:01:19.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/11/773411/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/11/773411/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/11/773411/13"}}},"commit":{"parents":[{"commit":"f8d9861c55532a265c24b1ecba3ec5267af60e41","subject":"Add documentation for multiple environments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f8d9861c55532a265c24b1ecba3ec5267af60e41"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-24 18:00:37.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-24 18:00:37.000000000","tz":60},"subject":"[DNM] Set environment for testing","message":"[DNM] Set environment for testing\n\nChange-Id: Id446d038f2a47b5ad9d8adeb223f3d132c7a56b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d86ccc38283aad20245bb885ff2ad509df8497f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d86ccc38283aad20245bb885ff2ad509df8497f3"}]},"branch":"refs/heads/master"},"0ff73aabcc4643aa9a127442d845b6b53592648a":{"kind":"REWORK","_number":14,"created":"2021-03-25 13:31:54.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/11/773411/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/11/773411/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/11/773411/14"}}},"commit":{"parents":[{"commit":"f8d9861c55532a265c24b1ecba3ec5267af60e41","subject":"Add documentation for multiple environments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f8d9861c55532a265c24b1ecba3ec5267af60e41"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-25 13:30:59.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-25 13:30:59.000000000","tz":60},"subject":"[DNM] Add jobs testing support for multiple environments","message":"[DNM] Add jobs testing support for multiple environments\n\nChange-Id: Id446d038f2a47b5ad9d8adeb223f3d132c7a56b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0ff73aabcc4643aa9a127442d845b6b53592648a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0ff73aabcc4643aa9a127442d845b6b53592648a"}]},"branch":"refs/heads/master"},"cd6329f32eb403308bed894f2aa9af127250cb06":{"kind":"REWORK","_number":15,"created":"2021-03-25 15:01:49.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/11/773411/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/11/773411/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/11/773411/15"}}},"commit":{"parents":[{"commit":"f8d9861c55532a265c24b1ecba3ec5267af60e41","subject":"Add documentation for multiple environments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f8d9861c55532a265c24b1ecba3ec5267af60e41"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-25 14:59:23.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-25 14:59:23.000000000","tz":60},"subject":"[DNM] Add jobs testing support for multiple environments","message":"[DNM] Add jobs testing support for multiple environments\n\nChange-Id: Id446d038f2a47b5ad9d8adeb223f3d132c7a56b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cd6329f32eb403308bed894f2aa9af127250cb06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cd6329f32eb403308bed894f2aa9af127250cb06"}]},"branch":"refs/heads/master"},"e340125290ea51cd20bf2c732cb1b3fe34d70a7f":{"kind":"REWORK","_number":16,"created":"2021-03-25 16:53:01.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/11/773411/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/11/773411/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/11/773411/16"}}},"commit":{"parents":[{"commit":"f8d9861c55532a265c24b1ecba3ec5267af60e41","subject":"Add documentation for multiple environments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f8d9861c55532a265c24b1ecba3ec5267af60e41"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-25 16:50:53.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-25 16:50:53.000000000","tz":60},"subject":"[DNM] Add jobs testing support for multiple environments","message":"[DNM] Add jobs testing support for multiple environments\n\nChange-Id: Id446d038f2a47b5ad9d8adeb223f3d132c7a56b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e340125290ea51cd20bf2c732cb1b3fe34d70a7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e340125290ea51cd20bf2c732cb1b3fe34d70a7f"}]},"branch":"refs/heads/master"},"bd2c25b8255409479f5753eed8aa2e90c4a1425b":{"kind":"REWORK","_number":17,"created":"2021-03-25 21:42:42.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/11/773411/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/11/773411/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/11/773411/17"}}},"commit":{"parents":[{"commit":"838ef6a9a544fb5c41c302ba9ce4babd99ec1b2d","subject":"Exit when environment clashes are detected","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/838ef6a9a544fb5c41c302ba9ce4babd99ec1b2d"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-25 21:41:23.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-25 21:41:23.000000000","tz":60},"subject":"[DNM] Add jobs testing support for multiple environments","message":"[DNM] Add jobs testing support for multiple environments\n\nChange-Id: Id446d038f2a47b5ad9d8adeb223f3d132c7a56b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bd2c25b8255409479f5753eed8aa2e90c4a1425b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bd2c25b8255409479f5753eed8aa2e90c4a1425b"}]},"branch":"refs/heads/master"},"e79e9efe6278c9c60336204ec3874cf431c20365":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2021-03-26 08:34:09.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/11/773411/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/11/773411/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/11/773411/18"}}},"commit":{"parents":[{"commit":"7bcea5edc6e57dd82b386126821cf6f6001d95aa","subject":"Exit when environment clashes are detected","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7bcea5edc6e57dd82b386126821cf6f6001d95aa"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-26 08:33:59.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-26 08:33:59.000000000","tz":60},"subject":"[DNM] Add jobs testing support for multiple environments","message":"[DNM] Add jobs testing support for multiple environments\n\nChange-Id: Id446d038f2a47b5ad9d8adeb223f3d132c7a56b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e79e9efe6278c9c60336204ec3874cf431c20365"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e79e9efe6278c9c60336204ec3874cf431c20365"}]},"branch":"refs/heads/master"},"f636fd17d30d8202c11e5cb3a0dac0587d8dd569":{"kind":"REWORK","_number":19,"created":"2021-05-07 16:45:15.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/11/773411/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/11/773411/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/773411/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/11/773411/19"}}},"commit":{"parents":[{"commit":"9c0bec4d3c36dfdd02f36f077f9f561d2ec83b6b","subject":"Merge \"Ubuntu: add documentation \u0026 release note\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9c0bec4d3c36dfdd02f36f077f9f561d2ec83b6b"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-05-07 16:43:48.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-05-07 16:43:48.000000000","tz":120},"subject":"[DNM] Add jobs testing support for multiple environments","message":"[DNM] Add jobs testing support for multiple environments\n\nChange-Id: Id446d038f2a47b5ad9d8adeb223f3d132c7a56b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f636fd17d30d8202c11e5cb3a0dac0587d8dd569"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f636fd17d30d8202c11e5cb3a0dac0587d8dd569"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
