)]}'
{"id":"openstack%2Fkayobe~773719","triplet_id":"openstack%2Fkayobe~master~Ieb2b7ff07cd43029399e9e19002b82940246f8c4","project":"openstack/kayobe","branch":"master","topic":"multiple-environments","hashtags":[],"change_id":"Ieb2b7ff07cd43029399e9e19002b82940246f8c4","subject":"Detect multiple inventories automatically","status":"MERGED","created":"2021-02-02 16:04:09.000000000","updated":"2021-04-01 17:54:10.000000000","submitted":"2021-04-01 17:51:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"773719-multiple-environments","meta_rev_id":"37697ffca8323f49fbeb36dfd505c2288aab55d0","_number":773719,"virtual_id_number":773719,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-01 17:51:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2021-04-01 17:54:10.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2021-04-01 17:54:10.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-04-01 17:54:10.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2021-04-01 14:39:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2021-04-01 17:54:10.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-02 17:00:54.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-02-02 17:57:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-26 09:52:34.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-03-30 10:28:55.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"c34da91a7885e6ba2a38913e3ff6b908edc37f51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-02 16:04:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"647b9ad2b50a256d94360b5112b936c0213ad7b4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-02 17:00:54.000000000","message":"Patch Set 1: Code-Review+1\n\nLooks sensible.","accounts_in_message":[],"_revision_number":1},{"id":"5f356a276efbaae1f7ad588777500f783666163c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-02 17:04:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1e8a5d2a3f739d8284c8b821434399fe993e858f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 17:57:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b337edd4625d43a2af1664cdf6d7e1ad : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b65605aee907423d9a68658e461e9847 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6a4920b49854924ace3127e20a4b840 : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f5b0996630884b58b60061d3b79d6e40 : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8bcd69c06b354ec7b5ade6345a8e083a : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cf4e8b412d24b4692e869c4f9d8d95c : SUCCESS in 7m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ff6821ed2ca043428e06ccf6fb10e3fa : SUCCESS in 5m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/394b92082ccc4a78ae794284e524c00d : SUCCESS in 9m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/86f15e1438d74f068f0bba21e574b889 : SUCCESS in 17m 34s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/5da1439506d243e3b52b22935342b478 : FAILURE in 30m 25s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/38e25437487a423a81815f99ca312fa3 : SUCCESS in 1h 00m 40s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/c9caaffe031b481b92585732d72d19a9 : SUCCESS in 16m 54s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/0dcd76ac957d4bb2880bf0a74cf286bf : SUCCESS in 1h 40m 18s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/4d25ad2ca52340e8867742e564e8c172 : SUCCESS in 15m 04s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/ec1380cf6aaa4c61aee28ba133e824d0 : SUCCESS in 20m 58s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/65edfbd2d8d747d6abf3a24b0e870a4a : SUCCESS in 12m 26s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"ce7acd30c8f8ddb6af5b559fd7967865558bb1ed","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-02 20:53:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"900e3f9a9bd6ddbed8106407f0c4d49c1040925b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-09 15:16:12.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"db30adb2daf29fa216a643e12b586c3ccee2e647","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 17:27:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b375347914284d1492625fc9254f80ce : SUCCESS in 8m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/727eb694764042d2919c0726730422f8 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/71bc679b09de46bead8007378ffaafa0 : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d90c3a64a5f49fea4b1833e68692f2c : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/864b65904b9e46e8b287e39a42e63182 : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1145291fd21c4f97934628a24ef47399 : SUCCESS in 6m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fe839edc7f8b46d9ad5ada92e310ed22 : SUCCESS in 5m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/10a4629248614ecfa7e768e175cdfa55 : SUCCESS in 8m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9f27438f772d424fbaa982a7973217f1 : SUCCESS in 20m 42s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/282ea56fb5d540cca6cb27d373840595 : SUCCESS in 1h 08m 25s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0529da7c15104aecaf0bbcc3e6eba2a4 : SUCCESS in 57m 37s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/1fb58edfc40340459101e663a53f2956 : SUCCESS in 1h 11m 18s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/4eb6107e67054251a76cfaccc42b7d71 : SUCCESS in 14m 13s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c4fe324057d04339b45a25a4d019bae9 : SUCCESS in 10m 50s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/737c89a5c4314952b75ffa6cf68b0579 : SUCCESS in 2h 06m 19s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/a58c84b3074b4f87bc2461fdabe26552 : SUCCESS in 25m 49s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/269070762aa94aefb0a470b8e7b49686 : SUCCESS in 22m 59s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/c4b1a7789cfb42b5b7b0fb9db0448509 : SUCCESS in 10m 34s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/77fb8bd610544da09ee478211d893b62 : SUCCESS in 8m 53s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"29ebe61cb8985dc67c4f4ecfec614e32e3d574ec","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-10 12:21:55.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dd62e380eeec3612450d64d7ae0d1503cea86146","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-15 14:18:44.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"6f6a187759af816f63cedef843072ef326e2fbf6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 16:28:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4606134185104b83a528c2904da6b222 : SUCCESS in 7m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5639710968ac4a39a5a6781a8ac4f035 : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f25ef30361ec412ea31fce02b3ebcc3a : SUCCESS in 7m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8169f3fa99e4f6e9f338489cce01b62 : SUCCESS in 6m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/406b3a78656a45d086a4f4ebac5d38cb : SUCCESS in 7m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f919cf5dbfda49718e4f38b0c746c6df : SUCCESS in 9m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d36f109417a44fa88ff42fdb61b3f930 : SUCCESS in 7m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/12a8836fe9f6410089461128a9fc5b04 : SUCCESS in 8m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c396ecf1144d4461866820a771bb89c2 : SUCCESS in 22m 31s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/88ba904aee9b471f8fc237fb468ca2c6 : SUCCESS in 1h 10m 48s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9377d0372a644047bfebc95ee69f6f5e : SUCCESS in 47m 24s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/a957537f13934604b2aa7e41e2f63c39 : SUCCESS in 1h 11m 52s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/f71cde284d15412787c91bce3309acc2 : SUCCESS in 19m 33s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb3bd84028f547f0a3d7d3b8532ee874 : SUCCESS in 13m 40s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/d8403364850e461aa020807919ce06ba : SUCCESS in 1h 31m 46s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/0021a015ffba496c90aeb97d0c0fee00 : SUCCESS in 32m 00s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f8e2f141c3dd4f52a0d1072675bf7a49 : SUCCESS in 24m 42s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/65cff3ff5b6f4fba95c889ee53ec87c9 : SUCCESS in 25m 51s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/81d9dbe9dc8f4a1c925e0a5c159713f9 : SUCCESS in 12m 12s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8aeff3d51bb44bc296632b35e6d63169 : SUCCESS in 12m 54s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":3},{"id":"f54215b5b6baa61ab16c7ad2b9695666384a961d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-25 13:19:13.000000000","message":"Patch Set 3: Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"f2e9479ca0c31c273230d2892c53c9bfd9495006","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-03-26 09:52:34.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"7f4f29bab64fca3e0971b98a9d9d78119e219673","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-03-30 10:28:55.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"ccadb9f9ea70ae37527377b36cd0648c2f6d0684","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 09:41:37.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"169c44687fe3f8c4f9fa2efef7898df4568e28f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 14:48:43.000000000","message":"Patch Set 3: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"c45c1b101c0852fa8864a4e9d242376e65f76a01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-31 14:58:47.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"325c741c6d8215bfdb47011a0b67e47a252ec47a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-31 15:01:48.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"7e86592de08fcf007f9bfe63e3dc7b950727a15b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 17:43:35.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5d40c10531834f91b888ab0b823e6f65 : SUCCESS in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46a826271a8a4705b95eab7928dbe882 : SUCCESS in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/46d054f4b95043cdad776d5d5770ddae : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/924b3ca8326145fc84bc757677aa0404 : SUCCESS in 7m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd3e274701944c43b84b0df16dc3b0ec : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/587e99602aa64391aec2c2a97fa97bfd : SUCCESS in 9m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d0dd611af57f4bdd9784d2c0bee50112 : SUCCESS in 7m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6258a6147f9e4b3684636272ce62de90 : SUCCESS in 11m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/33be04cb998d4906a84be8c5e3064655 : SUCCESS in 22m 06s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/2d5747c66dd14c7aa9aaf7eaa9a0217a : SUCCESS in 55m 59s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d9bb44dc66b4bb7aa22a606074fe0d0 : SUCCESS in 43m 52s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/d16920ebec024b8c863b7e5b528145ea : SUCCESS in 57m 07s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/55ab1ea938374f849c14567f8eff18aa : SUCCESS in 18m 11s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/281faf62fb9741f39402a689134bd479 : SUCCESS in 28m 15s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/d16adb43b1a944c3b97a2eb9433d02c8 : SUCCESS in 2h 15m 58s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/4733329e95a34f74a3bb47da14fd32a5 : SUCCESS in 34m 17s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/32972cce7d05409495bc58c51ffb1b5a : SUCCESS in 24m 19s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/1bd290db90734cbeba790278c2f5da38 : SUCCESS in 25m 58s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/13773b05fe5640dbb7ab774cf8a10cc4 : SUCCESS in 13m 17s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b76a3c2b2f5a4887a74e995da38c8e1d : SUCCESS in 12m 00s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":5},{"id":"a1886d1b0efba63fd33e6770d1e6c1cb50d61dee","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-01 14:39:16.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"7ee9ac5f24544fa096f658830ab41cea8b599c61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 14:40:39.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"a887443fb1e1b1a78f82a26082ff83413d7b5390","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 15:44:25.000000000","message":"Patch Set 5:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"b901bc7f4f08b2f881fcf0b2782b123cfc676bae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 17:51:19.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5856bd2436224fae892e2d9f75c4e8cc : SUCCESS in 9m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2aff95af28ed468dbf4d8c022443aac0 : SUCCESS in 8m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e609507b491c4226b916392fa250a63c : SUCCESS in 7m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a936dac68e64163a5b13260edacba22 : SUCCESS in 11m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0d19edf7b1f84930b9ebd92a4d20f9b1 : SUCCESS in 8m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7a458d4e39a446feb3b0645819484aba : SUCCESS in 11m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/33c597623b81499db3ff3006138f9a45 : SUCCESS in 19m 13s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/4e816278d6764b3ab04972d82f6fd4b8 : SUCCESS in 1h 00m 41s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c6731584311468b85032c3ef20e07c0 : SUCCESS in 47m 17s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/9a424822f069429e9d2d67acb725ca5f : SUCCESS in 1h 01m 45s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/9d8869e0035441ffa16772f32412cccb : SUCCESS in 26m 54s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3af8633c98bc4e68a29edb362c1dabdb : SUCCESS in 17m 46s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/237ae112c9ee480ca65908712add7de5 : SUCCESS in 1h 21m 44s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/904c926a61e54b61934ebaab9d6c4fd3 : SUCCESS in 29m 19s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d6951fefdbea40dfae310fbfcf5e71a6 : SUCCESS in 25m 27s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/ecf2aed328df4d699441237a2a86d334 : SUCCESS in 26m 26s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/5958c516704c49aca1487595944e1504 : SUCCESS in 13m 08s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0745d50a491f49de9170fe94354c836d : SUCCESS in 12m 20s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":5},{"id":"a835f8ab0cd9ec61f4b00c9a8227127e0411eaee","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 17:51:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"7646aac4e797b124a0b64c5b7e5809a11bcb91e2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 17:54:10.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c83a9555f02744faba862bf2f8546617 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"ea54f822999e6bc43e195f38cfd66caf05fc10ef","revisions":{"772f85e376450004386d82d8106e12971148147d":{"kind":"REWORK","_number":1,"created":"2021-02-02 16:04:09.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/19/773719/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/19/773719/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/773719/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/773719/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/773719/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/19/773719/1"}}},"commit":{"parents":[{"commit":"dceabffbcd3dbf4a21e9baab9b578eb3257027e2","subject":"Support passing multiple Kayobe inventory arguments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dceabffbcd3dbf4a21e9baab9b578eb3257027e2"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-02 16:02:38.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-02 16:02:38.000000000","tz":60},"subject":"Detect multiple inventories automatically","message":"Detect multiple inventories automatically\n\nWhen inventory arguments are not passed but an environment is specified,\ndetect if there are inventory directories present in the shared Kayobe\nconfiguration directory and in the environment directory. Using\nos.path.exists() is sufficient because we validate these paths later\nwith utils.is_readable_dir().\n\nChange-Id: Ieb2b7ff07cd43029399e9e19002b82940246f8c4\nStory: 2002009\nTask: 41739\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/772f85e376450004386d82d8106e12971148147d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/772f85e376450004386d82d8106e12971148147d"}]},"branch":"refs/heads/master"},"a7a50d2ac37a779285315038cbcc1204944f6c44":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-03-09 15:16:12.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/19/773719/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/19/773719/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/773719/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/773719/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/773719/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/19/773719/2"}}},"commit":{"parents":[{"commit":"c5f9e23579488dea9b2fac07676a6196d01a1606","subject":"Support passing multiple Kayobe inventory arguments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c5f9e23579488dea9b2fac07676a6196d01a1606"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-02 16:02:38.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-09 15:15:12.000000000","tz":60},"subject":"Detect multiple inventories automatically","message":"Detect multiple inventories automatically\n\nWhen inventory arguments are not passed but an environment is specified,\ndetect if there are inventory directories present in the shared Kayobe\nconfiguration directory and in the environment directory. Using\nos.path.exists() is sufficient because we validate these paths later\nwith utils.is_readable_dir().\n\nChange-Id: Ieb2b7ff07cd43029399e9e19002b82940246f8c4\nStory: 2002009\nTask: 41739\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a7a50d2ac37a779285315038cbcc1204944f6c44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a7a50d2ac37a779285315038cbcc1204944f6c44"}]},"branch":"refs/heads/master"},"3f475638c77aee08b75770fdab3aee65923f5c0d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-03-15 14:18:44.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/19/773719/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/19/773719/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/773719/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/773719/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/773719/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/19/773719/3"}}},"commit":{"parents":[{"commit":"97a1e9143f28b9b4365ef2db374b2eb2822c2086","subject":"Support passing multiple Kayobe inventory arguments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/97a1e9143f28b9b4365ef2db374b2eb2822c2086"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-02 16:02:38.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-15 14:17:28.000000000","tz":60},"subject":"Detect multiple inventories automatically","message":"Detect multiple inventories automatically\n\nWhen inventory arguments are not passed but an environment is specified,\ndetect if there are inventory directories present in the shared Kayobe\nconfiguration directory and in the environment directory. Using\nos.path.exists() is sufficient because we validate these paths later\nwith utils.is_readable_dir().\n\nChange-Id: Ieb2b7ff07cd43029399e9e19002b82940246f8c4\nStory: 2002009\nTask: 41739\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3f475638c77aee08b75770fdab3aee65923f5c0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3f475638c77aee08b75770fdab3aee65923f5c0d"}]},"branch":"refs/heads/master"},"70e9a2cacfc9ea594b552a730417ff10e5e99b1f":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-03-31 14:58:47.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/19/773719/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/19/773719/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/773719/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/773719/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/773719/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/19/773719/4"}}},"commit":{"parents":[{"commit":"42513e05dd2964dafeb7808365774c3724cbd87f","subject":"Support passing multiple Kayobe inventory arguments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/42513e05dd2964dafeb7808365774c3724cbd87f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-02 16:02:38.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-31 14:50:11.000000000","tz":120},"subject":"Detect multiple inventories automatically","message":"Detect multiple inventories automatically\n\nWhen inventory arguments are not passed but an environment is specified,\ndetect if there are inventory directories present in the shared Kayobe\nconfiguration directory and in the environment directory. Using\nos.path.exists() is sufficient because we validate these paths later\nwith utils.is_readable_dir().\n\nChange-Id: Ieb2b7ff07cd43029399e9e19002b82940246f8c4\nStory: 2002009\nTask: 41739\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/70e9a2cacfc9ea594b552a730417ff10e5e99b1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/70e9a2cacfc9ea594b552a730417ff10e5e99b1f"}]},"branch":"refs/heads/master"},"ea54f822999e6bc43e195f38cfd66caf05fc10ef":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-03-31 15:01:48.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/19/773719/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/19/773719/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/773719/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/773719/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/773719/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/19/773719/5"}}},"commit":{"parents":[{"commit":"c15ed69b18dfbf4de69d8bf2ca01c934cb4fbf54","subject":"Support passing multiple Kayobe inventory arguments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c15ed69b18dfbf4de69d8bf2ca01c934cb4fbf54"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-02 16:02:38.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-31 15:01:35.000000000","tz":120},"subject":"Detect multiple inventories automatically","message":"Detect multiple inventories automatically\n\nWhen inventory arguments are not passed but an environment is specified,\ndetect if there are inventory directories present in the shared Kayobe\nconfiguration directory and in the environment directory. Using\nos.path.exists() is sufficient because we validate these paths later\nwith utils.is_readable_dir().\n\nChange-Id: Ieb2b7ff07cd43029399e9e19002b82940246f8c4\nStory: 2002009\nTask: 41739\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ea54f822999e6bc43e195f38cfd66caf05fc10ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ea54f822999e6bc43e195f38cfd66caf05fc10ef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
