)]}'
{"id":"openstack%2Fkayobe~773853","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~I4719b4d397c01b35c78cb84c6d686dd27742d1c0","project":"openstack/kayobe","branch":"stable/train","topic":"fix-centos7-ipa-train","hashtags":[],"change_id":"I4719b4d397c01b35c78cb84c6d686dd27742d1c0","subject":"Test building seed deployment images in the seed job (CentOS 8 only)","status":"MERGED","created":"2021-02-03 09:12:51.000000000","updated":"2021-03-10 13:38:40.000000000","submitted":"2021-03-10 13:36:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"773853-fix-centos7-ipa-train","meta_rev_id":"fad96b34121124e0cb1bbeb8446042b7dc9952ee","_number":773853,"virtual_id_number":773853,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-10 13:36:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-03-10 12:01:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-03-10 13:38:40.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-03-10 12:01:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-03 11:11:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-17 10:15:50.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2021-02-22 10:16:40.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-03-10 12:01:36.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"d8ed83433f58ddfabe6e17ad36d52156eaa08f24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-03 09:12:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"df828c08e519b6844ddd1054fd9d24f416ed208f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-03 09:27:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"31474b8d5f81a85ddf7520f5f5d44c328b9d066c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-03 09:34:24.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"52d81713055f411e96889fc2564d7711f4cb81c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 11:11:05.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e0ced0f2734d4600aa7b3b462ed00f59 : SUCCESS in 3m 53s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5253966635a041dc8fb069804bd8cebc : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/26fea2ced32945ad97263f1bf7d37ca9 : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23514a5cd2df4ff48300c91969f42c25 : SUCCESS in 3m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/44493ca339134251a5e424e3e3824c98 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a611ed98dbe4867b5c6923bb6831df4 : SUCCESS in 6m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/049bd91e6e714d15bf9cbb474b4c7caa : SUCCESS in 3m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4d3aa981c53f4c5e85e2e6a8a7d102ab : SUCCESS in 7m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/65a1fb73ba7c4970990c4dba4b66ba24 : SUCCESS in 16m 25s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/205371be6c4d439c82d44ec2c3ecfed8 : SUCCESS in 1h 16m 02s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/c466cfc017d646e398f63a25cde59340 : FAILURE in 22m 58s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/f0bde12b026447869987264e0480b22d : SUCCESS in 13m 31s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/99957dce89d94a189643b9cf64b32fba : SUCCESS in 13m 49s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/6529182916614a909d82d31738b85e91 : FAILURE in 27m 25s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/81df0109df9441b68ab9a7516f035182 : FAILURE in 21m 45s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/eab15ec6693345e2a968cd3dab054dff : FAILURE in 26m 51s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/b55a629f6dbc4c44acfdc8c01fc1ff46 : FAILURE in 24m 26s","accounts_in_message":[],"_revision_number":3},{"id":"5b66d80c8a1085d047f1a0ffe977c5bca4f76ed6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-12 08:51:00.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"f9929b2a3377ee57096fee50d80918d16bd2aa12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-16 21:36:32.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"719a59521446a9360e4322a8b93f936f25db13f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 23:14:50.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0bdd5c6bd10740dc91e42bc9a59b8afc : SUCCESS in 5m 18s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b893cd00c54a4264ade4d759318bafcf : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf1e1f7961404e35bad55541040b7b7a : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a9c9d4be16b477a8afecae2e7c984fc : SUCCESS in 5m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6d4e1e78123842dc860cd182e5dcfbcd : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de49893ea9f444c0bb47bf49b94a3259 : SUCCESS in 5m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f470dab482cf448c802ec36b022b5142 : SUCCESS in 4m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ead6755157fc41fa9166c8db75e92968 : SUCCESS in 8m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a558359d5b3346ecbaf0bad7a351fce7 : SUCCESS in 17m 45s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/da67b7d000bc4090a0f88d5807d88e20 : SUCCESS in 1h 04m 15s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/c84aa3c1bc4c42448c66ab16539d9568 : SUCCESS in 59m 36s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/a37e6b6870a44bc39a5d623197943f97 : SUCCESS in 13m 21s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/0bdc401b8c194336876ec1fdc4128ceb : SUCCESS in 14m 36s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/9c2d9221e9c24e84923159cc639f606a : SUCCESS in 1h 26m 53s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/c3e539d0abfa4343b1cb5e4f2b7f5f81 : FAILURE in 25m 07s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/49b9c6374fd3408da86b666d5aaa191a : FAILURE in 20m 49s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/43a27f3eba9d44fb954c4cb148458939 : SUCCESS in 32m 14s","accounts_in_message":[],"_revision_number":4},{"id":"95fefcfe71d71e86ed0da3c418cac052a13a3d6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-17 07:07:15.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"1a7c3076274901fd068af416644cf6ae3a127fe3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-17 07:26:13.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"d525a47a8e08989eca5c3278b0a4adffd1ddc95c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 09:02:54.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/43fef0535d924d7bab0602a8fa605df3 : SUCCESS in 4m 39s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/32735d8333174d5fb0d317afd69d3fda : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b68d991285e14a74b981c349329bef1d : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/656a2f13997d4117bd11d6911f370454 : SUCCESS in 4m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d759fb028f6b41d388fdde3c05878ff9 : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8152c9a31c1479b8dc2e14b34a5c30b : SUCCESS in 6m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b4b344773ee5407ab97aea6db5242b4b : SUCCESS in 4m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a88416955efc4eeca5616d9141841dd6 : SUCCESS in 8m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/aebefab4a4c54ba99d1cfebd831e390c : SUCCESS in 15m 43s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/ab48243fd00d4be194c0055b6608900e : SUCCESS in 1h 05m 09s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/f98a28f437e5400fad67c9a63b9f703c : SUCCESS in 1h 06m 43s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/e96d66052eb146728ec685df97dcdc3e : SUCCESS in 14m 08s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/0dad65ff382544b7b29d873087aed9cd : SUCCESS in 16m 36s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/003242f3eb9f4572ad45c2ff8e346456 : SUCCESS in 1h 35m 34s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/5d778073134248e791891f4e40ea5dca : SUCCESS in 16m 56s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/e9625985eb0447a79b606fbc3538f411 : FAILURE in 25m 39s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/dfc4b659a5974fe2b73da1473bcbcd76 : SUCCESS in 32m 54s","accounts_in_message":[],"_revision_number":6},{"id":"a4930865106e4dc8bbbf0f87af53230493e10e81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-17 09:10:10.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"37ecca2d6d412647067266573c4f3293cb2baef7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-17 09:37:54.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"dc693e1152b0d4932db76bae51c42c15565a3c0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 10:09:56.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/be5bbb4dc3384a6f9213f7c0df0b5ff5 : FAILURE in 11m 23s","accounts_in_message":[],"_revision_number":8},{"id":"ffb47768f83cefc131fb3f4e82ff625a8a60d7fe","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-17 10:15:50.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"9921f3f1f2ecc85cdf86f4059ddc50572fe44dc5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-17 10:23:07.000000000","message":"Patch Set 8: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"51818255bf639fe0c5e7c8bbef8b10982efde81e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-17 11:19:52.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"aa7fd68d657289f4ade2f42d0606daaa8063c41b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 11:47:31.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/b996b2fb023642fe8fdca3a87e1d5238 : SUCCESS in 23m 15s","accounts_in_message":[],"_revision_number":9},{"id":"8ef472a5abc7de91a7e14e15d3b7fdffb20db67d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-17 12:10:52.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"a817fc8e38b2ca139e01bdd9b026ef788f1b7eab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-17 12:10:57.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"65c018d2fbd4595ce6a814a2022d8f130181d5e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-17 12:18:34.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"bd56b0ea8b6ce8ceec186ef7a8e26a491b863727","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 14:29:35.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/15d291a9b3ac4addb806334ac805dbbd : SUCCESS in 4m 35s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/09a8773f7ff24a8988ba017f6d5d07ff : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c6b55cca8eb4775a10b196dadc64f1c : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02471fde562f4d2d972eeaf7cbdb534f : SUCCESS in 4m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a5fcd089864f4e1aaa3d9b88c17abf4f : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/155d9b1cb7dc46289fd006e6c9b51967 : SUCCESS in 7m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a32a7307c1ac42249614fa0198cdeff5 : SUCCESS in 4m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/14e2ed32749745c3888980bff0a4a564 : SUCCESS in 9m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d93259ec0bfb4d1da2bbe12f7689d058 : SUCCESS in 17m 13s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/69800f9269244e9eb8f50751e757b393 : SUCCESS in 1h 11m 06s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/862e7a32762f4044bfb62e483b151cac : TIMED_OUT in 2h 01m 30s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/6ac58a4dbf064eccad3fa0a6a2a8f1e2 : SUCCESS in 14m 34s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/3d48b6ee9e0f43c7b5f9800d7f1d2a48 : SUCCESS in 16m 08s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/7b5993e665434fc3bb09e1f6fa369128 : SUCCESS in 1h 23m 54s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/39b17eb84e5d4cbd9e46f34c0545eda4 : SUCCESS in 19m 13s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/1a5716253fe7426ba00ef24c4e371a96 : FAILURE in 15m 38s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/f9e248a11e5a4e9d98845649012c677e : SUCCESS in 31m 50s","accounts_in_message":[],"_revision_number":11},{"id":"b651841954d61e164daa061f396f891669c0ee27","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-17 14:35:57.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"9af0935117c0a22d6b88fae33c55a7da65f350bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 15:46:38.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/63417963025943279a3f41aa1998bc91 : SUCCESS in 4m 56s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/59656196f85d4fee83c771d60377681b : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/65ce59112ddd46ea979fa16a00306850 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc5ea230ecea4f608a45dcc48a215e9a : SUCCESS in 4m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f6326147ad5246e7ba4cb019177b5f32 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a968283f855840f2811eff5f8fd5eb7f : SUCCESS in 7m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f8488cf558024ed897cfa2109daae70c : SUCCESS in 4m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a5fca8366df24f8e8c722cd9984cb23a : SUCCESS in 9m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3baeee7311dc40b98db7d0aaea0ca914 : SUCCESS in 17m 02s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/aa118af4650347f4a7c7da8b23405886 : SUCCESS in 1h 06m 54s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/7fb6b521911340988048c92471920412 : SUCCESS in 1h 01m 36s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/7a3441ef887247c7a229627db4129d2a : SUCCESS in 14m 10s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/c53e6a21cc8f439f878444b9fe5845c0 : SUCCESS in 16m 15s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/57eb9aa2d9744b859971b0608a5064f6 : FAILURE in 57m 11s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/1bd90462e9a24565b1fb5c78d7411c06 : SUCCESS in 20m 47s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/9b26d3d5130f4d2688f3b89733302140 : SUCCESS in 27m 43s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/12073f6af351401f9e7d01ff29030773 : SUCCESS in 32m 41s","accounts_in_message":[],"_revision_number":11},{"id":"1b245aee37b81c0d9e6d3ba4bad2046bc5a87c70","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-18 14:07:42.000000000","message":"Patch Set 11:\n\nrecheck\n\nDocker pull limits during upgrade","accounts_in_message":[],"_revision_number":11},{"id":"9a6eb5bd1fca178c820f59e2550665ba946b16ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 15:42:33.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd3a0a532b124845a497bb059f1921c9 : SUCCESS in 5m 58s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b4383d5c0bbc4c33a2ff35760eae5298 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15a3fada1b364a96a7c0d2ca63a91b4f : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/424d20e8c881445da6169ea84d7a169c : SUCCESS in 5m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ee9236eabd9c4e8cbdb29d87d10b095d : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f54c201517854f5c84bb7ec291991f2f : SUCCESS in 9m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7d711279eda4426288f0340743d46f62 : SUCCESS in 4m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8a4eb0269f8e419c85337af30b5dfce1 : SUCCESS in 10m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/da4a9c6de89047e9bfedaa73743c321c : SUCCESS in 17m 58s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/b4beae98637b4186b2acb8b8d91cf748 : SUCCESS in 1h 08m 17s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/6b1b8c004f0f4cf096af410b9160be6e : SUCCESS in 1h 04m 14s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/67148505fff6426b912bf224771c6f0e : SUCCESS in 15m 01s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/f02464c362a248a1ace924d31ba6417a : SUCCESS in 13m 05s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/636330410a2f4c17a41a10a709924ce0 : FAILURE in 1h 00m 35s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/3b5da12977c9470baa8c906dd0667562 : SUCCESS in 23m 36s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/60287053907a43c68826c3816eff6994 : SUCCESS in 30m 38s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/0d5c7f531be74b918294daa61c5bbc0f : SUCCESS in 38m 31s","accounts_in_message":[],"_revision_number":11},{"id":"689d7a026bf55bf7de3503a6c01ae9f5e95fb260","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-19 13:58:45.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"53f19b958df250f9253c4b8da1fe754917705679","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 16:08:00.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/288c73b96b5648fcafed81050c827fe4 : SUCCESS in 4m 39s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3234ddf247bf4730a740941220215f94 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/39ec9bdf8b8b427ab2906f64bb9968cd : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/124eeb663434472ea6e7512c3dfaa516 : SUCCESS in 5m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4c32a7aff4bc4036a19f52fa1252d9ef : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abafaf0630a5419cb34a8d3f8ecdcb1f : SUCCESS in 6m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/072de364e7bb49318df599258cf8b44d : SUCCESS in 4m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a23b8ddcd46443e5b51cc893051d4b24 : SUCCESS in 9m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8500dd96b757478a8f200d26c04040c6 : SUCCESS in 17m 21s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/12f948f2b1d34aedab8385a39b273cf8 : SUCCESS in 1h 06m 53s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/9d4de27672474e4892089bc38ef3ae8f : SUCCESS in 1h 00m 48s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/fbb76ae90b8e471b918a908261fb2f00 : SUCCESS in 16m 19s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/9639b2b825894de4830bbc4c945540e9 : SUCCESS in 23m 49s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/2ccaea7436fe446497d23647de4017bc : SUCCESS in 1h 55m 19s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/4505cd63ec3e412aa8a57d6d46e17af2 : SUCCESS in 30m 50s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/06bc67202d874e789924f72143cb6d72 : SUCCESS in 25m 03s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/beddbd946c1c4fd2809c1d376e7cef26 : SUCCESS in 42m 26s","accounts_in_message":[],"_revision_number":11},{"id":"2d04dfaae519d47a614ea13ba5281d36e481d657","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-22 10:16:40.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"b164465f4044c3e899b703071abd96aa009931ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-05 14:57:13.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"0a73090936ffc1afe5be0a954cc8c810bd4960e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 16:32:35.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1edadd97bc54574a8bbe72e3f933de4 : SUCCESS in 10m 27s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4b6c90000d3342d2b6d3d62d52b4547f : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/12debf61b7254b28819dd03f038e4455 : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/881561809cd943bba6155dee11f9ef5c : SUCCESS in 4m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b940ca92c36d4fd8a969bec88c714bd7 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3343562a1e0c46d68d61270fd6fa9a67 : SUCCESS in 6m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f4b2dfb60cc4402ba89fc490b991b340 : SUCCESS in 4m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cd5d1ffd73d84fabbb0a73a218749470 : SUCCESS in 8m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6cff3cb00ee346018473cc0cd62aec9b : SUCCESS in 20m 11s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/bd7dec727eef4f4896dc3f2c9f4ca780 : SUCCESS in 1h 14m 17s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/d06aa0cdc78c41f195db227595998137 : SUCCESS in 1h 08m 39s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/0bf0d346bc8246459805f53d81aa43e9 : SUCCESS in 14m 50s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/90b154f94a194d5aafcc5a87f45c7962 : SUCCESS in 14m 11s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/51747157117c47babcccf74317a84dcb : SUCCESS in 1h 22m 54s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/3f976d01bf1746deadbd7c5088f27e81 : SUCCESS in 17m 37s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/5858e64d853d40d4b42167114b1b746c : SUCCESS in 24m 11s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/097e624d3fd34110a6053311cfd33b63 : SUCCESS in 28m 28s","accounts_in_message":[],"_revision_number":12},{"id":"40357f4a57c77df822a818284eb2eda9d65798b6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-03-10 12:01:36.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"34ea570cf5ad6f6913a62a4c2dcd5470e9e33677","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 12:01:52.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"311332d6b19e506a13c00f5ef2a207a43b482c0b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 13:36:18.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/dc65b9fd912f4dcd9555d19e2ec2bf52 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac54581c09ca4931aa10bf27e7b4f0b1 : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f69b6b6758f24452aec810d87a457eea : SUCCESS in 4m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/afbde22cd4ed48e08be83f4697ddaf13 : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2e3ba40b7764b37a033bfcbb06ebe06 : SUCCESS in 8m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bd9c1fec1fd34f26996593f344b63e1e : SUCCESS in 4m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/55571200f21642449e71e5df1c6d0236 : SUCCESS in 8m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1eeb653b45bc44de9125741c90ee4f3c : SUCCESS in 18m 10s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/8c5bb712f14347ab813e2d2cae4cd416 : SUCCESS in 1h 06m 47s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/8ecd862556284bada5b076a5c3cc5d18 : SUCCESS in 52m 48s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/f8e25c45f2b9499cadf93144e3eb98c3 : SUCCESS in 12m 32s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/b274ce95ef5546719000affd1579883d : SUCCESS in 16m 22s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/cab16412d16f4547bb27c75f26005d9e : SUCCESS in 1h 30m 04s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/946f2e3b2c104c4198c0c408317cd704 : SUCCESS in 22m 29s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/f0ad621b42da47348a02d524225225c4 : SUCCESS in 27m 25s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/fe7d5edee2274de386d678fb3249bcd4 : SUCCESS in 38m 12s","accounts_in_message":[],"_revision_number":12},{"id":"bbef0e3fe67b56ce4b1686b7df2758988af47a5d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 13:36:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":12},{"id":"23071c23037f4b1026f6e1ae17ebfaf7d3e5bb98","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 13:38:40.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0236cbbc306445bab148fdda03b6287f : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"7640af70d07c9f8f7a31d3545e3d8e763057bd45","revisions":{"5e3efcaf36b891b5e58dc5fdff6ded29a318f5ca":{"kind":"REWORK","_number":1,"created":"2021-02-03 09:12:51.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/53/773853/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/53/773853/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/773853/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/773853/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/773853/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/53/773853/1"}}},"commit":{"parents":[{"commit":"e0e0db459d3c99fbc1cb266739fa7e6e17f0c00e","subject":"Fix Python setup when venv is not used","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e0e0db459d3c99fbc1cb266739fa7e6e17f0c00e"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-01-28 09:40:27.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-03 09:12:08.000000000","tz":60},"subject":"Test building seed deployment images in the seed job","message":"Test building seed deployment images in the seed job\n\nThis requires stackhpc.os-images v1.10.0 or newer, for compatibility\nwith CentOS 8 when SELinux is enabled: we disable SELinux, but without\nrebooting it stays enabled. This Ansible role is updated to v1.10.2 by\nthe following change, for another reason. (772774)\n\nFinally, it requires building seed deployment images only after\ndeploying Bifrost, because the task copying images onto the seed expects\n/etc/kolla/bifrost to exist.\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe/+/772774\n\nChange-Id: I4719b4d397c01b35c78cb84c6d686dd27742d1c0\n(cherry picked from commit 1a012675cf73763ac42bed1cfc7117ff77071f35)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5e3efcaf36b891b5e58dc5fdff6ded29a318f5ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5e3efcaf36b891b5e58dc5fdff6ded29a318f5ca"}]},"branch":"refs/heads/stable/train"},"e5468defbe086ed4a08879ac91fa2c6a22abbf7f":{"kind":"REWORK","_number":2,"created":"2021-02-03 09:27:12.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/53/773853/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/53/773853/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/773853/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/773853/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/773853/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/53/773853/2"}}},"commit":{"parents":[{"commit":"e0e0db459d3c99fbc1cb266739fa7e6e17f0c00e","subject":"Fix Python setup when venv is not used","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e0e0db459d3c99fbc1cb266739fa7e6e17f0c00e"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-03 09:25:35.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-03 09:25:59.000000000","tz":60},"subject":"Test building seed deployment images in the seed job","message":"Test building seed deployment images in the seed job\n\nThis requires stackhpc.os-images v1.10.0 or newer, for compatibility\nwith CentOS 8 when SELinux is enabled: we disable SELinux, but without\nrebooting it stays enabled. Update the role to v1.10.3 which is the\nlatest version available.\n\nIt also requires building seed deployment images only after deploying\nBifrost, because the task copying images onto the seed expects\n/etc/kolla/bifrost to exist.\n\nChange-Id: I4719b4d397c01b35c78cb84c6d686dd27742d1c0\n(cherry picked from commit 1a012675cf73763ac42bed1cfc7117ff77071f35)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e5468defbe086ed4a08879ac91fa2c6a22abbf7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e5468defbe086ed4a08879ac91fa2c6a22abbf7f"}]},"branch":"refs/heads/stable/train"},"a68a2911e42a66d0cb53361abe573a028b4c89d7":{"kind":"REWORK","_number":3,"created":"2021-02-03 09:34:24.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/53/773853/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/53/773853/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/773853/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/773853/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/773853/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/53/773853/3"}}},"commit":{"parents":[{"commit":"e0e0db459d3c99fbc1cb266739fa7e6e17f0c00e","subject":"Fix Python setup when venv is not used","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e0e0db459d3c99fbc1cb266739fa7e6e17f0c00e"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-03 09:25:35.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-03 09:34:17.000000000","tz":60},"subject":"Test building seed deployment images in the seed job","message":"Test building seed deployment images in the seed job\n\nThis requires stackhpc.os-images v1.10.0 or newer, for compatibility\nwith CentOS 8 when SELinux is enabled: we disable SELinux, but without\nrebooting it stays enabled. Update the role to v1.10.3 which is the\nlatest version available.\n\nIt also requires building seed deployment images only after deploying\nBifrost, because the task copying images onto the seed expects\n/etc/kolla/bifrost to exist.\n\nChange-Id: I4719b4d397c01b35c78cb84c6d686dd27742d1c0\n(cherry picked from commit 1a012675cf73763ac42bed1cfc7117ff77071f35)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a68a2911e42a66d0cb53361abe573a028b4c89d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a68a2911e42a66d0cb53361abe573a028b4c89d7"}]},"branch":"refs/heads/stable/train"},"3f324f192b6e2d1b4532ae91f602759576e5f99f":{"kind":"REWORK","_number":4,"created":"2021-02-16 21:36:32.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/53/773853/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/53/773853/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/773853/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/773853/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/773853/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/53/773853/4"}}},"commit":{"parents":[{"commit":"70f2cb0f60b037ad2923626d9aaccd56a29bc7b7","subject":"Merge \"Fix /tmp/swift-rings/backups/ files deletion\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/70f2cb0f60b037ad2923626d9aaccd56a29bc7b7"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-16 21:32:16.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-16 21:32:16.000000000","tz":60},"subject":"Test building seed deployment images in the seed job","message":"Test building seed deployment images in the seed job\n\nThis requires stackhpc.os-images v1.10.0 or newer, for compatibility\nwith CentOS 8 when SELinux is enabled: we disable SELinux, but without\nrebooting it stays enabled.\n\nThis Ansible role was updated to v1.10.2 in master and stable/victoria\nby I5efdbd52556721914fe69d7c6ba454b2c721b643, for another reason.\nRemember to bump the requirement when backporting to earlier releases.\n\nFinally, it requires building seed deployment images only after\ndeploying Bifrost, because the task copying images onto the seed expects\n/etc/kolla/bifrost to exist.\n\nWe also copy log files to identify issues when the job fails.\n\nChange-Id: I4719b4d397c01b35c78cb84c6d686dd27742d1c0\n(cherry picked from commit c84a9757dd449dbd8863bb1dcc55837abe26a716)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3f324f192b6e2d1b4532ae91f602759576e5f99f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3f324f192b6e2d1b4532ae91f602759576e5f99f"}]},"branch":"refs/heads/stable/train"},"23766109809b99aaecc088307d3351fd427e57d6":{"kind":"REWORK","_number":5,"created":"2021-02-17 07:07:15.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/53/773853/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/53/773853/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/773853/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/773853/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/773853/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/53/773853/5"}}},"commit":{"parents":[{"commit":"40d410907be8fe1af9b77bd2c5ad842f48c82f0c","subject":"Fix building CentOS 8 IPA images on stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/40d410907be8fe1af9b77bd2c5ad842f48c82f0c"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-16 21:32:16.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-17 07:07:05.000000000","tz":60},"subject":"Test building seed deployment images in the seed job (CentOS 8 only)","message":"Test building seed deployment images in the seed job (CentOS 8 only)\n\nThis requires stackhpc.os-images v1.10.0 or newer, for compatibility\nwith CentOS 8 when SELinux is enabled: we disable SELinux, but without\nrebooting it stays enabled.\n\nThis Ansible role was updated to v1.10.2 in master and stable/victoria\nby I5efdbd52556721914fe69d7c6ba454b2c721b643, for another reason.\nRemember to bump the requirement when backporting to earlier releases.\n\nFinally, it requires building seed deployment images only after\ndeploying Bifrost, because the task copying images onto the seed expects\n/etc/kolla/bifrost to exist.\n\nWe also copy log files to identify issues when the job fails.\n\nChange-Id: I4719b4d397c01b35c78cb84c6d686dd27742d1c0\n(cherry picked from commit c84a9757dd449dbd8863bb1dcc55837abe26a716)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/23766109809b99aaecc088307d3351fd427e57d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/23766109809b99aaecc088307d3351fd427e57d6"}]},"branch":"refs/heads/stable/train"},"8f886f23e30622559b0fb95fbce5e79c4e90824f":{"kind":"REWORK","_number":6,"created":"2021-02-17 07:26:13.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/53/773853/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/53/773853/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/773853/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/773853/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/773853/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/53/773853/6"}}},"commit":{"parents":[{"commit":"40d410907be8fe1af9b77bd2c5ad842f48c82f0c","subject":"Fix building CentOS 8 IPA images on stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/40d410907be8fe1af9b77bd2c5ad842f48c82f0c"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-16 21:32:16.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-17 07:25:26.000000000","tz":60},"subject":"Test building seed deployment images in the seed job (CentOS 8 only)","message":"Test building seed deployment images in the seed job (CentOS 8 only)\n\nThis requires stackhpc.os-images v1.10.0 or newer, for compatibility\nwith CentOS 8 when SELinux is enabled: we disable SELinux, but without\nrebooting it stays enabled.\n\nThis Ansible role was updated to v1.10.2 in master and stable/victoria\nby I5efdbd52556721914fe69d7c6ba454b2c721b643, for another reason.\nRemember to bump the requirement when backporting to earlier releases.\n\nFinally, it requires building seed deployment images only after\ndeploying Bifrost, because the task copying images onto the seed expects\n/etc/kolla/bifrost to exist.\n\nWe also copy log files to identify issues when the job fails.\n\nChange-Id: I4719b4d397c01b35c78cb84c6d686dd27742d1c0\n(cherry picked from commit c84a9757dd449dbd8863bb1dcc55837abe26a716)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8f886f23e30622559b0fb95fbce5e79c4e90824f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8f886f23e30622559b0fb95fbce5e79c4e90824f"}]},"branch":"refs/heads/stable/train"},"1c70c1b00a2f5e328fe65ab58067a4ecf2677bb5":{"kind":"REWORK","_number":7,"created":"2021-02-17 09:10:10.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/53/773853/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/53/773853/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/773853/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/773853/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/773853/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/53/773853/7"}}},"commit":{"parents":[{"commit":"40d410907be8fe1af9b77bd2c5ad842f48c82f0c","subject":"Fix building CentOS 8 IPA images on stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/40d410907be8fe1af9b77bd2c5ad842f48c82f0c"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-16 21:32:16.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-17 09:08:22.000000000","tz":60},"subject":"Test building seed deployment images in the seed job (CentOS 8 only)","message":"Test building seed deployment images in the seed job (CentOS 8 only)\n\nThis requires stackhpc.os-images v1.10.0 or newer, for compatibility\nwith CentOS 8 when SELinux is enabled: we disable SELinux, but without\nrebooting it stays enabled.\n\nThis Ansible role was updated to v1.10.2 in master and stable/victoria\nby I5efdbd52556721914fe69d7c6ba454b2c721b643, for another reason.\nRemember to bump the requirement when backporting to earlier releases.\n\nFinally, it requires building seed deployment images only after\ndeploying Bifrost, because the task copying images onto the seed expects\n/etc/kolla/bifrost to exist.\n\nWe also copy log files to identify issues when the job fails.\n\nChange-Id: I4719b4d397c01b35c78cb84c6d686dd27742d1c0\n(cherry picked from commit c84a9757dd449dbd8863bb1dcc55837abe26a716)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1c70c1b00a2f5e328fe65ab58067a4ecf2677bb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1c70c1b00a2f5e328fe65ab58067a4ecf2677bb5"}]},"branch":"refs/heads/stable/train"},"8045048fb8dff39bce8283a738276f3448c43334":{"kind":"REWORK","_number":8,"created":"2021-02-17 09:37:54.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/53/773853/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/53/773853/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/773853/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/773853/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/773853/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/53/773853/8"}}},"commit":{"parents":[{"commit":"40d410907be8fe1af9b77bd2c5ad842f48c82f0c","subject":"Fix building CentOS 8 IPA images on stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/40d410907be8fe1af9b77bd2c5ad842f48c82f0c"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-16 21:32:16.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-17 09:37:43.000000000","tz":60},"subject":"Test building seed deployment images in the seed job (CentOS 8 only)","message":"Test building seed deployment images in the seed job (CentOS 8 only)\n\nThis requires stackhpc.os-images v1.10.0 or newer, for compatibility\nwith CentOS 8 when SELinux is enabled: we disable SELinux, but without\nrebooting it stays enabled.\n\nThis Ansible role was updated to v1.10.2 in master and stable/victoria\nby I5efdbd52556721914fe69d7c6ba454b2c721b643, for another reason.\nRemember to bump the requirement when backporting to earlier releases.\n\nFinally, it requires building seed deployment images only after\ndeploying Bifrost, because the task copying images onto the seed expects\n/etc/kolla/bifrost to exist.\n\nWe also copy log files to identify issues when the job fails.\n\nChange-Id: I4719b4d397c01b35c78cb84c6d686dd27742d1c0\n(cherry picked from commit c84a9757dd449dbd8863bb1dcc55837abe26a716)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8045048fb8dff39bce8283a738276f3448c43334"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8045048fb8dff39bce8283a738276f3448c43334"}]},"branch":"refs/heads/stable/train"},"6118b33951cb897dfee8a5222b7dcc0a6acc4d34":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2021-02-17 11:19:52.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/53/773853/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/53/773853/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/773853/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/773853/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/773853/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/53/773853/9"}}},"commit":{"parents":[{"commit":"970721d2d53fc10d84ba2c0dc5c96a58ecf5d60e","subject":"Fix building CentOS 8 IPA images on stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/970721d2d53fc10d84ba2c0dc5c96a58ecf5d60e"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-16 21:32:16.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-17 11:19:17.000000000","tz":60},"subject":"Test building seed deployment images in the seed job (CentOS 8 only)","message":"Test building seed deployment images in the seed job (CentOS 8 only)\n\nThis requires stackhpc.os-images v1.10.0 or newer, for compatibility\nwith CentOS 8 when SELinux is enabled: we disable SELinux, but without\nrebooting it stays enabled.\n\nThis Ansible role was updated to v1.10.2 in master and stable/victoria\nby I5efdbd52556721914fe69d7c6ba454b2c721b643, for another reason.\nRemember to bump the requirement when backporting to earlier releases.\n\nFinally, it requires building seed deployment images only after\ndeploying Bifrost, because the task copying images onto the seed expects\n/etc/kolla/bifrost to exist.\n\nWe also copy log files to identify issues when the job fails.\n\nChange-Id: I4719b4d397c01b35c78cb84c6d686dd27742d1c0\n(cherry picked from commit c84a9757dd449dbd8863bb1dcc55837abe26a716)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6118b33951cb897dfee8a5222b7dcc0a6acc4d34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6118b33951cb897dfee8a5222b7dcc0a6acc4d34"}]},"branch":"refs/heads/stable/train"},"255dd3cb998fc77ed9c00fc87ce07c5e136eb152":{"kind":"REWORK","_number":10,"created":"2021-02-17 12:10:57.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/53/773853/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/53/773853/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/773853/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/773853/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/773853/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/53/773853/10"}}},"commit":{"parents":[{"commit":"970721d2d53fc10d84ba2c0dc5c96a58ecf5d60e","subject":"Fix building CentOS 8 IPA images on stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/970721d2d53fc10d84ba2c0dc5c96a58ecf5d60e"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-16 21:32:16.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-17 12:10:06.000000000","tz":60},"subject":"Test building seed deployment images in the seed job (CentOS 8 only)","message":"Test building seed deployment images in the seed job (CentOS 8 only)\n\nThis requires stackhpc.os-images v1.10.0 or newer, for compatibility\nwith CentOS 8 when SELinux is enabled: we disable SELinux, but without\nrebooting it stays enabled.\n\nThis Ansible role was updated to v1.10.2 in master and stable/victoria\nby I5efdbd52556721914fe69d7c6ba454b2c721b643, for another reason.\nRemember to bump the requirement when backporting to earlier releases.\n\nFinally, it requires building seed deployment images only after\ndeploying Bifrost, because the task copying images onto the seed expects\n/etc/kolla/bifrost to exist.\n\nWe also copy log files to identify issues when the job fails.\n\nChange-Id: I4719b4d397c01b35c78cb84c6d686dd27742d1c0\n(cherry picked from commit c84a9757dd449dbd8863bb1dcc55837abe26a716)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/255dd3cb998fc77ed9c00fc87ce07c5e136eb152"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/255dd3cb998fc77ed9c00fc87ce07c5e136eb152"}]},"branch":"refs/heads/stable/train"},"5ea0377089a05c233fa2a30f180904d0ff9f6b86":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2021-02-17 12:18:34.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/53/773853/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/53/773853/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/773853/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/773853/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/773853/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/53/773853/11"}}},"commit":{"parents":[{"commit":"739a4bac416192ad6454f1f247a8ea4afeb77462","subject":"Fix building CentOS 8 IPA images on stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/739a4bac416192ad6454f1f247a8ea4afeb77462"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-16 21:32:16.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-17 12:17:55.000000000","tz":60},"subject":"Test building seed deployment images in the seed job (CentOS 8 only)","message":"Test building seed deployment images in the seed job (CentOS 8 only)\n\nThis requires stackhpc.os-images v1.10.0 or newer, for compatibility\nwith CentOS 8 when SELinux is enabled: we disable SELinux, but without\nrebooting it stays enabled.\n\nThis Ansible role was updated to v1.10.2 in master and stable/victoria\nby I5efdbd52556721914fe69d7c6ba454b2c721b643, for another reason.\nRemember to bump the requirement when backporting to earlier releases.\n\nFinally, it requires building seed deployment images only after\ndeploying Bifrost, because the task copying images onto the seed expects\n/etc/kolla/bifrost to exist.\n\nWe also copy log files to identify issues when the job fails.\n\nChange-Id: I4719b4d397c01b35c78cb84c6d686dd27742d1c0\n(cherry picked from commit c84a9757dd449dbd8863bb1dcc55837abe26a716)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5ea0377089a05c233fa2a30f180904d0ff9f6b86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5ea0377089a05c233fa2a30f180904d0ff9f6b86"}]},"branch":"refs/heads/stable/train"},"7640af70d07c9f8f7a31d3545e3d8e763057bd45":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2021-03-05 14:57:13.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/53/773853/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/53/773853/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/773853/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/773853/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/773853/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/53/773853/12"}}},"commit":{"parents":[{"commit":"41c4c68788a4b6d86daa9ce66b498f70c3a8cfa3","subject":"Fix building CentOS 8 IPA images on stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/41c4c68788a4b6d86daa9ce66b498f70c3a8cfa3"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-16 21:32:16.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-05 14:56:48.000000000","tz":60},"subject":"Test building seed deployment images in the seed job (CentOS 8 only)","message":"Test building seed deployment images in the seed job (CentOS 8 only)\n\nThis requires stackhpc.os-images v1.10.0 or newer, for compatibility\nwith CentOS 8 when SELinux is enabled: we disable SELinux, but without\nrebooting it stays enabled.\n\nThis Ansible role was updated to v1.10.2 in master and stable/victoria\nby I5efdbd52556721914fe69d7c6ba454b2c721b643, for another reason.\nRemember to bump the requirement when backporting to earlier releases.\n\nFinally, it requires building seed deployment images only after\ndeploying Bifrost, because the task copying images onto the seed expects\n/etc/kolla/bifrost to exist.\n\nWe also copy log files to identify issues when the job fails.\n\nChange-Id: I4719b4d397c01b35c78cb84c6d686dd27742d1c0\n(cherry picked from commit c84a9757dd449dbd8863bb1dcc55837abe26a716)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7640af70d07c9f8f7a31d3545e3d8e763057bd45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7640af70d07c9f8f7a31d3545e3d8e763057bd45"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
